)]}'
{"id":"openstack%2Fmanila-tempest-plugin~928263","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I5c73ef42d8f01560ea85bffc2283ffaee001f48f","project":"openstack/manila-tempest-plugin","branch":"master","topic":"cleanup-ceph-installer","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-09-06 13:00:34.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"32594":{"account":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"last_update":"2024-09-11 18:47:49.000000000","reason":"\u003cGERRIT_ACCOUNT_32594\u003e replied on the change","reason_account":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-09-17 09:25:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5c73ef42d8f01560ea85bffc2283ffaee001f48f","subject":"Disable Ceph for Nova in cephfs-nfs multinode job","status":"MERGED","created":"2024-09-05 16:39:11.000000000","updated":"2024-09-17 09:48:31.000000000","submitted":"2024-09-17 09:25:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928263-cleanup-ceph-installer","meta_rev_id":"0b50ec758a9d9104dbdf996a7122f91f51992219","_number":928263,"virtual_id_number":928263,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-17 09:25:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2024-09-06 13:00:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2024-09-11 18:47:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":2,"date":"2024-09-10 17:56:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":1,"date":"2024-09-10 17:56:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-05 17:39:52.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-05 18:22:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-05 18:39:59.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-09-05 21:23:51.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-09-05 21:30:40.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2024-09-05 21:30:40.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"state":"REVIEWER"},{"updated":"2024-09-10 17:56:03.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"f4ea13b624061ef7c405e89fc5addb2c0a360037","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-05 16:39:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4a54e6db6cca8ab21796050f308cc3b82ffacfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-05 16:41:10.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"00a3f2bcde08dffa9a1bc34e15fc8d9948dfaeaa","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-05 17:39:52.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/63/928263/2/check/DellEMC_PowerFlex_Manila_dhss_false/69a441b/DellEMC_PowerFlex_Manila_dhss_false/1637 : SUCCESS in 31m 47s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"19ac12b251d3f657e4c1c3a8e6a214f821dd88ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 18:22:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/144cb7b8e61049bea632cf2650166069\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cb8b01813624da3af865b208d4af25c : SUCCESS in 2m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/110b2b5821e24d1a82b1b025ba4a9d30 : SUCCESS in 53m 24s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2a0371ba5f3d418199a55d1b3b3aeeed : FAILURE in 1h 39m 44s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9468ef1eb0ca48cd992ec63c4511a54e : FAILURE in 52m 49s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f95a8034895042d5bb1c4f04aa23fe41 : SUCCESS in 1h 06m 34s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8d978bca7f32425d806507b009907203 : FAILURE in 6m 39s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/73549b61e3274cd6a8abb915e505102a : SUCCESS in 53m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/df0c85dafbcd44cc9079dac627f24565 : SUCCESS in 23m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/61c36273a81a4ced90af7a36f157c372 : SUCCESS in 28m 13s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/eae156cb52d7464aac049bead0390969 : SUCCESS in 53m 01s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/6c3d109dffd64ef3bef1199ca4580ef3 : SUCCESS in 1h 01m 24s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/314c3b524e9143fda2e9b28cc34aa395 : SUCCESS in 58m 45s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/68c20d41d6204ef6a344dd433984b7cf : SUCCESS in 39m 20s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fe9d98b449444f7fbf9b75c22d083af8 : FAILURE in 50m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2c485dd4a18446c7b5ffc0fce27eadaf : FAILURE in 9m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dd16092791e4427e03452205cd418cf8c3089745","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-05 18:29:30.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7ef5b686fc635eeeba939059f5ea8800b466b422","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-05 18:39:59.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"f3bfb28cea0962c7de801c5d4ccbde7fba0bc473","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-05 19:50:54.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/63/928263/2/check/DellEMC_PowerFlex_Manila_dhss_false/1ff85ef/DellEMC_PowerFlex_Manila_dhss_false/1641 : SUCCESS in 31m 37s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"51c3263a931b3634dc50f6aab244f50caee9b869","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 20:11:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66265d3f70e7475c9a20b47a2f850d44\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c620cb64c64a465db48ba4a57dc32459 : SUCCESS in 3m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/26ef0daabebd4c9e93983ef0cfc4fa86 : SUCCESS in 58m 45s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8836cf0357f64725a4b5a40827bf89a6 : SUCCESS in 1h 39m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bf4dbf83789f43a091199e0487b1343e : FAILURE in 1h 00m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/6c89f8f0027c41ae88ef7f383d28f7e6 : SUCCESS in 1h 19m 19s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/429a9b0146834de2b656f1df758d41a0 : SUCCESS in 56m 43s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b7c6b2d633df478a8238e67c74fc99ac : SUCCESS in 1h 23m 01s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9784b70a44ee4b45acfe2314f4531d3d : SUCCESS in 34m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/890b697ce0d4445b864edae95c67e001 : SUCCESS in 31m 10s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/3a0356089f7c472aa2ccdf90769a4e0e : SUCCESS in 57m 32s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/341b37c68d8f45119bd989397f462869 : SUCCESS in 1h 15m 23s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/a8a6c90a30aa401fa2eef3b79b3e71e5 : SUCCESS in 1h 17m 14s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/7dc045da238f455c8ba8294ecf5c58e5 : SUCCESS in 37m 11s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2f3b51f5dccd41f8830b0cfb216772bf : FAILURE in 27m 34s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3009eebef2eb4616ab066b0b5382775b : FAILURE in 21m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8be184ad22038dbdbd0e2943e8b9b1a94b6c6ace","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-05 20:26:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/928263/2/upstream-check/manila-tempest-plugin-ontap-dhss/cee30c8/ : RETRY_LIMIT in 14m 15s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/928263/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/e686463/ : RETRY_LIMIT in 12m 36s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"abf7cd4af9a6403cc0c5714f8ac4126924df5fe6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-05 20:42:57.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/63/928263/2/check/DellEMC_PowerFlex_Manila_dhss_false/4047a94/DellEMC_PowerFlex_Manila_dhss_false/1642 : SUCCESS in 31m 35s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"29afe9a8882ff2611697acbd69926d5eedfcdd16","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-05 21:23:51.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/63/928263/2/check/DellEMC_Unity_Manila_dhss_true/913b203/DellEMC_Unity_Manila_dhss_true/1636 : FAILURE in 11m 55s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/63/928263/2/check/DellEMC_Unity_Manila_dhss_false/c5b9f8f/DellEMC_Unity_Manila_dhss_false/1744 : SUCCESS in 1h 17m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"7631b52073d562d527d74bc10af3785b8eea9a64","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-09-06 13:00:34.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"38561f71cc01570c2d482a3ca26be8a1418ea427","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2024-09-10 17:56:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1408fdb7c16283c90af731b06c2e46ae01240252","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 17:56:15.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/045c45e68a094c76a28672bf5b395f41\n\n\nWarning:\n  Change 918952 in project openstack/devstack-plugin-ceph does not share a change queue with 928263 in project openstack/manila-tempest-plugin","accounts_in_message":[],"_revision_number":2},{"id":"8eeb48ba475d0551486d7ddb678ce5b22a6a5797","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-09-11 18:47:49.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8507b1173776b5c8867d83474b9bfa05348d3a55","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-09-16 18:16:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"22f2c66fa93a9cb244b8187602e900fc771f988b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 21:10:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63629218f61647ff9e6221c08188414c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/412477fe618d49cfabcf09ea53e68a94 : SUCCESS in 2m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/275d9c9a1ce547f1a93030099c0605c6 : SUCCESS in 1h 09m 41s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/35c9b5dca2dc49348e25f54002612a42 : FAILURE in 2h 46m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/da8ce48fd2d94bb5b4c1335b19ab7907 : SUCCESS in 1h 33m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f61354d8f13a4fda8b7c9e4cffd9a3d9 : SUCCESS in 1h 05m 28s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/035fa75e2f3f4cff81fd7f1ea6a7ebc5 : SUCCESS in 52m 54s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b8c308b9faad4e338bac9cd506211a4a : FAILURE in 24m 07s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9dd4ab0cac7848d0a8b4db786b95c0f6 : SUCCESS in 36m 01s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1767bc418fee46d7b8fad6fea4080c86 : SUCCESS in 40m 39s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/5db3071a427a442787c56b5ee35a1dbe : SUCCESS in 51m 43s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/0d024dfa4a374dfa9463858ae7b5759d : SUCCESS in 1h 10m 54s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/4cbc3d1cddef4fb69312c11faaf997f1 : SUCCESS in 1h 02m 29s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/47ad751004cb481b9680d5a33d11742a : SUCCESS in 35m 56s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/43b850780c304f86b1a06fe6da105f4a : FAILURE in 12m 05s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b5413b6138744629a23d86356a2c7b48 : FAILURE in 13m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"41d3f6dafdeebdc7586bdcd3828ee49a4e132205","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 21:11:11.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb730969d4584c3f988220825f1e225b\n\n\nWarning:\n  Change 929192 in project openstack/devstack-plugin-ceph does not share a change queue with 928263 in project openstack/manila-tempest-plugin","accounts_in_message":[],"_revision_number":2},{"id":"889d5ab449925cc340c2d79455b7439363e2f2d2","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-16 21:27:50.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/928263/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/b1385e7/ : RETRY_LIMIT in 12m 31s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"3a9d2f80a9ae87140e87adc7d5fdb8ed95f3ae23","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-16 21:41:45.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/63/928263/2/check/DellEMC_PowerFlex_Manila_dhss_false/93b2d8a/DellEMC_PowerFlex_Manila_dhss_false/1700 : SUCCESS in 30m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"96cd7324c08b20d441f6105cef7d85f42cb96445","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-16 22:45:03.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/63/928263/2/check/DellEMC_Unity_Manila_dhss_true/7de8b6f/DellEMC_Unity_Manila_dhss_true/1689 : FAILURE in 13m 27s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/63/928263/2/check/DellEMC_Unity_Manila_dhss_false/cfd0692/DellEMC_Unity_Manila_dhss_false/1796 : SUCCESS in 1h 11m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"3e25466157da7e5d07028d84be11d95fb1b81c52","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-17 05:27:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4cd79ed6b1c81f086d50b0ded6056f8bd496bd0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 08:03:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc4b6b810fc141c287ed4f301b42fbb0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2e580f59b1840d5b39cb02886c45536 : SUCCESS in 2m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b9f93c9be2d04e79ad5aaeb38064314a : SUCCESS in 1h 13m 44s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a0f9b36e41a64e4c84a57d3fdd64a826 : FAILURE in 2h 30m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b0e3280bc2b546c69493e3a717e85502 : FAILURE in 56m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/6e18275aac644580a5934ab0303c8296 : SUCCESS in 42m 43s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/6069c1baa5db4e09953f1f2b6521f824 : SUCCESS in 58m 47s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/060fd6a3edf34242a389a96cb24d6465 : FAILURE in 24m 52s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1477e3072e584294ba02f912ec9ed4ba : SUCCESS in 33m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3ada1a85dd6143bca6c54888abcd98bd : SUCCESS in 25m 55s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/5c97af815648441d852e9258a908be97 : SUCCESS in 1h 12m 04s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/d7888af0d5be4a7b963551386ddcbc8f : SUCCESS in 48m 46s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/51fc5db3b86349458a09c79ceee88d89 : SUCCESS in 58m 18s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a948c00fd7cf47dabfcabd76e95c5599 : SUCCESS in 27m 50s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9de808f9a17b47d6a1026e693a0b1d4e : FAILURE in 23m 47s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/58c070f41bdb4c25b35ba0a0049057f6 : FAILURE in 19m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7fdd057c562bfeb717efd26bec959c2fbaeb5a84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 08:04:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"46bed6c862022ef9dcbf8a1b62f09f43145335b5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-17 08:38:08.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/63/928263/2/check/DellEMC_PowerFlex_Manila_dhss_false/14b808c/DellEMC_PowerFlex_Manila_dhss_false/1702 : SUCCESS in 33m 58s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"c85ce70e67ca3c574fb65fe06ea760c39d46c7ce","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-17 09:25:34.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/63/928263/2/check/DellEMC_Unity_Manila_dhss_true/4ea7a1e/DellEMC_Unity_Manila_dhss_true/1691 : FAILURE in 15m 04s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/63/928263/2/check/DellEMC_Unity_Manila_dhss_false/e1f7896/DellEMC_Unity_Manila_dhss_false/1798 : SUCCESS in 1h 21m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"86a9cb5e9f569ab5489638b0a1bf1a6a9079cc4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 09:25:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58d31c389dc9442f958251dc5424c878\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d4c3b5b35904d4d8ffe7fddb02fda2e : SUCCESS in 2m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e26c2f3f47774f0daa47e812ccd47102 : SUCCESS in 1h 10m 21s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e2fc7e48e5484933a03e3996697b1e02 : SUCCESS in 27m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e67dcadee26545348b7df86fd1810e4a : SUCCESS in 43m 34s","accounts_in_message":[],"_revision_number":2},{"id":"6ef6b6265154b6b21a27b69e66f48c3414a64cd6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 09:25:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0b50ec758a9d9104dbdf996a7122f91f51992219","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-17 09:48:31.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/928263/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/7119759/ : RETRY_LIMIT in 11m 55s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"94d5164e55225de87ddf6f9b7843c9440a0ba5ab","revisions":{"9cf1e9876399917e74526b9ecda60f4d4bd10a59":{"kind":"REWORK","_number":1,"created":"2024-09-05 16:39:11.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/63/928263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/63/928263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/928263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/928263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/928263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/928263/1"}}},"commit":{"parents":[{"commit":"11618100f18026e8178d2da4f9698f88eee3658d","subject":"Add waiter to access create on resource setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/11618100f18026e8178d2da4f9698f88eee3658d"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-09-05 16:36:55.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-09-05 16:36:55.000000000","tz":-420},"subject":"Disable Ceph for Nova in cephfs-nfs multinode job","message":"Disable Ceph for Nova in cephfs-nfs multinode job\n\nThe ENABLE_CEPH_NOVA variable needs to be set\nto False in the compute node because its disabled\nin the controller.\n\nChange-Id: I5c73ef42d8f01560ea85bffc2283ffaee001f48f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9cf1e9876399917e74526b9ecda60f4d4bd10a59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9cf1e9876399917e74526b9ecda60f4d4bd10a59"}]},"branch":"refs/heads/master"},"94d5164e55225de87ddf6f9b7843c9440a0ba5ab":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-09-05 16:41:10.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/63/928263/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/63/928263/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/928263/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/928263/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/928263/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/928263/2"}}},"commit":{"parents":[{"commit":"11618100f18026e8178d2da4f9698f88eee3658d","subject":"Add waiter to access create on resource setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/11618100f18026e8178d2da4f9698f88eee3658d"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-09-05 16:36:55.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-09-05 16:40:06.000000000","tz":-420},"subject":"Disable Ceph for Nova in cephfs-nfs multinode job","message":"Disable Ceph for Nova in cephfs-nfs multinode job\n\nThe ENABLE_CEPH_NOVA variable needs to be set\nto False in the compute node because its disabled\nin the controller.\n\nDepends-On: I5f75cb829383d7acd536e24c70cc4418d93c13bc\nChange-Id: I5c73ef42d8f01560ea85bffc2283ffaee001f48f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/94d5164e55225de87ddf6f9b7843c9440a0ba5ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/94d5164e55225de87ddf6f9b7843c9440a0ba5ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
