)]}'
{"id":"openstack%2Fmanila-tempest-plugin~930510","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I112b4682b36fd8bfe9b0b2325578bedc1cd20dee","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-12-19 14:57:24.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"36178":{"account":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"last_update":"2024-12-19 16:10:25.000000000","reason":"Change was submitted"},"32594":{"account":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"last_update":"2024-12-19 16:10:25.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2024-12-12 16:30:19.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"33038":{"account":{"_account_id":33038,"name":"Chuan Miao","email":"chuan137@gmail.com","username":"chuan137"},"last_update":"2024-12-18 09:57:32.000000000","reason":"\u003cGERRIT_ACCOUNT_33038\u003e replied on the change","reason_account":{"_account_id":33038,"name":"Chuan Miao","email":"chuan137@gmail.com","username":"chuan137"}}},"hashtags":[],"change_id":"I112b4682b36fd8bfe9b0b2325578bedc1cd20dee","subject":"IPV4 with Multinode Ceph-NFS Job","status":"MERGED","created":"2024-09-25 19:33:32.000000000","updated":"2024-12-19 16:10:25.000000000","submitted":"2024-12-19 16:10:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930510","meta_rev_id":"7081753e083a24d7e1bb929b834bfb667f95bfc1","_number":930510,"virtual_id_number":930510,"owner":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-19 16:10:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33038,"name":"Chuan Miao","email":"chuan137@gmail.com","username":"chuan137"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2024-12-19 14:57:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":2,"date":"2024-12-12 16:30:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":1,"date":"2024-12-17 15:05:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-18 09:57:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33038,"name":"Chuan Miao","email":"chuan137@gmail.com","username":"chuan137"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2024-12-19 14:57:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33038,"name":"Chuan Miao","email":"chuan137@gmail.com","username":"chuan137"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"_account_id":33038,"name":"Chuan Miao","email":"chuan137@gmail.com","username":"chuan137"},{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-25 20:06:29.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-25 20:57:08.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-09-25 21:21:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-25 21:38:13.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-10-02 18:29:01.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2024-12-17 15:05:42.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2024-12-17 15:44:44.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":33038,"name":"Chuan Miao","email":"chuan137@gmail.com","username":"chuan137"},"state":"REVIEWER"},{"updated":"2024-12-17 15:44:44.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"state":"REVIEWER"},{"updated":"2024-12-17 15:44:44.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"69a4c4a85b8d4665d03637a9ba77c2cca5b5ec45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-09-25 19:33:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99d65ca56255ae512f1f3ed360464c9fb6d3f130","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-25 20:06:29.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/1/check/DellEMC_PowerFlex_Manila_dhss_false/8b2d5af/DellEMC_PowerFlex_Manila_dhss_false/1740 : SUCCESS in 32m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"4036b82cebba6c4f423889cf2c3b91edb71481fe","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-25 20:57:08.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/10/930510/1/check/DellEMC_Unity_Manila_dhss_true/f2812b4/DellEMC_Unity_Manila_dhss_true/1728 : FAILURE in 14m 32s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/1/check/DellEMC_Unity_Manila_dhss_false/a2c0f7b/DellEMC_Unity_Manila_dhss_false/1833 : SUCCESS in 1h 23m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"7dcb900e05440ae636796b5d98f993429671041b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 21:21:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74c8a7d1657441de9aaf3d893cd39d30\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8facaacd25d47b1ab176ec997431a33 : SUCCESS in 3m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e2d6916950cf4e77a1ad4154df9e40f5 : SUCCESS in 1h 07m 12s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3a02bd730af94a1099367c9dd1bb66a1 : SUCCESS in 1h 45m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6b442f1df31e4c4e903d9ae55b2446c4 : FAILURE in 28m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/4d9a3d90c11243bca670b7667605d757 : SUCCESS in 1h 14m 35s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c2b866955f3840128286cd9a9f412973 : FAILURE in 18m 31s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/23fcb4b108d94bce861b81b73edf762c : FAILURE in 28m 06s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/74c9b22bc2d44b9b9ac357c41da02f50 : SUCCESS in 24m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1815305586f44f268a8ead6d960d8759 : SUCCESS in 40m 29s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/6b0b92930a4447d9ba57cf2060fc98fc : SUCCESS in 44m 18s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/8f8898b2a8d9425cae53811bdfdec339 : SUCCESS in 1h 22m 33s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/30fb5bab80ca44d995f779caabeeb06a : SUCCESS in 1h 10m 30s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/20e00c21c9194ce08e727971a56633c2 : SUCCESS in 31m 28s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ec93a8f476d54f6e8655b0bd7af87433 : FAILURE in 13m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/954938f4bac547e893f3d0daf6019978 : FAILURE in 23m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"25a2a7a8652ef7df5d46efaedf70e96ae1bac999","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-25 21:38:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/10/930510/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/48d93ce/ : RETRY_LIMIT in 8m 28s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"509343114faafcff1020df62cf97debd2c6fd974","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-25 21:54:11.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/1/check/DellEMC_PowerFlex_Manila_dhss_false/e48d83a/DellEMC_PowerFlex_Manila_dhss_false/1742 : SUCCESS in 32m 41s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"cf3579ffaf1a5e730b5fc90df5600a63032785b2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-25 23:33:17.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/10/930510/1/check/DellEMC_Unity_Manila_dhss_true/e165770/DellEMC_Unity_Manila_dhss_true/1730 : FAILURE in 13m 44s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/1/check/DellEMC_Unity_Manila_dhss_false/3de7817/DellEMC_Unity_Manila_dhss_false/1835 : SUCCESS in 1h 16m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"f1c2fdbeacf866b4156fde821733cd35133fa301","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-09-26 14:54:48.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0fbe2c0d91c66d7d941c70ce171979e35df3101d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 14:55:29.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/manila-tempest-plugin on branch master.  The\nproblem was:\n\n  The key \"voting\" appears more than once; duplicate keys are not permitted.\n\n  voting: false\n          #- manila-tempest-plugin-zfsonlinux:\n              voting: false\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 874","accounts_in_message":[],"_revision_number":2},{"id":"0dd505639b0eb570bcf032a292cd96f399bca7de","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-26 15:28:45.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/2/check/DellEMC_PowerFlex_Manila_dhss_false/a5c385f/DellEMC_PowerFlex_Manila_dhss_false/1747 : SUCCESS in 33m 41s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"de82f276faa679f0eac0c298cf0f4fb793cac947","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-26 16:13:23.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/10/930510/2/check/DellEMC_Unity_Manila_dhss_true/93d7af6/DellEMC_Unity_Manila_dhss_true/1735 : FAILURE in 13m 40s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/2/check/DellEMC_Unity_Manila_dhss_false/3fa8f21/DellEMC_Unity_Manila_dhss_false/1840 : SUCCESS in 1h 18m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"86cc9cd65fde155cfbb4d797b37f45268ab4a3b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-09-26 18:42:21.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4c77afd1dbac22781ce1fc9f837b0192f5988c9f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-26 19:16:48.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/3/check/DellEMC_PowerFlex_Manila_dhss_false/970a990/DellEMC_PowerFlex_Manila_dhss_false/1748 : SUCCESS in 34m 11s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"1be1cee7c5715adde0a8925176f1b09939147741","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-26 20:00:51.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/10/930510/3/check/DellEMC_Unity_Manila_dhss_true/6bf1671/DellEMC_Unity_Manila_dhss_true/1736 : FAILURE in 14m 41s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/3/check/DellEMC_Unity_Manila_dhss_false/26f1030/DellEMC_Unity_Manila_dhss_false/1841 : SUCCESS in 1h 18m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"6754e902d8b130abf4ca4d105ab7452b4e6c452d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 20:11:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcdbf5d77086404d9b58a8ee0790e5db\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de913b9f6c90449e9c45e31c77cc6f72 : SUCCESS in 4m 12s\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a97bee11a0624a9ba345e5522115c1b1 : SUCCESS in 54m 30s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b3dba20eb6df49b397c794354a01fe75 : SUCCESS in 1h 06m 41s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b4531c11da0a49269875aafc80111dd6 : SUCCESS in 29m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/65e8a64f35564454aa9990ae6803606f : SUCCESS in 33m 43s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/17150c2e66b149d5a6e41a85579cf963 : SUCCESS in 1h 21m 24s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/acef1eef204a45349985f5ac2b5ac5fe : SUCCESS in 54m 30s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/8766166cd0b04f138e319ae8bd02f9f7 : SUCCESS in 1h 12m 18s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/97ce534fabec4bfe8a4be1a1d275bc3c : SUCCESS in 30m 15s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fa6b9843f4fe4df48af9f181d80ffde7 : FAILURE in 26m 59s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c1d8bf38e4c14552a352fccfb74e2748 : FAILURE in 12m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ddbe17821df725886bbad898bdeb70323c3dfa6d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-26 20:26:21.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/10/930510/3/upstream-check/manila-tempest-plugin-ontap-dhss/3318bf8/ : RETRY_LIMIT in 9m 10s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/10/930510/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/afb3782/ : RETRY_LIMIT in 8m 33s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"da4e1c8da8566af4212af132e72f557fea493c24","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-26 20:44:52.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/3/check/DellEMC_PowerFlex_Manila_dhss_false/a1f2c24/DellEMC_PowerFlex_Manila_dhss_false/1749 : SUCCESS in 33m 11s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"0f716c11975caf1e8b42799a34c2bc4fbfb52499","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-26 21:48:06.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/10/930510/3/check/DellEMC_Unity_Manila_dhss_true/612d310/DellEMC_Unity_Manila_dhss_true/1737 : FAILURE in 14m 35s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/3/check/DellEMC_Unity_Manila_dhss_false/bd09553/DellEMC_Unity_Manila_dhss_false/1842 : SUCCESS in 1h 36m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"6db47e2e1da47c341355da07a2be2285dadec09c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-09-27 17:20:34.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"ee73a62b0a05d534606c5c0c964ea9780eb6d397","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-27 17:52:07.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/4/check/DellEMC_PowerFlex_Manila_dhss_false/d61c8a5/DellEMC_PowerFlex_Manila_dhss_false/1763 : SUCCESS in 31m 19s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"10134a0f0d4534965f4f0427303ba714435a6e7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 18:38:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7b830e1a99b4f58ad0fae696f860d70\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fade623440f45ba9a4b943b80991348 : SUCCESS in 5m 11s\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c82f8cf034dc4bdb94645ffb25606ef1 : FAILURE in 49m 18s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8fa308e8c12946f2bed53e6eb80e45bc : FAILURE in 50m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/eb8c9b4cb1e644d89ab85dd233d1aa8a : SUCCESS in 40m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c1a6c72dfc3e423c8d758a23d6e1638e : SUCCESS in 45m 25s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/a0c37dce3f764fb59c2e1f5bc49be0a9 : SUCCESS in 1h 09m 25s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/9813b11ea3034de794f44068bb92decd : SUCCESS in 1h 00m 24s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/7d137deff7b14180994551e328ae2732 : SUCCESS in 1h 02m 25s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/cbc0793340a746ee8fb7eb9fefc6036d : SUCCESS in 35m 24s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0591cc190bde42d8a04510373b56189c : FAILURE in 25m 13s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ed05c77e82aa408bb48c735fa77b55cc : FAILURE in 9m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9f5c1c7676903427230872dfa7c20d709cda6ecc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-27 18:52:49.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/10/930510/4/check/DellEMC_Unity_Manila_dhss_true/9eab513/DellEMC_Unity_Manila_dhss_true/1750 : FAILURE in 13m 44s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/4/check/DellEMC_Unity_Manila_dhss_false/b914ace/DellEMC_Unity_Manila_dhss_false/1853 : SUCCESS in 1h 32m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"175f10d8015156008d8d97e64cd3926e2d023975","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-27 18:55:16.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/10/930510/4/upstream-check/manila-tempest-plugin-ontap-dhss/da49740/ : RETRY_LIMIT in 8m 46s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"ae53be2a77c9af3ae7cbaaa393ab4b6e7a322005","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-27 19:17:31.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/4/check/DellEMC_PowerFlex_Manila_dhss_false/8b35d8f/DellEMC_PowerFlex_Manila_dhss_false/1766 : FAILURE in 20m 24s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"491471700f741b6994c203f133e4ae59a98416d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-09-27 20:07:13.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"6658682128e80074b0c7b9cdeb387bec7c1a2797","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-27 20:37:18.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/5/check/DellEMC_PowerFlex_Manila_dhss_false/655b62d/DellEMC_PowerFlex_Manila_dhss_false/1768 : SUCCESS in 29m 32s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"5dabcc945fad7bdb1ac7a204026590a5170e6b18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 21:28:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2afbcbfb629946e2bccfefa735814d4d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a07cf7dd4324ef685db9089d3a02e28 : SUCCESS in 3m 36s\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/7ffbd5e5de0a4b65b67f5c90e35e5d30 : FAILURE in 1h 09m 17s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5fa3da3d9ce34284bc541cbdd3855d39 : SUCCESS in 56m 01s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/44f482c0499e473dade818738bdbe1ff : SUCCESS in 28m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ab7aee5b38c24448b41e0999daa4fcca : SUCCESS in 38m 04s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/a329a89dc3ff4954b65e19944f756a1e : SUCCESS in 51m 50s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/2fa65364216c41fe9e8cb6c9503303b2 : SUCCESS in 1h 15m 23s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/7948b0862c744941ad9a5e3aba942bfe : SUCCESS in 54m 44s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/4c6d4ce563a049af84ca86d066d93f4a : SUCCESS in 26m 00s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a6f65695cf8a46abaf32721de991fc45 : FAILURE in 25m 32s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/adf7e6e12e13448cadac06a0aa1ebcec : FAILURE in 9m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c83ee591317bd31640b05c5379d8fab23918aa9d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-27 21:46:09.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/10/930510/5/upstream-check/manila-tempest-plugin-ontap-dhss/7fe6548/ : RETRY_LIMIT in 8m 33s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"ef5d9d903bdd11ccdc0a7d80eca8b1d08508d6f8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-27 22:01:24.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/5/check/DellEMC_PowerFlex_Manila_dhss_false/7dd023d/DellEMC_PowerFlex_Manila_dhss_false/1769 : SUCCESS in 32m 18s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"fbb9bb48774dd1767cdd828d4e76cf552383b283","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-27 22:42:05.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/10/930510/5/check/DellEMC_Unity_Manila_dhss_true/3282565/DellEMC_Unity_Manila_dhss_true/1753 : FAILURE in 13m 36s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/5/check/DellEMC_Unity_Manila_dhss_false/90d7105/DellEMC_Unity_Manila_dhss_false/1856 : SUCCESS in 1h 21m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"be400464ab9cffbcc1278e23cdab6a215f230dbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-10-02 16:21:00.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"2efb833344fc0423dd432f59a7cd91e73512a950","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-02 16:51:36.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/6/check/DellEMC_PowerFlex_Manila_dhss_false/975950c/DellEMC_PowerFlex_Manila_dhss_false/1803 : SUCCESS in 30m 21s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"29cad01beb395e52334d47932676af509cad61c9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-02 16:57:09.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/10/930510/6/check/DellEMC_Unity_Manila_dhss_true/b12a28b/DellEMC_Unity_Manila_dhss_true/1786 : FAILURE in 13m 29s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/6/check/DellEMC_Unity_Manila_dhss_false/f41e004/DellEMC_Unity_Manila_dhss_false/1889 : FAILURE in 35m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"d9923e71b32c82fe6157ef17253a15ebecc4b92d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 17:46:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90f4e89ad4da4a22979b0b687bcbf68d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a5c557712cd47f6b5c2e5302876f110 : SUCCESS in 2m 59s\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/71e38948959d4d049c9ee85b48196512 : FAILURE in 54m 44s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/fcf11f8edace4057b22fefbe73ddb8ac : SUCCESS in 1h 17m 17s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/928a15465950430bbe0a738cd2502d00 : SUCCESS in 24m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/22e1ec50b56146a4b05a925d78ab190a : SUCCESS in 50m 05s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/9842b800e34d4bd68ca87c13263a2b07 : SUCCESS in 1h 13m 20s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/830074308de34e0b86a43a9929c6eee1 : SUCCESS in 1h 10m 47s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/b7e58049345844ecac9cd460c6c7d8c8 : SUCCESS in 1h 15m 53s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/b92790d5790d46c59de143bfabc69621 : SUCCESS in 34m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8fc0a76e999642199fa0859ab1e11aca : FAILURE in 25m 13s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4f9f900d433646d98c4b433b6d31b819 : FAILURE in 21m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3fc11e0ce4feb34d2767b05320215f3492c51112","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-10-02 18:02:44.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/10/930510/6/upstream-check/manila-tempest-plugin-ontap-dhss/c45523a/ : RETRY_LIMIT in 9m 08s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/10/930510/6/upstream-check/manila-tempest-plugin-ontap-no-dhss/b73befe/ : RETRY_LIMIT in 9m 12s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"e6558af04dbbdfc2297229097578e855fdc2afd5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-02 18:21:18.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/10/930510/6/check/DellEMC_Unity_Manila_dhss_true/cc471f8/DellEMC_Unity_Manila_dhss_true/1788 : FAILURE in 13m 10s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/6/check/DellEMC_Unity_Manila_dhss_false/850ca3e/DellEMC_Unity_Manila_dhss_false/1891 : FAILURE in 34m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"38c794f01c3db85c84a037402a0a6036f4d77810","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-02 18:21:53.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/6/check/DellEMC_PowerFlex_Manila_dhss_false/fc718a0/DellEMC_PowerFlex_Manila_dhss_false/1805 : SUCCESS in 34m 53s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"01fb6b3ffde0e09a8a1ba6be5fafa9026523d1fe","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2024-10-02 18:29:01.000000000","message":"Patch Set 6: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"fa322308728caf0081b6447cba16d6f81aecc5a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-10-02 19:29:58.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"659f118fd07f28008d0a7042821d32c412984167","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-02 20:02:23.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/7/check/DellEMC_PowerFlex_Manila_dhss_false/907c608/DellEMC_PowerFlex_Manila_dhss_false/1807 : SUCCESS in 32m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"63d97fccc0fe416ec9213fc924e19f460147f705","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-02 20:07:44.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/10/930510/7/check/DellEMC_Unity_Manila_dhss_true/dbed32d/DellEMC_Unity_Manila_dhss_true/1790 : FAILURE in 17m 34s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/7/check/DellEMC_Unity_Manila_dhss_false/ee8694b/DellEMC_Unity_Manila_dhss_false/1893 : FAILURE in 37m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"3a36e9cfa87fe703c3cee17e93e5f12e9c96cb75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 20:54:46.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/342423c640574212bc343c2fca45c807\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81860a2115704118ae7fe8e72b028d7a : SUCCESS in 2m 14s\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/aca28e1e3af14e86a38c75ddc46715ac : RETRY_LIMIT in 2m 45s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/fdf6dcc2561646a9bc3d1efb88a2b979 : FAILURE in 45m 04s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a5ccbe2dbf174f1eb495236374699753 : SUCCESS in 25m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/862b803730e34df793002bb0c6532b17 : SUCCESS in 27m 22s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/216217e3d6644af88f7730e274ecca91 : SUCCESS in 1h 18m 25s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/ac64429c3c48490e9abee04126d306dd : SUCCESS in 55m 51s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/f9ab83b408664d0c8d5a2978204237c7 : SUCCESS in 42m 27s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/1674d831531243d9b04a3e4f541826c2 : SUCCESS in 25m 49s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/40f1e3cbac154e8ca0bd6176d8a28616 : FAILURE in 17m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a8179530456c498a969910524da0c775 : FAILURE in 11m 42s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8d932caf96fa3dd9c43005b8b7b28d5a94e53e5b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-10-02 21:15:12.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/10/930510/7/upstream-check/manila-tempest-plugin-ontap-dhss/050971a/ : RETRY_LIMIT in 8m 43s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/10/930510/7/upstream-check/manila-tempest-plugin-ontap-no-dhss/c3dba25/ : RETRY_LIMIT in 8m 47s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"461f3d752ce409b7f06317b749811f8fec2d19ea","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-02 21:25:07.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/7/check/DellEMC_PowerFlex_Manila_dhss_false/0989f5d/DellEMC_PowerFlex_Manila_dhss_false/1808 : SUCCESS in 30m 08s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"34bc74e095f1b8a6567b2d4ee1c72c7249499f64","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-02 21:27:21.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/10/930510/7/check/DellEMC_Unity_Manila_dhss_true/2864625/DellEMC_Unity_Manila_dhss_true/1791 : FAILURE in 13m 20s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/7/check/DellEMC_Unity_Manila_dhss_false/210667a/DellEMC_Unity_Manila_dhss_false/1894 : FAILURE in 32m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"257e3cfecc2647a20db02756406c34f73a2aeef4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-10-03 13:13:45.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"f6544dfb0c5a4631dcc6b6cc2fc610e5ddb77e70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-10-03 13:30:09.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"c598906a0d30ebd0e89f08cb60a7b2489fede2c2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-03 13:31:22.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/9/check/DellEMC_PowerFlex_Manila_dhss_false/86eb7d7/DellEMC_PowerFlex_Manila_dhss_false/1813 : FAILURE in 59s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"bcfb665004ef0c7f129fadb2de89335cdd1f279f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-03 14:02:30.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/10/930510/9/check/DellEMC_Unity_Manila_dhss_true/e97f726/DellEMC_Unity_Manila_dhss_true/1796 : FAILURE in 13m 34s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/9/check/DellEMC_Unity_Manila_dhss_false/54c792d/DellEMC_Unity_Manila_dhss_false/1899 : FAILURE in 32m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"3d88476293d42220167d4396493122c8aef02b5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 14:58:00.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e7b273c6f7949e79879825a6637439d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9bebf303bf243228ccfa03c2b4802c3 : SUCCESS in 2m 22s\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/7b2e4875d8c5430dad7d373a13da1ef2 : FAILURE in 50m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/79050aa4b9d04563bc5f3d0338e4c16f : SUCCESS in 1h 21m 13s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9deaa6ea524940018b7669c5fe47d4c7 : SUCCESS in 38m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2d3cfc468d2b461f83eeb9777749bc0d : SUCCESS in 45m 49s\n- manila-tempest-plugin-lvm-caracal https://zuul.opendev.org/t/openstack/build/dab4e6ffc4384c788a100fb35875f070 : SUCCESS in 1h 17m 00s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/bf89d85253b14773a3f629a5539be750 : SUCCESS in 1h 14m 30s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/abc8c2b249b24a40b5e864efa1c51c90 : SUCCESS in 1h 03m 32s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/4e9729542b7c470ca805e6d75362cb59 : SUCCESS in 34m 26s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a7118061a87e4486a99c80a52d2ad600 : POST_FAILURE in 14m 32s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d8a9b97288d74182943952168a000a35 : FAILURE in 10m 39s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b03b2007a77c1647540f3cc3f4c14bf8940c3728","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-10-03 15:09:10.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/10/930510/9/upstream-check/manila-tempest-plugin-ontap-dhss/10bf1fd/ : RETRY_LIMIT in 9m 09s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/10/930510/9/upstream-check/manila-tempest-plugin-ontap-no-dhss/452552a/ : RETRY_LIMIT in 9m 29s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"55abfb90634a54e0f05a23a32c8be42526302097","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-03 15:27:26.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/9/check/DellEMC_PowerFlex_Manila_dhss_false/358e65d/DellEMC_PowerFlex_Manila_dhss_false/1814 : SUCCESS in 29m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"688b5b5fb825d1ef8bcdfe1a0bfcefc2b86f8521","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-03 15:35:29.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/10/930510/9/check/DellEMC_Unity_Manila_dhss_true/65810c7/DellEMC_Unity_Manila_dhss_true/1797 : FAILURE in 13m 26s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/9/check/DellEMC_Unity_Manila_dhss_false/d2bb973/DellEMC_Unity_Manila_dhss_false/1900 : FAILURE in 37m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"74c2e02dfaadb1794357934e9f51c8b8c6e50c57","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-10-03 19:45:47.000000000","message":"Patch Set 9: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6db9105419763eabf7c862c9b3706a2e5bc1e282","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-12-06 21:46:42.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":10},{"id":"67638f1ad32c16cf70868a39c7dccf76de0ccd29","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-12-06 21:46:59.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"640524506a19174695e4e158d7d405a9bdca9566","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-12-06 21:47:26.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"af36c640edde92f467e9c69fea44f49be3039dc4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-06 22:20:32.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/11/check/DellEMC_PowerFlex_Manila_dhss_false/3301595/DellEMC_PowerFlex_Manila_dhss_false/2187 : SUCCESS in 32m 48s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"a497e653b06abee3aa214de43267bdefc6a8f08d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-07 00:27:09.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3a02972ac7e47bc90b964703586a4e4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a10481e5dce4333abc9fe1726c5e433 : SUCCESS in 2m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/809ff1603cd2459094e6e93b768d571a : FAILURE in 37m 07s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/091e3c6b6582495c8b25f96cbe957b31 : SUCCESS in 1h 14m 43s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8d172c10e898458a895619353e266891 : SUCCESS in 2h 20m 05s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2c3b6ec2118f47578f41b960d6791583 : FAILURE in 54m 42s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ce878fbbfb9a4eedb24bb9e8b95a74c6 : SUCCESS in 1h 04m 36s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/98d4babf0a0346cd8e5b74dcecdccf41 : FAILURE in 1h 23m 09s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/93b5c4c3477648878ce32d531f98caaf : FAILURE in 1h 21m 00s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/be80d15811b243298958079806c6b5bc : SUCCESS in 31m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/017d82be01ea469589ee3fb4bbe3be4b : SUCCESS in 46m 17s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/228a2a0dad954a76bb53e961c26c5de6 : SUCCESS in 1h 12m 41s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/11b659ab3a554856a45526d037189017 : SUCCESS in 1h 14m 38s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/51b8a9580f864cacbb2c588f28871849 : SUCCESS in 1h 08m 20s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/2542bfc6a93e44fa95384c0527d4cec4 : SUCCESS in 19m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/05d0cd45a6ee44969022399c7479adf9 : FAILURE in 53m 33s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"3ec3eb341ae78af64ff59ed4466ecba6b63735e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-12-10 13:47:23.000000000","message":"Patch Set 12: Published edit on patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"f1b32f909b732956b350a9de32f990c8d495f2ad","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-10 14:55:28.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/12/check/DellEMC_PowerFlex_Manila_dhss_false/3d17f11/DellEMC_PowerFlex_Manila_dhss_false/2203 : SUCCESS in 33m 16s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"1a90439ce61e22573a3ad2ffdf526866a03071d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 16:23:22.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/68bc73715b04439781d90495cd33bfb8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d260bf5465f64e86a07e691fa3f4d50c : SUCCESS in 2m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4f02054d95b4437b943c7fe8d882dd07 : SUCCESS in 1h 19m 38s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/82ce6d86dee640fdb4db0a60815e7b5c : SUCCESS in 1h 10m 56s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/eb06250173a442a39861c78242c1e647 : SUCCESS in 2h 19m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/76d1b8cdccaa4d81b47fdefef2852713 : FAILURE in 57m 11s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c2ccebcc339e464daeac64ed21eb2a3a : SUCCESS in 36m 58s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8e1fc0b2d6f44bf1b6a8d8f401aee7cf : FAILURE in 1h 24m 34s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8ce8b78507e94b83b6781fd16a27bddd : FAILURE in 1h 10m 12s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5edcd24e69b8473b81c7ad29eb463234 : SUCCESS in 25m 45s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/158ecd4ca78141079ba3b59a1101aa45 : SUCCESS in 39m 27s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/d9adbe0049f44b218cdccd8f9aa94f4e : SUCCESS in 55m 27s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/7336109c00414ee6b642882fe8adfd92 : SUCCESS in 1h 16m 01s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/07c8075fcc834ca0a0d7b3f2528747f6 : SUCCESS in 1h 08m 12s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/afdc24a96d394f7f9ce19a3503687400 : POST_FAILURE in 49m 32s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1eb841c8e8bd461f882bbc7ea42c1051 : FAILURE in 43m 54s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"56cc3f0fb877ee53d4b53f8223a4a278ca4c0965","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-12-10 17:35:03.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"b1cde3b4938e4e6eb5ad9df9ebccb0500ecb9f33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 20:17:27.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44da4c8d692d4f4ea79e6e05f9b2a7f5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a597b49ae67d4f159f2b6023b7136301 : SUCCESS in 3m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f1e7ca876ca440ed9d7dc583426abd76 : SUCCESS in 1h 15m 30s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/d789ce4666bc476fad03c296a0d8aaea : SUCCESS in 1h 04m 25s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/028394784578481a91c2ec9d3a42c3d1 : SUCCESS in 2h 35m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2f45909af9df421abf6a47cfb383d30f : FAILURE in 28m 57s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/0f59d25f9c104a73996a2c940e676d67 : SUCCESS in 1h 18m 09s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/3642575886e84eaea82cca389dd5ba7e : FAILURE in 1h 32m 48s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d72a17872ab44739a02ff54eb7e5a0d1 : SUCCESS in 1h 39m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4a87aa22a78d452596fa90310552a6ae : SUCCESS in 33m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6ce99489eac94f88b634c9d042d900d7 : SUCCESS in 42m 12s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/545cfeaeb7974bdbaa55a8a8c6b7ef54 : SUCCESS in 1h 16m 42s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/1e3a4c079b5a484e87f1bd32e1331a40 : SUCCESS in 1h 19m 57s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/5bc56c25111e4608a56a87a01a3e6d9f : SUCCESS in 1h 04m 38s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/7b584c4814da4326a9dc09c0aeb75d02 : SUCCESS in 39m 08s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a3cdafde24ad45b7b74b54aa8f420e49 : FAILURE in 39m 41s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"dcef5e8540bdbba74bb4933b8d593b5345f2d296","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-12-10 20:34:30.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss finger://managesf.services.openstack.netapp.com:7979/c8df39c523f540b480e9f991a60cd1ad : RETRY_LIMIT in 4m 43s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"aa9afd57e4d2992358c367f3b915b643fc5e3831","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-10 20:52:13.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/12/check/DellEMC_PowerFlex_Manila_dhss_false/66cfcf7/DellEMC_PowerFlex_Manila_dhss_false/2206 : SUCCESS in 34m 33s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"a2c9ae99a1122b1bb995cc1b4fdc934e69ea6251","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2024-12-11 12:19:19.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"ba084996db76ad5b90c38bac99bc22de471f4c89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-12-11 15:17:57.000000000","message":"Patch Set 13: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"dc05e951adfd78a25a375c6e51008d75bcc8119f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-11 15:53:29.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/13/check/DellEMC_PowerFlex_Manila_dhss_false/9f7d7f2/DellEMC_PowerFlex_Manila_dhss_false/2218 : SUCCESS in 35m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":13},{"id":"ece5043072031f67b42de0732cd2fba7be014fbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 18:50:49.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59fe9d20b2064232a5ca9d8208ec6530\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04be32ccb96c4455a902a1214109e678 : SUCCESS in 2m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2381fa9333534ac1a07ec4aa73aaeff3 : SUCCESS in 1h 11m 42s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/4e5d7185b32a44ee8811570de74e80f7 : SUCCESS in 1h 25m 03s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9436544198c440c68dd98f771cf624f6 : FAILURE in 2h 52m 57s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7bab496a2a8242ad9d19b8baa010b1cb : FAILURE in 1h 12m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a980fc2b92064e5db25a0bfb91afb4ec : SUCCESS in 1h 00m 07s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/207a23ba338341398bc321434145b4a1 : FAILURE in 51m 28s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e27f71a7326443f2a4f4c0c3b40d6e03 : FAILURE in 1h 26m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/32fb12507b5f4e15875631ef1213c2d5 : SUCCESS in 45m 45s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ae0bbf058c5943b49f2b174d2e5479f7 : SUCCESS in 42m 24s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/8defd7358ba54c7fafb2b3f83232e36c : SUCCESS in 1h 03m 54s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/8f827b6bc4f4431d86260ccf761deed2 : SUCCESS in 1h 18m 00s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/673d33f3e09c4a39b3196ed199534cd9 : SUCCESS in 1h 17m 24s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/628fe2a29ca94b81b5f93843b013e8fb : SUCCESS in 36m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9e32b52b826c404cbad9e313f01b881f : FAILURE in 45m 49s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"f6bb9e9a7307f3b5f487b328c45616c4c59b5576","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-11 19:26:28.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/10/930510/13/check/DellEMC_PowerFlex_Manila_dhss_false/72d2f4c/DellEMC_PowerFlex_Manila_dhss_false/2219 : SUCCESS in 35m 25s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":13},{"id":"84f93aee2fc450e78954b7106a8d830fbc7137c2","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-12-11 21:31:41.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"2d7319401e99e72c8840eb0a4bead2d18599bced","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2024-12-12 16:30:19.000000000","message":"Patch Set 13: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"36cc068ff07e663cdb52964824461387633251e1","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-12-17 15:05:42.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"4344bc54b0c82535d53f69c439cf154f4e46fbfc","author":{"_account_id":33038,"name":"Chuan Miao","email":"chuan137@gmail.com","username":"chuan137"},"date":"2024-12-18 09:57:32.000000000","message":"Patch Set 13: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"d06b08ab8de5041ccb7dca85fdd9b1005bffa539","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-12-19 14:57:24.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"92c3d0421c7432b21ff472862004a3e9df928306","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 14:58:14.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"68b6152bae6b698bced76106cbf35556afb11da9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 16:10:25.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06943427f87f403abe306db281447d52\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4f1aa1fc6414bad8ea0cf701fee9a24 : SUCCESS in 2m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b5f7e74be69645158e85c290f28e6a45 : SUCCESS in 1h 10m 23s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/0af2ce34fe4642a89a5dafba9e50d1f4 : SUCCESS in 1h 11m 11s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0c7cc3db249c4a8f8d333438fa3fb776 : SUCCESS in 28m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1af96a5e66804e88a0da04c2c4cceff5 : SUCCESS in 48m 17s","accounts_in_message":[],"_revision_number":13},{"id":"7081753e083a24d7e1bb929b834bfb667f95bfc1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 16:10:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"10945e54664adcd52101280e71a45477f33e3e29","revisions":{"e800310292a8b340f6a54cf2218cd7cc4b20bc73":{"kind":"REWORK","_number":1,"created":"2024-09-25 19:33:32.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/10/930510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/10/930510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/1"}}},"commit":{"parents":[{"commit":"960f9f6e2ca363f034ede3c02310bb4e3e5e7169","subject":"Merge \"Disable Ceph for Nova in cephfs-nfs multinode job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/960f9f6e2ca363f034ede3c02310bb4e3e5e7169"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2024-09-25 19:23:30.000000000","tz":0},"committer":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2024-09-25 19:23:30.000000000","tz":0},"subject":"IPV4 with Standalone Ceph-NFS Job","message":"IPV4 with Standalone Ceph-NFS Job\n\nChanging VIP to IPV4\n\nChange-Id: I112b4682b36fd8bfe9b0b2325578bedc1cd20dee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e800310292a8b340f6a54cf2218cd7cc4b20bc73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e800310292a8b340f6a54cf2218cd7cc4b20bc73"}]},"branch":"refs/heads/master"},"b8aa1f164d55c27155e6c5af4bb28cdcbc92f703":{"kind":"REWORK","_number":2,"created":"2024-09-26 14:54:48.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/10/930510/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/10/930510/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/2"}}},"commit":{"parents":[{"commit":"960f9f6e2ca363f034ede3c02310bb4e3e5e7169","subject":"Merge \"Disable Ceph for Nova in cephfs-nfs multinode job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/960f9f6e2ca363f034ede3c02310bb4e3e5e7169"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2024-09-25 19:23:30.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2024-09-26 14:54:48.000000000","tz":0},"subject":"IPV4 with Standalone Ceph-NFS Job","message":"IPV4 with Standalone Ceph-NFS Job\n\nChanging VIP to IPV4\n\nChange-Id: I112b4682b36fd8bfe9b0b2325578bedc1cd20dee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b8aa1f164d55c27155e6c5af4bb28cdcbc92f703"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b8aa1f164d55c27155e6c5af4bb28cdcbc92f703"}]},"branch":"refs/heads/master"},"8cf6023500c99e6de9b936d1d9bdd138ebce4286":{"kind":"REWORK","_number":3,"created":"2024-09-26 18:42:21.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/10/930510/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/10/930510/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/3"}}},"commit":{"parents":[{"commit":"960f9f6e2ca363f034ede3c02310bb4e3e5e7169","subject":"Merge \"Disable Ceph for Nova in cephfs-nfs multinode job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/960f9f6e2ca363f034ede3c02310bb4e3e5e7169"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2024-09-25 19:23:30.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2024-09-26 18:42:21.000000000","tz":0},"subject":"IPV4 with Standalone Ceph-NFS Job","message":"IPV4 with Standalone Ceph-NFS Job\n\nChanging VIP to IPV4\n\nChange-Id: I112b4682b36fd8bfe9b0b2325578bedc1cd20dee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8cf6023500c99e6de9b936d1d9bdd138ebce4286"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8cf6023500c99e6de9b936d1d9bdd138ebce4286"}]},"branch":"refs/heads/master"},"c8972b5b31789d34987980e557ea1577da6425e4":{"kind":"REWORK","_number":4,"created":"2024-09-27 17:20:34.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/10/930510/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/10/930510/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/4"}}},"commit":{"parents":[{"commit":"960f9f6e2ca363f034ede3c02310bb4e3e5e7169","subject":"Merge \"Disable Ceph for Nova in cephfs-nfs multinode job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/960f9f6e2ca363f034ede3c02310bb4e3e5e7169"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2024-09-25 19:23:30.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2024-09-27 17:20:34.000000000","tz":0},"subject":"IPV4 with Standalone Ceph-NFS Job","message":"IPV4 with Standalone Ceph-NFS Job\n\nChanging VIP to IPV4\n\nChange-Id: I112b4682b36fd8bfe9b0b2325578bedc1cd20dee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c8972b5b31789d34987980e557ea1577da6425e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c8972b5b31789d34987980e557ea1577da6425e4"}]},"branch":"refs/heads/master"},"cad2809f7c325d2338be68c578f601f37e0be138":{"kind":"REWORK","_number":5,"created":"2024-09-27 20:07:13.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/10/930510/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/10/930510/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/5"}}},"commit":{"parents":[{"commit":"960f9f6e2ca363f034ede3c02310bb4e3e5e7169","subject":"Merge \"Disable Ceph for Nova in cephfs-nfs multinode job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/960f9f6e2ca363f034ede3c02310bb4e3e5e7169"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2024-09-25 19:23:30.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2024-09-27 20:07:12.000000000","tz":0},"subject":"IPV4 with Standalone Ceph-NFS Job","message":"IPV4 with Standalone Ceph-NFS Job\n\nChanging VIP to IPV4\n\nChange-Id: I112b4682b36fd8bfe9b0b2325578bedc1cd20dee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cad2809f7c325d2338be68c578f601f37e0be138"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cad2809f7c325d2338be68c578f601f37e0be138"}]},"branch":"refs/heads/master"},"773ab31cb2bfe870e2c95bd13c17adb6c700e21b":{"kind":"REWORK","_number":6,"created":"2024-10-02 16:21:00.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/10/930510/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/10/930510/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/6"}}},"commit":{"parents":[{"commit":"960f9f6e2ca363f034ede3c02310bb4e3e5e7169","subject":"Merge \"Disable Ceph for Nova in cephfs-nfs multinode job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/960f9f6e2ca363f034ede3c02310bb4e3e5e7169"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2024-09-25 19:23:30.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2024-10-02 16:19:42.000000000","tz":0},"subject":"IPV4 with Multinode Ceph-NFS Job","message":"IPV4 with Multinode Ceph-NFS Job\n\nChanging VIP to IPV4\n\nChange-Id: I112b4682b36fd8bfe9b0b2325578bedc1cd20dee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/773ab31cb2bfe870e2c95bd13c17adb6c700e21b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/773ab31cb2bfe870e2c95bd13c17adb6c700e21b"}]},"branch":"refs/heads/master"},"fb28b6dc94d8d20500229401ec1a1c4cf05d7802":{"kind":"REWORK","_number":7,"created":"2024-10-02 19:29:58.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/10/930510/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/10/930510/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/7"}}},"commit":{"parents":[{"commit":"960f9f6e2ca363f034ede3c02310bb4e3e5e7169","subject":"Merge \"Disable Ceph for Nova in cephfs-nfs multinode job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/960f9f6e2ca363f034ede3c02310bb4e3e5e7169"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2024-09-25 19:23:30.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2024-10-02 19:29:58.000000000","tz":0},"subject":"[DNM] IPV4 with Multinode Ceph-NFS Job","message":"[DNM] IPV4 with Multinode Ceph-NFS Job\n\nChanging VIP to IPV4\n\nChange-Id: I112b4682b36fd8bfe9b0b2325578bedc1cd20dee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fb28b6dc94d8d20500229401ec1a1c4cf05d7802"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fb28b6dc94d8d20500229401ec1a1c4cf05d7802"}]},"branch":"refs/heads/master"},"f5e3e9bcd86a989a03d418173ba6a0141055a113":{"kind":"REWORK","_number":8,"created":"2024-10-03 13:13:45.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/10/930510/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/10/930510/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/8"}}},"commit":{"parents":[{"commit":"960f9f6e2ca363f034ede3c02310bb4e3e5e7169","subject":"Merge \"Disable Ceph for Nova in cephfs-nfs multinode job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/960f9f6e2ca363f034ede3c02310bb4e3e5e7169"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2024-09-25 19:23:30.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2024-10-03 13:13:45.000000000","tz":0},"subject":"[DNM] IPV4 with Multinode Ceph-NFS Job","message":"[DNM] IPV4 with Multinode Ceph-NFS Job\n\nChanging VIP to IPV4\n\nChange-Id: I112b4682b36fd8bfe9b0b2325578bedc1cd20dee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f5e3e9bcd86a989a03d418173ba6a0141055a113"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f5e3e9bcd86a989a03d418173ba6a0141055a113"}]},"branch":"refs/heads/master"},"d2151def8ef616af7bb69ad18a20b6a6fc0af838":{"kind":"REWORK","_number":9,"created":"2024-10-03 13:30:09.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/10/930510/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/10/930510/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/9"}}},"commit":{"parents":[{"commit":"960f9f6e2ca363f034ede3c02310bb4e3e5e7169","subject":"Merge \"Disable Ceph for Nova in cephfs-nfs multinode job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/960f9f6e2ca363f034ede3c02310bb4e3e5e7169"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2024-09-25 19:23:30.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2024-10-03 13:30:09.000000000","tz":0},"subject":"[DNM] IPV4 with Multinode Ceph-NFS Job","message":"[DNM] IPV4 with Multinode Ceph-NFS Job\n\nChanging VIP to IPV4\n\nChange-Id: I112b4682b36fd8bfe9b0b2325578bedc1cd20dee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2151def8ef616af7bb69ad18a20b6a6fc0af838"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2151def8ef616af7bb69ad18a20b6a6fc0af838"}]},"branch":"refs/heads/master"},"249d75ac48ed3bd82555143063e11fb21014d69f":{"kind":"REWORK","_number":10,"created":"2024-12-06 21:46:42.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/10/930510/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/10/930510/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/10"}}},"commit":{"parents":[{"commit":"ee3079db74bd8ce6be5115a99263907fd3958da3","subject":"Merge \"Add nfs_versions option for NFS protocol version selection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ee3079db74bd8ce6be5115a99263907fd3958da3"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2024-09-25 19:23:30.000000000","tz":0},"committer":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2024-12-06 21:42:52.000000000","tz":0},"subject":"IPV4 with Multinode Ceph-NFS Job","message":"IPV4 with Multinode Ceph-NFS Job\n\nChanging VIP to IPV4\n\nChange-Id: I112b4682b36fd8bfe9b0b2325578bedc1cd20dee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/249d75ac48ed3bd82555143063e11fb21014d69f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/249d75ac48ed3bd82555143063e11fb21014d69f"}]},"branch":"refs/heads/master"},"ca1846dd10815b506b3bc88a4a747c5b679044ce":{"kind":"REWORK","_number":11,"created":"2024-12-06 21:47:26.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/10/930510/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/10/930510/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/11"}}},"commit":{"parents":[{"commit":"ee3079db74bd8ce6be5115a99263907fd3958da3","subject":"Merge \"Add nfs_versions option for NFS protocol version selection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ee3079db74bd8ce6be5115a99263907fd3958da3"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2024-09-25 19:23:30.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2024-12-06 21:47:26.000000000","tz":0},"subject":"IPV4 with Multinode Ceph-NFS Job","message":"IPV4 with Multinode Ceph-NFS Job\n\nChanging VIP to IPV4\n\nChange-Id: I112b4682b36fd8bfe9b0b2325578bedc1cd20dee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ca1846dd10815b506b3bc88a4a747c5b679044ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ca1846dd10815b506b3bc88a4a747c5b679044ce"}]},"branch":"refs/heads/master"},"0bb2da041d73ab80c8c55b74cb0900dfb386dbef":{"kind":"REWORK","_number":12,"created":"2024-12-10 13:47:23.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/10/930510/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/10/930510/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/12"}}},"commit":{"parents":[{"commit":"ee3079db74bd8ce6be5115a99263907fd3958da3","subject":"Merge \"Add nfs_versions option for NFS protocol version selection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ee3079db74bd8ce6be5115a99263907fd3958da3"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2024-09-25 19:23:30.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2024-12-10 13:47:23.000000000","tz":0},"subject":"IPV4 with Multinode Ceph-NFS Job","message":"IPV4 with Multinode Ceph-NFS Job\n\nChanging VIP to IPV4\n\nChange-Id: I112b4682b36fd8bfe9b0b2325578bedc1cd20dee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0bb2da041d73ab80c8c55b74cb0900dfb386dbef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0bb2da041d73ab80c8c55b74cb0900dfb386dbef"}]},"branch":"refs/heads/master"},"10945e54664adcd52101280e71a45477f33e3e29":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2024-12-11 15:17:57.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/10/930510/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/10/930510/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/10/930510/13"}}},"commit":{"parents":[{"commit":"ee3079db74bd8ce6be5115a99263907fd3958da3","subject":"Merge \"Add nfs_versions option for NFS protocol version selection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ee3079db74bd8ce6be5115a99263907fd3958da3"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2024-09-25 19:23:30.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2024-12-11 15:17:57.000000000","tz":0},"subject":"IPV4 with Multinode Ceph-NFS Job","message":"IPV4 with Multinode Ceph-NFS Job\n\nChanging VIP from IPV6 to IPV4\n\nPartial-Bug: 2082617\nChange-Id: I112b4682b36fd8bfe9b0b2325578bedc1cd20dee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/10945e54664adcd52101280e71a45477f33e3e29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/10945e54664adcd52101280e71a45477f33e3e29"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
