)]}'
{"id":"openstack%2Fmanila-tempest-plugin~931648","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ib861e1a707109daecb2cdd189de40da24bb829cb","project":"openstack/manila-tempest-plugin","branch":"master","topic":"stable/2024.2-jobs","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-10-07 16:18:35.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-07 20:23:35.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2024-10-07 19:04:54.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}},"hashtags":[],"change_id":"Ib861e1a707109daecb2cdd189de40da24bb829cb","subject":"Add stable/2024.2 job","status":"MERGED","created":"2024-10-07 12:06:40.000000000","updated":"2024-10-07 20:23:35.000000000","submitted":"2024-10-07 20:23:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931648-stable/2024.2-jobs","meta_rev_id":"e95b6e6ce1f616d2210fc02faf9ee40914c97e65","_number":931648,"virtual_id_number":931648,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-07 20:23:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2024-10-07 16:18:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2024-10-07 19:04:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2024-10-07 19:04:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-07 12:41:29.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-07 12:47:23.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-10-07 14:04:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-07 14:20:28.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-10-07 15:08:44.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-10-07 15:08:44.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"18504ee41b219079ac8e2961f3d5b32ed1396aac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-07 12:06:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56fef421460f86050ccba497cf32fa5f807356ac","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-07 12:41:29.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/48/931648/1/check/DellEMC_PowerFlex_Manila_dhss_false/c2d11dd/DellEMC_PowerFlex_Manila_dhss_false/1826 : SUCCESS in 34m 33s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"edd2a84d351248067f76c0670a5e9af91df78277","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-07 12:47:23.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/48/931648/1/check/DellEMC_Unity_Manila_dhss_true/4cd0f87/DellEMC_Unity_Manila_dhss_true/1809 : FAILURE in 17m 37s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/48/931648/1/check/DellEMC_Unity_Manila_dhss_false/88c8313/DellEMC_Unity_Manila_dhss_false/1912 : FAILURE in 38m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"8d310dbd8df60bf02c00c0c90324decf4c619208","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 14:04:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/429102ba71114cbb9928c8b6cca7fed0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81a2743d468a43fabb0b0f5a70872495 : SUCCESS in 2m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5b3f2c54e120439881b0c49689119fe1 : SUCCESS in 1h 02m 07s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2f85d827d46841acac0e6bae0661984d : FAILURE in 1h 55m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9c1233b2d94f4100a8b62e859d1ad612 : FAILURE in 43m 30s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/5538e25d035b480a924dde353fcdc009 : SUCCESS in 1h 08m 41s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/756164514ac0429d88aa3473580d1757 : SUCCESS in 1h 07m 20s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6e85f0360e084bb2ad5c4e02a6f51a13 : FAILURE in 13m 17s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ab93bfcdb17e45a2aafd118a12489607 : SUCCESS in 27m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c36d1ccd60ce4ee4ae3e6fe28daf66f9 : SUCCESS in 43m 53s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/4bdc40ffc4b3457d83708284a23fa84b : SUCCESS in 41m 08s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/0919c34c0b6a4e6e899f4d30c5a72ee3 : SUCCESS in 1h 07m 17s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/eb7d9479456b4b838cbacc5404a5ea18 : SUCCESS in 42m 11s\n- manila-tempest-plugin-lvm-2023.1 https://zuul.opendev.org/t/openstack/build/cff418822a604be3be3748bccc477da7 : SUCCESS in 1h 13m 15s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/25c18fff05cd4baa82ecf7c707a0c92d : SUCCESS in 21m 01s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d911e04023624a62804b03b5451aab66 : FAILURE in 25m 51s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a672f8567cac48bf8202c68a8eadacf2 : FAILURE in 21m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a9a3fe8641f8e60729125d33991108a55bb8276","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-10-07 14:20:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/931648/1/upstream-check/manila-tempest-plugin-ontap-dhss/9d5cf59/ : RETRY_LIMIT in 9m 33s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/931648/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/98ce350/ : RETRY_LIMIT in 14m 34s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"2b8ceabbe26587e1b771c4d44d3ff543f3e7d74c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-07 14:38:47.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/48/931648/1/check/DellEMC_PowerFlex_Manila_dhss_false/b76aead/DellEMC_PowerFlex_Manila_dhss_false/1827 : SUCCESS in 33m 48s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"9dc29418696408ae54b650a5a6014414166a7903","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-07 14:40:39.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/48/931648/1/check/DellEMC_Unity_Manila_dhss_true/0d7eee9/DellEMC_Unity_Manila_dhss_true/1810 : FAILURE in 13m 42s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/48/931648/1/check/DellEMC_Unity_Manila_dhss_false/23a7e61/DellEMC_Unity_Manila_dhss_false/1913 : FAILURE in 35m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"72fe5646562204bd64e8c9414f6fdbf55c216819","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-10-07 15:08:44.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"706990b793b1614e196114bb328d4b8c636d4b76","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2024-10-07 15:46:02.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d80465ab04baf679ffa720aec75a8e7bb8e93b03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-07 15:55:53.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7e7e966c7708fd1ff55e4452091a4e3dd0f07a50","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-07 15:56:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b97b53100a60b3a3e3bf37be445a414a30c25c8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-10-07 16:18:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"61bc6031affe93deed3823854e7c393651483369","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-07 16:27:47.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/48/931648/2/check/DellEMC_PowerFlex_Manila_dhss_false/628dd58/DellEMC_PowerFlex_Manila_dhss_false/1828 : SUCCESS in 31m 38s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"7da4fb0ba2904c9003ae539cceae987fce90c056","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-07 16:33:33.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/48/931648/2/check/DellEMC_Unity_Manila_dhss_true/b5f7da8/DellEMC_Unity_Manila_dhss_true/1811 : FAILURE in 13m 38s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/48/931648/2/check/DellEMC_Unity_Manila_dhss_false/69fbbc0/DellEMC_Unity_Manila_dhss_false/1914 : FAILURE in 37m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"8b9ef544cfcd3df57feca6cb84ec780d20a412fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 18:54:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8ba925b8df2428591b7c1be206d3cf1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9543c66bb254916a1a1e53920681287 : SUCCESS in 2m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e5514030afb94a06b6ab2faca23c8d16 : SUCCESS in 1h 08m 35s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/0cd6e60f5ced4d5ba98d380e786c2742 : FAILURE in 2h 51m 55s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a282bbc88d484496bd02420be68a6e87 : FAILURE in 1h 03m 02s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/73c7e3c4d79c49aba1a4675fad93496e : SUCCESS in 1h 08m 07s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a959d9a577014eb88d7f0bbcbe81e952 : SUCCESS in 57m 18s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d024f90300364167afcbb2c2d1926418 : FAILURE in 14m 14s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d646436042c940519b5d10b50022b0d4 : SUCCESS in 41m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fa67ca9a20d049f8b6cc820673e7d60e : SUCCESS in 39m 36s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/462eb67825304a6f89941593e7b2c697 : SUCCESS in 42m 54s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/35b48e81cf224d159e57defc3b40ae77 : SUCCESS in 1h 11m 12s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/aa01dc3dad114a33805debdfea7d3305 : SUCCESS in 1h 13m 22s\n- manila-tempest-plugin-lvm-2023.1 https://zuul.opendev.org/t/openstack/build/146402ce7c0e41cb92a679fa92c51e5c : SUCCESS in 1h 02m 18s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/52ee5dee8daa4cb282ee5f2483460e07 : SUCCESS in 26m 25s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6772ca488a214095a58225d5b1fa18c4 : FAILURE in 16m 16s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9fbf8ab565e44500b4db7f0b6198c604 : FAILURE in 13m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0891d7a3e49a9ec3e34f426503a9f01d8c6ffa87","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2024-10-07 19:04:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e07de826532c2faf67718d4b07ba4bc1c7d44ce5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 19:05:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3434b85525fe31b5c0a521701642ac7c318e9ad8","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-10-07 19:14:45.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/931648/2/upstream-check/manila-tempest-plugin-ontap-dhss/17336d6/ : RETRY_LIMIT in 8m 52s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/931648/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/eba27e3/ : RETRY_LIMIT in 8m 17s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"4e040455bd2b0a2daa43de4e0d512dc667bbf478","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-07 19:28:10.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/48/931648/2/check/DellEMC_PowerFlex_Manila_dhss_false/ccf0e07/DellEMC_PowerFlex_Manila_dhss_false/1829 : SUCCESS in 33m 08s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"4c3f2112bb3a8822e3103df03437a822384af040","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-07 19:30:46.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/48/931648/2/check/DellEMC_Unity_Manila_dhss_true/0d7d890/DellEMC_Unity_Manila_dhss_true/1812 : FAILURE in 15m 13s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/48/931648/2/check/DellEMC_Unity_Manila_dhss_false/adcc39e/DellEMC_Unity_Manila_dhss_false/1915 : FAILURE in 35m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"cd731418855e3b2b8713432f7ea4638ee71bf183","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 20:23:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5daa56861e1e4a81a31c62220dd44d88\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55a506c804a440d8b50d2d6292c60cda : SUCCESS in 2m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3243718ff542444284146183ec2466e7 : SUCCESS in 55m 46s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d146b32790084456a235cf00b5ff6bb8 : SUCCESS in 37m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c89f65d4f93b4aa293b02ef4730ec740 : SUCCESS in 32m 22s","accounts_in_message":[],"_revision_number":2},{"id":"e95b6e6ce1f616d2210fc02faf9ee40914c97e65","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 20:23:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b1f749a046ad09f89a6d8ed74dedbc7540d693b0","revisions":{"4d9213c46b5ef51262e714573d8a918701738bcd":{"kind":"REWORK","_number":1,"created":"2024-10-07 12:06:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/931648/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/931648/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/931648/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/931648/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/931648/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/931648/1"}}},"commit":{"parents":[{"commit":"960f9f6e2ca363f034ede3c02310bb4e3e5e7169","subject":"Merge \"Disable Ceph for Nova in cephfs-nfs multinode job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/960f9f6e2ca363f034ede3c02310bb4e3e5e7169"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-07 12:06:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-07 12:06:11.000000000","tz":540},"subject":"Add stable/2024.2 job","message":"Add stable/2024.2 job\n\nstable/2024.2 was released, so should be tested now.\n\nChange-Id: Ib861e1a707109daecb2cdd189de40da24bb829cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d9213c46b5ef51262e714573d8a918701738bcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d9213c46b5ef51262e714573d8a918701738bcd"}]},"branch":"refs/heads/master"},"b1f749a046ad09f89a6d8ed74dedbc7540d693b0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-10-07 15:55:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/931648/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/931648/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/931648/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/931648/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/931648/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/931648/2"}}},"commit":{"parents":[{"commit":"960f9f6e2ca363f034ede3c02310bb4e3e5e7169","subject":"Merge \"Disable Ceph for Nova in cephfs-nfs multinode job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/960f9f6e2ca363f034ede3c02310bb4e3e5e7169"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-07 12:06:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-07 15:55:53.000000000","tz":0},"subject":"Add stable/2024.2 job","message":"Add stable/2024.2 job\n\nstable/2024.2 was released, so should be tested now.\n\nAlso replace codenames by official release names.\n\nChange-Id: Ib861e1a707109daecb2cdd189de40da24bb829cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b1f749a046ad09f89a6d8ed74dedbc7540d693b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b1f749a046ad09f89a6d8ed74dedbc7540d693b0"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
