)]}'
{"id":"openstack%2Fmanila-tempest-plugin~931712","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ieb8dfc8813a3366b4252b6816397632ffb6a20bf","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bp/metadata-for-share-resources","attention_set":{},"removed_from_attention_set":{"33728":{"account":{"_account_id":33728,"name":"Ashley Rodriguez","email":"ashrodri@redhat.com"},"last_update":"2024-10-09 16:02:29.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-10-09 16:02:29.000000000","reason":"Change was submitted"},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2024-10-09 05:01:32.000000000","reason":"\u003cGERRIT_ACCOUNT_32919\u003e replied on the change","reason_account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2024-10-09 16:02:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ieb8dfc8813a3366b4252b6816397632ffb6a20bf","subject":"Fix test failures with api version 2.87+","status":"MERGED","created":"2024-10-07 22:06:42.000000000","updated":"2024-10-09 16:02:29.000000000","submitted":"2024-10-09 16:02:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931712-bp/metadata-for-share-resources","meta_rev_id":"1851fd9a7734da4f0e35df01d5573f9922b65cf1","_number":931712,"virtual_id_number":931712,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-09 16:02:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2024-10-09 14:31:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2024-10-09 05:01:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2024-10-09 14:31:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":33728,"name":"Ashley Rodriguez","email":"ashrodri@redhat.com"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":33728,"name":"Ashley Rodriguez","email":"ashrodri@redhat.com"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-07 22:07:05.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2024-10-07 22:07:05.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"},{"updated":"2024-10-07 22:07:05.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":33728,"name":"Ashley Rodriguez","email":"ashrodri@redhat.com"},"state":"REVIEWER"},{"updated":"2024-10-07 22:45:49.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-07 22:52:43.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-10-07 23:51:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-08 22:14:18.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-10-09 14:31:05.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"ba51dd167341733c62cee919af6358ce892df830","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-10-07 22:06:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14462f55b5d296b9a8a4398f2a8cb4c5edc94e67","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-07 22:45:49.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/12/931712/1/check/DellEMC_PowerFlex_Manila_dhss_false/d5c4889/DellEMC_PowerFlex_Manila_dhss_false/1831 : SUCCESS in 33m 21s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"13feb51643c4da204fc0c8fed0a53cd0cf40cce9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-07 22:52:43.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/12/931712/1/check/DellEMC_Unity_Manila_dhss_true/27e9614/DellEMC_Unity_Manila_dhss_true/1814 : FAILURE in 13m 37s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/12/931712/1/check/DellEMC_Unity_Manila_dhss_false/9b68c8e/DellEMC_Unity_Manila_dhss_false/1917 : FAILURE in 35m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"ea7e34b9acc2a81e84cabfa1daf090088f1dc9da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 23:51:40.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1dbbe050b54346bd9722f5b7cb853a65\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0a533834b85408b993e146f979a3be9 : FAILURE in 2m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/42eb0ceb808140bc800bbbdb4987c813 : SUCCESS in 54m 59s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f056c42df4dc44a58e09fb29f5f6c06d : FAILURE in 1h 44m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e2e5d7c1f9214d399e124104ed4f79a6 : FAILURE in 1h 10m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/feaa7fe04b9a4ed0bb7ab34d715bdec4 : FAILURE in 46m 33s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/97547887333c43ccbe05afd6733355df : FAILURE in 9m 48s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4248ea1631c44d3ca5fa4030a2bc59cf : FAILURE in 13m 18s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d99a0c27f0934e00b7731a74b07ca6aa : FAILURE in 24m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9fecd72302fa46e2ba85f8137fe484d7 : FAILURE in 39m 03s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/0909891b7d254fe4b5ba0ba1969ff268 : SUCCESS in 1h 11m 23s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/2af2749e8aec4091be529ff1dd24da67 : SUCCESS in 1h 15m 25s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/fed8bacd867c416bbcf3031a75820b9c : SUCCESS in 50m 27s\n- manila-tempest-plugin-lvm-2023.1 https://zuul.opendev.org/t/openstack/build/994bc2ed699e4024b76700134a95c22b : SUCCESS in 37m 16s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3e561f8b27694194baca87266f245871 : SUCCESS in 34m 25s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/519cca7d2bf64e35950dbc5e4864d851 : FAILURE in 27m 34s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/108d119ec59d463194766691f27ce6c2 : FAILURE in 10m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"73c15317dcfa459d36829ee09badbda3bc2adee3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-10-08 20:04:00.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b8ddb2a289333365ab5604ccf35bf4c9b4548414","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-08 20:35:27.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/12/931712/2/check/DellEMC_PowerFlex_Manila_dhss_false/f80b37d/DellEMC_PowerFlex_Manila_dhss_false/1841 : SUCCESS in 31m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"97d83b98d2b483b48e8affee25d9ccc97dbef330","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-08 20:39:48.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/12/931712/2/check/DellEMC_Unity_Manila_dhss_true/6800890/DellEMC_Unity_Manila_dhss_true/1824 : FAILURE in 15m 13s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/12/931712/2/check/DellEMC_Unity_Manila_dhss_false/4fa38ee/DellEMC_Unity_Manila_dhss_false/1927 : FAILURE in 35m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"6f1ee2568ce9b5bea3cd369b1b76184f3888d45b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-08 21:57:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1b2f251d0614098827732d28dc3efb3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7e91871bc4d467b8be5f0b458c32d5d : SUCCESS in 3m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c7e9f2044b404839b383d3838685313a : SUCCESS in 54m 30s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b3994d6e7bc24c32986cdaa79e5150bc : FAILURE in 1h 51m 03s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/66d4c8c1fae24b418fee13dd1964962a : FAILURE in 27m 29s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e87cf09445c24d31af1b7ab4c1d69ef8 : SUCCESS in 51m 32s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ecb4249507214bdbbd6639ff56bec67a : SUCCESS in 49m 48s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/071a086b8fb347e69b03d1071ba4d205 : FAILURE in 13m 37s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/da2ff0e473dc4e1197572c7595c91244 : SUCCESS in 27m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9553f9259d6a41769a3ecee62aee3fe1 : SUCCESS in 29m 43s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/c67835782cae43788cbaf0619f9f2499 : SUCCESS in 51m 37s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/7cf28fdc1d77455086d90d5a701e83f4 : SUCCESS in 1h 01m 20s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/4da0d97e06f0470997ef2b5211ff4d3e : SUCCESS in 1h 11m 20s\n- manila-tempest-plugin-lvm-2023.1 https://zuul.opendev.org/t/openstack/build/78f79bcf35ef4bb78d3885bdb35c627a : SUCCESS in 53m 36s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/b8e5dff4c56a43fe8d2846328fdb35a7 : SUCCESS in 24m 40s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/328202a2f8fb4a91bb7a197a4155edad : FAILURE in 25m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0e8f9c44006a4455875f5f043b5fa012 : FAILURE in 22m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4a82ca471e6d4dfdc966ceb8f6a9fec58bbd448a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-08 22:11:17.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/12/931712/2/check/DellEMC_Unity_Manila_dhss_true/aaf08ec/DellEMC_Unity_Manila_dhss_true/1825 : FAILURE in 13m 43s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/12/931712/2/check/DellEMC_Unity_Manila_dhss_false/7d1ca1a/DellEMC_Unity_Manila_dhss_false/1928 : FAILURE in 12m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"464555e8f043898f2209b7065147a40addde64eb","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-10-08 22:14:18.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/12/931712/2/upstream-check/manila-tempest-plugin-ontap-dhss/c739d78/ : RETRY_LIMIT in 9m 01s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"b98bb16c4f7b27c473014b128f19b7468004235c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-08 22:27:38.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/12/931712/2/check/DellEMC_PowerFlex_Manila_dhss_false/7a465cb/DellEMC_PowerFlex_Manila_dhss_false/1842 : SUCCESS in 30m 06s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"2406e0f3c3793084ce2cad887133c2953b929e60","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-10-09 05:01:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7e81b795079b6545dedb01bd9edf0c51bb5ed603","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-10-09 14:31:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eec29762055713ecad72b49cdceac9105a5d81c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 14:31:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7e8d253567f5c238521d151381fb15cf1148c9f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 16:02:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccd54de1eb194cc398e5249e71ecca16\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6106d9c58a94c8ebd46921572bc9b07 : SUCCESS in 2m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a6e4799334984ed6b4c019979f02a091 : SUCCESS in 1h 18m 11s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f34927ae93cd418bb1079d3d0d61a6f4 : SUCCESS in 39m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f7a2b2bb7f8e438dbcda5f45915efb26 : SUCCESS in 38m 36s","accounts_in_message":[],"_revision_number":2},{"id":"1851fd9a7734da4f0e35df01d5573f9922b65cf1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 16:02:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"986a0d7ba39c4d2657a099a0110de40cbedc15af","revisions":{"0be0526b3a1fbd8c0e1f143b3e7f406051e1531c":{"kind":"REWORK","_number":1,"created":"2024-10-07 22:06:42.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/12/931712/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/12/931712/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/931712/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/931712/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/931712/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/931712/1"}}},"commit":{"parents":[{"commit":"b1f749a046ad09f89a6d8ed74dedbc7540d693b0","subject":"Add stable/2024.2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b1f749a046ad09f89a6d8ed74dedbc7540d693b0"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-10-07 22:05:56.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-10-07 22:05:56.000000000","tz":-420},"subject":"Fix test failures with api version 2.87+","message":"Fix test failures with api version 2.87+\n\nexport location structure has changed in 2.87.\n\nNew test cases are being added in\nI113ea3095cdc9d5d2903ba83641afd3ee112c17d.\n\nChange-Id: Ieb8dfc8813a3366b4252b6816397632ffb6a20bf\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0be0526b3a1fbd8c0e1f143b3e7f406051e1531c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0be0526b3a1fbd8c0e1f143b3e7f406051e1531c"}]},"branch":"refs/heads/master"},"986a0d7ba39c4d2657a099a0110de40cbedc15af":{"kind":"REWORK","_number":2,"created":"2024-10-08 20:04:00.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/12/931712/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/12/931712/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/931712/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/931712/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/931712/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/931712/2"}}},"commit":{"parents":[{"commit":"b1f749a046ad09f89a6d8ed74dedbc7540d693b0","subject":"Add stable/2024.2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b1f749a046ad09f89a6d8ed74dedbc7540d693b0"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-10-07 22:05:56.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-10-08 20:03:59.000000000","tz":0},"subject":"Fix test failures with api version 2.87+","message":"Fix test failures with api version 2.87+\n\nexport location structure has changed in 2.87.\n\nNew test cases are being added in\nI113ea3095cdc9d5d2903ba83641afd3ee112c17d.\n\nChange-Id: Ieb8dfc8813a3366b4252b6816397632ffb6a20bf\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/986a0d7ba39c4d2657a099a0110de40cbedc15af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/986a0d7ba39c4d2657a099a0110de40cbedc15af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
