)]}'
{"id":"openstack%2Fmanila-tempest-plugin~931731","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I87905f1358ecfb8958af3b85fe587be98b415e9d","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-08 05:34:16.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I87905f1358ecfb8958af3b85fe587be98b415e9d","subject":"Change [share] catalog_type default","status":"NEW","created":"2024-10-08 03:24:10.000000000","updated":"2024-10-08 05:56:32.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2cd0014543e8a45a73d1973a7e9361f09111744c","_number":931731,"virtual_id_number":931731,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-08 05:18:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2024-10-08 04:12:21.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-08 03:57:54.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-08 04:00:30.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-10-08 05:18:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-08 05:34:16.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"41640eda9dd54690c2f61db3582f2e89ebe9492a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-08 03:24:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d294db6eba887787e1343e280ffe608ca409a23d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-08 03:57:54.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/31/931731/1/check/DellEMC_PowerFlex_Manila_dhss_false/442055a/DellEMC_PowerFlex_Manila_dhss_false/1836 : SUCCESS in 33m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"1197b428e67c7fd0eec3c0f8cd2aeaff50ef5fb7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-08 04:00:30.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/31/931731/1/check/DellEMC_Unity_Manila_dhss_true/76a944f/DellEMC_Unity_Manila_dhss_true/1819 : FAILURE in 13m 33s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/31/931731/1/check/DellEMC_Unity_Manila_dhss_false/3aee03c/DellEMC_Unity_Manila_dhss_false/1922 : FAILURE in 36m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"2f739bc3a5f4e6f0c7492f73a190e546546897fe","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-08 04:12:21.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"7b77f8de6890a1ed34856e01fabff68e5fe7b1ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-08 05:18:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5deb60a6cca4a02bddf7e3d77980c79\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cce83ad1ef2420aa0fa35df986428f1 : SUCCESS in 2m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/57f35b37562e4727b198ddab2d2f3ebb : SUCCESS in 57m 23s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/37aaa92c25204f66abe8ce2aff6e58de : SUCCESS in 1h 51m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/255e8e84c19d4962b8089a9ad5cddf0e : FAILURE in 1h 24m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/48599e0b318847ac968ca3d36e1777c8 : SUCCESS in 59m 13s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/198b42d9377542daaf173431d1589ae9 : SUCCESS in 1h 00m 39s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/01ce0c8d1f0648569a504cb9e7254681 : FAILURE in 26m 13s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0862a23c5b0d4e3e96e9da2acda50796 : SUCCESS in 35m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ac69b5e2d5e44ee79cd1c3469400c482 : SUCCESS in 48m 35s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/e70f9402537c4f9486d0000803c668bb : SUCCESS in 1h 07m 46s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/49deede9a85642e39383c29fb616262b : SUCCESS in 55m 45s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/97ecd14e784744c3b9cbefb6b875e600 : SUCCESS in 57m 05s\n- manila-tempest-plugin-lvm-2023.1 https://zuul.opendev.org/t/openstack/build/bf269345de594d8eaf967f119a273172 : SUCCESS in 52m 50s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ac30193e752e4a239729aae77d3d72c7 : SUCCESS in 22m 46s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8e9928ea60774306a9145ce28a0724dd : FAILURE in 13m 04s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/546803c0563647f9979f2cbbd6bb39c6 : FAILURE in 9m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f85c3a8f1efd285416a6b0e8f9073a7e6ada162f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-10-08 05:34:16.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/931731/1/upstream-check/manila-tempest-plugin-ontap-dhss/52d04f7/ : RETRY_LIMIT in 8m 47s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/931731/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/5c5c0d9/ : RETRY_LIMIT in 9m 10s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"cef00f9fee5be6eeea97a818a81e1b36fe033a4e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-08 05:51:02.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/31/931731/1/check/DellEMC_PowerFlex_Manila_dhss_false/595f277/DellEMC_PowerFlex_Manila_dhss_false/1837 : SUCCESS in 31m 52s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"2cd0014543e8a45a73d1973a7e9361f09111744c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-08 05:56:32.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/31/931731/1/check/DellEMC_Unity_Manila_dhss_true/4227329/DellEMC_Unity_Manila_dhss_true/1820 : FAILURE in 13m 38s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/31/931731/1/check/DellEMC_Unity_Manila_dhss_false/1e12acc/DellEMC_Unity_Manila_dhss_false/1923 : FAILURE in 37m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"24662dd4293a5d0dd64182cd70d4938ebc6f9390","revisions":{"24662dd4293a5d0dd64182cd70d4938ebc6f9390":{"kind":"REWORK","_number":1,"created":"2024-10-08 03:24:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/931731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/31/931731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/931731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/931731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/931731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/931731/1"}}},"commit":{"parents":[{"commit":"960f9f6e2ca363f034ede3c02310bb4e3e5e7169","subject":"Merge \"Disable Ceph for Nova in cephfs-nfs multinode job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/960f9f6e2ca363f034ede3c02310bb4e3e5e7169"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-08 03:22:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-08 03:24:00.000000000","tz":540},"subject":"Change [share] catalog_type default","message":"Change [share] catalog_type default\n\nFollowing the change[1] recently merged into tempest, update\nthe default catalog type to the name defined in service name authority.\n\n[1] https://review.opendev.org/c/openstack/tempest/+/930296\n\nChange-Id: I87905f1358ecfb8958af3b85fe587be98b415e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/24662dd4293a5d0dd64182cd70d4938ebc6f9390"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/24662dd4293a5d0dd64182cd70d4938ebc6f9390"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
