)]}'
{"id":"openstack%2Fmanila-tempest-plugin~932956","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Id7ddf6e39ae8c0de72b2d48d6beed718fcd760ce","project":"openstack/manila-tempest-plugin","branch":"master","topic":"migrate-to-noble","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-12-04 11:31:47.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-12-04 11:31:47.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2024-12-04 11:31:47.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-12-04 11:31:47.000000000","reason":"Change was submitted"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2024-12-04 10:18:16.000000000","reason":"\u003cGERRIT_ACCOUNT_19262\u003e replied on the change","reason_account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}},"hashtags":[],"change_id":"Id7ddf6e39ae8c0de72b2d48d6beed718fcd760ce","subject":"Update gate jobs as per the 2025.1 cycle testing runtime","status":"MERGED","created":"2024-10-22 02:36:08.000000000","updated":"2024-12-04 11:31:47.000000000","submitted":"2024-12-04 11:31:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"932956-migrate-to-noble","meta_rev_id":"24b8bdf32278fa55238e3379d1a7172f990bc739","_number":932956,"virtual_id_number":932956,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-04 11:31:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2024-11-19 06:03:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-04 10:18:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-04 10:18:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-22 03:08:35.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-22 03:10:26.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-10-22 05:22:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-23 19:16:42.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2024-11-15 22:19:09.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-11-19 06:03:29.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-12-03 23:12:20.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2024-12-03 23:12:20.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2024-12-03 23:12:20.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"194be227d6747df18cde352e3d8f84ec821490ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-22 02:36:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4e8418c82f604d799f02016564eedda9c6dd55c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-22 03:08:35.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/56/932956/1/check/DellEMC_PowerFlex_Manila_dhss_false/f433fe6/DellEMC_PowerFlex_Manila_dhss_false/1911 : SUCCESS in 32m 11s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"30dd7ee237258b3761ec5a36ceb00975700506a6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-22 03:10:26.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/56/932956/1/check/DellEMC_Unity_Manila_dhss_true/b541c7e/DellEMC_Unity_Manila_dhss_true/1894 : FAILURE in 13m 24s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/56/932956/1/check/DellEMC_Unity_Manila_dhss_false/5c5a19f/DellEMC_Unity_Manila_dhss_false/1997 : FAILURE in 34m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"0af6bcdab4bb47bdf64e9217f78e275079c594c0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-22 03:14:50.000000000","message":"Topic set to migrate-to-noble","accounts_in_message":[],"_revision_number":1},{"id":"8cdbb0462e8c13e66f51ee507c54a76448645d3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-22 05:22:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/26016dc4c36c49f1a08acf8de9c7aae4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dc0acfc220c497cbb1f2a9930b21020 : SUCCESS in 4m 23s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/078205cd57b34140b1e37e01107bf328 : SUCCESS in 1h 15m 05s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/791691675a494d23bdc4710796a322d6 : SUCCESS in 1h 05m 18s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/af616ea72ae241a7b565668d01736393 : SUCCESS in 2h 15m 02s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0457f14b7e6e4d75bcb8aaf763599dd0 : FAILURE in 56m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/02f839bb14974ce89543cbdf28285179 : FAILURE in 1h 26m 07s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/469ce01f165f413cb355037fb9af26f1 : FAILURE in 1h 22m 15s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9b95e7d85fdb43f9bcbeff05ef7d3e88 : FAILURE in 25m 57s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/540673d040a842edb07c7f3522d19f73 : FAILURE in 48m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d629d2e5578f4bb986dcafaccf786542 : FAILURE in 52m 09s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/d40285a83f8c4826bfcac6c90a6ed84b : SUCCESS in 1h 13m 05s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/7b0c6210fd184ae693e3838aea1ae626 : SUCCESS in 1h 00m 14s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/efe93e14cd384906892faf919845107e : SUCCESS in 1h 00m 33s\n- manila-tempest-plugin-lvm-2023.1 https://zuul.opendev.org/t/openstack/build/518478488d2b4a4ebf6026e218c99dd2 : SUCCESS in 1h 08m 24s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/6842707a6a0c4f0cbcbb578b60c1bc59 : SUCCESS in 34m 47s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/72fe3fe571384a1a8d116e022359d482 : FAILURE in 28m 55s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/93a59dfbead84fae97dc4a8b59b87336 : FAILURE in 12m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d77344c62cc7d06edb8bd92cd216a0c1289fb333","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-10-23 19:16:42.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3c2ca397c2650cb8c65f36d6df95b7734f383a29","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-24 20:21:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0cb5e3c59e2cad5a3fd030aaf7e9b5e0398d800c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-10-25 20:06:13.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_8556\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"_revision_number":2},{"id":"a227fc71309d31528bec04a919e54571d0ac1d40","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-25 20:39:13.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/56/932956/2/check/DellEMC_PowerFlex_Manila_dhss_false/632331e/DellEMC_PowerFlex_Manila_dhss_false/1941 : SUCCESS in 32m 42s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"ebe8d9c56d0d09f8a5c57a8e62a978526df92983","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-25 20:49:39.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/56/932956/2/check/DellEMC_Unity_Manila_dhss_true/f67aadd/DellEMC_Unity_Manila_dhss_true/1924 : FAILURE in 22m 03s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/56/932956/2/check/DellEMC_Unity_Manila_dhss_false/1220fad/DellEMC_Unity_Manila_dhss_false/2028 : FAILURE in 40m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"3cf9ab286b993f38f97a790c79d9a9340b8c6fcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 21:48:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/22b7aeb462a84c7f8622a629037b81bc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7245d58da004b83b46ac08c2a4b027b : SUCCESS in 3m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/289945b5bf824275aaadc16efcc5c190 : SUCCESS in 54m 48s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/0521ddbea8c3431ea052b4b5e9522d69 : SUCCESS in 1h 11m 37s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/aa4ed73a3d1a4a58ade59a2919c36a05 : SUCCESS in 1h 39m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f9a48956773b478bb565a088cd2826d8 : FAILURE in 57m 28s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e7e8ce08655347ca9b14a652e1275b13 : SUCCESS in 50m 41s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/63c3c14d35d54c3d856f2a7f6b220ea6 : SUCCESS in 1h 06m 21s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c1f792f3aa424add81b820d08169dc3e : FAILURE in 23m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ed4b14b15061433cb4168e8bf41c0224 : SUCCESS in 31m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0aee11607bac4b1b95db5ea2a5d6f7c6 : FAILURE in 31m 12s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/b37043dcd6d3416b89ff218ef769e880 : SUCCESS in 53m 41s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/69cfd6ffe63140068f7bf64dfcc1b850 : SUCCESS in 41m 08s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/66a83cee559d4626ba75a54233998d09 : SUCCESS in 1h 07m 45s\n- manila-tempest-plugin-lvm-2023.1 https://zuul.opendev.org/t/openstack/build/4170d8843e1245b6bb78d5cd6c8e398f : SUCCESS in 1h 21m 59s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/586ee28bb18d406fbf7d473dd33a4d1c : SUCCESS in 21m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d98412f2199b4d68b6c5bfe1e5d1938b : FAILURE in 39m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"18d23da85265614fb36c5a4aafeabd9c87ebe751","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-11-07 20:26:55.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_8556\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"_revision_number":3},{"id":"f2a115380622adf7376e634c32a3419c90f83e78","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-07 21:35:16.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/56/932956/3/check/DellEMC_PowerFlex_Manila_dhss_false/2838f40/DellEMC_PowerFlex_Manila_dhss_false/1991 : SUCCESS in 33m 11s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"b4ff79feddf5abbcc19318187953f02cb74f8ba3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-11-07 21:56:13.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/56/932956/3/check/DellEMC_Unity_Manila_dhss_true/a7e2c14/DellEMC_Unity_Manila_dhss_true/1974 : FAILURE in 14m 04s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/56/932956/3/check/DellEMC_Unity_Manila_dhss_false/cd539ab/DellEMC_Unity_Manila_dhss_false/2078 : FAILURE in 35m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"35dd574a21ff2c48484c54053ede2067c64f27a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 22:24:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3da0199fba3c4b9088565e2bf18bb3c1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25d6c844140b434d9412042267a7872d : SUCCESS in 3m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7eb16ffd81c44096a12f837998636c1d : SUCCESS in 51m 27s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/86ec61c26319442a977ac10fc3df3b02 : FAILURE in 58m 13s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/6ca0ff2d46c04ba297f1e562cd6c79ba : SUCCESS in 1h 55m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5e806c8639aa4103ae9a5030964f9cf1 : FAILURE in 1h 03m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/70bd5697b065445583358a8860de5769 : FAILURE in 39m 14s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/7bb642336e604f67bc2edfc833f5063c : FAILURE in 6m 44s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e80d5950a65f4cb3a9a4f94b5fc72b62 : FAILURE in 11m 42s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0dc78d50e3b64d98a7028f7113d0e1c9 : SUCCESS in 31m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a89cbb0367e2464ab699f1d6f771ee28 : SUCCESS in 32m 47s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/614026d4c01d48048fdd89b6a36b8433 : SUCCESS in 55m 02s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/89c02b16d24f45d9b236ba07f78a33a2 : SUCCESS in 1h 14m 23s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/0707670012694771b5cd2749f358e8ec : SUCCESS in 38m 00s\n- manila-tempest-plugin-lvm-2023.1 https://zuul.opendev.org/t/openstack/build/c04e2d7c42434d83b43b69137509e170 : SUCCESS in 54m 43s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/5d560a7b063641e7a73c1971c4e8a7b6 : SUCCESS in 22m 28s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5e8674656a654259a4bcfacb0bcc7ba9 : FAILURE in 48m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1e163c85e80d43c98a0e09c084e6b710919cfd28","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-11-14 19:27:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"42c04b5cc56d6a8694812a06371ac6a8dc2da436","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-11-14 19:30:07.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"844659f14498263139045b69242b9b2978370c13","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-14 20:03:42.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/56/932956/4/check/DellEMC_PowerFlex_Manila_dhss_false/44aead7/DellEMC_PowerFlex_Manila_dhss_false/2037 : SUCCESS in 33m 15s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"1c219d5df92556d7d93471987acb79306d88f462","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-11-14 20:06:15.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/56/932956/4/check/DellEMC_Unity_Manila_dhss_true/f23a242/DellEMC_Unity_Manila_dhss_true/2020 : FAILURE in 15m 15s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/56/932956/4/check/DellEMC_Unity_Manila_dhss_false/2f2e23c/DellEMC_Unity_Manila_dhss_false/2124 : FAILURE in 35m 35s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"c855c2fa935f61b2ca7dfdad7c6a4ee5865311f9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-11-15 19:03:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"91dc29ef4ecb45a740a4d4723ad6228a0a4c99c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 21:50:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebe4807e4708448b96e6e193b7f537d4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dade5963b7544ab3b3b306058bbf10ea : SUCCESS in 3m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b092d32a9f0144a8b0f372ee01ceffe4 : SUCCESS in 1h 10m 05s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/28ed7bbf3e6d4d9cb31316305cbe6b49 : SUCCESS in 39m 20s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ea44a15a73944ad78b4596018b9382b1 : FAILURE in 2h 28m 17s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5fbe43ce85744c24bddfde9680e8ec26 : FAILURE in 1h 05m 35s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9cebb567314646db93c6c51fd3336099 : SUCCESS in 49m 14s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/bb514b59de0a42139436881e136b4092 : FAILURE in 42m 41s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/70c6aa512ec2482bb766a48b08bd4c60 : SUCCESS in 1h 05m 39s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4602bbabd4fe463eb0a47ec0be8d1bff : SUCCESS in 38m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b846c2b2c453489da98dc7bb470e3511 : SUCCESS in 47m 14s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/0a545681db554917bfcc1354de10dab0 : SUCCESS in 1h 17m 10s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/9dd746443c63402a8312724dbd916894 : SUCCESS in 55m 57s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/0257b43794f8432e8470fb213475e43e : SUCCESS in 1h 23m 46s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3380b3a579e645cf9b3bb0584bfe2fd2 : SUCCESS in 44m 08s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b82018ad6f1c4602a37a2826511c8b44 : FAILURE in 54m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d3178ed227fb7998333597ca389cd9744a2c578d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-11-15 22:19:09.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss finger://managesf.services.openstack.netapp.com:7979/37031c2db65046f6adaef813eb434d72 : RETRY_LIMIT in 11m 00s\n- manila-tempest-plugin-ontap-no-dhss finger://managesf.services.openstack.netapp.com:7979/483ddf7492c54991bb98e29919d693ed : RETRY_LIMIT in 10m 26s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"7ed0257fbfa7ed26bbf6b3354a7e1847d6258777","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-11-15 22:23:44.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/56/932956/4/check/DellEMC_Unity_Manila_dhss_true/b1370b3/DellEMC_Unity_Manila_dhss_true/2033 : FAILURE in 1s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/56/932956/4/check/DellEMC_Unity_Manila_dhss_false/f8b8c59/DellEMC_Unity_Manila_dhss_false/2137 : FAILURE in 32m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"6c8ce1469bbb448515e61e63f478c5962f684401","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-15 22:24:49.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/56/932956/4/check/DellEMC_PowerFlex_Manila_dhss_false/bdc3880/DellEMC_PowerFlex_Manila_dhss_false/2050 : SUCCESS in 34m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"5a5e570c63478a87e3a8b278c0217080158c0b9e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-11-19 06:03:29.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"26b3b1073428d87d4cb779d3d2503a48d46904ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 08:32:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f3f77b835744c1c80ca348dde22dfde\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a87b29ab19df4e6fbcd82808e1bb4c00 : SUCCESS in 3m 07s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/91515563acc644f98ed5b00b46193c7d : SUCCESS in 1h 24m 40s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/12202b630fe044b8abde154bcde60fd0 : SUCCESS in 1h 10m 19s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/13825501b64b4c598dddd2a828afcb64 : SUCCESS in 2h 21m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ec07b115d4db40f1bafd3603a86441a3 : FAILURE in 1h 26m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d0fe9db4b6f746ea8f32fdc09818fb04 : SUCCESS in 1h 07m 34s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/aaa8abd1f74343d3800b536e3d0b3547 : SUCCESS in 1h 06m 21s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2f780a847bb5402cb8bc413b87229a2c : SUCCESS in 1h 17m 40s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/08eafce614be4bf89bf2867cb3db3745 : SUCCESS in 26m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/47b8dea3fdba45fd998419f37ea8c1ef : SUCCESS in 53m 47s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/b6c631483dc94472bd20e731709d8a1b : SUCCESS in 1h 18m 38s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/408f0ece4b0b43de8995375bfa05a722 : SUCCESS in 1h 10m 09s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/655884fd0e3049379d701fa907712208 : SUCCESS in 1h 16m 10s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/977caea67ffa48aa8b9f9f7738603a53 : SUCCESS in 21m 25s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9fe216be65874a99a5751fcb10d93713 : FAILURE in 39m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fea94bef3c537c70cde982364518aa5b3bc96409","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-11-19 08:50:06.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss finger://managesf.services.openstack.netapp.com:7979/1e30ffc2fde34356b3849b030e357f23 : RETRY_LIMIT in 5m 13s\n- manila-tempest-plugin-ontap-no-dhss finger://managesf.services.openstack.netapp.com:7979/b506faa748a245b988cfe883ab422c7f : RETRY_LIMIT in 5m 29s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"3c77d2dd3a7a3269f5277ea95f6ec984d5b2a555","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-11-19 09:05:32.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/56/932956/4/check/DellEMC_Unity_Manila_dhss_true/bc4b346/DellEMC_Unity_Manila_dhss_true/2049 : FAILURE in 1s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/56/932956/4/check/DellEMC_Unity_Manila_dhss_false/854aa4c/DellEMC_Unity_Manila_dhss_false/2153 : FAILURE in 32m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"1e800cace7923b48700ba7b36bcea5913bf368e5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-19 09:07:09.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/56/932956/4/check/DellEMC_PowerFlex_Manila_dhss_false/f673a9a/DellEMC_PowerFlex_Manila_dhss_false/2066 : SUCCESS in 34m 37s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"c5331d46981179884dc7b9723f71ac002f8e6430","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-30 19:16:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c60049ae98c9f753bf258438e772957f331577ed","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-12-04 10:18:16.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5d13fd7a509b828c20b1f670b8084f6427c66e15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 10:18:55.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"afdf0bf6b2b9e70fca34b408e61621e83aaec359","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 11:31:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d1b4a29bc5c4631b611d74f05b51f99\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4499475cdd9d40e0bd54d3a1f8a7615d : SUCCESS in 3m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f3a2b4aeccb54bd794e12e4f39eaad61 : SUCCESS in 1h 12m 48s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/69526def26724d7d850f8cf8e17ef396 : SUCCESS in 57m 04s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/86fdeb1809044896a38762149f416c9c : SUCCESS in 33m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e72c3ed6e4704d1caaab9acebc3c90a1 : SUCCESS in 38m 55s","accounts_in_message":[],"_revision_number":4},{"id":"24b8bdf32278fa55238e3379d1a7172f990bc739","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 11:31:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"32da8e27c72ee73caa390a83785fc3d19b6392dc","revisions":{"f71fdaec004d68c2889add22d9a2d36591ff0f74":{"kind":"REWORK","_number":1,"created":"2024-10-22 02:36:08.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/56/932956/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/56/932956/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/56/932956/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/56/932956/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/56/932956/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/56/932956/1"}}},"commit":{"parents":[{"commit":"1cf281f9878f3ece0d64e84dd6885d3c4de8f336","subject":"Merge \"Drop unnecessary \u0027x\u0027 bit from doc config file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1cf281f9878f3ece0d64e84dd6885d3c4de8f336"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-22 02:35:47.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-22 02:36:07.000000000","tz":-420},"subject":"Update gate jobs as per the 2025.1 cycle testing runtime","message":"Update gate jobs as per the 2025.1 cycle testing runtime\n\nAs per 2025.1 testing runtime[1], we need to test on Ubuntu\nNoble (which will be taken care by depends-on tempest and devstack\npatches to move base jobs to Noble) and at least single job to run on\nUbuntu Jammy (for smooth upgrade from previous releases).\n\nThis commit adds a new job to run on Jammy which can be removed\nin future cycle when testing runtime test next version of Ubuntu\nas default.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/932156\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: Id7ddf6e39ae8c0de72b2d48d6beed718fcd760ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f71fdaec004d68c2889add22d9a2d36591ff0f74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f71fdaec004d68c2889add22d9a2d36591ff0f74"}]},"branch":"refs/heads/master"},"4e369bdb2de7c32e37d2c9db52ffba348bffb20d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-25 20:06:13.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/56/932956/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/56/932956/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/56/932956/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/56/932956/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/56/932956/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/56/932956/2"}}},"commit":{"parents":[{"commit":"c1f3f37990840be797927c826fc638a661ebb5f6","subject":"Replace the use of assertDictContainsSubset","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c1f3f37990840be797927c826fc638a661ebb5f6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-22 02:35:47.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-10-25 20:06:13.000000000","tz":0},"subject":"Update gate jobs as per the 2025.1 cycle testing runtime","message":"Update gate jobs as per the 2025.1 cycle testing runtime\n\nAs per 2025.1 testing runtime[1], we need to test on Ubuntu\nNoble (which will be taken care by depends-on tempest and devstack\npatches to move base jobs to Noble) and at least single job to run on\nUbuntu Jammy (for smooth upgrade from previous releases).\n\nThis commit adds a new job to run on Jammy which can be removed\nin future cycle when testing runtime test next version of Ubuntu\nas default.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/932156\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: Id7ddf6e39ae8c0de72b2d48d6beed718fcd760ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4e369bdb2de7c32e37d2c9db52ffba348bffb20d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4e369bdb2de7c32e37d2c9db52ffba348bffb20d"}]},"branch":"refs/heads/master","description":"Rebase"},"41279a68cea92eda146919da78c8ae1dbc787183":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-11-07 20:26:55.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/56/932956/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/56/932956/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/56/932956/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/56/932956/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/56/932956/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/56/932956/3"}}},"commit":{"parents":[{"commit":"f4b555705867d95ab0b24845c91195c22d410235","subject":"Fix last use of assertDictContainsSubset","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f4b555705867d95ab0b24845c91195c22d410235"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-22 02:35:47.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-11-07 20:26:55.000000000","tz":0},"subject":"Update gate jobs as per the 2025.1 cycle testing runtime","message":"Update gate jobs as per the 2025.1 cycle testing runtime\n\nAs per 2025.1 testing runtime[1], we need to test on Ubuntu\nNoble (which will be taken care by depends-on tempest and devstack\npatches to move base jobs to Noble) and at least single job to run on\nUbuntu Jammy (for smooth upgrade from previous releases).\n\nThis commit adds a new job to run on Jammy which can be removed\nin future cycle when testing runtime test next version of Ubuntu\nas default.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/932156\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: Id7ddf6e39ae8c0de72b2d48d6beed718fcd760ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/41279a68cea92eda146919da78c8ae1dbc787183"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/41279a68cea92eda146919da78c8ae1dbc787183"}]},"branch":"refs/heads/master","description":"Rebase"},"32da8e27c72ee73caa390a83785fc3d19b6392dc":{"kind":"REWORK","_number":4,"created":"2024-11-14 19:30:07.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/56/932956/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/56/932956/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/56/932956/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/56/932956/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/56/932956/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/56/932956/4"}}},"commit":{"parents":[{"commit":"f4b555705867d95ab0b24845c91195c22d410235","subject":"Fix last use of assertDictContainsSubset","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f4b555705867d95ab0b24845c91195c22d410235"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-22 02:35:47.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-11-14 19:30:06.000000000","tz":0},"subject":"Update gate jobs as per the 2025.1 cycle testing runtime","message":"Update gate jobs as per the 2025.1 cycle testing runtime\n\nAs per 2025.1 testing runtime[1], we need to test on Ubuntu\nNoble (which will be taken care by depends-on tempest and devstack\npatches to move base jobs to Noble) and at least single job to run on\nUbuntu Jammy (for smooth upgrade from previous releases).\n\nThis commit adds a new job to run on Jammy which can be removed\nin future cycle when testing runtime test next version of Ubuntu\nas default.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/932156\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: Id7ddf6e39ae8c0de72b2d48d6beed718fcd760ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/32da8e27c72ee73caa390a83785fc3d19b6392dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/32da8e27c72ee73caa390a83785fc3d19b6392dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
