)]}'
{"id":"openstack%2Fmanila-tempest-plugin~933198","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Iabd59b683ac998a784eccc7fbc634ad1eb20e889","project":"openstack/manila-tempest-plugin","branch":"master","topic":"pre-commit-skip-functional","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-24 22:07:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iabd59b683ac998a784eccc7fbc634ad1eb20e889","subject":"Skip functional tests on pre-commit config update","status":"MERGED","created":"2024-10-24 01:53:49.000000000","updated":"2024-10-24 22:07:45.000000000","submitted":"2024-10-24 22:07:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933198-pre-commit-skip-functional","meta_rev_id":"3767c00e6ec81a466584d91f38d1f3f1cbbe052c","_number":933198,"virtual_id_number":933198,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-24 22:07:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2024-10-24 21:10:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2024-10-24 17:33:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2024-10-24 21:10:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-24 02:49:47.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-24 02:50:49.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-10-24 03:50:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-24 11:37:54.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-10-24 17:33:47.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2024-10-24 21:10:56.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"2181c3c52381abe5b5a6b98a710834ae1c4ed8bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-24 01:53:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d483576cc3de908a012caefd12bd3f3f79e4db34","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-24 02:49:47.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/98/933198/1/check/DellEMC_PowerFlex_Manila_dhss_false/d247993/DellEMC_PowerFlex_Manila_dhss_false/1927 : SUCCESS in 31m 35s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"5361e7eba3450a3bd4f5ebfca01aa015aa564e67","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-24 02:50:49.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/98/933198/1/check/DellEMC_Unity_Manila_dhss_true/d73204d/DellEMC_Unity_Manila_dhss_true/1910 : FAILURE in 13m 39s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/98/933198/1/check/DellEMC_Unity_Manila_dhss_false/0e736e6/DellEMC_Unity_Manila_dhss_false/2013 : FAILURE in 33m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"6b956af453bdb883b5e5d4067d0d80897f308fca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 03:50:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6705ae49ede48ab9e33f39a833e2dc0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e04e45ebe5f45b488ad0ffb038e87e2 : SUCCESS in 3m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2c84d7d036bd40c2bb08adbe22883dc2 : SUCCESS in 1h 01m 37s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f38b546e27004a29974c745216e93787 : SUCCESS in 1h 55m 02s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/877e18167f67465180989bd925b1c5c5 : FAILURE in 40m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/65979ccd7bda44b898e583f37bdb1a4b : SUCCESS in 1h 11m 34s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/2b4f57a5dfa141c080e7c52173bf8de2 : SUCCESS in 57m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/de1b814a2f9a4a7691b357c26c5cc691 : FAILURE in 15m 10s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fcc14ca6ab67432ab8bdb06db1d284c3 : SUCCESS in 38m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d642733d4bdc4b9e9433e343ba9189a6 : SUCCESS in 33m 24s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/39e04aa0b1df44408ccf7e7511e95dd1 : SUCCESS in 1h 17m 33s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/365a672d478c4feda7931c5b32635b11 : FAILURE in 1h 21m 06s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/b86b92ab0a6d44888f01b271a7aeb952 : SUCCESS in 1h 19m 51s\n- manila-tempest-plugin-lvm-2023.1 https://zuul.opendev.org/t/openstack/build/730359d89a5a4217aae45e4bf0a3ca52 : SUCCESS in 1h 10m 59s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/0254073e086d4b0ab70941b802db7822 : SUCCESS in 33m 37s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/78fcb6470da44c249e7f4ca62cb578ee : FAILURE in 24m 04s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/64d12fa02cf34b46bf98338afe27043e : FAILURE in 9m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"13d006d33ebbd7f46d90b91ea2b9b7cc73606c84","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-24 08:53:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"701f9b6888aaaa7b85b67c65b3880f30a68ce27a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 11:22:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b7c508b9ce24c99aee5f6cc682370d4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/822409f011bc4d16b52d49e1ea1836e4 : SUCCESS in 2m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a5b7fac78985415c8e68cc7166b59bb7 : SUCCESS in 1h 05m 43s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ca7bcdf165d3477f90510dde30cf6326 : FAILURE in 2h 22m 38s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/85db56a3949143499085bd7eab26df6f : FAILURE in 1h 01m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e40f8b6ba23949f8bd1176ea04ca8e1e : SUCCESS in 1h 03m 03s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f81447322d4d4ddbba49478ef17390f4 : SUCCESS in 1h 04m 39s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f905eae7ee5c4a81b1b26f3b0c128962 : FAILURE in 26m 44s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/31421df7078f44a1b085175f35ec65b9 : SUCCESS in 33m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9df24e622ce64df8a4f5ffdb8cffccbf : SUCCESS in 45m 58s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/7d717af74bad4b93a32d89b0ef442c7a : SUCCESS in 1h 19m 29s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/5bd647d3df0e4a5599f04f7f71e30380 : SUCCESS in 1h 08m 43s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/5aaceceb80c6411e83410533b125edda : SUCCESS in 1h 10m 50s\n- manila-tempest-plugin-lvm-2023.1 https://zuul.opendev.org/t/openstack/build/7d2e3b7cd13544b4aee01782b73a79ea : SUCCESS in 1h 19m 45s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a6f28896d72e4ad798c977da5f730f8e : SUCCESS in 37m 53s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/563b00de1bc5410d8438b47db24de497 : FAILURE in 19m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/fd51e0e2da5a4dd48128c216188731a8 : FAILURE in 10m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d9e2643c3d38ffb76d21119433520cef5d48c3a5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-10-24 11:37:54.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss finger://managesf.services.openstack.netapp.com:7979/540c5dc08fbd4666b91ddb1e01cb21ce : RETRY_LIMIT in 14m 24s\n- manila-tempest-plugin-ontap-no-dhss finger://managesf.services.openstack.netapp.com:7979/08c733052c6a44029ec2aef3dbee51fb : RETRY_LIMIT in 14m 05s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"5cb10c529e2cd9ab51ddde43d9be74acded99753","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-24 11:54:41.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/98/933198/1/check/DellEMC_PowerFlex_Manila_dhss_false/7799314/DellEMC_PowerFlex_Manila_dhss_false/1933 : SUCCESS in 31m 49s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"479e9ad25294997c5b3dc84b5c9a4d981904c46a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-24 11:57:16.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/98/933198/1/check/DellEMC_Unity_Manila_dhss_true/eff075f/DellEMC_Unity_Manila_dhss_true/1916 : FAILURE in 13m 34s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/98/933198/1/check/DellEMC_Unity_Manila_dhss_false/a62c0d9/DellEMC_Unity_Manila_dhss_false/2020 : FAILURE in 34m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"0c857621c7f4356cb4d9e83942c53d1a2afd2bb4","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2024-10-24 17:33:47.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"57195e476412089384ac5d1afd2880aebfa09335","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-10-24 21:10:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6e4de0d7a498e7f8fd94a0c62f44bab834f3fe48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 21:11:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a95cfdc59a9b092635de6417b207b96bf558e864","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 22:07:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5470fe3bb3864fb9a5b9c7b166c6d690\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d53117b597bb46408611bd272fbe4d57 : SUCCESS in 2m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7a07510d49c743048165968126e2c46c : SUCCESS in 55m 45s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/42385c11c5464c26ba4caf6e4e62908e : SUCCESS in 22m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/97ec37ab32d14db6bbd267262e335f9e : SUCCESS in 29m 27s","accounts_in_message":[],"_revision_number":1},{"id":"3767c00e6ec81a466584d91f38d1f3f1cbbe052c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 22:07:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78bd6b81641480b1fb17a7e10b78178bf4df1827","revisions":{"78bd6b81641480b1fb17a7e10b78178bf4df1827":{"kind":"REWORK","_number":1,"created":"2024-10-24 01:53:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/933198/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/98/933198/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/98/933198/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/98/933198/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/98/933198/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/98/933198/1"}}},"commit":{"parents":[{"commit":"331beaeaf787eeaf396527ebf118267b20e2e745","subject":"Drop unnecessary \u0027x\u0027 bit from doc config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/331beaeaf787eeaf396527ebf118267b20e2e745"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 01:53:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 01:53:21.000000000","tz":540},"subject":"Skip functional tests on pre-commit config update","message":"Skip functional tests on pre-commit config update\n\nThis file is used only in code style checks.\n\nChange-Id: Iabd59b683ac998a784eccc7fbc634ad1eb20e889\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/78bd6b81641480b1fb17a7e10b78178bf4df1827"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/78bd6b81641480b1fb17a7e10b78178bf4df1827"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
