)]}'
{"id":"openstack%2Fmanila-tempest-plugin~934540","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I048d4e954329d2d4f28604a8bf02c615703ac43c","project":"openstack/manila-tempest-plugin","branch":"master","topic":"nfs_version","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-12-04 16:25:49.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-12-04 19:35:50.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2024-12-04 13:24:21.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2024-12-04 19:35:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I048d4e954329d2d4f28604a8bf02c615703ac43c","subject":"Add nfs_versions option for NFS protocol version selection","status":"MERGED","created":"2024-11-09 17:56:26.000000000","updated":"2024-12-04 19:35:50.000000000","submitted":"2024-12-04 19:35:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"934540-nfs_version","meta_rev_id":"e732567e51fc2d97eaa2e18f96353ee89e1668c4","_number":934540,"virtual_id_number":934540,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-04 19:35:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2024-12-04 16:25:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2024-12-04 17:55:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":2,"date":"2024-12-04 13:24:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":1,"date":"2024-12-04 18:27:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-09 17:58:53.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2024-11-09 17:58:53.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2024-11-09 17:58:53.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-11-09 18:30:21.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-09 18:32:35.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-11-09 20:37:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-09 20:44:43.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-12-04 17:55:37.000000000","updated_by":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"reviewer":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"state":"REVIEWER"}],"messages":[{"id":"138977ee0083b73df2e400562eee36ddabc09437","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-11-09 17:56:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96eb076a2c349f5fc59891e4dba70b3db8710802","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-11-09 17:57:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b8c6ab960064cfa0901c1eb363c25be7772f1c1b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-09 18:30:21.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/40/934540/2/check/DellEMC_PowerFlex_Manila_dhss_false/54d85fc/DellEMC_PowerFlex_Manila_dhss_false/2000 : SUCCESS in 32m 45s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"a24ab5b8a83cb6173336181346e860f2748b207a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-11-09 18:32:35.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/40/934540/2/check/DellEMC_Unity_Manila_dhss_true/9b16ffc/DellEMC_Unity_Manila_dhss_true/1983 : FAILURE in 14m 29s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/40/934540/2/check/DellEMC_Unity_Manila_dhss_false/259b5d3/DellEMC_Unity_Manila_dhss_false/2087 : FAILURE in 34m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"f1e6e673a956f2e9b0ef984f7d6e13ed80e9e804","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-11-09 19:12:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fcb2cb80fa1711d7618ebb983ca4a0c544cb66cb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-09 19:46:26.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/40/934540/3/check/DellEMC_PowerFlex_Manila_dhss_false/b24d9a7/DellEMC_PowerFlex_Manila_dhss_false/2001 : SUCCESS in 33m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"fe03bdb95b2d996776269f29c365dd14140e6a42","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-11-09 19:48:46.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/40/934540/3/check/DellEMC_Unity_Manila_dhss_true/c520fd2/DellEMC_Unity_Manila_dhss_true/1984 : FAILURE in 16m 11s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/40/934540/3/check/DellEMC_Unity_Manila_dhss_false/097188c/DellEMC_Unity_Manila_dhss_false/2088 : FAILURE in 35m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"97328aa6bf4eddf026a115cf197a426661735f4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-09 20:37:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cf9687189be4a249e298f356f281998\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af3471a129304d37a414c4b8c1a35a24 : SUCCESS in 2m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/02d30826809644ccaf4d446e12961c2e : SUCCESS in 45m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/dba409bc573a41319f9d335cc0e37ff0 : SUCCESS in 1h 23m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/aecba104a1c94daa9ad4f7c10e5f080c : FAILURE in 35m 28s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c76b7195cd104e9cb3e04798d270bcff : SUCCESS in 40m 40s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/44f81928d70b48808afd30766a67f0af : FAILURE in 6m 21s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d28bb5b8b8a144f3ba1341edc71cf3cf : SUCCESS in 51m 52s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ecd11effc103430b87f8d37646ae5889 : SUCCESS in 28m 42s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b2ac8085088b4f3e89ea4ea884bf0f41 : SUCCESS in 27m 34s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/fe98f43705a24cfe8072fd579dea35ec : SUCCESS in 1h 12m 17s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/80a95ba9949143f1b19817b9ea9bdae7 : SUCCESS in 37m 16s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/71789815a42349768e93af81d95a78c6 : SUCCESS in 1h 13m 09s\n- manila-tempest-plugin-lvm-2023.1 https://zuul.opendev.org/t/openstack/build/ae89ed5fa7be4270af2cc8f16b056c8b : SUCCESS in 49m 28s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d65584ac3236460faed03facbbe549ac : SUCCESS in 22m 37s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ab2466e612db4974851cc01265ca910f : FAILURE in 50m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cbe6144fb1968bc8590d7fdcb3cdc30feba4ca80","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-11-09 20:44:43.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss finger://managesf.services.openstack.netapp.com:7979/27e8196acbf0430796ceb3243f29a545 : RETRY_LIMIT in 5m 35s\n- manila-tempest-plugin-ontap-no-dhss finger://managesf.services.openstack.netapp.com:7979/75e4b1394a964a399f13a6f39ba066f8 : RETRY_LIMIT in 6m 03s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"5521d289976e6a1629e7052a9807db57edc9e40a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-09 21:12:56.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/40/934540/3/check/DellEMC_PowerFlex_Manila_dhss_false/5153f77/DellEMC_PowerFlex_Manila_dhss_false/2002 : SUCCESS in 34m 56s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"2bd25e9da19f9cf39a85bd27a37b6def64ceffdb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-11-09 21:13:30.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/40/934540/3/check/DellEMC_Unity_Manila_dhss_true/efa6714/DellEMC_Unity_Manila_dhss_true/1985 : FAILURE in 14m 06s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/40/934540/3/check/DellEMC_Unity_Manila_dhss_false/2723db3/DellEMC_Unity_Manila_dhss_false/2089 : FAILURE in 35m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"ddd0692f386211853abe2e586b9f6c54bc3f03da","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2024-11-11 11:50:11.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8e5e993dc6d226f536a11e1aa7e921d65ea327fa","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-11-11 19:17:58.000000000","message":"Patch Set 3: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e2f8a01d0947b1e4adffbf7616dd575bb7dd2fff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-11-14 06:59:43.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2a9c5868315d8511cf30b894c65b3ff072aadb2c","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-11-14 07:00:52.000000000","message":"Patch Set 3:\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"384242cb006cdfbdd0657618eed619853d8b34c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-11-14 07:02:25.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"d83b2ea555bf01f23ea4b94e15124cc2b6882cc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-11-14 07:06:32.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent tree, and commit message as Patch Set 5.","accounts_in_message":[],"_revision_number":6},{"id":"64050b7b29bb9a3b981b6bedb1a83b0148ad0100","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-11-14 07:07:59.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0819ec5ea1c44a66807d16394cde920268e5cc76","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-14 07:27:59.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/40/934540/6/check/DellEMC_PowerFlex_Manila_dhss_false/15b8afd/DellEMC_PowerFlex_Manila_dhss_false/2029 : FAILURE in 21m 14s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"aca1a05f09f29d61483c68a25b0f673fd088cda2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-11-14 07:30:23.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/40/934540/6/check/DellEMC_Unity_Manila_dhss_true/8991a1a/DellEMC_Unity_Manila_dhss_true/2012 : FAILURE in 12m 55s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/40/934540/6/check/DellEMC_Unity_Manila_dhss_false/f833f88/DellEMC_Unity_Manila_dhss_false/2116 : FAILURE in 23m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"7a2e3e213bdfda695f4ebb30a3bb4dc214d3a3ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 07:48:45.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/92a4ea2b4b25434a9ae9bc8327b281b9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee9562f2805249338f409defff61b818 : FAILURE in 2m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/42a3a0649801458ba097eac48223ff0b : FAILURE in 27m 32s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/cdacdd61c428421785edaf8ac507e43b : FAILURE in 35m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7345427686b7499f89f47afc53fd23e5 : FAILURE in 27m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e993e30767054b718c0c040177a1ddee : FAILURE in 31m 32s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/9a62ce2cc6f14735b16fe62fd7124df4 : FAILURE in 35m 47s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/825825f87f1949de918b3912eebe817d : FAILURE in 31m 55s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fc4c92231f57492fa28bb4da67394e0a : FAILURE in 11m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/39e7dfc15ea744819656d06df0f3a73e : FAILURE in 10m 44s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/418278f681ec43bea8de61f494560eb9 : FAILURE in 27m 07s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/b3e5b333669c49feb63d47d9f54a3305 : FAILURE in 31m 03s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/0b25dba761a049e6b0adbd88971ac0e0 : FAILURE in 21m 37s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3e27fb764527462db10d60c165cfd2ff : FAILURE in 13m 00s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/845b3989a1bb440884b6b126b34cbc23 : FAILURE in 34m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"97ab017e765a530f091a9193d01086feed6fcd25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-11-14 08:37:50.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"2812e43c2b49723445c444bdfdc388eb5af7dc18","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-14 09:12:13.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/40/934540/7/check/DellEMC_PowerFlex_Manila_dhss_false/933c5bb/DellEMC_PowerFlex_Manila_dhss_false/2031 : SUCCESS in 34m 07s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"f2603f0101d91a353441341ef7f720bd6481d7a3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-11-14 09:14:07.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/40/934540/7/check/DellEMC_Unity_Manila_dhss_true/c30c2a8/DellEMC_Unity_Manila_dhss_true/2014 : FAILURE in 13m 45s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/40/934540/7/check/DellEMC_Unity_Manila_dhss_false/0c8885b/DellEMC_Unity_Manila_dhss_false/2118 : FAILURE in 36m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"904f8b9ace4b05966faedd5a1bd93f741aa02ad5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 11:43:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/997364943d334ad8b681ad6c7ca903bf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5c764b12f844221a32bee177ed287b9 : SUCCESS in 2m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/17a31bd9338a41fb9e3290d509a0dddb : SUCCESS in 1h 16m 49s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b9756a47338f4c5cb4ea0b1a2078b116 : FAILURE in 2h 55m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a54164326a6b463592ad8d241ea124c6 : FAILURE in 1h 07m 30s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/2efd84758b304a56add9ad6ee9fc3530 : SUCCESS in 1h 06m 25s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/2642cf6a284848a8a3bb5afafc7fd0c7 : SUCCESS in 1h 08m 29s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/412e3cfc0aa042668c7efc800f5c6d20 : FAILURE in 1h 10m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/93581acf0ff6449ca08beb696e30815d : SUCCESS in 26m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/70316394fa8040958a573df06bb90ee2 : SUCCESS in 32m 14s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/88d67243be3a45c697d7ca29cfdfc102 : SUCCESS in 57m 31s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/28122794ebd54dbc927080bcd15d6175 : SUCCESS in 1h 06m 51s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/693b8a4b0f724fd79dc08ada64f2290b : SUCCESS in 59m 46s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/b16c77872a5145f1b8b98d32b8731158 : SUCCESS in 32m 58s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/679e2b82efca4b08bfa904cdfd00e89e : FAILURE in 41m 57s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"debf0922f432c3b95a5a32d4cc98521512d6a3a4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-11-14 11:59:51.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss finger://managesf.services.openstack.netapp.com:7979/5de56dd332b1427aa3c91218124ea677 : RETRY_LIMIT in 4m 41s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"69a3d93952d62acf2971218a40521d6b4d2017df","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-14 12:21:52.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/40/934540/7/check/DellEMC_PowerFlex_Manila_dhss_false/b3f5c8c/DellEMC_PowerFlex_Manila_dhss_false/2033 : SUCCESS in 33m 46s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"c4e59330a708690f3d2dae12f3cb35e6b32761cb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-11-14 12:24:56.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/40/934540/7/check/DellEMC_Unity_Manila_dhss_true/2edafe7/DellEMC_Unity_Manila_dhss_true/2016 : FAILURE in 14m 12s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/40/934540/7/check/DellEMC_Unity_Manila_dhss_false/c5f8da1/DellEMC_Unity_Manila_dhss_false/2120 : FAILURE in 34m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"1e9cbb31b7cab9618ea7c750bfccdf227ffb69a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-11-25 19:37:33.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"d0f8f248cf0369adf30d37bff84ad926f711fd6d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-25 20:11:12.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/40/934540/8/check/DellEMC_PowerFlex_Manila_dhss_false/3215be4/DellEMC_PowerFlex_Manila_dhss_false/2131 : SUCCESS in 33m 23s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"b1d081561c692eced38a8195822d896a4ecf80b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-11-25 20:18:13.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"4be5314f9056192ea975abe76044e4397a34ab95","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-25 20:51:01.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/40/934540/9/check/DellEMC_PowerFlex_Manila_dhss_false/d5a246c/DellEMC_PowerFlex_Manila_dhss_false/2132 : SUCCESS in 32m 34s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"b9a0f67fd2df175e995e1d854d37563ce7835b89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 23:02:48.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d86196231640403cbf32f8a746a22955\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e041766d866d4de995ce4b87d330b31d : SUCCESS in 2m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3c59d01a28164cb587ed8739885f1405 : SUCCESS in 54m 37s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/84a7e0c0f89b4df1aeae5b3adf317270 : SUCCESS in 2h 37m 14s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2ad26c5d6ba947b29712734a55bfb5a7 : FAILURE in 1h 25m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8f46bcf0fe574cc782a92bbe28f89c13 : SUCCESS in 1h 06m 53s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/5609c7de3634439a8fdac7b0ce96991f : SUCCESS in 1h 05m 48s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5d07f0468c1244d78aeebcad5d14b46b : FAILURE in 1h 19m 07s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d1843d66c8ae4858bbb3880493651f6d : SUCCESS in 24m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4afd077abe8142908a7ca3ba45905b54 : SUCCESS in 29m 56s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/84f8431178d241858547331b7e457f6e : SUCCESS in 53m 03s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/814be55f390c4b7cb306a0f593d89fb5 : SUCCESS in 57m 07s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/6f4bc76925b34a88b782c33f2e4aaae4 : SUCCESS in 56m 40s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/cda92993b5224d06b54fec455dde9048 : SUCCESS in 20m 22s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/180a62b0bf064175b95ecb0d3c7df31f : FAILURE in 32m 11s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7c63d0e2dcf1826cbe2cb05af3c1d9f340014a90","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-11-25 23:19:57.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss finger://managesf.services.openstack.netapp.com:7979/126633c9a14a46fab8e774ce574cab52 : RETRY_LIMIT in 5m 06s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"5bb7026515e12633b4f4033c3fa3c064fb616eba","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-25 23:37:18.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/40/934540/9/check/DellEMC_PowerFlex_Manila_dhss_false/56b04f4/DellEMC_PowerFlex_Manila_dhss_false/2133 : SUCCESS in 34m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"bb811e2d87c1dc032984dbfd951ae1ae6407e010","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2024-12-04 13:24:21.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"58cb46855a3261361c0267aea3fb1ab2b3169c32","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-12-04 16:25:49.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"1f5cd71e6441ed77e87bdf26a03fc778f2c179c0","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-12-04 17:55:37.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"fa878de153374156a94c659fb18e32b21fa28b2c","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2024-12-04 18:27:28.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"6be02e27ba9f53e61dabb692a942378bfcbdc829","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 18:27:59.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"d209bb451c8c4de22934502353fac77e4c638b2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 19:35:49.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d39af8ce5be34fedb705b88ce17d206a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/050aff8f56404110ab9aeb62f7fd9af1 : SUCCESS in 2m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9454758e78a54c4488231814efd68774 : SUCCESS in 53m 43s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/4ea4b1e537ba4420a5a5c1d115280e9d : SUCCESS in 1h 05m 18s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/63f8451880f644769fb55082f79f4d21 : SUCCESS in 24m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3e670910973f4fe0abcd52b2c43fe44e : SUCCESS in 31m 06s","accounts_in_message":[],"_revision_number":9},{"id":"e732567e51fc2d97eaa2e18f96353ee89e1668c4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 19:35:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"b5d6f2997f325bd3b59f663bd9def88f068d8f9b","revisions":{"d3da1b22ea577c7c56f4b4b6f1d280f184faa505":{"kind":"REWORK","_number":1,"created":"2024-11-09 17:56:26.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/40/934540/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/40/934540/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/1"}}},"commit":{"parents":[{"commit":"31df75e18b2cfad59322c20575a608fdbb592de1","subject":"Merge \"Support py3.12 and drop py3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/31df75e18b2cfad59322c20575a608fdbb592de1"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-11-09 17:54:47.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-11-09 17:54:47.000000000","tz":120},"subject":"Add nfs_version option for NFS protocol version selection","message":"Add nfs_version option for NFS protocol version selection\n\nThis commit introduces a new configuration option, `nfs_version`,\nto allow specifying the NFS protocol version used when mounting NFS\nshares in Tempest tests. The option supports selecting specific\nversions, such as NFSv3 or NFSv4, enabling compatibility with environments\nwhere certain NFS versions are required.\n\n- Option Name: `nfs_version`\n- Default: Blank (uses system default NFS version)\n\nThis enhancement increases flexibility in testing scenarios with\nNFS version dependencies.\n\nChange-Id: I048d4e954329d2d4f28604a8bf02c615703ac43c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d3da1b22ea577c7c56f4b4b6f1d280f184faa505"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d3da1b22ea577c7c56f4b4b6f1d280f184faa505"}]},"branch":"refs/heads/master"},"a1a6d20fff327316785cb24c558b537fd02c9d0b":{"kind":"REWORK","_number":2,"created":"2024-11-09 17:57:22.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/40/934540/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/40/934540/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/2"}}},"commit":{"parents":[{"commit":"31df75e18b2cfad59322c20575a608fdbb592de1","subject":"Merge \"Support py3.12 and drop py3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/31df75e18b2cfad59322c20575a608fdbb592de1"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-11-09 17:54:47.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-11-09 17:57:13.000000000","tz":120},"subject":"Add nfs_version option for NFS protocol version selection","message":"Add nfs_version option for NFS protocol version selection\n\nThis commit introduces a new configuration option, `nfs_version`,\nto allow specifying the NFS protocol version used when mounting NFS\nshares in Tempest tests. The option supports selecting specific\nversions, such as NFSv3 or NFSv4, enabling compatibility with environments\nwhere certain NFS versions are required.\n\n- Option Name: `nfs_version`\n- Default: Blank (uses system default NFS version)\n\nThis enhancement increases flexibility in testing scenarios with\nNFS version dependencies.\n\nChange-Id: I048d4e954329d2d4f28604a8bf02c615703ac43c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a1a6d20fff327316785cb24c558b537fd02c9d0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a1a6d20fff327316785cb24c558b537fd02c9d0b"}]},"branch":"refs/heads/master"},"38da69c0c9d253c2ec2a34021ca5832b65229265":{"kind":"REWORK","_number":3,"created":"2024-11-09 19:12:55.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/40/934540/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/40/934540/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/3"}}},"commit":{"parents":[{"commit":"31df75e18b2cfad59322c20575a608fdbb592de1","subject":"Merge \"Support py3.12 and drop py3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/31df75e18b2cfad59322c20575a608fdbb592de1"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-11-09 17:54:47.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-11-09 19:12:44.000000000","tz":120},"subject":"Add nfs_version option for NFS protocol version selection","message":"Add nfs_version option for NFS protocol version selection\n\nThis commit introduces a new configuration option, `nfs_version`,\nto allow specifying the NFS protocol version used when mounting NFS\nshares in Tempest tests. The option supports selecting specific\nversions, such as NFSv3 or NFSv4, enabling compatibility with environments\nwhere certain NFS versions are required.\n\n- Option Name: `nfs_version`\n- Default: Blank (uses system default NFS version)\n\nThis enhancement increases flexibility in testing scenarios with\nNFS version dependencies.\n\nChange-Id: I048d4e954329d2d4f28604a8bf02c615703ac43c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/38da69c0c9d253c2ec2a34021ca5832b65229265"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/38da69c0c9d253c2ec2a34021ca5832b65229265"}]},"branch":"refs/heads/master"},"35baebdf43caeeca8f849eb1d20128f86448d719":{"kind":"REWORK","_number":4,"created":"2024-11-14 06:59:43.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/40/934540/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/40/934540/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/4"}}},"commit":{"parents":[{"commit":"31df75e18b2cfad59322c20575a608fdbb592de1","subject":"Merge \"Support py3.12 and drop py3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/31df75e18b2cfad59322c20575a608fdbb592de1"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-11-09 17:54:47.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-11-14 06:59:23.000000000","tz":120},"subject":"Add nfs_version option for NFS protocol version selection","message":"Add nfs_version option for NFS protocol version selection\n\nThis commit introduces a new configuration option, `nfs_version`,\nto allow specifying the NFS protocol version used when mounting NFS\nshares in Tempest tests. The option supports selecting specific\nversions, such as NFSv3 or NFSv4, enabling compatibility with environments\nwhere certain NFS versions are required.\n\n- Option Name: `nfs_version`\n- Default: Blank (uses system default NFS version)\n\nThis enhancement increases flexibility in testing scenarios with\nNFS version dependencies.\n\nChange-Id: I048d4e954329d2d4f28604a8bf02c615703ac43c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/35baebdf43caeeca8f849eb1d20128f86448d719"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/35baebdf43caeeca8f849eb1d20128f86448d719"}]},"branch":"refs/heads/master"},"3bc446e20dea98f0b533bbcc007a21d792fbf0d8":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-11-14 07:02:25.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/40/934540/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/40/934540/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/5"}}},"commit":{"parents":[{"commit":"31df75e18b2cfad59322c20575a608fdbb592de1","subject":"Merge \"Support py3.12 and drop py3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/31df75e18b2cfad59322c20575a608fdbb592de1"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-11-09 17:54:47.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-11-14 07:01:42.000000000","tz":120},"subject":"Add nfs_version option for NFS protocol version selection","message":"Add nfs_version option for NFS protocol version selection\n\nThis commit introduces a new configuration option, `nfs_version`,\nto allow specifying the NFS protocol version used when mounting NFS\nshares in Tempest tests. The option supports selecting specific\nversions, such as NFSv3 or NFSv4, enabling compatibility with\nenvironments where certain NFS versions are required.\n\n- Option Name: `nfs_version`\n- Default: Blank (uses system default NFS version)\n\nThis enhancement increases flexibility in testing scenarios with\nNFS version dependencies.\n\nChange-Id: I048d4e954329d2d4f28604a8bf02c615703ac43c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3bc446e20dea98f0b533bbcc007a21d792fbf0d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3bc446e20dea98f0b533bbcc007a21d792fbf0d8"}]},"branch":"refs/heads/master"},"eb717b3454c8e194571a08cc914663a8e7f1cfb0":{"kind":"NO_CHANGE","_number":6,"created":"2024-11-14 07:06:32.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/40/934540/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/40/934540/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/6"}}},"commit":{"parents":[{"commit":"31df75e18b2cfad59322c20575a608fdbb592de1","subject":"Merge \"Support py3.12 and drop py3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/31df75e18b2cfad59322c20575a608fdbb592de1"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-11-09 17:54:47.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-11-14 07:05:03.000000000","tz":120},"subject":"Add nfs_version option for NFS protocol version selection","message":"Add nfs_version option for NFS protocol version selection\n\nThis commit introduces a new configuration option, `nfs_version`,\nto allow specifying the NFS protocol version used when mounting NFS\nshares in Tempest tests. The option supports selecting specific\nversions, such as NFSv3 or NFSv4, enabling compatibility with\nenvironments where certain NFS versions are required.\n\n- Option Name: `nfs_version`\n- Default: Blank (uses system default NFS version)\n\nThis enhancement increases flexibility in testing scenarios with\nNFS version dependencies.\n\nChange-Id: I048d4e954329d2d4f28604a8bf02c615703ac43c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/eb717b3454c8e194571a08cc914663a8e7f1cfb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/eb717b3454c8e194571a08cc914663a8e7f1cfb0"}]},"branch":"refs/heads/master"},"861c6a5ba0a28086a41124c9dad811e1e1ffd1d7":{"kind":"REWORK","_number":7,"created":"2024-11-14 08:37:50.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/40/934540/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/40/934540/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/7"}}},"commit":{"parents":[{"commit":"31df75e18b2cfad59322c20575a608fdbb592de1","subject":"Merge \"Support py3.12 and drop py3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/31df75e18b2cfad59322c20575a608fdbb592de1"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-11-09 17:54:47.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-11-14 08:37:40.000000000","tz":120},"subject":"Add nfs_version option for NFS protocol version selection","message":"Add nfs_version option for NFS protocol version selection\n\nThis commit introduces a new configuration option, `nfs_version`,\nto allow specifying the NFS protocol version used when mounting NFS\nshares in Tempest tests. The option supports selecting specific\nversions, such as NFSv3 or NFSv4, enabling compatibility with\nenvironments where certain NFS versions are required.\n\n- Option Name: `nfs_version`\n- Default: Blank (uses system default NFS version)\n\nThis enhancement increases flexibility in testing scenarios with\nNFS version dependencies.\n\nChange-Id: I048d4e954329d2d4f28604a8bf02c615703ac43c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/861c6a5ba0a28086a41124c9dad811e1e1ffd1d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/861c6a5ba0a28086a41124c9dad811e1e1ffd1d7"}]},"branch":"refs/heads/master"},"f19b115fd67878fcd8faf358eef7a66fbc905638":{"kind":"REWORK","_number":8,"created":"2024-11-25 19:37:33.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/40/934540/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/40/934540/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/8"}}},"commit":{"parents":[{"commit":"31df75e18b2cfad59322c20575a608fdbb592de1","subject":"Merge \"Support py3.12 and drop py3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/31df75e18b2cfad59322c20575a608fdbb592de1"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-11-09 17:54:47.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-11-25 19:34:42.000000000","tz":120},"subject":"Add nfs_versions option for NFS protocol version selection","message":"Add nfs_versions option for NFS protocol version selection\n\nThis commit introduces a new configuration option, `nfs_versions`,\nto allow specifying the NFS protocol version used when mounting NFS\nshares in Tempest tests. The option supports selecting specific\nversions, such as NFSv3 or NFSv4, enabling compatibility with\nenvironments where certain NFS versions are required.\n\n- Option Name: `nfs_versions`\n- Default: [\"4\", ] (NFSv4 is used by default)\n\nThis enhancement increases flexibility in testing scenarios with\nNFS version dependencies.\n\nChange-Id: I048d4e954329d2d4f28604a8bf02c615703ac43c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f19b115fd67878fcd8faf358eef7a66fbc905638"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f19b115fd67878fcd8faf358eef7a66fbc905638"}]},"branch":"refs/heads/master"},"b5d6f2997f325bd3b59f663bd9def88f068d8f9b":{"kind":"REWORK","_number":9,"created":"2024-11-25 20:18:13.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/40/934540/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/40/934540/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/40/934540/9"}}},"commit":{"parents":[{"commit":"31df75e18b2cfad59322c20575a608fdbb592de1","subject":"Merge \"Support py3.12 and drop py3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/31df75e18b2cfad59322c20575a608fdbb592de1"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-11-09 17:54:47.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-11-25 20:18:02.000000000","tz":120},"subject":"Add nfs_versions option for NFS protocol version selection","message":"Add nfs_versions option for NFS protocol version selection\n\nThis commit introduces a new configuration option, `nfs_versions`,\nto allow specifying the NFS protocol version used when mounting NFS\nshares in Tempest tests. The option supports selecting specific\nversions, such as NFSv3 or NFSv4, enabling compatibility with\nenvironments where certain NFS versions are required.\n\n- Option Name: `nfs_versions`\n- Default: [\"4\", ] (NFSv4 is used by default)\n\nThis enhancement increases flexibility in testing scenarios with\nNFS version dependencies.\n\nChange-Id: I048d4e954329d2d4f28604a8bf02c615703ac43c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b5d6f2997f325bd3b59f663bd9def88f068d8f9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b5d6f2997f325bd3b59f663bd9def88f068d8f9b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
