)]}'
{"id":"openstack%2Fmanila-tempest-plugin~937243","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I8500f71b527b4e5037f128906f68385f821ebf42","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/2066871","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2025-01-16 15:16:02.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-01-16 15:16:02.000000000","reason":"Change was submitted"},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2025-01-16 15:16:02.000000000","reason":"Change was submitted"},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2025-01-16 15:16:02.000000000","reason":"Change was submitted"},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2025-01-08 06:33:37.000000000","reason":"\u003cGERRIT_ACCOUNT_32919\u003e replied on the change","reason_account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2025-01-07 21:11:40.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2025-01-08 06:27:36.000000000","reason":"\u003cGERRIT_ACCOUNT_19262\u003e replied on the change","reason_account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}},"hashtags":[],"change_id":"I8500f71b527b4e5037f128906f68385f821ebf42","subject":"Add tests for allow to update share access rule","status":"MERGED","created":"2024-12-06 10:34:42.000000000","updated":"2025-01-16 15:16:02.000000000","submitted":"2025-01-16 15:16:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":52,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937243-bug/2066871","meta_rev_id":"f888018f2a5ff8b426ad80639fb5e10d4aadc01f","_number":937243,"virtual_id_number":937243,"owner":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-16 15:16:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":2,"date":"2025-01-07 21:11:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-08 06:27:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-08 06:27:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-06 11:01:22.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-06 13:56:24.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2024-12-06 13:56:24.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2024-12-06 13:56:24.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-12-06 15:20:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-06 15:37:01.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-12-13 15:17:32.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2024-12-17 15:20:11.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2025-01-02 10:26:31.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"}],"messages":[{"id":"8ede846a26a863d8dfbca73d0297c1c679e6122f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-12-06 10:34:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce79dc816cf00ff47c38470c3742fc3b4871de0c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-06 11:01:22.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/43/937243/1/check/DellEMC_PowerFlex_Manila_dhss_false/1c23cd8/DellEMC_PowerFlex_Manila_dhss_false/2180 : FAILURE in 26m 26s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"514e448a13b2098c1ddbc57d6dc7a3d0f57d37ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-12-06 12:10:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"57d5aa489e9c0fea627512f5d87b07e53419b6a3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-06 12:53:27.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/43/937243/2/check/DellEMC_PowerFlex_Manila_dhss_false/553a07e/DellEMC_PowerFlex_Manila_dhss_false/2182 : FAILURE in 26m 06s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"6f13aecde1d506f6d8f29e59136ac876d6227b31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 15:20:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ece95c551fd4641bb3ea30be40633f6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ae475601c334892ae4ac1d3d07023f5 : SUCCESS in 3m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/40a85a3817e6449b93232914a99bf5a2 : SUCCESS in 1h 28m 25s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/6eb89f77dd274944a6727f71d7dfe15f : SUCCESS in 1h 30m 52s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/91de0331279e4c2a913be1331b5370f0 : TIMED_OUT in 3h 03m 12s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a18ce41d3b434037a84c5a7157d4d753 : FAILURE in 1h 00m 12s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/08b3d1f5c9004754814071aae7e0cad1 : FAILURE in 1h 40m 41s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ab488d27f8d94789a8315d176d217085 : FAILURE in 10m 29s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/163471a33b6e452f8281dee2610ae41b : SUCCESS in 1h 16m 33s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c9e48b894acd4250b0bf76c044086f9e : SUCCESS in 43m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/525e73fe867f4ec7822dd8f021c39d2d : SUCCESS in 27m 27s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/582bef2631b645c8af6044ae03a0793d : SUCCESS in 1h 07m 50s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/d58cfa1573f14f1282d2bfcb2ab141a1 : SUCCESS in 1h 16m 39s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/9d3c8d97c1fd4a0c9fb3949b89bdd1c7 : SUCCESS in 41m 31s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/49bd9b3786b24e19a3ce3eaf3a2c536d : SUCCESS in 34m 29s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7a89b41a3cb143ceac8594bbd46166fc : FAILURE in 1h 09m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3957ca0d34d8f27bba75663a5ed9cca5cbe6cb39","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-12-06 15:37:01.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss finger://managesf.services.openstack.netapp.com:7979/d0eda0d8f8fd48f7bdd10df6244b8f94 : RETRY_LIMIT in 5m 18s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"a4d5867acc8e3083ca039a6f497b71d40930b3d4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-06 15:49:25.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/43/937243/2/check/DellEMC_PowerFlex_Manila_dhss_false/bad3781/DellEMC_PowerFlex_Manila_dhss_false/2183 : FAILURE in 28m 54s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"7d9d390e46d3d07ecc20ce2342a61b7a426c399d","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-12-13 15:17:32.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d55818fc71d3f89800213a166d2bed4ca453d37d","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2024-12-17 15:20:11.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3d8213f0f6384ec6d7c7c791adb34e0c18ef407c","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2024-12-18 03:27:49.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6a3fef1544d61c414b3a2bdc161fa443c7b5359c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-12-18 15:49:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"5d688cf169ebb93c7337c5076e58837373d0a781","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-12-18 15:49:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8ae0c23acdb9b4f7e935ccc4033d5c9236988bf2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-18 16:30:03.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/43/937243/3/check/DellEMC_PowerFlex_Manila_dhss_false/631e628/DellEMC_PowerFlex_Manila_dhss_false/2250 : FAILURE in 28m 13s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"e2e6f28e15b9a8fef239c73a846b04efe1f87b35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 18:55:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e47549831a3041338e13a78e30315b8b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/069fe0ff67cb406cb9127fd08257e836 : SUCCESS in 3m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cc0ddca82de44c979ebbfeb0b6f277f2 : SUCCESS in 1h 09m 13s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/e26c452fa1da46f3bf19dcefbb6fc18f : SUCCESS in 1h 06m 06s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ed415b0ec169460cba4e27e147872893 : TIMED_OUT in 3h 02m 26s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8617750885eb4f0094624fb2d685b82e : FAILURE in 1h 14m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d894cc6b51f549599b9f7b694e2c7dd8 : FAILURE in 1h 26m 17s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e93c4b7ca9a84a65a97b32f9d283207a : FAILURE in 1h 10m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/69769546c3e24c228fd1269aedf934a2 : FAILURE in 1h 20m 04s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/25340530fe664393a8b19cf0efca6480 : SUCCESS in 30m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8f33d1107711439fa4a670b023dba7ac : SUCCESS in 37m 59s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/34d00e91c5d6437098ece3f4edf5c69c : SUCCESS in 1h 46m 34s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/a9f307b2e3814fdb9ec6f3d6d2f8f525 : SUCCESS in 1h 19m 46s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/467e1d0474e24c018672d3b9eb685c51 : SUCCESS in 1h 16m 31s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/758af7fdbe624141a156c0f7af7c64f3 : FAILURE in 23m 53s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8c37c180a9474ac3857eaa347240f982 : FAILURE in 38m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fd1b5adef6a4b2ec4c192be55791c7dc25d4bee7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-12-19 04:52:53.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"94812bd1d8d337e44aa4718fbbf2947120c659b5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-19 05:36:55.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/43/937243/4/check/DellEMC_PowerFlex_Manila_dhss_false/e1ad56b/DellEMC_PowerFlex_Manila_dhss_false/2259 : FAILURE in 29m 03s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"9ff6cfc9212868c6c73d2c724efc36bfd3f4434e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 08:01:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1be30a4e89b43dfaeaace6287f940a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/744cff5b023c4d9d90d9771bb04a4d61 : SUCCESS in 2m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3a8665a26913400f9eae55376c455a58 : SUCCESS in 45m 20s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/1d3e7643b5ea445fbf5211f5521f6baa : SUCCESS in 1h 22m 59s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/90c32d9197a04fc4bca2bc250ca8087a : TIMED_OUT in 3h 03m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c04908c947e74776966e7007a812dcb6 : FAILURE in 1h 33m 42s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/826b6cc3c6cd4708928701e7d01a84b0 : FAILURE in 1h 30m 06s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e9aeeda444e644788809b7cd218b8481 : FAILURE in 1h 26m 52s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b170b47f34f34beda59b123a3b13e782 : SUCCESS in 1h 22m 06s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b64df77d2514403780547f37aeb3fd94 : FAILURE in 36m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7412a88833c543558bf875d6592b031d : SUCCESS in 44m 51s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/1da1e6e323014115819e5a874c058c7a : SUCCESS in 1h 09m 49s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/8f8b5e2f6e9c465da142b634733c390c : SUCCESS in 1h 20m 26s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/5ca133262f14414d92b24cb92214f713 : SUCCESS in 1h 24m 28s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/6e066505750140b186e81852326c1cba : SUCCESS in 33m 49s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/047a61201a0848daa2bc8296c9405915 : FAILURE in 30m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2f1defbdea07ab1f92c107e48a99fa5b90f299fe","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-12-19 08:08:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2c9577ca1ab4db228540fe3f55281d8363dd4fd8","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-12-19 09:51:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d6c8cc7bba7ea8e0aad0fe633eeacd5862b1dd41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 11:45:02.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a0e735f5ae3412b917d8a77a2dec64c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01db09e8d09b412f8e6a5179f39245ea : SUCCESS in 2m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bc1046bd90604a928ffd33edffa10b8b : SUCCESS in 1h 35m 49s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/d27fa31db3dc45f390d408f1a6da04be : SUCCESS in 1h 14m 20s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/dfc86e2780db4904873fdd4eb6af6f7c : TIMED_OUT in 3h 03m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e6e0e890a9284d2293eedfca034f88ef : FAILURE in 1h 04m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/163b53e1072e4d26a29fc4285f5ee2b4 : FAILURE in 1h 17m 54s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/948b9c80b2a5429e93b6217fb0d4e7cb : FAILURE in 10m 59s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/635455880d764e12976b3e8004ee96e8 : SUCCESS in 1h 20m 04s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b03b16b76b584e2993c8b68768a58178 : SUCCESS in 26m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/aee3d16a7072446f944c69103f275a06 : TIMED_OUT in 1h 04m 05s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/d0c6a0356b99482eb87655c2c592edfa : SUCCESS in 1h 19m 17s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/1bc5365eeaaf497e92d81169820453e0 : SUCCESS in 1h 15m 44s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/6843101d5e914503a403dac04f2144cd : SUCCESS in 1h 18m 11s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a0ae77a507a44c31b97ae2c89cc36c00 : SUCCESS in 26m 17s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2487b262815446369df43ca4483420fc : FAILURE in 36m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bb1de231ad8c9b30eedf9b4a3652c7e9203e7a92","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-12-19 11:47:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"081e28958293a5f7df749e7c2e5f1d200bd6d648","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 14:56:23.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71c4c7f3a59f4463ac2831a3be9bc8d0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d85fa8ea3764cf285ac0130f12c8b95 : SUCCESS in 2m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/92c151d34cc34dbf952d51c3930ea85c : SUCCESS in 1h 11m 48s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/7a8b0f179bd44de7b2fc67219a3c0416 : SUCCESS in 1h 16m 51s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9e49d1acf7dd46bdac8a133eafb1c4d1 : TIMED_OUT in 3h 02m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/696cfdc209bf44f0b912e39b4a32acbd : FAILURE in 1h 28m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/da110585c8ec44acb5af7528a8530048 : FAILURE in 1h 38m 56s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a0040ad083c04180acc8b4d001ad1ccd : FAILURE in 1h 40m 18s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/32c484faad194d519c5f943c8596cc58 : SUCCESS in 1h 16m 30s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6049416392d14b01ad9e78459d7e3983 : SUCCESS in 39m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e4ea5e4ae0f94717aa416c22277f8b75 : SUCCESS in 44m 12s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/2892cc4020a94ce6a133ab41137da27d : SUCCESS in 1h 07m 52s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/8f22358b82b64d3aad8d9b001a4dc2ef : SUCCESS in 1h 16m 36s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/5a59cde5e4e94304aa11790400307da8 : FAILURE in 1h 06m 55s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/27d999d3b8684156bc3f93a6707a25fa : SUCCESS in 35m 45s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/df5a053b54264df8915da37efb02fc02 : FAILURE in 38m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"043a819cf99b98f4bd9a397daf91f1043f96a56c","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-12-19 15:39:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bd62c478ac92854e9dbb105675c3e7f87e5bc1e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 18:43:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30630025b82c4ec6835b393454d6573d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91d01a04ceb44b00bcabf3d9fa37b35f : SUCCESS in 3m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/54aa5e1bccac4397a6cada25f75019e2 : SUCCESS in 1h 03m 03s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/c1da3b54a6884361aa279401162e207d : SUCCESS in 1h 00m 33s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1b33433f44c94f8d9260463048f9ef88 : TIMED_OUT in 3h 02m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e4bc59bbf8e74ed385b55daf53d9d591 : FAILURE in 46m 07s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/1bbf3ce7b2b24e999dc99cfd062e4abf : FAILURE in 1h 18m 21s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e205180cdae14c498362b19396a83d2e : FAILURE in 7m 25s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/be800d20187442138d4436cea8c16d31 : FAILURE in 1h 12m 01s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e0468db064f44b72a624a90fef623d56 : SUCCESS in 27m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3ded3c5441024f6f9c6073b43f616638 : SUCCESS in 31m 35s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/23582379429941dba2ae52dacd150606 : SUCCESS in 39m 24s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/f0d71387244245e8b9d8f42e889ea5e6 : SUCCESS in 39m 15s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/7bc1d6a237994e72b2a2aaf0fb082fa2 : SUCCESS in 1h 23m 24s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/c03f7908212b4ea394f277505f3ab03a : SUCCESS in 39m 27s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/87289235b819487fb902cefbb19c72a3 : FAILURE in 51m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"12903588918bb62123ef1cb57e9cde89944b4c1e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-19 19:09:53.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/43/937243/4/check/DellEMC_PowerFlex_Manila_dhss_false/01abae9/DellEMC_PowerFlex_Manila_dhss_false/2268 : FAILURE in 25m 53s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"5af09015f3d414d0897deaf167726929c9481527","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2024-12-20 01:07:15.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"960dbc2491db6044072e4b6cf8058c0e92ae5051","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-12-23 19:24:34.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ae0c48ee339e2b96a18b768ee219f302a262c5ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-12-27 15:53:14.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"f40b134c60694da270bdbabf499917748a5a1759","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-12-27 16:31:22.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"fa28344fab104d3c75c04ef31ec9cb803b5821cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-27 19:40:32.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e5f902236394ec6aabf938f7f190a86\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaa289fed9be4feea36f53a199bb0494 : SUCCESS in 4m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/79d145c8547548cbafcb067011d5e423 : SUCCESS in 1h 11m 29s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/c22be75455d848ea9c8fd9e80620cf91 : SUCCESS in 57m 38s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c51d5ee17ea24df0812e1944aaa94691 : TIMED_OUT in 3h 02m 55s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b6e68f6063124765842339415873cad6 : FAILURE in 55m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b2f7177c255b45c7a6e29623a4727e7b : FAILURE in 1h 48m 24s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/09786a5c547f4f77bd3ca9bf63d2a9db : FAILURE in 2h 47m 43s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e0eea6ba6a7c4265907391db2ba36760 : FAILURE in 1h 00m 53s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/be799e68685c425daa87187a9ae2db10 : SUCCESS in 28m 12s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/da9a0897f5b2453fb9cab824821f4251 : SUCCESS in 35m 56s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/996840e359654f5eb045d5351b94e3b0 : SUCCESS in 1h 09m 41s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/4e97652fdc7c4458bf2e9ccef5a76e70 : SUCCESS in 56m 19s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/89939eb892674df4b089be70384dfe44 : SUCCESS in 1h 04m 08s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/713e862bbd424aed8b23b972cddc8992 : SUCCESS in 32m 09s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/683b545c18064e64955583151ad03873 : FAILURE in 36m 13s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e1188a1c41ca9a2daf764ccb06e7fb7187a8c190","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-12-29 04:56:28.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"fcf8fb380e93654648216e8b05f7540c3676c6fe","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2024-12-30 02:24:26.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1f1f47ff7a3e4721bfc8bc0d8238891120c34aa9","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-01-02 10:26:31.000000000","message":"Patch Set 6: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"22138dbcae6f356e4b33ce7b8d4882e384c08d34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-01-02 11:23:11.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"7ac6accef7f11edc76ea6995e9aa00136e1a8b4a","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-01-02 11:24:16.000000000","message":"Patch Set 6:\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"303cea129766b301c47a6e59e565a5fed6f70448","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-01-02 13:18:52.000000000","message":"Patch Set 7: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"890ed13a6f7967e4aef4ed417143c17474eda463","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-01-02 14:18:07.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":8},{"id":"4d5fef40171aa8a6bcb4c98de62a76878dc5d47c","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-01-02 14:18:10.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"78f9bffa0e43d81ef459e825774c7687b896bdfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-02 14:35:21.000000000","message":"Patch Set 8: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/931050 is needed.","accounts_in_message":[],"_revision_number":8},{"id":"6d99f89329a1af936414352022bb47a94901965a","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-01-02 16:57:53.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5cba55cd522ee290e8e4c91e4865d28736334542","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-02 20:03:55.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e476c09e24845e6ae58bf377ce4a69e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be9bd1c8483648289c0ffd97c255269b : SUCCESS in 4m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6a64234b39a742288de0d203ee0a27b8 : FAILURE in 23m 56s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/32f4810ee22f43158c4c8f72fff60fc4 : SUCCESS in 59m 42s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/657205d6ec5443af8fd40be2356e0fd2 : TIMED_OUT in 3h 04m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/cd526affb3984f16a7d0a2cba7ad6444 : FAILURE in 1h 02m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/39ebac8bd65a449f888b7eaf5b1aa6bf : FAILURE in 1h 29m 18s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8b367a412ecb4611a8d036390af52c1b : FAILURE in 2h 30m 07s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/73970894b6404c3891b3c66a6b33e0e5 : SUCCESS in 1h 12m 36s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c779cf746d1e411681ad2bdd05719eb6 : SUCCESS in 28m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b9e24535db7448de9de849e258a3375e : SUCCESS in 35m 38s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/7453643315f5489b9ec09badf2c03ec1 : SUCCESS in 1h 16m 24s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/db46715ac2ea477994f65c136caf870a : SUCCESS in 47m 26s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/21ba832600434ac9aaf8a3c0a7c48824 : SUCCESS in 1h 19m 24s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/58c1f14cef2444e2a052ffe7400baee7 : SUCCESS in 30m 14s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ed6e84a380a848b0963c001fabf0ee59 : FAILURE in 1h 05m 33s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"06e8025ccfa4cfa7b2aceb8fdde782525cae5b28","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-01-03 03:49:20.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c1557a112f18d02d852402d75c856a0a9b1bc222","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 07:01:59.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cb9ce3c8bc64adab2f33a91da74d31c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc9a4194071343fcaa887753db3bff0c : SUCCESS in 2m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/05c39ecfc4a84abe824f2c55742278dd : SUCCESS in 1h 25m 12s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/5d8fa4e9e3f94bdfb94a5d21e600856b : SUCCESS in 1h 02m 33s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ff17de7e817248e1b9c814cbd663a435 : TIMED_OUT in 3h 04m 04s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/89100767435c4e22ab80ea9fee8d41dd : FAILURE in 49m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/16c225d6d1c5439391294fa0402460ee : FAILURE in 1h 44m 44s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a7eac80bf94f423195e1d841a6f344cd : FAILURE in 2h 30m 17s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0d0fa8adb2a446c4913bc7ce57fcf41f : FAILURE in 41m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8d2b9175818a4ac184a5988ade5d1272 : SUCCESS in 39m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/63cdc7bac29e4bf7b0f4d0cc3887fc4c : SUCCESS in 31m 31s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/55f51b46615046078e8c130c125f8aa2 : SUCCESS in 42m 16s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/eefcb6bf244847a0af0d533b6c39dfc7 : SUCCESS in 53m 48s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/460cb991c1f74231819356784a1f814b : SUCCESS in 1h 04m 50s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3e944d5ff3d949f8be358f8a989162b5 : SUCCESS in 21m 14s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a774da66e3b946c8a7988ed0d53a5b9d : FAILURE in 31m 16s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2ad2d4d0e60fe424f9b52686122b28a090e9193b","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-01-05 09:07:28.000000000","message":"Patch Set 8:\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"9b5cef7cb2ffb09a24ca8d4dd3a8770f5a44ece3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-01-06 13:52:42.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"2206fbb514986d0609499c52f465ae5a2f6634a2","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-01-06 13:53:38.000000000","message":"Patch Set 8:\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"bbf77660e21678f93e098dea78b57a70d1e581f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-01-06 15:06:34.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"20622b1500814ce12183926624e8b6f38cf91202","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-01-06 16:22:42.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"1e8e8c51aefb7908e7290633e55c763e4b22dd2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 19:33:47.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/08f49a4b0ccf4d5f943ee4ee5fb3031f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4244939d9ed34faf8659ee6f8dff1eb3 : SUCCESS in 3m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/207e23737143486db3f8a6901ad622d2 : FAILURE in 1h 28m 41s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/f90c0aa0a7ea4a8aae436768a89ba3fd : SUCCESS in 1h 31m 04s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c080d390a4d8426b9c3ad5349d011159 : TIMED_OUT in 3h 03m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0b9a9d3fccdc4562b211c2fac70bcac6 : FAILURE in 1h 33m 16s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/fe51fb62649c498b991f49957e2cc19c : FAILURE in 1h 50m 49s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/4bede3faff634777baff1ffb471951bc : FAILURE in 2h 01m 59s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/03d92cdc28ba448999ef25ef0c13c1cf : SUCCESS in 1h 29m 01s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8621f4cd5e9a49cb8eadf1990a248eab : SUCCESS in 25m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9f5561e7b01c442fb3d4de29c483ba84 : SUCCESS in 42m 07s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/6505fedaa31e42cb82e964ac44827f46 : SUCCESS in 1h 05m 36s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/e3309d7e557545888d95e67bc6ab8537 : SUCCESS in 1h 03m 13s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/e2e8727243ee4b0595bb91612a74e8fc : SUCCESS in 40m 09s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/1621ed0b849c44c691e4b55eda020027 : SUCCESS in 28m 29s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/96215991bc5147d7b34032cfb2816ca0 : FAILURE in 27m 32s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"20b05e1449923359d9e3735420c04bb3677030c4","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-01-06 20:04:23.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"9fb4490d22853dc08bdcd5ef9d574cd687da43ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-01-07 12:44:30.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"21b89ea3ffef395f2f6a1c828502c534ebded65d","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-01-07 12:44:33.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"038711bd2e040a6386e50a8917539211649a10e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 15:50:58.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a6fc9fa448340c5994b0aeaa8112f51\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/493e7e023e9b42aeb8ddc1a101ece7cd : SUCCESS in 3m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9bfc772346844c6eac9472f0dea7365b : SUCCESS in 1h 22m 45s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/0219942eabdf44a78894f4bc46fd4832 : SUCCESS in 1h 10m 57s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d5c8486d3df6479d9573d1b08ad69223 : TIMED_OUT in 3h 04m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/142ac77e7c1044e695111d0874ce024c : FAILURE in 28m 30s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/820ee6253eb147fb987853755eb26b32 : FAILURE in 1h 04m 46s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/5a6d642267af466bb6b86a4c333eb673 : FAILURE in 2h 30m 25s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/58c372967e484800826ee1bebb7089ab : FAILURE in 1h 02m 50s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c5321daa9d8541eaacd820000818b91c : SUCCESS in 42m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/80464d64a08d46639521c02d331b8eaf : SUCCESS in 42m 21s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/59923834a5a141e0b60b51660f7b05c2 : SUCCESS in 51m 46s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/5573a27eda8b4bb69da14c7ed2999a7e : SUCCESS in 1h 05m 17s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/0fcfcff002e54de28d9148ec45e3af8b : SUCCESS in 1h 06m 29s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/4bb9e5f3bfd14a10ba5929a78267c3ca : SUCCESS in 32m 07s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6ab99dcf5ac846408523ea164f14ea84 : FAILURE in 50m 51s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"79fb148c55685cb45a32ad5970216092ed8cc3e7","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2025-01-07 21:11:40.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"a958e66c3d0ae1c88280f2df617b586f11887226","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-01-08 06:27:36.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"f816eb26a890ac9ae6caec329e6e0dca86351618","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-01-08 06:33:37.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"ae3929f37cd71a1dd13040073d6cec463190f10f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 13:49:18.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"94e0ae08f1b6cd7fd3509a6388927d2aca40f053","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 15:16:02.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a698af4db6c04f74a0ed8bdaf5879b5d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4e74800719f496c99aa80dbc15886ef : SUCCESS in 2m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c932b11c7ad1426696a9668413385ddb : SUCCESS in 59m 34s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/37ba843dbd154763a68db3d1f179e3ea : SUCCESS in 36m 42s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/037ca1d9e78e4be7a680d977f22c21c3 : SUCCESS in 24m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a1d8ef2beb634487805836876fe735f9 : SUCCESS in 36m 19s","accounts_in_message":[],"_revision_number":12},{"id":"f888018f2a5ff8b426ad80639fb5e10d4aadc01f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 15:16:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"7ff4509dbd643216175a62e84342510d65e6bcd4","revisions":{"4fc0842b6bb55fb8460b658b6c97f34d96456bc3":{"kind":"REWORK","_number":1,"created":"2024-12-06 10:34:42.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/43/937243/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/43/937243/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/1"}}},"commit":{"parents":[{"commit":"ee3079db74bd8ce6be5115a99263907fd3958da3","subject":"Merge \"Add nfs_versions option for NFS protocol version selection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ee3079db74bd8ce6be5115a99263907fd3958da3"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-12-06 10:30:34.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-12-06 10:32:36.000000000","tz":0},"subject":"Add tests for allow to update share access rule","message":"Add tests for allow to update share access rule\n\nConsider only share access rule with type \u0027ip\u0027.\n\nDepends-on: https://review.opendev.org/c/openstack/manila/+/931050\nRelated-Bug: #2066871\nChange-Id: I8500f71b527b4e5037f128906f68385f821ebf42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4fc0842b6bb55fb8460b658b6c97f34d96456bc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4fc0842b6bb55fb8460b658b6c97f34d96456bc3"}]},"branch":"refs/heads/master"},"21620a340454ca6f3c66bc85d99fd83dac57866c":{"kind":"REWORK","_number":2,"created":"2024-12-06 12:10:33.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/43/937243/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/43/937243/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/2"}}},"commit":{"parents":[{"commit":"ee3079db74bd8ce6be5115a99263907fd3958da3","subject":"Merge \"Add nfs_versions option for NFS protocol version selection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ee3079db74bd8ce6be5115a99263907fd3958da3"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-12-06 10:30:34.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-12-06 12:10:16.000000000","tz":0},"subject":"Add tests for allow to update share access rule","message":"Add tests for allow to update share access rule\n\nConsider only share access rule with type \u0027ip\u0027.\n\nDepends-on: https://review.opendev.org/c/openstack/manila/+/931050\nRelated-Bug: #2066871\nChange-Id: I8500f71b527b4e5037f128906f68385f821ebf42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/21620a340454ca6f3c66bc85d99fd83dac57866c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/21620a340454ca6f3c66bc85d99fd83dac57866c"}]},"branch":"refs/heads/master"},"3cb3d171cd539e2113192fbccc597cefe3f8517d":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2024-12-18 15:49:45.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/43/937243/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/43/937243/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/3"}}},"commit":{"parents":[{"commit":"1e8acf1170814519a5a82cc5f8aba952f5ca7f5f","subject":"Merge \"[RBAC] Generate unique cephx IDs for each operation in tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1e8acf1170814519a5a82cc5f8aba952f5ca7f5f"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-12-06 10:30:34.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-12-18 15:48:47.000000000","tz":0},"subject":"Add tests for allow to update share access rule","message":"Add tests for allow to update share access rule\n\nStarting micro-version 2.88, Manila will allow to update\naccess_level of share access rule.\n\nDepends-on: https://review.opendev.org/c/openstack/manila/+/931050\nRelated-Bug: #2066871\nChange-Id: I8500f71b527b4e5037f128906f68385f821ebf42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3cb3d171cd539e2113192fbccc597cefe3f8517d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3cb3d171cd539e2113192fbccc597cefe3f8517d"}]},"branch":"refs/heads/master"},"16322e4fb4efe00c16ac661039ab3db8fa0831a7":{"kind":"REWORK","_number":4,"created":"2024-12-19 04:52:53.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/43/937243/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/43/937243/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/4"}}},"commit":{"parents":[{"commit":"1e8acf1170814519a5a82cc5f8aba952f5ca7f5f","subject":"Merge \"[RBAC] Generate unique cephx IDs for each operation in tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1e8acf1170814519a5a82cc5f8aba952f5ca7f5f"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-12-06 10:30:34.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-12-19 04:02:20.000000000","tz":330},"subject":"Add tests for allow to update share access rule","message":"Add tests for allow to update share access rule\n\nStarting micro-version 2.88, Manila will allow to update\naccess_level of share access rule.\n\nDepends-on: https://review.opendev.org/c/openstack/manila/+/931050\nRelated-Bug: #2066871\nChange-Id: I8500f71b527b4e5037f128906f68385f821ebf42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/16322e4fb4efe00c16ac661039ab3db8fa0831a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/16322e4fb4efe00c16ac661039ab3db8fa0831a7"}]},"branch":"refs/heads/master"},"3de1ef484710d706ba9812490eb061de7ddfeac0":{"kind":"REWORK","_number":5,"created":"2024-12-27 15:53:14.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/43/937243/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/43/937243/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/5"}}},"commit":{"parents":[{"commit":"9c743992f835de0ba9e3b73daa2b3e5b783690f2","subject":"Merge \"IPV4 with Multinode Ceph-NFS Job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c743992f835de0ba9e3b73daa2b3e5b783690f2"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-12-06 10:30:34.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-12-27 15:53:08.000000000","tz":0},"subject":"Add tests for allow to update share access rule","message":"Add tests for allow to update share access rule\n\nStarting micro-version 2.88, Manila will allow to update\naccess_level of share access rule.\n\nDepends-on: https://review.opendev.org/c/openstack/manila/+/931050\nRelated-Bug: #2066871\nChange-Id: I8500f71b527b4e5037f128906f68385f821ebf42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3de1ef484710d706ba9812490eb061de7ddfeac0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3de1ef484710d706ba9812490eb061de7ddfeac0"}]},"branch":"refs/heads/master"},"38cd20b9aca6797628bebd6743310011f4eeac5d":{"kind":"REWORK","_number":6,"created":"2024-12-27 16:31:22.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/43/937243/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/43/937243/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/6"}}},"commit":{"parents":[{"commit":"9c743992f835de0ba9e3b73daa2b3e5b783690f2","subject":"Merge \"IPV4 with Multinode Ceph-NFS Job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c743992f835de0ba9e3b73daa2b3e5b783690f2"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-12-06 10:30:34.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-12-27 16:30:51.000000000","tz":0},"subject":"Add tests for allow to update share access rule","message":"Add tests for allow to update share access rule\n\nStarting micro-version 2.88, Manila will allow to update\naccess_level of share access rule.\n\nDepends-on: https://review.opendev.org/c/openstack/manila/+/931050\nRelated-Bug: #2066871\nChange-Id: I8500f71b527b4e5037f128906f68385f821ebf42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/38cd20b9aca6797628bebd6743310011f4eeac5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/38cd20b9aca6797628bebd6743310011f4eeac5d"}]},"branch":"refs/heads/master"},"2745aa62419ea39a7db1ec02d5cf4889928a0357":{"kind":"REWORK","_number":7,"created":"2025-01-02 11:23:11.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/43/937243/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/43/937243/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/7"}}},"commit":{"parents":[{"commit":"9c743992f835de0ba9e3b73daa2b3e5b783690f2","subject":"Merge \"IPV4 with Multinode Ceph-NFS Job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c743992f835de0ba9e3b73daa2b3e5b783690f2"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-12-06 10:30:34.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-01-02 11:22:31.000000000","tz":0},"subject":"Add tests for allow to update share access rule","message":"Add tests for allow to update share access rule\n\nStarting micro-version 2.88, Manila will allow to update\naccess_level of share access rule.\n\nDepends-on: https://review.opendev.org/c/openstack/manila/+/931050\nRelated-Bug: #2066871\nChange-Id: I8500f71b527b4e5037f128906f68385f821ebf42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2745aa62419ea39a7db1ec02d5cf4889928a0357"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2745aa62419ea39a7db1ec02d5cf4889928a0357"}]},"branch":"refs/heads/master"},"1adaa26f27cb9d4c37b436e253d35fe62f4ad747":{"kind":"REWORK","_number":8,"created":"2025-01-02 14:18:07.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/43/937243/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/43/937243/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/8"}}},"commit":{"parents":[{"commit":"9c743992f835de0ba9e3b73daa2b3e5b783690f2","subject":"Merge \"IPV4 with Multinode Ceph-NFS Job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c743992f835de0ba9e3b73daa2b3e5b783690f2"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-12-06 10:30:34.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-01-02 14:18:01.000000000","tz":0},"subject":"Add tests for allow to update share access rule","message":"Add tests for allow to update share access rule\n\nStarting micro-version 2.88, Manila will allow to update\naccess_level of share access rule.\n\nDepends-on: https://review.opendev.org/c/openstack/manila/+/931050\nRelated-Bug: #2066871\nChange-Id: I8500f71b527b4e5037f128906f68385f821ebf42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1adaa26f27cb9d4c37b436e253d35fe62f4ad747"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1adaa26f27cb9d4c37b436e253d35fe62f4ad747"}]},"branch":"refs/heads/master"},"9cbdc71f51fde5dd368734d73f7599f60addc9d4":{"kind":"REWORK","_number":9,"created":"2025-01-06 13:52:42.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/43/937243/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/43/937243/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/9"}}},"commit":{"parents":[{"commit":"9c743992f835de0ba9e3b73daa2b3e5b783690f2","subject":"Merge \"IPV4 with Multinode Ceph-NFS Job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c743992f835de0ba9e3b73daa2b3e5b783690f2"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-12-06 10:30:34.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-01-06 13:52:29.000000000","tz":0},"subject":"Add tests for allow to update share access rule","message":"Add tests for allow to update share access rule\n\nStarting micro-version 2.88, Manila will allow to update\naccess_level of share access rule.\n\nDepends-on: https://review.opendev.org/c/openstack/manila/+/931050\nRelated-Bug: #2066871\nChange-Id: I8500f71b527b4e5037f128906f68385f821ebf42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9cbdc71f51fde5dd368734d73f7599f60addc9d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9cbdc71f51fde5dd368734d73f7599f60addc9d4"}]},"branch":"refs/heads/master"},"805159a73f58c9e0a5f2dede47e3abd47eb3be13":{"kind":"REWORK","_number":10,"created":"2025-01-06 15:06:34.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/43/937243/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/43/937243/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/10"}}},"commit":{"parents":[{"commit":"9c743992f835de0ba9e3b73daa2b3e5b783690f2","subject":"Merge \"IPV4 with Multinode Ceph-NFS Job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c743992f835de0ba9e3b73daa2b3e5b783690f2"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-12-06 10:30:34.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-01-06 15:06:28.000000000","tz":0},"subject":"Add tests for allow to update share access rule","message":"Add tests for allow to update share access rule\n\nStarting micro-version 2.88, Manila will allow to update\naccess_level of share access rule.\n\nDepends-on: https://review.opendev.org/c/openstack/manila/+/931050\nRelated-Bug: #2066871\nChange-Id: I8500f71b527b4e5037f128906f68385f821ebf42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/805159a73f58c9e0a5f2dede47e3abd47eb3be13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/805159a73f58c9e0a5f2dede47e3abd47eb3be13"}]},"branch":"refs/heads/master"},"e78a59a265fe35ea6813a05a0d609f8b6c896f11":{"kind":"REWORK","_number":11,"created":"2025-01-06 16:22:42.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/43/937243/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/43/937243/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/11"}}},"commit":{"parents":[{"commit":"9c743992f835de0ba9e3b73daa2b3e5b783690f2","subject":"Merge \"IPV4 with Multinode Ceph-NFS Job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c743992f835de0ba9e3b73daa2b3e5b783690f2"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-12-06 10:30:34.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-01-06 16:22:12.000000000","tz":0},"subject":"Add tests for allow to update share access rule","message":"Add tests for allow to update share access rule\n\nStarting micro-version 2.88, Manila will allow to update\naccess_level of share access rule.\n\nDepends-on: https://review.opendev.org/c/openstack/manila/+/931050\nRelated-Bug: #2066871\nChange-Id: I8500f71b527b4e5037f128906f68385f821ebf42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e78a59a265fe35ea6813a05a0d609f8b6c896f11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e78a59a265fe35ea6813a05a0d609f8b6c896f11"}]},"branch":"refs/heads/master"},"7ff4509dbd643216175a62e84342510d65e6bcd4":{"kind":"REWORK","_number":12,"created":"2025-01-07 12:44:30.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/43/937243/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/43/937243/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/937243/12"}}},"commit":{"parents":[{"commit":"9c743992f835de0ba9e3b73daa2b3e5b783690f2","subject":"Merge \"IPV4 with Multinode Ceph-NFS Job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c743992f835de0ba9e3b73daa2b3e5b783690f2"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-12-06 10:30:34.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-01-07 12:44:08.000000000","tz":0},"subject":"Add tests for allow to update share access rule","message":"Add tests for allow to update share access rule\n\nStarting micro-version 2.88, Manila will allow to update\naccess_level of share access rule.\n\nDepends-on: https://review.opendev.org/c/openstack/manila/+/931050\nRelated-Bug: #2066871\nChange-Id: I8500f71b527b4e5037f128906f68385f821ebf42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7ff4509dbd643216175a62e84342510d65e6bcd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7ff4509dbd643216175a62e84342510d65e6bcd4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
