)]}'
{"id":"openstack%2Fmanila-tempest-plugin~937407","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I74ac4eafa41f0d6355716e7fe26d09b7dfcbc982","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/2089798","attention_set":{},"removed_from_attention_set":{"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2024-12-12 06:29:33.000000000","reason":"\u003cGERRIT_ACCOUNT_19262\u003e replied on the change","reason_account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}},"hashtags":[],"change_id":"I74ac4eafa41f0d6355716e7fe26d09b7dfcbc982","subject":"Fix floating IP association by filtering port list using IP address","status":"MERGED","created":"2024-12-09 20:44:22.000000000","updated":"2024-12-12 09:54:43.000000000","submitted":"2024-12-12 09:54:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937407-bug/2089798","meta_rev_id":"5766ab59e82049e16c45922c230d304ed1f81cfb","_number":937407,"virtual_id_number":937407,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-12 09:54:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2024-12-11 21:11:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2024-12-11 14:52:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":2,"date":"2024-12-11 15:07:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2024-12-11 14:20:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":1,"date":"2024-12-11 15:07:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-09 21:14:59.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-09 22:56:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-09 23:14:04.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-12-11 14:20:20.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2024-12-11 14:52:58.000000000","updated_by":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"reviewer":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"state":"REVIEWER"},{"updated":"2024-12-11 15:07:08.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-12-11 21:11:10.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"3c15dd6c5ff214ed6bb426c430831fa5e32a8749","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-12-09 20:44:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9f5ad78eade0442976290acbe59c3bcf38b2382","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-12-09 20:45:39.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0b8649e20f078b776f56da0e96c7f57e79113f6e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-09 21:14:59.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/07/937407/2/check/DellEMC_PowerFlex_Manila_dhss_false/d51c50c/DellEMC_PowerFlex_Manila_dhss_false/2200 : SUCCESS in 29m 07s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"d1d0a3dff4998528d3b7a313131feaad76955cfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 22:56:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83e9f7f11b28455890b09495fb817d35\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4dedb32431c473db09cbd3553acd22f : SUCCESS in 4m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fd29493f039a4aa0a820027cc8234532 : SUCCESS in 1h 07m 28s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/088fd5ffa0ae481ab153bab0b7bafb2e : SUCCESS in 55m 05s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c6c075898ae2424cbdda8f1ec9efe151 : SUCCESS in 1h 49m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2f7d533392364060a8fedd68da832154 : FAILURE in 49m 45s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a687167daa984775a53293847cf9edec : SUCCESS in 49m 49s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/5daff50013b04bfebb3c016d1e6363fb : FAILURE in 7m 14s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c28a58a8006946bc85b66f05470f4bea : SUCCESS in 58m 24s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1a9d4df980bb4dbca88f188ed60d63d3 : SUCCESS in 27m 01s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f2a9fa8b52274f0096ade0909b8168e5 : SUCCESS in 48m 26s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/9e1c7c61b1ac4d8e84839abdb58f286b : SUCCESS in 1h 16m 49s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/d4c3155fa0754c8c8af23234c14c9eec : SUCCESS in 1h 16m 52s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/3c0bd98e2ddd448fb53496fc37f60aa1 : SUCCESS in 1h 20m 10s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/8f61f4e79b5a41a69cddc597323a3489 : SUCCESS in 23m 03s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e08c692cd2c54b5091c938eaf82b6c31 : FAILURE in 49m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c83559594bfb49080056c4f8168c563d00ff3f98","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-12-09 23:14:04.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss finger://managesf.services.openstack.netapp.com:7979/c89c728a1a044493a0c1afaccebfeb5a : RETRY_LIMIT in 5m 17s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"4da31a5e5c2f948f56ac7244b068c118ba3e26b0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-09 23:31:22.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/07/937407/2/check/DellEMC_PowerFlex_Manila_dhss_false/50c5719/DellEMC_PowerFlex_Manila_dhss_false/2201 : SUCCESS in 34m 46s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"6dd7eb258e0097f0e83ea4d38c75219fa4c46cfe","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2024-12-11 14:20:20.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dab3c9c52000e349d56523678a9d09bb6d2699e0","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-12-11 14:52:58.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a05117b6a921493ebbe8e9e70eb7c527b29e4301","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-12-11 15:07:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d366e69ba9a4e1d85df4ac9955727305c50e696f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 15:08:14.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/210625f13dc147aeb36f457176bb20bb\n\n\nWarning:\n  Change 937406 in project openstack/tempest does not share a change queue with 937407 in project openstack/manila-tempest-plugin","accounts_in_message":[],"_revision_number":2},{"id":"2bc0b33fafd9f8059156932d8592f80788313e37","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-12-11 21:11:10.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3845db5fe59e816f51c0ad7233948157f65607aa","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-12-11 21:49:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8be214ccc673453a52be4c76c8d82bdd87d2e238","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-12-11 21:49:47.000000000","message":"Topic ports removed","accounts_in_message":[],"_revision_number":2},{"id":"3a6a525753bdd9a5562582a964320096f048d053","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-12-11 21:49:51.000000000","message":"Topic set to bug/2089798","accounts_in_message":[],"_revision_number":2},{"id":"ffb23abee515d1f257e9be99136541c21dd920fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 23:07:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b0f35fca41a488982c7dbddae7f705b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c1f246a59854702b0f7f399a15db29a : SUCCESS in 2m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4e90fe2e7c57494e9fb367daf2aed8a2 : SUCCESS in 37m 38s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/b6a957569ed044bf83ae2487291f6405 : SUCCESS in 54m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c19fd0581d454e0d80d7224b52d4e4b8 : SUCCESS in 1h 16m 03s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9b1dc14718304d76bc133f09755ffee0 : FAILURE in 48m 43s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3f667b126ae947fcb4f6c649c5397294 : SUCCESS in 54m 09s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/7d60c5cb34bb46cbbb3ab20a0d128653 : SUCCESS in 1h 07m 38s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1148f9b342ef4090bdc91e28da8d262d : FAILURE in 53m 00s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2e1ff2e96bc24d099c11e10501e0bf3a : SUCCESS in 43m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/51d7936e8bfe44f9b6b6317d7fff489d : SUCCESS in 41m 35s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/8dd86b8986e1475a9116c328bdf272aa : SUCCESS in 54m 58s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/078da9ece65841a2b6adecfaf341c958 : SUCCESS in 1h 11m 24s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/e4b39273b24a46ac864257ebc741af8b : SUCCESS in 52m 42s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/90cea017ed2d486e9a6e831f3b98c13b : SUCCESS in 23m 05s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7983b8279cf946ca9443fcb73ffc95d9 : FAILURE in 46m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3ced4d6d2fa5b09e2eb42ea19c963dea1d23b070","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 23:07:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2fd9b1349082c4234a6e68595b359334ec6f200e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-11 23:43:01.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/07/937407/2/check/DellEMC_PowerFlex_Manila_dhss_false/3017b5b/DellEMC_PowerFlex_Manila_dhss_false/2220 : SUCCESS in 35m 47s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"0a690bce6e84f7ee823adedeee9606c84e6b532b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 00:02:01.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c50021cc25848eba0b872ff104a1d72\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31e805c241ed416e8876ff7431f29005 : SUCCESS in 3m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/82ad05774aca41fab80d050975aaac32 : SUCCESS in 49m 21s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/ac126da41a9a4b138e1a4d87dfe588db : FAILURE in 53m 12s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d069d6c55ca54627a162737e500d627a : SUCCESS in 29m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4f8a15f8938e4313a7864fda45e066db : SUCCESS in 39m 27s","accounts_in_message":[],"_revision_number":2},{"id":"917fed83bff1285ee12f36e92f8f2aad24b05f03","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-12-12 06:29:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"be582e5e478689833c3a12e182e2088453971aeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 08:28:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0310fd9c19d4041a0aad87d00dd28a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa4e259daa1d4d658f2fa2a1ecb547df : SUCCESS in 2m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/80645a1ca5fb425eb634a3530ec4dd4e : SUCCESS in 1h 08m 42s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/4e9080c678c44e6d881d71e2fa696fa6 : SUCCESS in 47m 01s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/bb79b0b79d914d46846e47010a1420da : SUCCESS in 1h 56m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8a9ad9dd87fc4b42ba22875173ff9c6a : FAILURE in 52m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/6862d7e9e4084564a676b47d4fd9ff03 : SUCCESS in 50m 22s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/65d605c9b3ed4ced82940f61a13d0751 : SUCCESS in 1h 01m 14s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3a90c177280d42728d070b191f863cd3 : FAILURE in 1h 03m 27s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/85e7a33af42846d2806d1937e13ccc6a : SUCCESS in 28m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e49fb3a8eaa849609c20a1723609fc9d : SUCCESS in 42m 04s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/387fb9113f6d4dc6a722a364fd1ab538 : SUCCESS in 1h 19m 13s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/b61cf041ebc44aa281a06b1da42c4757 : SUCCESS in 1h 14m 15s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/ca837912c868484980f1fb7596644939 : SUCCESS in 1h 16m 27s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/0be9c2087b4046eb8ba63a869d9fe27f : SUCCESS in 25m 34s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/dbf2a1fbe77a430eb600c3bcc8ca3c2a : FAILURE in 1h 04m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0ec2aea474b79f34d590cfbd1535a09302c3c20d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 08:29:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0dcaa7817ed06abab4bed247c2ab54a683b4f3ea","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-12 08:40:32.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/07/937407/2/check/DellEMC_PowerFlex_Manila_dhss_false/6073d56/DellEMC_PowerFlex_Manila_dhss_false/2221 : FAILURE in 11m 45s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"9057af0426ce96fe647bacddfc69e1f0d93650b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 09:54:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cb72ea9b6ae4da899b6034f35c88fd6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8c19a0b531e40b792b4635b4055b952 : SUCCESS in 2m 18s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5c09c336aace47a1a4ef4992ae450d0f : SUCCESS in 1h 20m 31s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/3fd7ec3e02f84646876f181b1f2574d0 : SUCCESS in 1h 18m 09s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/55f9aa45c6664b4b958a30f92def9101 : SUCCESS in 27m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ed8425a82696475c97dad20afb6470f5 : SUCCESS in 42m 28s","accounts_in_message":[],"_revision_number":2},{"id":"5766ab59e82049e16c45922c230d304ed1f81cfb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 09:54:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bebefa776d58f26c048b434004adbaef0b699a2e","revisions":{"218888a1fc233c8a260b41edfd6bc5acd32bb0e3":{"kind":"REWORK","_number":1,"created":"2024-12-09 20:44:22.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/07/937407/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/07/937407/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/07/937407/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/07/937407/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/07/937407/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/07/937407/1"}}},"commit":{"parents":[{"commit":"ee3079db74bd8ce6be5115a99263907fd3958da3","subject":"Merge \"Add nfs_versions option for NFS protocol version selection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ee3079db74bd8ce6be5115a99263907fd3958da3"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-12-09 20:19:01.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-12-09 20:43:55.000000000","tz":120},"subject":"Fix floating IP association by filtering port list using IP address","message":"Fix floating IP association by filtering port list using IP address\n\nThe recent change in Tempest introduced a step to ensure all VM ports are\nactive before associating a floating IP.\nThis caused issues in the manila-tempest-plugin scenario tests, where the\ncorrect port couldn\u0027t be identified for floating IP association.\n\nThis patch fixes the issue by filtering the port list using the IP address\nof the fixed IP to narrow down the result to the specific port associated\nwith the given address.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/937406\nRelated-Bug: #2089798\nChange-Id: I74ac4eafa41f0d6355716e7fe26d09b7dfcbc982\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/218888a1fc233c8a260b41edfd6bc5acd32bb0e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/218888a1fc233c8a260b41edfd6bc5acd32bb0e3"}]},"branch":"refs/heads/master"},"bebefa776d58f26c048b434004adbaef0b699a2e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-12-09 20:45:39.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/07/937407/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/07/937407/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/07/937407/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/07/937407/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/07/937407/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/07/937407/2"}}},"commit":{"parents":[{"commit":"ee3079db74bd8ce6be5115a99263907fd3958da3","subject":"Merge \"Add nfs_versions option for NFS protocol version selection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ee3079db74bd8ce6be5115a99263907fd3958da3"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2024-12-09 20:19:01.000000000","tz":120},"committer":{"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","date":"2024-12-09 20:45:39.000000000","tz":0},"subject":"Fix floating IP association by filtering port list using IP address","message":"Fix floating IP association by filtering port list using IP address\n\nThe recent change in Tempest introduced a step to ensure all VM ports are active before associating a floating IP.\nThis caused issues in the manila-tempest-plugin scenario tests, where the correct port couldn\u0027t be identified for floating IP association.\n\nThis patch fixes the issue by filtering the port list using the IP address of the fixed IP to narrow down the result to the specific port associated with the given address.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/937406\nRelated-Bug: #2089798\nChange-Id: I74ac4eafa41f0d6355716e7fe26d09b7dfcbc982\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bebefa776d58f26c048b434004adbaef0b699a2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bebefa776d58f26c048b434004adbaef0b699a2e"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
