)]}'
{"id":"openstack%2Fmanila-tempest-plugin~938253","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I57598f07dc9fd55b00f043f564414f222b8765d9","project":"openstack/manila-tempest-plugin","branch":"master","topic":"add-endpoint-type","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2025-01-14 14:24:07.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2025-01-14 12:59:21.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"36686":{"account":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"last_update":"2025-01-14 14:24:07.000000000","reason":"Change was submitted"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2025-01-14 14:24:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I57598f07dc9fd55b00f043f564414f222b8765d9","subject":"Update for using endpoint_type to create client","status":"MERGED","created":"2024-12-26 08:22:23.000000000","updated":"2025-01-14 14:24:07.000000000","submitted":"2025-01-14 14:24:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"938253-add-endpoint-type","meta_rev_id":"69b94a8de2b08cf5b1dd090cab20852acedc4454","_number":938253,"virtual_id_number":938253,"owner":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-14 14:24:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2025-01-14 12:59:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2025-01-14 13:00:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"reviewers":{"REVIEWER":[{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-26 10:55:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-27 18:28:55.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2025-01-09 15:43:17.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2025-01-09 15:47:41.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"a99ba542d3d358f8e7f899cfbf3f848545b1e958","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"date":"2024-12-26 08:22:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"865cdd1e8b30ab68adcbd36c05e2e6e5b3d89b0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"date":"2024-12-26 08:24:05.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b5905e6127c64f7e24f52fb89527411dc434a3af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"date":"2024-12-26 08:25:05.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"6d029761786a3def7a4ffce575fdf8658fb338bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"date":"2024-12-26 08:42:28.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"16ef2299f0a9498c85157fd25274dcae83dae31f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-26 10:55:38.000000000","message":"Patch Set 4: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbd7a3de2a4e4b74bacaa5140489d42d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf0d5c04fab04557b9f437d174b56c47 : FAILURE in 2m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d3236f2e3cdf4d6598e32037ad91c889 : SUCCESS in 1h 20m 57s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/f358f35ddd184ce89a72fc26d4063946 : SUCCESS in 1h 08m 05s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f1a238253b29451e92219cb93a13d399 : SUCCESS in 2h 12m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fd548a1e1c614a9b9ea76ca5e0a75875 : FAILURE in 54m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/88a4a92c727043cd9f0303e4aa45d05d : SUCCESS in 1h 20m 03s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a6a2c4b65cfb45378bea4d324e462796 : FAILURE in 1h 37m 50s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/27b4b80dbf4a4a479fd675c5240a97d5 : SUCCESS in 1h 23m 22s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/72b58f1f481a4d468f0c3c24ca1e27ed : SUCCESS in 22m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/40cae0e054744957a57576968a28cee8 : SUCCESS in 29m 18s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/9c15cd28c2a84f4ebecd14f7011d50e7 : SUCCESS in 1h 23m 06s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/d7eedd086b1c492dab4c574b5cf56fbf : SUCCESS in 1h 14m 07s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/435135f3387843d1b68d1709cf9e25d6 : FAILURE in 1h 27m 46s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/9cb0ea067f034ae4a508169dc4efbc4e : SUCCESS in 37m 31s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4ab04b2b4e6e4b818de91ac9b4c1812a : FAILURE in 52m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"27374050fdd37136761d43c4d82ea4b158ef9f05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"date":"2024-12-27 03:32:20.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a6a23721fff6ef42ac05fd4cbaec8c303aca0409","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-27 06:06:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a74d4dcad2e649bc973b9b5f3cf7b7f9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/983c0f1dbc464d62869558a16d54ad0b : SUCCESS in 3m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d6fa490c176b453a8279ba2bb6360407 : SUCCESS in 1h 12m 24s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/d17303825c1e4ec4a3b671878ad84a47 : SUCCESS in 1h 08m 33s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/80af10784f1f41f6ae1d3d52042ad326 : SUCCESS in 2h 29m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b159f43f7c03438bb2d59348ff6fa0f1 : FAILURE in 39m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/cc960d57e3934998a3438cd0200310f0 : SUCCESS in 1h 08m 23s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/4888992df36541829f57a0b4cb0e2c35 : FAILURE in 1h 31m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2201e66aa3ae4e4fa5c9cd9e086ddca4 : FAILURE in 44m 44s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d0965f3dc01d4b258414b3457a7cdcc1 : SUCCESS in 27m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d3c337e0eef24397a350b92eaadf4e62 : SUCCESS in 40m 32s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/371f34c866b6416f86c294784641d901 : SUCCESS in 1h 16m 15s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/5605b973ba2d48d3ad2649da9ed38b33 : SUCCESS in 54m 55s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/2356c2e3b293498eba5877e5089654b8 : SUCCESS in 1h 13m 24s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/5d7a04eae7c54fbcb2934f5caf99ea11 : SUCCESS in 25m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fa7d3ea8cf3145f896bd60c1bdf81e38 : FAILURE in 36m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"37a3fab90af843b6fb1d64e38ac73d45baa4ee18","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-12-27 18:28:55.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"34fe272cdb99993450bc2c239cc8af113e70efd2","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2025-01-09 15:47:41.000000000","message":"Patch Set 5: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"7d9a0af6c9d64fafe564a2c6dc4f685d8d947eff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"date":"2025-01-10 04:53:47.000000000","message":"Uploaded patch set 6: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"353444107fc5b27fc453d6bc04e1a16e16f2b923","author":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"date":"2025-01-10 04:54:53.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b5d86b7e48fd0a40f85056885d1b4f3f507cab81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 07:25:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d22c8f58d4ea49c6882eb8d346398079\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/110660fefe94401e9ca74554e5ad3932 : SUCCESS in 2m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a9173206b77b4be59239cbe75467fd43 : SUCCESS in 1h 28m 15s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/38a38c0965114a94820561f45013c99c : SUCCESS in 1h 07m 51s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7576ea67a3e748bea545419946b36cb9 : SUCCESS in 2h 26m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/77cebb1e84ce4c4992edb8807d3cbf89 : FAILURE in 49m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8011a49171714ec390a9bf0e57fe9863 : SUCCESS in 56m 15s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f3ab74d2487044f3a0293b445e8b1792 : FAILURE in 1h 33m 43s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b845c35bc01641c6ba43a9b6650ed04c : FAILURE in 1h 04m 33s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c13e7d3c4fbe4380b270b68426964ed4 : SUCCESS in 28m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/761df593ffae449aa10725c2a35978b1 : SUCCESS in 36m 08s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/2c6f6da559d540a48eeb097d254a4a40 : SUCCESS in 49m 41s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/2fe57d70cff1441aac1b92bcca41fdef : SUCCESS in 35m 33s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/646cd413fa06450aaa5e9ce520a98eed : SUCCESS in 36m 23s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3e9e9b25c30b4624b43f1f5d46534d44 : SUCCESS in 23m 55s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/99d1efc100044aa8ae05e8a6bff73f2f : FAILURE in 50m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"745af193fbd0faa578344de82d6a0db29067ade5","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2025-01-14 12:59:21.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"54b7a586419b883908343a446bdeb638fce80ae1","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2025-01-14 13:00:14.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"1a62ec824d04a87d1fa2b5d7256458264254f29a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 13:00:47.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"3fa117a3e8f9adb8ebb92595997886c583daadd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 14:24:06.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2fa8e43a20140f5ae7f46185bb0759c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/744d03dee72742618b8827f29b7f08c9 : SUCCESS in 4m 14s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a1bf6562a5a64a36b2950da63013d5be : SUCCESS in 1h 23m 11s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/1c01cc90f16d421f9f1258d5c1b8d995 : SUCCESS in 1h 10m 44s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/24e7be1d329c448d8feccdde7e235bd1 : SUCCESS in 42m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f134f1aa958348a4a46989fdbf8278f8 : SUCCESS in 39m 50s","accounts_in_message":[],"_revision_number":6},{"id":"69b94a8de2b08cf5b1dd090cab20852acedc4454","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 14:24:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"5f585fb639ac54b906dbd14cc6fe1620406b6fe3","revisions":{"bb08ceb03c68b02c5314b090b4631dfe5e4e425e":{"kind":"REWORK","_number":1,"created":"2024-12-26 08:22:23.000000000","uploader":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"ref":"refs/changes/53/938253/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/53/938253/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/938253/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/938253/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/938253/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/938253/1"}}},"commit":{"parents":[{"commit":"9c743992f835de0ba9e3b73daa2b3e5b783690f2","subject":"Merge \"IPV4 with Multinode Ceph-NFS Job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c743992f835de0ba9e3b73daa2b3e5b783690f2"}]}],"author":{"name":"hongp","email":"inyong.hong@samsung.com","date":"2024-12-26 08:18:38.000000000","tz":540},"committer":{"name":"hongp","email":"inyong.hong@samsung.com","date":"2024-12-26 08:18:38.000000000","tz":540},"subject":"update for using event_type to create client","message":"update for using event_type to create client\n\nChange-Id: I57598f07dc9fd55b00f043f564414f222b8765d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bb08ceb03c68b02c5314b090b4631dfe5e4e425e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bb08ceb03c68b02c5314b090b4631dfe5e4e425e"}]},"branch":"refs/heads/master"},"b99510ad1f05d99e7343cf532f4255bdf9da6b2f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-12-26 08:24:05.000000000","uploader":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"ref":"refs/changes/53/938253/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/53/938253/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/938253/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/938253/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/938253/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/938253/2"}}},"commit":{"parents":[{"commit":"9c743992f835de0ba9e3b73daa2b3e5b783690f2","subject":"Merge \"IPV4 with Multinode Ceph-NFS Job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c743992f835de0ba9e3b73daa2b3e5b783690f2"}]}],"author":{"name":"hongp","email":"inyong.hong@samsung.com","date":"2024-12-26 08:18:38.000000000","tz":540},"committer":{"name":"Inyong Hong","email":"inyong.hong@samsung.com","date":"2024-12-26 08:24:05.000000000","tz":0},"subject":"update for using event_type to create client","message":"update for using event_type to create client\n\n\n\nRelated-Bug: #2092514\nChange-Id: I57598f07dc9fd55b00f043f564414f222b8765d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b99510ad1f05d99e7343cf532f4255bdf9da6b2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b99510ad1f05d99e7343cf532f4255bdf9da6b2f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1741afb911117e0a3f4cb714eb3a90c6e977187c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-12-26 08:25:05.000000000","uploader":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"ref":"refs/changes/53/938253/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/53/938253/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/938253/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/938253/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/938253/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/938253/3"}}},"commit":{"parents":[{"commit":"9c743992f835de0ba9e3b73daa2b3e5b783690f2","subject":"Merge \"IPV4 with Multinode Ceph-NFS Job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c743992f835de0ba9e3b73daa2b3e5b783690f2"}]}],"author":{"name":"hongp","email":"inyong.hong@samsung.com","date":"2024-12-26 08:18:38.000000000","tz":540},"committer":{"name":"Inyong Hong","email":"inyong.hong@samsung.com","date":"2024-12-26 08:25:05.000000000","tz":0},"subject":"Update for using event_type to create client","message":"Update for using event_type to create client\n\nTo create client using event_type conf value.\n\nRelated-Bug: #2092514\nChange-Id: I57598f07dc9fd55b00f043f564414f222b8765d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1741afb911117e0a3f4cb714eb3a90c6e977187c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1741afb911117e0a3f4cb714eb3a90c6e977187c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"16a0b70fa8bd7dddc803bbed4563584116627ad3":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-12-26 08:42:28.000000000","uploader":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"ref":"refs/changes/53/938253/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/53/938253/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/938253/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/938253/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/938253/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/938253/4"}}},"commit":{"parents":[{"commit":"9c743992f835de0ba9e3b73daa2b3e5b783690f2","subject":"Merge \"IPV4 with Multinode Ceph-NFS Job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c743992f835de0ba9e3b73daa2b3e5b783690f2"}]}],"author":{"name":"hongp","email":"inyong.hong@samsung.com","date":"2024-12-26 08:18:38.000000000","tz":540},"committer":{"name":"hongp","email":"inyong.hong@samsung.com","date":"2024-12-26 08:37:49.000000000","tz":540},"subject":"Update for using endpoint_type to create client","message":"Update for using endpoint_type to create client\n\nTo create client using endpoint_type conf value.\n\nChange-Id: I57598f07dc9fd55b00f043f564414f222b8765d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/16a0b70fa8bd7dddc803bbed4563584116627ad3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/16a0b70fa8bd7dddc803bbed4563584116627ad3"}]},"branch":"refs/heads/master"},"0d9520bd33fab89fbed66056f2d1a0b94ae9f25b":{"kind":"REWORK","_number":5,"created":"2024-12-27 03:32:20.000000000","uploader":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"ref":"refs/changes/53/938253/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/53/938253/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/938253/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/938253/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/938253/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/938253/5"}}},"commit":{"parents":[{"commit":"9c743992f835de0ba9e3b73daa2b3e5b783690f2","subject":"Merge \"IPV4 with Multinode Ceph-NFS Job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c743992f835de0ba9e3b73daa2b3e5b783690f2"}]}],"author":{"name":"hongp","email":"inyong.hong@samsung.com","date":"2024-12-26 08:18:38.000000000","tz":540},"committer":{"name":"hongp","email":"inyong.hong@samsung.com","date":"2024-12-27 03:28:41.000000000","tz":540},"subject":"Update for using endpoint_type to create client","message":"Update for using endpoint_type to create client\n\nTo create client using endpoint_type conf value.\n\nChange-Id: I57598f07dc9fd55b00f043f564414f222b8765d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0d9520bd33fab89fbed66056f2d1a0b94ae9f25b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0d9520bd33fab89fbed66056f2d1a0b94ae9f25b"}]},"branch":"refs/heads/master"},"5f585fb639ac54b906dbd14cc6fe1620406b6fe3":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-01-10 04:53:47.000000000","uploader":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"ref":"refs/changes/53/938253/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/53/938253/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/938253/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/938253/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/938253/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/938253/6"}}},"commit":{"parents":[{"commit":"9c743992f835de0ba9e3b73daa2b3e5b783690f2","subject":"Merge \"IPV4 with Multinode Ceph-NFS Job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c743992f835de0ba9e3b73daa2b3e5b783690f2"}]}],"author":{"name":"hongp","email":"inyong.hong@samsung.com","date":"2024-12-26 08:18:38.000000000","tz":540},"committer":{"name":"hongp","email":"inyong.hong@samsung.com","date":"2025-01-10 04:49:57.000000000","tz":540},"subject":"Update for using endpoint_type to create client","message":"Update for using endpoint_type to create client\n\nTo create client using endpoint_type conf value.\n\nRelated-bug: #2092514\nChange-Id: I57598f07dc9fd55b00f043f564414f222b8765d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5f585fb639ac54b906dbd14cc6fe1620406b6fe3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5f585fb639ac54b906dbd14cc6fe1620406b6fe3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
