)]}'
{"id":"openstack%2Fmanila-tempest-plugin~938643","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I601f13dbdadff52c7d89a67b35e6035f4fadeaab","project":"openstack/manila-tempest-plugin","branch":"master","topic":"ssh-exc","hashtags":[],"change_id":"I601f13dbdadff52c7d89a67b35e6035f4fadeaab","subject":"Fix debug_ssh decorator","status":"NEW","created":"2025-01-08 10:40:09.000000000","updated":"2025-01-08 13:07:27.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"04429b3cdd54a03ca13d4eb1d0ee335570df236b","_number":938643,"virtual_id_number":938643,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-01-08 13:07:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-08 13:07:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"81da216e151ab2343d086089235ffcfdd6e51c8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2025-01-08 10:40:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"621b55e06b9ded893fdb4f24491334f4232712fa","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2025-01-08 10:42:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a00095a21ae0b767a85f6813a34429915e6846fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2025-01-08 10:48:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"04429b3cdd54a03ca13d4eb1d0ee335570df236b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 13:07:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b3bc8b699774c758c7405f83bab9eb0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/194b2bbb92434a77b50fa54304b3e335 : SUCCESS in 2m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4ebbd6923f2449eda50c9c79d3f4f334 : SUCCESS in 1h 05m 45s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/8d55ab094a9f4b00846934aaa1d8716c : SUCCESS in 1h 08m 54s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/22fac064744d4f8a89dd9f6a5f558b8b : SUCCESS in 2h 17m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/62c6109e9061424f9bcc39fdc544ce52 : FAILURE in 1h 08m 49s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ece689026a364072a848f6fe238af015 : SUCCESS in 1h 02m 02s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/0e84ec4e3989446f900a2db9d48b7a6e : FAILURE in 1h 32m 31s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a5bd3cbd0212488f8ed9fda58bb82f87 : FAILURE in 1h 03m 55s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/da84fbadaade479ca0afecd60178b400 : SUCCESS in 27m 01s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9d0eb23043dc4862a47a549ab51a7ef1 : SUCCESS in 33m 05s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/80c5aee456644ff1a3351a8a3e2d7ea7 : SUCCESS in 1h 19m 43s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/36a9f28f4ef5481687107a4d5766f0c2 : SUCCESS in 1h 05m 45s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/997fb30fd25b4e50bb99abdfc65c601d : SUCCESS in 1h 02m 18s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f22a06ed1a684b83a1732b902eb1cb32 : SUCCESS in 35m 28s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/234dc8ef774d449c8b8b86b8931baec5 : FAILURE in 51m 11s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8d6feda3754bc5dc80bc6284f05375c1d5d7dae0","revisions":{"cefe2b539fe44a7d6935adf6327b1303a3942cff":{"kind":"REWORK","_number":1,"created":"2025-01-08 10:40:09.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/43/938643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/43/938643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/938643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/938643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/938643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/938643/1"}}},"commit":{"parents":[{"commit":"9c743992f835de0ba9e3b73daa2b3e5b783690f2","subject":"Merge \"IPV4 with Multinode Ceph-NFS Job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c743992f835de0ba9e3b73daa2b3e5b783690f2"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2025-01-08 10:36:41.000000000","tz":120},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2025-01-08 10:36:41.000000000","tz":120},"subject":"Fix debug_ssh decorator","message":"Fix debug_ssh decorator\n\nin Python3, `raise` statement stopped accepting a 3-tuple input\nproduced by sys.exc_info().\n\nUse empty `raise` to re-raise the active SSHTimeout exception as is.\n\nChange-Id: I601f13dbdadff52c7d89a67b35e6035f4fadeaab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cefe2b539fe44a7d6935adf6327b1303a3942cff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cefe2b539fe44a7d6935adf6327b1303a3942cff"}]},"branch":"refs/heads/master"},"8d6feda3754bc5dc80bc6284f05375c1d5d7dae0":{"kind":"REWORK","_number":2,"created":"2025-01-08 10:48:02.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/43/938643/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/43/938643/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/938643/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/938643/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/938643/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/43/938643/2"}}},"commit":{"parents":[{"commit":"9c743992f835de0ba9e3b73daa2b3e5b783690f2","subject":"Merge \"IPV4 with Multinode Ceph-NFS Job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c743992f835de0ba9e3b73daa2b3e5b783690f2"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2025-01-08 10:36:41.000000000","tz":120},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2025-01-08 10:47:36.000000000","tz":120},"subject":"Fix debug_ssh decorator","message":"Fix debug_ssh decorator\n\nin Python3, `raise` statement stopped accepting a 3-tuple input\nproduced by sys.exc_info().\n\nUse empty `raise` to re-raise the active SSHTimeout exception as is.\n\nChange-Id: I601f13dbdadff52c7d89a67b35e6035f4fadeaab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8d6feda3754bc5dc80bc6284f05375c1d5d7dae0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8d6feda3754bc5dc80bc6284f05375c1d5d7dae0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
