)]}'
{"id":"openstack%2Fmanila-tempest-plugin~940169","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I42ce5663a13c41c6e11e78edee4949519971ec1c","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/2019156","attention_set":{"32594":{"account":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"last_update":"2025-09-19 06:27:47.000000000","reason":"A robot voted negatively on a label"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-10-17 22:58:54.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I42ce5663a13c41c6e11e78edee4949519971ec1c","subject":"DNM Run share group test","status":"NEW","created":"2025-01-24 20:23:56.000000000","updated":"2026-03-02 16:18:33.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"70e4ab34c3610718df1447c0a4ec1e6f1d0f6bae","_number":940169,"virtual_id_number":940169,"owner":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-19 18:25:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-24 23:33:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-26 00:45:37.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-09-19 05:21:47.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2025-10-06 22:59:30.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"}],"messages":[{"id":"0e64bccefaf17b4fb10b37c3c8025b3205a2ba90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2025-01-24 20:23:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdc1677d77320483719b2f81db08febe73f12a67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2025-01-24 20:24:47.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"621e58571c61d24865fc00846ab3b22636fc4b6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 23:33:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/312e697e6cea40538f29acc61f0b7e72\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63871148266445999c3aae1e08288849 : SUCCESS in 2m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bb9c194a352c4d2e84cb2d22f7a4017b : SUCCESS in 1h 01m 12s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/e2c5fbf3a989438cb72fabe467d74b72 : SUCCESS in 1h 17m 12s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2481676cd43f4468b9640675a7744e59 : TIMED_OUT in 3h 02m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e837dbe685944f3c9399766b6461b575 : FAILURE in 1h 08m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/2e3cfe608ead4dd88c7b73f0f40ec2d6 : FAILURE in 1h 33m 03s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a72188006f0b4f5fb517f33381a0c4ee : FAILURE in 2h 19m 23s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b8aad264feba45fba46fecc1ed0f1213 : FAILURE in 57m 50s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fc9eb1adb8c840d69450c85e59066413 : SUCCESS in 33m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9321381ed0634a7bb892e4e844efa036 : SUCCESS in 34m 39s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/b71ab53e4fd64ef797b1af6000492d30 : SUCCESS in 1h 05m 44s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/0c15a00982584736a7a8e9a5eb5ee3a7 : SUCCESS in 1h 00m 34s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/e6f3c0283241442eab848d175fac94cc : SUCCESS in 1h 07m 05s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/4fa95af093164a08a582b5dddc380c1b : SUCCESS in 26m 36s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6c67e4dc5fc348eabfc6f20b1769f8cd : FAILURE in 38m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9948ae69e4cdda562a3c42f9350b426256ff00b1","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2025-01-31 18:50:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"04e01a5e271819545ed53300b83c301401c8e162","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 21:56:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab4fa83ad78f47beaf421be32ae27a4a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b2922344fe9412e909245c4b867965f : SUCCESS in 3m 07s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3aac4f4df15d47e0a1ae1318019863ed : SUCCESS in 1h 07m 53s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/8c7aced357bd49218be2f29374bc6e35 : SUCCESS in 1h 02m 37s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7713638c3c6c4bf598e9a1d055782d07 : TIMED_OUT in 3h 04m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bae1ad2b9b7d40eb810f659fbfe31f5c : FAILURE in 1h 06m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3492305842a0497ea521ef0f2815b677 : FAILURE in 1h 52m 08s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/93edecaeb6504191abeb102c559dc98e : FAILURE in 1h 55m 08s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/125d912a6800470e91a300ffae70ae82 : SUCCESS in 1h 12m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/73f23c5fa1a34496aae1fba35a38e0a8 : SUCCESS in 37m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/754d0a540d8c4e4dab914340d7421153 : SUCCESS in 1h 00m 27s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/3b00a0f6a63f4754a4cfff54a92f7690 : SUCCESS in 1h 25m 35s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/74386e133691484b958b9777a74d1a17 : SUCCESS in 1h 07m 14s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/32d2e0ef24a047838b2657bfd33a1a14 : SUCCESS in 58m 50s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ab9b6a0c2f0b403b896bbbba82d93ae3 : SUCCESS in 37m 11s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/de274bd0eaa84fb1ba3bdac0f0a1c19c : FAILURE in 51m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4aafd049ed4e50bd483117d5e37b9c058134b2ce","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2025-03-25 19:46:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4141c5010ab45149c0e00622c8d95f14d0b42b27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 21:44:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b8c19b8ee2e44e79edef29d08cc9791\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/513721469e324431943a48f0382547fb : SUCCESS in 2m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/74e99fff2b79488d81e9e848a2e369c7 : SUCCESS in 31m 36s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/077fbd8716b84609ae0a5175ff1f9a31 : SUCCESS in 43m 58s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2413bae61ec143c2a4590f583a59abd6 : SUCCESS in 1h 57m 00s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/69c56a339a994f2da7b32b3e2f12f198 : FAILURE in 52m 30s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/81b7a8e282db4ca98963c540c757724d : FAILURE in 55m 06s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/3bba32cc630445049eee2d76fd779f53 : FAILURE in 1h 09m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f3a9aeb3267048ccbcc16a8984f0eef2 : SUCCESS in 47m 41s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1a3ad760cb1644098450c9ae92154689 : SUCCESS in 27m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/791eacb7760f44bfa8744417c3866045 : SUCCESS in 28m 53s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/88daa6e4d432495eb9ea3ff2973fd25a : SUCCESS in 58m 01s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/7c86f4ead4f7444d8189cb4d56df2451 : SUCCESS in 53m 49s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/7358cecf6a374becbe164857e3c533e5 : SUCCESS in 55m 45s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/eb21f93e5016430ba3cd5e0ed50c7137 : SUCCESS in 30m 22s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2a9724dd2c6a4e039c219746cbd6f6ec : FAILURE in 44m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bd960fb6ae7a63f53c0f65b9f6523a66ab03fd99","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-26 00:45:37.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"11c18575e331202f1ea32127b11cbc5ab70fb377","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2025-06-26 17:06:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4f5b0e66ddcc81ada0b2c1622016e9e3ff3f1ad5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 19:00:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/55e529a89cdb4448811db845f5e2983a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3ddd470b05a49ea8d4fbad25a6bb883 : SUCCESS in 2m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8367a6fdf3a8449da11a8db563fc9b80 : SUCCESS in 1h 02m 52s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8090fbe2f21c438da85e0bb58fa41cf1 : FAILURE in 1h 53m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5d7f313760f84b7eb6ada58fcee50605 : FAILURE in 46m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b3d491069507415aa9171f9ecd493f85 : FAILURE in 1h 02m 07s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/71ce3e801da84f68be825acc1ba7392a : SUCCESS in 1h 12m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a61ebacb5b5646f0a03bd66b0f4b6717 : SUCCESS in 1h 01m 44s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/614450b1f42d48209b27da7383b42ae8 : SUCCESS in 29m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/16bcad7f34a74293b7e75f7b502d72e4 : SUCCESS in 36m 55s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/3e33404fb61a45f8a0c88e3f4847e2e3 : SUCCESS in 1h 05m 02s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/f4596d1bd748441aa59c68407a1af983 : FAILURE in 33m 25s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/4dd7cff28b5c4e24ac6996a84748c468 : SUCCESS in 37m 10s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/863361eeb1d14b219726c002e917f73e : SUCCESS in 16m 55s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/47969f34372e46c7ae85e3f72b1b95bc : FAILURE in 40m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"200875b0a1e96369793e3fbc6f52717699d79325","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2025-06-27 13:21:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b7141912095eb69d915876e721bf0363c1954783","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 15:43:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d744a30981d84bfead55d409fc2744e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c656e4afbbce4a79b8ff706fa0170e2c : SUCCESS in 3m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4c8c0dc57bff4f66a906a3e5e13de6fc : SUCCESS in 1h 09m 58s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/08d228458a3c4a41937bde0b70e44570 : SUCCESS in 2h 20m 00s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ce7d397f41ba4d06b8f68034b293e5e0 : FAILURE in 11m 49s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/999cc8614af84b3bb10da7a2c7587376 : FAILURE in 1h 00m 27s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/463895f58811435db579a201c5eaf4be : SUCCESS in 1h 32m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/580ec466af1f40a98cbb7a12536cd229 : SUCCESS in 1h 09m 55s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c7802152351a4ef3ac6fe817362eb4c2 : SUCCESS in 34m 11s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/66388ef09c004e07a3c05b28aa795214 : SUCCESS in 42m 01s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/0f418b69c27f4bdcb47a53192aa3fd24 : SUCCESS in 1h 13m 01s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/990d2de80130433b9a1e2ef02f2fdf90 : SUCCESS in 1h 08m 21s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/7f87e92ac614422f9e594824daece3ed : SUCCESS in 1h 06m 28s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/020d1f497cb543e597b1933da7c6ba37 : SUCCESS in 30m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0df384b4ee4e470ab9a589aa11487c0d : FAILURE in 47m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1d4404b666636b888ffa9d0b4beef16b2d8efbda","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-19 05:21:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d1b61238ecc3ba4a50b2707e35ea725cb5a7cd14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-19 05:24:13.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"5a4eed3dbb7444acda47857f89ff82e5efbfdd6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-19 05:24:16.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"d350812bb4fb99b9da4f037d46c121e833d0997c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-19 05:25:49.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"0d7d6886753d986b9d8753670e667778792a837c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 06:27:47.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee97492690844cb7bb8f2c1c8cca2076\n\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f83b5beeb43e497a81c1d98a5c1b20a9 : FAILURE in 41m 54s\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/65f724924b3e44ce90e2c68c4069d158 : SUCCESS in 1h 00m 36s","accounts_in_message":[],"_revision_number":5},{"id":"7dce2db30698818dded621b5c9526dfe874dbb6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-19 16:54:14.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"c6b7435f9ad9c871d52596a4a0fdca68c40885c7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-19 18:08:20.000000000","message":"Topic set to bug/2019156","accounts_in_message":[],"_revision_number":6},{"id":"2281aa700ebc8889d7d86e728d2a91a7ee0c65cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 18:25:37.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/781edc772a5b42d08bb1ef449a4473d0\n\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f141059fcf71452481c91f05a0734b6b : FAILURE in 56m 46s\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/bb36e55805c14063ac609de53c2f6028 : FAILURE in 1h 26m 01s","accounts_in_message":[],"_revision_number":6},{"id":"085d070f9726a901063d07275f691c8e2eb2a084","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-22 21:50:10.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"78d236a52b19d1054f40ad8648d000680c594c1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 22:47:05.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/95d94458b5144e52aac05ea51210402a\n\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/44b322ace8db48698765e6382c00ab8e : FAILURE in 55m 22s\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/26823295eb3340709cf29bc32bd1f5e5 : SUCCESS in 33m 42s","accounts_in_message":[],"_revision_number":6},{"id":"dcf4c35de6522b96fb623f4a85b96e72321e05af","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-10-06 22:59:30.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b594db92b08fa6ddc7a74378344c9519ab311cd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 23:45:50.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d3d46f6de59410fa7763c4fde96859b\n\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9ce4143dc4614b8e9c0f236cff4a584b : FAILURE in 41m 29s\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b9bcf2ed04e546f69822e927e4d4c558 : SUCCESS in 44m 55s","accounts_in_message":[],"_revision_number":6},{"id":"37cdd31a421231be14280721f025adc917d66374","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-17 21:39:32.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ec46d755d278d6ca67e8b57e99a1662f38527c6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 22:58:54.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ccedd13249c940f5939055bf09f0a045\n\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/4521b42ee505489b827e8d8f716b0885 : FAILURE in 50m 55s\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/664e171a751b40ef966949819a6168be : SUCCESS in 1h 14m 33s","accounts_in_message":[],"_revision_number":6},{"id":"542ef4b1df1c770a981a77e19924fd8975a5f2d6","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-03-02 15:46:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"70e4ab34c3610718df1447c0a4ec1e6f1d0f6bae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 16:18:33.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/manila-tempest-plugin for 940169,6","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"e10cf0f59a26b5c788fed88664c8cf09a2cb7853","revisions":{"46fb0ae186bd6d6456f2b7d148173b9904fc30c2":{"kind":"REWORK","_number":1,"created":"2025-01-24 20:23:56.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/69/940169/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/69/940169/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/940169/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/940169/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/940169/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/940169/1"}}},"commit":{"parents":[{"commit":"a2a770ec88d41e4f322df015269402123747287b","subject":"Merge \"Refactor CIFS protocol test skipping with decorators\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a2a770ec88d41e4f322df015269402123747287b"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2025-01-24 20:21:32.000000000","tz":-300},"committer":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2025-01-24 20:21:32.000000000","tz":-300},"subject":"DNM Run share group test","message":"DNM Run share group test\n\nenables share group test, check bug fix\n\nRelated-Bug: #2019156\nChange-Id: I42ce5663a13c41c6e11e78edee4949519971ec1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/46fb0ae186bd6d6456f2b7d148173b9904fc30c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/46fb0ae186bd6d6456f2b7d148173b9904fc30c2"}]},"branch":"refs/heads/master"},"5521c92af9cedadcf79a48b6eaebf07f1a75ecaa":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-01-24 20:24:47.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/69/940169/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/69/940169/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/940169/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/940169/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/940169/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/940169/2"}}},"commit":{"parents":[{"commit":"a2a770ec88d41e4f322df015269402123747287b","subject":"Merge \"Refactor CIFS protocol test skipping with decorators\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a2a770ec88d41e4f322df015269402123747287b"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2025-01-24 20:21:32.000000000","tz":-300},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2025-01-24 20:24:47.000000000","tz":0},"subject":"DNM Run share group test","message":"DNM Run share group test\n\nenables share group test, check bug fix\n\nRelated-Bug: #2019156\nDepends-On:  Id6e7591e3417ec2df69fc1856b37215958fcb66c\nChange-Id: I42ce5663a13c41c6e11e78edee4949519971ec1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5521c92af9cedadcf79a48b6eaebf07f1a75ecaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5521c92af9cedadcf79a48b6eaebf07f1a75ecaa"}]},"branch":"refs/heads/master","description":"Edit commit message"},"58dcd46a94ed09c930252c9efe50a6ad8e90a477":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-09-19 05:24:13.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/69/940169/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/69/940169/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/940169/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/940169/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/940169/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/940169/3"}}},"commit":{"parents":[{"commit":"a2a770ec88d41e4f322df015269402123747287b","subject":"Merge \"Refactor CIFS protocol test skipping with decorators\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a2a770ec88d41e4f322df015269402123747287b"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2025-01-24 20:21:32.000000000","tz":-300},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-09-19 05:24:13.000000000","tz":0},"subject":"DNM Run share group test","message":"DNM Run share group test\n\nenables share group test, check bug fix\n\nRelated-Bug: #2019156\nDepends-On:  Id6e7591e3417ec2df69fc1856b37215958fcb66c\nChange-Id: I42ce5663a13c41c6e11e78edee4949519971ec1c\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/58dcd46a94ed09c930252c9efe50a6ad8e90a477"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/58dcd46a94ed09c930252c9efe50a6ad8e90a477"}]},"branch":"refs/heads/master","description":"Edit commit message"},"66d85fcf597da8593a03a2a52716be76d61ea517":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-09-19 05:24:16.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/69/940169/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/69/940169/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/940169/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/940169/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/940169/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/940169/4"}}},"commit":{"parents":[{"commit":"5ca4e5190a7921f019b2c9101e1d5374a9f8ac80","subject":"Merge \"Fix share server migration with replica test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5ca4e5190a7921f019b2c9101e1d5374a9f8ac80"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2025-01-24 20:21:32.000000000","tz":-300},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-09-19 05:24:16.000000000","tz":0},"subject":"DNM Run share group test","message":"DNM Run share group test\n\nenables share group test, check bug fix\n\nRelated-Bug: #2019156\nDepends-On:  Id6e7591e3417ec2df69fc1856b37215958fcb66c\nChange-Id: I42ce5663a13c41c6e11e78edee4949519971ec1c\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/66d85fcf597da8593a03a2a52716be76d61ea517"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/66d85fcf597da8593a03a2a52716be76d61ea517"}]},"branch":"refs/heads/master","description":"Rebase"},"273a99ce28fe9e0cabffbe9cc0c28eb6c89b354a":{"kind":"REWORK","_number":5,"created":"2025-09-19 05:25:49.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/69/940169/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/69/940169/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/940169/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/940169/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/940169/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/940169/5"}}},"commit":{"parents":[{"commit":"5ca4e5190a7921f019b2c9101e1d5374a9f8ac80","subject":"Merge \"Fix share server migration with replica test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5ca4e5190a7921f019b2c9101e1d5374a9f8ac80"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2025-01-24 20:21:32.000000000","tz":-300},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-09-19 05:25:49.000000000","tz":0},"subject":"DNM Run share group test","message":"DNM Run share group test\n\nenables share group test, check bug fix\n\nRelated-Bug: #2019156\nDepends-On:  Id6e7591e3417ec2df69fc1856b37215958fcb66c\nChange-Id: I42ce5663a13c41c6e11e78edee4949519971ec1c\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/273a99ce28fe9e0cabffbe9cc0c28eb6c89b354a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/273a99ce28fe9e0cabffbe9cc0c28eb6c89b354a"}]},"branch":"refs/heads/master"},"e10cf0f59a26b5c788fed88664c8cf09a2cb7853":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-09-19 16:54:14.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/69/940169/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/69/940169/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/940169/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/940169/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/940169/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/940169/6"}}},"commit":{"parents":[{"commit":"5ca4e5190a7921f019b2c9101e1d5374a9f8ac80","subject":"Merge \"Fix share server migration with replica test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5ca4e5190a7921f019b2c9101e1d5374a9f8ac80"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2025-01-24 20:21:32.000000000","tz":-300},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-09-19 16:54:14.000000000","tz":0},"subject":"DNM Run share group test","message":"DNM Run share group test\n\nenables share group test, check bug fix\n\nRelated-Bug: #2019156\nDepends-On: Id6e7591e3417ec2df69fc1856b37215958fcb66c\nChange-Id: I42ce5663a13c41c6e11e78edee4949519971ec1c\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e10cf0f59a26b5c788fed88664c8cf09a2cb7853"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e10cf0f59a26b5c788fed88664c8cf09a2cb7853"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
