)]}'
{"id":"openstack%2Fmanila-tempest-plugin~941553","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ied7d75fc3e26c1963744df6e9e5af6ae0e605158","project":"openstack/manila-tempest-plugin","branch":"master","topic":"az","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2025-02-20 13:16:31.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"32594":{"account":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"last_update":"2025-02-25 02:19:44.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-02-25 00:55:13.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2025-02-24 11:39:20.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2025-02-25 02:19:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ied7d75fc3e26c1963744df6e9e5af6ae0e605158","subject":"Ensure Manila shares are created on the same backend host","status":"MERGED","created":"2025-02-13 13:01:35.000000000","updated":"2025-02-25 02:19:44.000000000","submitted":"2025-02-25 02:19:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"941553-az","meta_rev_id":"ec67be43c42512e10abab1cba28cfafb16697298","_number":941553,"virtual_id_number":941553,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-25 02:19:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2025-02-20 13:16:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2025-02-25 00:55:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2025-02-24 11:39:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2025-02-25 00:55:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-13 13:56:48.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"state":"REVIEWER"},{"updated":"2025-02-13 13:56:48.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2025-02-13 13:56:48.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2025-02-13 13:56:48.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2025-02-13 16:11:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5c14078f08e8da24a58ee4069a6424b6add00edf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-02-13 13:01:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4be9aed95f75017962b47c6b469c3d7508077de","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-02-13 14:13:58.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e555597e9e575aaf700584e524c1d5456e43a5b","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-02-13 14:27:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d28b7dd30f25fbb50cdb38afbcb48392def794d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 16:11:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c5f4689346b45c99eb0ffb96aadabc3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b7825cccc844ff9a0fbccc087455b95 : SUCCESS in 2m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1028675f61144dce96516991df275844 : SUCCESS in 57m 57s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/c4b6633aa1cc475eb8da011f7330f7bc : SUCCESS in 1h 02m 45s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/21dc585b3e404310bb6a045c6c6f48a8 : TIMED_OUT in 3h 03m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/858d722ef83542ea9b2dcceef92ab4ac : FAILURE in 1h 34m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/70ba353cf4d0414bb40465b30d04ed4f : FAILURE in 1h 45m 12s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/87a86edb44fd4f23949bf60b2ea02393 : FAILURE in 2h 40m 26s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8dbcaa39853e40088f717d23a4f0eb71 : SUCCESS in 1h 11m 30s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2b8a5c7d14e0409690e03302bab80806 : SUCCESS in 35m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/250f664969f4478f9e09e82d77d2aff4 : SUCCESS in 24m 52s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/759dfa3e1f794c0db993d1baacfb147b : SUCCESS in 1h 19m 11s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/862d662bdd714b72a9aa6c85e862ad4f : SUCCESS in 1h 13m 39s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/03a49cb280374afdb8c00477de618b2f : SUCCESS in 1h 21m 50s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f05b0724388845918be590453aba08fa : SUCCESS in 27m 26s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9e152a3187f3409186d41786837caf5a : FAILURE in 49m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"105ff8b14860b641a90de336e3e408a9e52141a2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-02-13 20:24:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5de3f1f1f3dc0625fe17c31d8eba0868e8a08da7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-02-13 20:26:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8bea8048c38aa7959b6606be3ebd200425101d08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-02-16 08:10:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"8e89bcc9fd499d0ad3f72d24dc66696f47c962b6","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-02-16 08:18:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ea02239ee33d3413e1bfb211e5ceb612cc6a21fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 11:14:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b3a65da450c473aa7d372e39713b94a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79164b1eee78491380dfb63e6d902599 : SUCCESS in 2m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2a45a8cd5960490b8db67ff5fedf576f : SUCCESS in 1h 01m 24s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/fd6565e5225d423c84be677e66ce1ea0 : SUCCESS in 59m 01s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/58d830e445ec49d1b40daa8b71224512 : TIMED_OUT in 3h 03m 04s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f1a7941409a5410a9786b05997feeb90 : FAILURE in 35m 43s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8b238bd5aafc414e851139f3a52beb51 : SUCCESS in 56m 58s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/d78eb427802d4db8a1cb274fe24f7411 : FAILURE in 52m 40s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b0a61ca11a5544c59dbdbf51c713725d : SUCCESS in 1h 00m 16s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/39f32fe471254f6fa0e21bbe5eb395f2 : SUCCESS in 24m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/98ee1bef1b87495dbef761fb212e5d3d : SUCCESS in 29m 54s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/5c742c43ca57412a9b00b52d57bd881f : SUCCESS in 50m 39s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/c26a7f418c3547cc9e493951190db5bf : SUCCESS in 1h 02m 19s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/164d175794dd42be89e22cb4b469256e : SUCCESS in 1h 12m 48s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/2796cfcc654243ad80d6f73c4be489df : SUCCESS in 21m 51s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7160a263db5e42a69fc48862cdca28bb : FAILURE in 39m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"13878c20f980a729cd3e4ceff3bdb31ed98e6392","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-02-20 13:16:31.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e41ad75d12ce3f12b33c1e5ebbcf98f54b1f650c","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2025-02-24 11:39:20.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"83a218c3390ec1ebfff17eac12a207d9d3a752b7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-02-25 00:55:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8802973d44e196baf24c49e9a4cc3a325cbec555","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 00:56:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b78db1e532830f70902f11ebc8a54d732b8e120c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 02:19:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b4f689ebfac47c78b6f3e0e5e274d09\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbd3d57dc1fe4377a541e0ad88be0346 : SUCCESS in 2m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9abea2f38eaf402c85cc38466e8d858a : SUCCESS in 1h 22m 29s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/9f78ad1711bb4e6281e171002a3f1935 : SUCCESS in 47m 06s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d54d70afa42c416ab92d002383014d40 : SUCCESS in 36m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9cc8917a96e24acb98b4bf294314f82e : SUCCESS in 40m 54s","accounts_in_message":[],"_revision_number":2},{"id":"ec67be43c42512e10abab1cba28cfafb16697298","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 02:19:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"58903bdd47ad33cc82fd29bd1066f4c4f1245143","revisions":{"6b39ec6e6a4481837f6568912b64eeb7ff5cc5d7":{"kind":"REWORK","_number":1,"created":"2025-02-13 13:01:35.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/53/941553/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/53/941553/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/941553/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/941553/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/941553/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/941553/1"}}},"commit":{"parents":[{"commit":"b616cb811c2e9ff2054de1cd760869502a1a69e6","subject":"skip tests for stable branches which don\u0027t support API 2.88","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b616cb811c2e9ff2054de1cd760869502a1a69e6"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2025-02-13 12:53:58.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2025-02-13 13:01:25.000000000","tz":120},"subject":"Ensure Manila shares are created in the same availability Zone","message":"Ensure Manila shares are created in the same availability Zone\n\nThis update ensures that Manila shares are consistently created\nwithin the same AZ, addressing issues where mismatched AZs lead\nto test failures.\n\nChange-Id: Ied7d75fc3e26c1963744df6e9e5af6ae0e605158\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6b39ec6e6a4481837f6568912b64eeb7ff5cc5d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6b39ec6e6a4481837f6568912b64eeb7ff5cc5d7"}]},"branch":"refs/heads/master"},"58903bdd47ad33cc82fd29bd1066f4c4f1245143":{"kind":"REWORK","_number":2,"created":"2025-02-16 08:10:36.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/53/941553/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/53/941553/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/941553/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/941553/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/941553/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/53/941553/2"}}},"commit":{"parents":[{"commit":"b616cb811c2e9ff2054de1cd760869502a1a69e6","subject":"skip tests for stable branches which don\u0027t support API 2.88","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b616cb811c2e9ff2054de1cd760869502a1a69e6"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2025-02-13 12:53:58.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2025-02-16 07:59:15.000000000","tz":120},"subject":"Ensure Manila shares are created on the same backend host","message":"Ensure Manila shares are created on the same backend host\n\nThis update ensures that Manila shares are consistently created\non the same backend host. This change addresses cases where CephX\nuser IDs are reused within a Ceph cluster across different OpenStack\nprojects, ensuring that access conflicts are accurately tested.\n\nChange-Id: Ied7d75fc3e26c1963744df6e9e5af6ae0e605158\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/58903bdd47ad33cc82fd29bd1066f4c4f1245143"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/58903bdd47ad33cc82fd29bd1066f4c4f1245143"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
