)]}'
{"id":"openstack%2Fmanila-tempest-plugin~942226","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I058475898d716ae41e576bd928d74f9720d55027","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"32594":{"account":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"last_update":"2025-08-06 19:53:17.000000000","reason":"Change was marked work in progress"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-08-06 19:53:17.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I058475898d716ae41e576bd928d74f9720d55027","subject":"[DNM] run test_share_extend on native","status":"NEW","created":"2025-02-19 17:11:20.000000000","updated":"2025-08-07 18:40:00.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"ad0e3d40dfbe0797706f3ec4b355a09ac3958c6d","_number":942226,"virtual_id_number":942226,"owner":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-08-07 18:40:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-02-19 18:17:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-01 17:41:38.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"9b70ecea118bd6ee4c1bbfbf9137be6f0e3d8595","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2025-02-19 17:11:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8725af369094d5af70f3cbf6ef32f9ab0dca2336","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 18:17:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d2e8b6ae83e14406bad804435b33f877\n\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/989eabf2de1549fc929829286abea57b : FAILURE in 44m 37s","accounts_in_message":[],"_revision_number":1},{"id":"88836c62698a7f92b7087c43ec46338bec29fe8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2025-02-19 20:18:50.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"36fb9a9e6dad150bb864be4d019a394a2ed16f56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 20:45:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ccd7957136cc468ba15397fd72aab2cb\n\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c6922da4164f4e1fa6f4bcf63088620b : FAILURE in 15m 26s","accounts_in_message":[],"_revision_number":2},{"id":"1d45391459c87d711135ba232079bb4bdb27aedc","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2025-02-20 14:08:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4e292a524c6fb552a13ac2087286bb5c9d420b2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 14:34:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9342b90e3e8d4a0891958f2c8e866ced\n\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/4e4926da64434301bd04a975d7d98c1e : FAILURE in 24m 33s","accounts_in_message":[],"_revision_number":2},{"id":"84788b8233cb0ee286b4b67155c5acf4432cbe80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-01 17:41:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"8d3831939f9afb40dcd8e5b66056aa4681906a14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 18:27:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2da0939eaa4846cf985c45cadf053389\n\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/0ad51b778a144c5ba01a6d4756bca28a : FAILURE in 40m 28s","accounts_in_message":[],"_revision_number":3},{"id":"9d28d598134024fc34ac37e5d6675f1fe03ada31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-06 18:21:31.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"cc8ddbbbab8ca4eeb7db1dc9b6679d1a17a83e05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 18:50:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1643c2dcb76c4735ab812adb44a60859\n\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ced079812a664bba9e9ac01e0d417f5e : FAILURE in 11m 25s","accounts_in_message":[],"_revision_number":4},{"id":"477f8c97a09963f333d41c5c0ce0163e0cae7199","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-06 19:53:17.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nThe following files contain Git conflicts:\n* zuul.d/manila-tempest-jobs.yaml\n* zuul.d/project.yaml\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a189d51b6370dbb47a2eb0d348d8cb05e6afb495","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 19:53:46.000000000","message":"Patch Set 5: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/manila-tempest-plugin on branch master.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 392, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 393, column 9","accounts_in_message":[],"_revision_number":5},{"id":"e89cab6ec42668fe425fa42dd57bd4abb2426c8d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-06 19:54:36.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"44f15d48189ddd14975f88617a3ddf73761b0547","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 19:55:02.000000000","message":"Patch Set 6: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/manila-tempest-plugin for 942226,6","accounts_in_message":[],"_revision_number":6},{"id":"4e72952157e0920c41506611e16cc726990adfcf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-07 18:21:41.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"ad0e3d40dfbe0797706f3ec4b355a09ac3958c6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 18:40:00.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/954eb2d62db044d4a9bbd8c4cac61c26\n\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/94701bb368fc41b38b33cd868bc2392b : FAILURE in 16m 33s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"e6ad550224deb9ab70714e2724d21ac6bba0c87c","revisions":{"c0be77b013fb9b44ca4c11a4737686e88671046e":{"kind":"REWORK","_number":1,"created":"2025-02-19 17:11:20.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/26/942226/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/26/942226/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/1"}}},"commit":{"parents":[{"commit":"b616cb811c2e9ff2054de1cd760869502a1a69e6","subject":"skip tests for stable branches which don\u0027t support API 2.88","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b616cb811c2e9ff2054de1cd760869502a1a69e6"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2025-02-19 17:06:49.000000000","tz":-300},"committer":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2025-02-19 17:09:42.000000000","tz":-300},"subject":"[DNM] run test_share_extend on native","message":"[DNM] run test_share_extend on native\n\nGet mds logs for failing test\n\nChange-Id: I058475898d716ae41e576bd928d74f9720d55027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c0be77b013fb9b44ca4c11a4737686e88671046e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c0be77b013fb9b44ca4c11a4737686e88671046e"}]},"branch":"refs/heads/master"},"b603cf49af9846557db7f77e95059d1ecbcb1d49":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-02-19 20:18:50.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/26/942226/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/26/942226/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/2"}}},"commit":{"parents":[{"commit":"b616cb811c2e9ff2054de1cd760869502a1a69e6","subject":"skip tests for stable branches which don\u0027t support API 2.88","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b616cb811c2e9ff2054de1cd760869502a1a69e6"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2025-02-19 17:06:49.000000000","tz":-300},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2025-02-19 20:18:50.000000000","tz":0},"subject":"[DNM] run test_share_extend on native","message":"[DNM] run test_share_extend on native\n\nGet mds logs for failing test\n\nDepends-On: Ibef35f88e81ebf47b772b19e9ed9f9655e0ecbe2\nChange-Id: I058475898d716ae41e576bd928d74f9720d55027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b603cf49af9846557db7f77e95059d1ecbcb1d49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b603cf49af9846557db7f77e95059d1ecbcb1d49"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7ee77d5b4443f0bbd762d26f573f0ea45f553f38":{"kind":"REWORK","_number":3,"created":"2025-08-01 17:41:38.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/26/942226/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/26/942226/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/3"}}},"commit":{"parents":[{"commit":"be1bb4bf6fc4092b18a23ba4b86d327e43f4c527","subject":"Fix share type used in negative quota test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/be1bb4bf6fc4092b18a23ba4b86d327e43f4c527"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2025-02-19 17:06:49.000000000","tz":-300},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-08-01 17:41:30.000000000","tz":-420},"subject":"[DNM] run test_share_extend on native","message":"[DNM] run test_share_extend on native\n\nGet mds logs for failing test\n\nDepends-On: Ibef35f88e81ebf47b772b19e9ed9f9655e0ecbe2\nChange-Id: I058475898d716ae41e576bd928d74f9720d55027\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7ee77d5b4443f0bbd762d26f573f0ea45f553f38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7ee77d5b4443f0bbd762d26f573f0ea45f553f38"}]},"branch":"refs/heads/master"},"54773077317a317bb09cff7c1b24d0cbc39827bd":{"kind":"REWORK","_number":4,"created":"2025-08-06 18:21:31.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/26/942226/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/26/942226/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/4"}}},"commit":{"parents":[{"commit":"be1bb4bf6fc4092b18a23ba4b86d327e43f4c527","subject":"Fix share type used in negative quota test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/be1bb4bf6fc4092b18a23ba4b86d327e43f4c527"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2025-02-19 17:06:49.000000000","tz":-300},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-08-06 18:21:31.000000000","tz":0},"subject":"[DNM] run test_share_extend on native","message":"[DNM] run test_share_extend on native\n\nGet mds logs for failing test\n\nDepends-On: Ibef35f88e81ebf47b772b19e9ed9f9655e0ecbe2\nChange-Id: I058475898d716ae41e576bd928d74f9720d55027\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/54773077317a317bb09cff7c1b24d0cbc39827bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/54773077317a317bb09cff7c1b24d0cbc39827bd"}]},"branch":"refs/heads/master"},"dc209214834f627d916b8008543afde0e83a30ab":{"kind":"REWORK","_number":5,"created":"2025-08-06 19:53:17.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/26/942226/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/26/942226/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/5"}}},"commit":{"parents":[{"commit":"91d2234aa465af4857c236e7b673368ba1293e43","subject":"[CI] Accept Ceph release defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/91d2234aa465af4857c236e7b673368ba1293e43"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2025-02-19 17:06:49.000000000","tz":-300},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-08-06 19:53:17.000000000","tz":0},"subject":"[DNM] run test_share_extend on native","message":"[DNM] run test_share_extend on native\n\nGet mds logs for failing test\n\nDepends-On: Ibef35f88e81ebf47b772b19e9ed9f9655e0ecbe2\nChange-Id: I058475898d716ae41e576bd928d74f9720d55027\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dc209214834f627d916b8008543afde0e83a30ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dc209214834f627d916b8008543afde0e83a30ab"}]},"branch":"refs/heads/master","description":"Rebase"},"88d9f5e8379282085207df464f2623062c0f7d9a":{"kind":"REWORK","_number":6,"created":"2025-08-06 19:54:36.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/26/942226/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/26/942226/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/6"}}},"commit":{"parents":[{"commit":"91d2234aa465af4857c236e7b673368ba1293e43","subject":"[CI] Accept Ceph release defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/91d2234aa465af4857c236e7b673368ba1293e43"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2025-02-19 17:06:49.000000000","tz":-300},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-08-06 19:54:28.000000000","tz":0},"subject":"[DNM] run test_share_extend on native","message":"[DNM] run test_share_extend on native\n\nGet mds logs for failing test\n\nDepends-On: Ibef35f88e81ebf47b772b19e9ed9f9655e0ecbe2\nChange-Id: I058475898d716ae41e576bd928d74f9720d55027\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/88d9f5e8379282085207df464f2623062c0f7d9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/88d9f5e8379282085207df464f2623062c0f7d9a"}]},"branch":"refs/heads/master"},"e6ad550224deb9ab70714e2724d21ac6bba0c87c":{"kind":"REWORK","_number":7,"created":"2025-08-07 18:21:41.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/26/942226/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/26/942226/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/942226/7"}}},"commit":{"parents":[{"commit":"b643b1237767132be466737e7c849a72b969d5e8","subject":"[CI] Accept Ceph release defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b643b1237767132be466737e7c849a72b969d5e8"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2025-02-19 17:06:49.000000000","tz":-300},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-08-07 18:21:28.000000000","tz":-420},"subject":"[DNM] run test_share_extend on native","message":"[DNM] run test_share_extend on native\n\nGet mds logs for failing test\n\nDepends-On: Ibef35f88e81ebf47b772b19e9ed9f9655e0ecbe2\nChange-Id: I058475898d716ae41e576bd928d74f9720d55027\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e6ad550224deb9ab70714e2724d21ac6bba0c87c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e6ad550224deb9ab70714e2724d21ac6bba0c87c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
