)]}'
{"id":"openstack%2Fmanila-tempest-plugin~942236","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I863e26b7255971eeaaa43ddfda13cbec1b4e4119","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2025-08-07 19:04:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I863e26b7255971eeaaa43ddfda13cbec1b4e4119","subject":"[DNM] Check share network metadata update change","status":"ABANDONED","created":"2025-02-19 19:38:10.000000000","updated":"2025-08-07 19:04:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2bedf4402756c754d0a1b87b90074aaa400b9924","_number":942236,"virtual_id_number":942236,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-20 18:38:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-19 22:42:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-21 14:56:19.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"daf75c3cdcae00465b4fbab7e5bb74bf8b2766fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-02-19 19:38:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"970c675a8838a55a607f00829bc75362b1a3a9c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 22:42:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4463f055aa144cd9553ace12a4abbb4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5afbc42ca09406fb8dce29c4694b9a1 : SUCCESS in 2m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/105febc25b98444aa7df616a68572273 : SUCCESS in 1h 19m 11s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/54f686a2ff3b47c9b5c85eb57d644b12 : SUCCESS in 1h 14m 57s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/789c3db313d047d18f4237e01cb25def : TIMED_OUT in 3h 03m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8dad507f62e5455cbe6f9db71ac3a14d : FAILURE in 1h 17m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/abdadd07e04a43adae7b5822c8a9fda1 : SUCCESS in 1h 12m 21s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/90e9c2da86f84aa49876b9be06998760 : FAILURE in 1h 10m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2d5f6352d42a4bf48d909aa09aa5bf7f : SUCCESS in 1h 23m 24s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/39add270664a4cfdb795ec0906680189 : SUCCESS in 40m 01s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f47e65faab5a4ceeadcf9d7fd25338e2 : SUCCESS in 32m 48s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/121ed95b07d44288a96e5f5591d2f660 : SUCCESS in 1h 31m 11s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/03f86095e01d4962915e1f8c9b330e91 : SUCCESS in 36m 53s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/bec5ef5c466349709494ad9e490af28f : SUCCESS in 1h 19m 51s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/b12af6d3af414ce29b3881c9c69404e3 : SUCCESS in 22m 36s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d39c9b88067e443aba95ef28629121bc : FAILURE in 55m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a5365f19d650d545cfd8f19b6e566401b277341","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-02-20 15:29:25.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"979c741b2ac13738c160d199f06501e52c0e71c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-02-20 15:29:58.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"59eddd2619e536987ed28e803f2272ea0068a8f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 18:38:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c19d6606adfc4b2ba5795f64fa5037bf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/913c9450ea254f28a235ec7567631049 : SUCCESS in 6m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/93405f7329da4fb4842bfc7be150fcab : SUCCESS in 1h 38m 55s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/bc01a260aba84cd1b2913cfb53b03215 : SUCCESS in 1h 40m 35s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8d8af4ce3ec0473ea7374efee3fcb27d : TIMED_OUT in 3h 02m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/83c39f7c2d184d4fb1c250e1d82d4b4b : FAILURE in 1h 53m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/7e557b79b70d43f0b2df9db7b243b09f : SUCCESS in 1h 26m 00s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/40a434ff2cc84be9bda6871141a565a3 : FAILURE in 1h 52m 51s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/06e59d0ec1664b07bcbedf6f5cc80fb9 : SUCCESS in 1h 14m 27s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/28f64771c3e4439caf4081e59837b58b : SUCCESS in 31m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/00e48e48bac2417784f6e1af48e2d594 : SUCCESS in 38m 50s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/40f7923aeea7498f864089635c641450 : SUCCESS in 1h 08m 34s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/fc65b38340af49e5b46399d5464b00b9 : SUCCESS in 1h 29m 38s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/eae2bfdbd4fa4de287aa675fdc4042bf : SUCCESS in 1h 07m 15s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/fe20dd853416487ba3a2da542dae4e18 : SUCCESS in 29m 45s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/99e67c45671b40f199d915d691b02190 : FAILURE in 53m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5290b73527d6edee82a70c8765c99eeb20425453","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-02-21 14:56:19.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"2bedf4402756c754d0a1b87b90074aaa400b9924","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-08-07 19:04:01.000000000","message":"Abandoned\n\nChange was submitted to test another change. We can drop this now.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6ba068cc759a3f16b5bd745eb078c676c8def866","revisions":{"04165a8f488d8f716120b33d784975c01c4a87ad":{"kind":"REWORK","_number":1,"created":"2025-02-19 19:38:10.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/36/942236/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/36/942236/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/36/942236/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/36/942236/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/36/942236/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/36/942236/1"}}},"commit":{"parents":[{"commit":"b616cb811c2e9ff2054de1cd760869502a1a69e6","subject":"skip tests for stable branches which don\u0027t support API 2.88","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b616cb811c2e9ff2054de1cd760869502a1a69e6"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-02-19 19:37:40.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-02-19 19:37:40.000000000","tz":-180},"subject":"[DNM] Check share network metadata update change","message":"[DNM] Check share network metadata update change\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/929091\nChange-Id: I863e26b7255971eeaaa43ddfda13cbec1b4e4119\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/04165a8f488d8f716120b33d784975c01c4a87ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/04165a8f488d8f716120b33d784975c01c4a87ad"}]},"branch":"refs/heads/master"},"5b950793b8f14b510231ebf6827ef9cf3e1a14f0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-02-20 15:29:25.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/36/942236/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/36/942236/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/36/942236/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/36/942236/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/36/942236/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/36/942236/2"}}},"commit":{"parents":[{"commit":"b616cb811c2e9ff2054de1cd760869502a1a69e6","subject":"skip tests for stable branches which don\u0027t support API 2.88","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b616cb811c2e9ff2054de1cd760869502a1a69e6"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-02-19 19:37:40.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2025-02-20 15:29:25.000000000","tz":0},"subject":"[DNM] Check share network metadata update change","message":"[DNM] Check share network metadata update change\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/929091\nDepends-On: https://review.opendev.org/c/openstack/manila/+/932149\nChange-Id: I863e26b7255971eeaaa43ddfda13cbec1b4e4119\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5b950793b8f14b510231ebf6827ef9cf3e1a14f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5b950793b8f14b510231ebf6827ef9cf3e1a14f0"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6ba068cc759a3f16b5bd745eb078c676c8def866":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-02-20 15:29:58.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/36/942236/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/36/942236/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/36/942236/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/36/942236/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/36/942236/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/36/942236/3"}}},"commit":{"parents":[{"commit":"b616cb811c2e9ff2054de1cd760869502a1a69e6","subject":"skip tests for stable branches which don\u0027t support API 2.88","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b616cb811c2e9ff2054de1cd760869502a1a69e6"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-02-19 19:37:40.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2025-02-20 15:29:58.000000000","tz":0},"subject":"[DNM] Check share network metadata update change","message":"[DNM] Check share network metadata update change\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/929091\nChange-Id: I863e26b7255971eeaaa43ddfda13cbec1b4e4119\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6ba068cc759a3f16b5bd745eb078c676c8def866"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6ba068cc759a3f16b5bd745eb078c676c8def866"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
