)]}'
{"id":"openstack%2Fmanila-tempest-plugin~942862","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I3dbab947bcf5a1df39ee16e308096b3ef75b6d51","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2025-03-26 20:24:29.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"32594":{"account":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"last_update":"2025-03-27 19:52:23.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-03-27 16:36:04.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2025-03-26 21:21:07.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}},"hashtags":[],"change_id":"I3dbab947bcf5a1df39ee16e308096b3ef75b6d51","subject":"Disable Ingress","status":"MERGED","created":"2025-02-26 22:11:35.000000000","updated":"2025-03-27 22:11:41.000000000","submitted":"2025-03-27 19:52:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942862","meta_rev_id":"68a14ac9b24d2e340c2cc9f9048e2a69f0ef63e6","_number":942862,"virtual_id_number":942862,"owner":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-27 19:52:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2025-03-26 20:24:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2025-03-26 21:21:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2025-03-26 21:21:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-26 23:44:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-03 16:59:58.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-03-21 14:23:52.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2025-03-21 18:44:50.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2025-03-21 18:44:50.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2025-03-21 18:44:50.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"9016100a9831240bf65a6702ad97671ac45e60a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2025-02-26 22:11:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38c75dc5e23f661568c771228e631c0b992711a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 23:44:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8014acc1201d45deb3e4301bb3c648db\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc4e903e2fb541dfa911a0b0d7f45a14 : SUCCESS in 3m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/de153404cda04bd793ca8ab923568662 : SUCCESS in 41m 48s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/e7247f6893b14900bdce739505d14306 : SUCCESS in 1h 18m 05s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/03c05f1f84c54d07ba77afaca4aae959 : SUCCESS in 1h 27m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2e67a2e12fd742d88ca17157296f9bef : FAILURE in 1h 13m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ccc6997d40544d59ad7f19e8dfcaa172 : SUCCESS in 1h 08m 25s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/5dc5bd0cc62841b4b83d1b5a9a66cc4e : FAILURE in 1h 23m 01s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/96dd7120d6bf409da82e5e060d7ae5f9 : SUCCESS in 1h 15m 32s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/db2dc757844b4b748e660e02ff017fe7 : SUCCESS in 27m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6c51244c639e40219df3632174b64a55 : SUCCESS in 29m 13s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/147349ee63f54352900298f568fc8338 : FAILURE in 1h 27m 23s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/9dc47d232c304783befc6d824ad9b5b0 : SUCCESS in 1h 03m 45s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/93524e9ca22b4917bd5d602eeae36cc6 : SUCCESS in 1h 16m 58s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/9d08237693e648769bb7ee87c8d73629 : SUCCESS in 22m 23s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a3a4837dbf174bef994b77cd0861c91b : FAILURE in 1h 02m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c6d7799f718f20afd87a33d0dcbcfd262f07afe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2025-02-28 17:42:25.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f093d13b25db4316846344e80b47c0c64acae217","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 17:44:27.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/942861 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"947b81e8ec1186ac47f77482c242ccb78db4fd69","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2025-02-28 19:05:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e4519cb890ddd90cdb605e0391a6a7358169f1fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 19:08:07.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/942861 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"ba72f461f81241104cc646756204e838a60d96bd","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2025-03-03 14:38:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"330ccc19304798ecdb4a2e23b720bd889ab7ffe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 16:52:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ab532d9adb642109be4dcf849102ce9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aaa4b9836f84393ae22beab28e28d6e : SUCCESS in 2m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7604bf32334445a49f6b98131f673fe0 : SUCCESS in 1h 13m 41s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/66caf953b3de48569957eff551ddf25d : SUCCESS in 1h 02m 06s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e9a76c69913d4d258e947d2032a26325 : SUCCESS in 2h 04m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/eed0603740f743c9912d7351f64e51a6 : FAILURE in 1h 34m 53s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/94bcddad139c49548b2d320e6bfa7ed7 : SUCCESS in 1h 01m 21s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/6b1410a292cc43b89af134ad7a1d8d98 : SUCCESS in 1h 20m 47s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/41f0b5a47f2f44e8840409e09f86955b : SUCCESS in 1h 13m 15s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/969b9e26437c415f97625026bcbb92e8 : SUCCESS in 35m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/88493359e2c543388d654039c1d6e031 : SUCCESS in 37m 33s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/4a055b5963114a63b699306c4c5ad79e : SUCCESS in 1h 03m 08s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/23913965a9344831af7341831315c417 : SUCCESS in 1h 09m 25s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/78ae912265044fd3b5e2820358fd268f : SUCCESS in 1h 08m 43s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f82a03ca6d684f01a828e914c5629e62 : SUCCESS in 36m 53s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d919eda152f84ac1994ea54e894c119b : FAILURE in 40m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"28b7e409d2d618813e1c8cd2414d2d1cb941aa3b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-03 16:59:58.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss finger://managesf.services.openstack.netapp.com:7979/d2b21e672eb64d148881695610d851b1 : RETRY_LIMIT in 6m 06s\n- manila-tempest-plugin-ontap-no-dhss finger://managesf.services.openstack.netapp.com:7979/89785398fe084c89ac33a4b3a6933ae7 : RETRY_LIMIT in 6m 05s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"9c450bc1b602d708a1e76616f8bdf3e9dfc32d83","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-03-21 14:23:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"531fe8056764e2ae66c2675f66e78c74ca120f97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 16:45:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/77e6337f47be493984cfadc498e0e386\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2802f21be784005a522305f569de8d2 : SUCCESS in 2m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a55b12d00a8d4606b434d648bc3442e9 : SUCCESS in 31m 56s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/cf7af230f0004eba9cf8cbfe0b6cd6df : SUCCESS in 31m 05s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/34b54ed6b7f84224b5ef50bc8632683a : FAILURE in 2h 19m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bbcdcb1875b9474db8f4752515605452 : SUCCESS in 47m 17s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/59ae4d199d244eddb68dfd3f69febc99 : SUCCESS in 28m 53s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/adeca89d2b424f43b478951262a6952b : SUCCESS in 47m 13s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4e16224abb8f45f08667001219b30beb : SUCCESS in 1h 07m 38s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7f81042acf904ee5aa876a2b3ea4d73c : SUCCESS in 39m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a1aa0a1c55354401b2524f13ee14bd81 : SUCCESS in 34m 00s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/0845b31b93cd47eebbd2470ab95dac0c : SUCCESS in 1h 11m 08s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/3a2b14688b464d09bf4bb8fbbf0b1ac7 : SUCCESS in 50m 36s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/4c0c3f0aacc54aa4acda9161fe99cdb9 : SUCCESS in 1h 10m 51s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/fb55313750174fd59a626ad3d3f7a233 : FAILURE in 5m 48s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2a1553b160dd49acb47d67cf46623313 : FAILURE in 47m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"72bad5108038dcb0f956e78764c548415b8d1532","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-03-21 18:44:50.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dd9dd523f7a48ac823d6664a717499542c8d65af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 20:53:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/96eb881085ed4c05a15d64dd0960d6b1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/488b6a8b12244c40978c919fb991c413 : SUCCESS in 2m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4186471ceb18441993cb92d16a505e1d : SUCCESS in 1h 16m 23s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/9d1f425b22ba4a59a1a2ce766e407d75 : SUCCESS in 52m 51s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/13f8b1760ff14301ab7a91f97e84239a : SUCCESS in 1h 53m 30s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4e9b4a20ac6e4210b2835f10d97173d8 : FAILURE in 1h 06m 57s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/30ffa0e8d4a74c93aa5f3e6b29835f37 : SUCCESS in 1h 07m 36s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/9b4e1134588b477cb605d9eebd454bd7 : SUCCESS in 1h 11m 06s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/608b2041f84544688a36e608461cb5fd : SUCCESS in 38m 06s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fb3c0e51d04446d69658cec989eb3cd3 : FAILURE in 5m 39s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a3fb133637dc4479bb61132b5b65061d : SUCCESS in 37m 08s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/4de4af870ff34ab689707f5521509948 : FAILURE in 1h 03m 34s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/bbd6858ecaa4467aad444f8251670973 : SUCCESS in 1h 03m 11s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/e30d92acd1b4424785bcaeebaf77d843 : SUCCESS in 55m 08s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/16abbaa683c84a889a8c04e61c10b5c5 : SUCCESS in 27m 42s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/680284d537b0453cbaf9d4ffcb17ae5a : FAILURE in 36m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"60efe4509548a8c56c288108767d7bfbdc584746","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2025-03-24 14:35:40.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"36cc35252b75bf08c5f9d3b1199a1f6f11ab94a2","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-03-24 17:05:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"04f851d280e977e51d0a5ff43350ebbd6f57e2e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 18:57:07.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e23cb802a6a54c2e9e89cabcc295aec4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecbdd8e9f4564193abaa66d8f7dfdd72 : SUCCESS in 3m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/be5ef4d50ab34b20bb27422dbaeac868 : SUCCESS in 54m 12s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/67554c47260b4584b0ec8bd7706d8ce8 : SUCCESS in 52m 54s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8791a54ec0e845949190154d02b1e384 : SUCCESS in 1h 49m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/891e94dee8ca4a52a07089ad9b8f9809 : FAILURE in 59m 51s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/87644fa80edc4662967b02cf948c801a : SUCCESS in 47m 05s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/19ad292040a542b3852deb00e6a0c123 : SUCCESS in 1h 20m 31s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/24c57c2788e04bd890e1f8b3b96c661b : SUCCESS in 55m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1d14234677a541198195ef3f0ced9405 : SUCCESS in 26m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e2fbf3148ba449c6ae5e0ec3f45c3318 : FAILURE in 5m 58s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/3682364125e6475aacf6c8fe524ba316 : SUCCESS in 54m 47s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/66302298cbb04574b23a037c5e1ef383 : SUCCESS in 38m 53s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/374bddc0573c4ee5ab40194f6486b96c : SUCCESS in 56m 20s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f0853d95e2a94dfe9593ae6dcb9f76d2 : SUCCESS in 23m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/81e5b3da964f4663a450bcf562ec0e26 : FAILURE in 35m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fc512eafcf6a898c314891fdc28aa400ef495d9b","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-03-25 17:37:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4f2eedfee794860a2b124da3dd534b0bd6c357a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 20:41:06.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/edb47f46ac6c419a87e78f85195628b1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e36f6f094aa4dc2b342e2b248184ed9 : SUCCESS in 2m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e4d8731b69e8465b9aa5646262db6b14 : SUCCESS in 1h 21m 12s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/0aa0b58ffcfc466997b8347e7500a16c : SUCCESS in 1h 14m 47s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2ad603f5b2134439af39531bbb020404 : FAILURE in 2h 56m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6e2afa5019764e4fa0ee72445d186e7e : FAILURE in 1h 11m 45s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/96e7a765e3fb47a19c363c63fdc1a35d : SUCCESS in 1h 58m 51s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a94f93c844dd4f5498085621db22356e : SUCCESS in 43m 20s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fbfeb97acd664ef382c17fdb7480b647 : FAILURE in 1h 09m 32s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8d657c9bd42e403198cfbf783703297a : FAILURE in 6m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/bd89eb55ef7c4d12a0542ed29e902fb8 : SUCCESS in 29m 52s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/50b905bbc9c04036955bbb7a11e23fa2 : SUCCESS in 1h 04m 36s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/09fc6ed1ccf648959048898e989f0098 : SUCCESS in 34m 01s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/607dedf13f684a4ab2ad2279cad40b88 : SUCCESS in 1h 12m 37s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/c0dead5f13f749a0bddc7bf079861f3f : FAILURE in 5m 58s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/41b63c8d545e49ddae6c0057b3d66503 : FAILURE in 40m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5792331dc04ff0383e934a23f1036a641b201177","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-03-26 17:37:55.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a09448271bee2b8fc362d50d3d25cd09c49d050e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 19:51:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d23b12c250e048b9a5f70f2723e8f28c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d010ca1af0bf4acc9841d88845d9359b : SUCCESS in 1m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1bf0152b0bf44ad59ab0a3e4661f2c34 : SUCCESS in 39m 53s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/8fe6456a596749faa0f7a82629715a23 : SUCCESS in 1h 02m 41s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8feb6f8a5be3483aa24d4ef5624fb770 : SUCCESS in 1h 34m 39s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4c5123f0a9894c14a6ff7d23dff6d4a2 : FAILURE in 34m 28s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d9fad0c4b3a44cd5bfa1a1f9ab371734 : SUCCESS in 35m 31s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/75d066e61b8e47cfbe1c41c7b6f95cbf : SUCCESS in 2h 03m 51s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1ea44b77deee482ab989ad998d4a1b4e : SUCCESS in 1h 11m 15s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/dd81647ee723417691b0a8f900695d58 : SUCCESS in 39m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/68eb421e4b6c4256ae019efc039c557c : SUCCESS in 36m 45s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/2fc0cc6ea1154d1585111bbf82e5bea1 : SUCCESS in 1h 02m 59s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/5062ac0d9d9c4c3e93328a4fd8c181a5 : SUCCESS in 1h 32m 45s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/d7c40984d2144478b126ff7d54d105d9 : SUCCESS in 1h 02m 09s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/98cabf81cd074ea98317035cb5ebb753 : SUCCESS in 28m 53s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a4e40e96033d4893b70e657a1afb2749 : FAILURE in 50m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a7b28efb70852fb5b3af7b6cbcb8409483a66439","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-03-26 20:24:29.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"53c9ba723aaf79c3082eef25fb9bc08dc03f6bba","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2025-03-26 21:21:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b32d4c57e96dd62845ba56de02211d1c4eb240fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 21:21:18.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/def5a8c8baa942549ab48d28b0addc23\n\n\nWarning:\n  Change 945637 in project openstack/devstack does not share a change queue with 942862 in project openstack/manila-tempest-plugin","accounts_in_message":[],"_revision_number":3},{"id":"e7a85bd7e981b0d9a36e5f3538919a369c568520","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-26 22:52:47.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"bb8d64df442307dcfadcfb622dfd53aebd10ead5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-03-27 16:36:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"36e808f92678cb888a3820cc8ec8c8d265546897","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 19:10:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c9aa33980f849a6b76ed1f4fe2c3d36\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9c028aef0b54a35b954521f1538727a : SUCCESS in 2m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/92b1239a186848d0aefaf178774f1542 : SUCCESS in 36m 51s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/7ae635b6d4dd44578fd56f38ab4de398 : SUCCESS in 33m 23s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/def3c892092d459a91c8e82ecd852b86 : FAILURE in 2h 23m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/071e3a8158c34adc9ec0d4b1283be772 : SUCCESS in 1h 53m 17s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9784eb5933a54361b7d20a2d6f9e561d : SUCCESS in 1h 02m 16s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/4e8432edc6374b19b256333d893c8ae3 : SUCCESS in 1h 15m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b4474eae47fd4a9ba8d65bae40d3d6c0 : SUCCESS in 1h 08m 52s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e6aa38a4e1be467d8a3a0fb735d15fb1 : SUCCESS in 30m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/17a6dc4611414bed8c31b015a3f7106b : SUCCESS in 37m 08s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/609837ed6943471eb1e2ca260bd148b3 : SUCCESS in 1h 19m 48s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/548e72c2ddee45a0ae9cc40e8c62b287 : SUCCESS in 1h 07m 33s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/c9560a5056534113beac01c15b7c7b6e : SUCCESS in 1h 13m 26s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/18fa1a961cd84658adb1fbc6f90014fe : SUCCESS in 27m 32s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fae93bbe2686436c93db202a9e31350f : FAILURE in 47m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a12b899920e5db37fdb245ecb08d86fed39f582d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 19:10:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"19543db684b41a6e7ae443b9082c2437fcde1863","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 19:52:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3aeabd3555a24da79f7ad0cbc6b074a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d40b168564674287ac2582589e5743c3 : SUCCESS in 2m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/87b31e4698e3427f8be9ad1e0fa3ba09 : SUCCESS in 40m 42s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/7b10a8a6ee1246ddbdf18abd59dd9757 : SUCCESS in 39m 44s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/842d7967822848b999a62d638c4e3298 : SUCCESS in 32m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/654b8492f5be4e58922778dd831391b2 : SUCCESS in 30m 24s","accounts_in_message":[],"_revision_number":3},{"id":"975090bcbf796108454e8434b39e6aa9931a8c46","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 19:52:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"68a14ac9b24d2e340c2cc9f9048e2a69f0ef63e6","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-27 22:11:41.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bad5272a3384510ace3d5449fe6673a9a582e350","revisions":{"bb12eff6b905e8665fdd8723d416505b97a6f518":{"kind":"REWORK","_number":1,"created":"2025-02-26 22:11:35.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/62/942862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/62/942862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/62/942862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/62/942862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/62/942862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/62/942862/1"}}},"commit":{"parents":[{"commit":"b616cb811c2e9ff2054de1cd760869502a1a69e6","subject":"skip tests for stable branches which don\u0027t support API 2.88","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b616cb811c2e9ff2054de1cd760869502a1a69e6"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2025-02-26 22:10:01.000000000","tz":-300},"committer":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2025-02-26 22:10:01.000000000","tz":-300},"subject":"Disable Ingress","message":"Disable Ingress\n\nDisables ingress in multinode-cephfs-nfs-cephadm job\npending nfs-ganesha bug fix for haproxy issues.\n\nChange-Id: I3dbab947bcf5a1df39ee16e308096b3ef75b6d51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bb12eff6b905e8665fdd8723d416505b97a6f518"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bb12eff6b905e8665fdd8723d416505b97a6f518"}]},"branch":"refs/heads/master"},"74542828d15d2a6499d18ec7c7848e47e841b843":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-02-28 17:42:25.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/62/942862/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/62/942862/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/62/942862/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/62/942862/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/62/942862/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/62/942862/2"}}},"commit":{"parents":[{"commit":"b616cb811c2e9ff2054de1cd760869502a1a69e6","subject":"skip tests for stable branches which don\u0027t support API 2.88","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b616cb811c2e9ff2054de1cd760869502a1a69e6"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2025-02-26 22:10:01.000000000","tz":-300},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2025-02-28 17:42:25.000000000","tz":0},"subject":"Disable Ingress","message":"Disable Ingress\n\nDisables ingress in multinode-cephfs-nfs-cephadm job\npending nfs-ganesha bug fix for haproxy issues.\n\nDepends-On: I5f03698ed09ff906488711d5d5a6feb74146b6ca\nChange-Id: I3dbab947bcf5a1df39ee16e308096b3ef75b6d51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/74542828d15d2a6499d18ec7c7848e47e841b843"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/74542828d15d2a6499d18ec7c7848e47e841b843"}]},"branch":"refs/heads/master","description":"Edit commit message"},"bad5272a3384510ace3d5449fe6673a9a582e350":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-03-26 17:37:55.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/62/942862/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/62/942862/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/62/942862/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/62/942862/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/62/942862/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/62/942862/3"}}},"commit":{"parents":[{"commit":"b616cb811c2e9ff2054de1cd760869502a1a69e6","subject":"skip tests for stable branches which don\u0027t support API 2.88","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b616cb811c2e9ff2054de1cd760869502a1a69e6"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2025-02-26 22:10:01.000000000","tz":-300},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-03-26 17:37:54.000000000","tz":0},"subject":"Disable Ingress","message":"Disable Ingress\n\nDisables ingress in multinode-cephfs-nfs-cephadm job\npending nfs-ganesha bug fix for haproxy issues.\n\nDepends-On: I425bb9eec525d82372f05edc63e4fb931e5a4887\nDepends-On: I5f03698ed09ff906488711d5d5a6feb74146b6ca\nChange-Id: I3dbab947bcf5a1df39ee16e308096b3ef75b6d51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bad5272a3384510ace3d5449fe6673a9a582e350"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bad5272a3384510ace3d5449fe6673a9a582e350"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
