)]}'
{"id":"openstack%2Fmanila-tempest-plugin~946622","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I14b040c0abb931b08771e0823616fd5565bb9b0c","project":"openstack/manila-tempest-plugin","branch":"master","topic":"mount_point_name","attention_set":{"37843":{"account":{"_account_id":37843,"name":"Olamide Ojo","display_name":"olamidepeterojo","email":"peterojoolamide@gmail.com","username":"olamidepeterojo"},"last_update":"2025-04-08 13:47:11.000000000","reason":"A robot voted negatively on a label"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-04-08 12:17:03.000000000","reason":"\u003cGERRIT_ACCOUNT_37843\u003e replied on the change","reason_account":{"_account_id":37843,"name":"Olamide Ojo","display_name":"olamidepeterojo","email":"peterojoolamide@gmail.com","username":"olamidepeterojo"}}},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2025-05-28 20:45:22.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}},"hashtags":["outreachy-2026-prospect"],"change_id":"I14b040c0abb931b08771e0823616fd5565bb9b0c","subject":"add functional test","status":"NEW","created":"2025-04-08 12:16:27.000000000","updated":"2026-03-31 20:07:03.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a2d0703ea784ad5007b3a8ecfe78fcfe991e9b4e","_number":946622,"virtual_id_number":946622,"owner":{"_account_id":37843,"name":"Olamide Ojo","display_name":"olamidepeterojo","email":"peterojoolamide@gmail.com","username":"olamidepeterojo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-04-08 13:47:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-08 12:17:03.000000000","updated_by":{"_account_id":37843,"name":"Olamide Ojo","display_name":"olamidepeterojo","email":"peterojoolamide@gmail.com","username":"olamidepeterojo"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2025-04-08 12:17:03.000000000","updated_by":{"_account_id":37843,"name":"Olamide Ojo","display_name":"olamidepeterojo","email":"peterojoolamide@gmail.com","username":"olamidepeterojo"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2025-04-08 13:47:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"016401d8f3e2b6c1b8a62e9d0a1e74901c4a7fdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37843,"name":"Olamide Ojo","display_name":"olamidepeterojo","email":"peterojoolamide@gmail.com","username":"olamidepeterojo"},"date":"2025-04-08 12:16:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7344054b3cb235a9d8da4ea0d893b3ffa8c5288","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 13:47:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0fc431c8dfe14bd99ec4bbd1e8f2e052\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/138d497a5d0849f98dd1bed5c115daf5 : SUCCESS in 1m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/81bc96885c074b58a0bd48fcdd6e0bd7 : FAILURE in 1h 12m 35s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/979a8f1809fc499f911950dc6d0dfca6 : FAILURE in 1h 08m 00s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/fd233d68f51f4433bbb34fe945f3668f : SUCCESS in 58m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e90ac6a2f23f46e68bff3cb5f49a56ae : FAILURE in 48m 35s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c24d466e1a2d4a2f824f14bbc2aa6583 : FAILURE in 35m 50s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ee084f8742de46ffacc9c806cb57eee8 : FAILURE in 1h 25m 17s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/16f7628fc5e14ff284c073154e19f22d : FAILURE in 1h 08m 27s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/417b1099a94b4b17ac95a9e72c63ba3e : FAILURE in 24m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7adfe92bcd5d4c4daa95458f9ddb573c : FAILURE in 32m 33s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/49b09a947a614c09b4f014a2891cf7ff : FAILURE in 35m 18s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/ef6f1e448f82426abecff0649f424673 : FAILURE in 28m 08s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/13bd71f06924448fa2034fc164e17b67 : FAILURE in 30m 14s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3eeb764d9e6b4ed3987f4f73d148fe3e : SUCCESS in 31m 27s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7fc8e2a64b8c4e8daed45495e3ecb7f5 : FAILURE in 39m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5163aed43ec7c24ee5fa004639d8b13839ea457","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-05-28 20:45:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"267587d2824f79c1d989b7d67f19af01a03b977a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 22:42:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d184f53d64c748d1a4c9ea3242e29075\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/774d76ebc69646f08d81eab1cc031d4d : SUCCESS in 3m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/10ada7642a814a90915f8ae36e63be43 : FAILURE in 1h 01m 06s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/941e25c912d7421a8427c7ee5f53edb5 : SUCCESS in 1h 55m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f0629eca90a54f869fa49bb78174571e : FAILURE in 1h 12m 20s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/2980c10a108e42c883ad7530653c0eaa : FAILURE in 1h 01m 56s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/39e8aa6ed34b43569bf9658596605e58 : FAILURE in 1h 28m 58s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/99543fa4958a4021a12a598919aa5a0a : FAILURE in 1h 00m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/11a827956fac4565b55069c887c0421e : FAILURE in 35m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/74059ec10abb49429cc9e06570339169 : FAILURE in 35m 23s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/989a9ad351444f368749f92517de7023 : FAILURE in 1h 07m 59s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/de2774c8e3da4461ac9edbbe4b8982a1 : FAILURE in 1h 03m 55s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/3120b620160e4c46aba78c5102aa447a : FAILURE in 1h 02m 16s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/966e8fc10f354647a5995ff9c4b3e1ca : FAILURE in 1h 07m 56s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/9dc145dee9e9475d98215d7ff2851937 : SUCCESS in 35m 03s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2d64e268da4e4991a12438997375185f : FAILURE in 30m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a2d0703ea784ad5007b3a8ecfe78fcfe991e9b4e","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-31 20:07:03.000000000","message":"Hashtag added: outreachy-2026-prospect","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"90b77a85744f006cfd82b81d0ad80dd9e4fe7ecc","revisions":{"90b77a85744f006cfd82b81d0ad80dd9e4fe7ecc":{"kind":"REWORK","_number":1,"created":"2025-04-08 12:16:27.000000000","uploader":{"_account_id":37843,"name":"Olamide Ojo","display_name":"olamidepeterojo","email":"peterojoolamide@gmail.com","username":"olamidepeterojo"},"ref":"refs/changes/22/946622/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/22/946622/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/22/946622/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/22/946622/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/22/946622/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/22/946622/1"}}},"commit":{"parents":[{"commit":"65ca50c79641214d982239ff59c942bec6f92c31","subject":"Merge \"Disable Ingress\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/65ca50c79641214d982239ff59c942bec6f92c31"}]}],"author":{"name":"olamidepeterojo","email":"peterojoolamide@gmail.com","date":"2025-04-08 12:08:11.000000000","tz":0},"committer":{"name":"olamidepeterojo","email":"peterojoolamide@gmail.com","date":"2025-04-08 12:14:22.000000000","tz":0},"subject":"add functional test","message":"add functional test\n\nThis change will have a test case in manila tempest plugin that ensures\nthe export path has a custom mount point name.\n\nCloses-Bug: #2057703\n\nChange-Id: I14b040c0abb931b08771e0823616fd5565bb9b0c\nSigned-off-by: olamidepeterojo \u003cpeterojoolamide@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/90b77a85744f006cfd82b81d0ad80dd9e4fe7ecc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/90b77a85744f006cfd82b81d0ad80dd9e4fe7ecc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
