)]}'
{"id":"openstack%2Fmanila-tempest-plugin~946714","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Iafe99618d96e28b4dcfbcf10692abd21b20ad63c","project":"openstack/manila-tempest-plugin","branch":"master","topic":"migrate-to-noble","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2025-04-17 17:41:17.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-04-17 18:50:12.000000000","reason":"Change was submitted"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2025-04-17 18:50:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iafe99618d96e28b4dcfbcf10692abd21b20ad63c","subject":"Stop testing the LVM driver with Jammy","status":"MERGED","created":"2025-04-08 21:07:09.000000000","updated":"2025-04-17 18:50:12.000000000","submitted":"2025-04-17 18:50:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"946714-migrate-to-noble","meta_rev_id":"bfc4c72a8b3fd0a6ee1d8a622f1f82dbfbd8276e","_number":946714,"virtual_id_number":946714,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-17 18:50:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2025-04-17 17:41:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2025-04-16 15:30:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2025-04-17 17:41:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"reviewers":{"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-08 22:19:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-09 03:54:04.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-04-16 15:30:35.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2025-04-16 16:05:28.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2025-04-16 16:05:28.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"}],"messages":[{"id":"73667ffc4d5e87921a148c276e9ab1405041a688","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-04-08 21:07:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6dfad4a6ea3f1a373e4ff3a192794f4d41cd3fff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 22:19:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4deb9cfb9674330b98c2d20b10da7c1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/931ee1486d5e44e88e4aec6474446d3a : SUCCESS in 3m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/40bafcb5a94b4de695688b454bd07662 : SUCCESS in 1h 07m 27s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e8d7df2b33074a7bbb697081700d417b : SUCCESS in 58m 39s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5d70d6033341487381d2c8c94e7d50cc : FAILURE in 45m 35s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/903d303a1e2d4ce9950592aa1de15827 : SUCCESS in 48m 24s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/7f4ace65c54c4e44866aa16329e7cb7a : SUCCESS in 1h 11m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/375f4ddc6b1d4d618a8bb2cd8489d257 : FAILURE in 1h 01m 13s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b809051083d44feb8f57e89f36c60c59 : SUCCESS in 29m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/389778e7d2614c628abe28cd4ac9a319 : SUCCESS in 37m 59s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/1cd5d4207f4842baa514b43ae3542078 : SUCCESS in 57m 07s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/9cc93e32f2a14967bd405b2f583bcd40 : SUCCESS in 1h 05m 32s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/64bf767d2f6b4a7a84833dd4bc167236 : FAILURE in 1h 00m 22s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/215d4983ac5748058cc8dad84fa073cb : SUCCESS in 22m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/19a33032b3c14dfeb9b569102efa3788 : FAILURE in 37m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ffa73e56f62845481096b0978b15451e1a416f4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-04-08 22:24:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c9d3e4fb1924378118555435d4f51488115af4a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 00:52:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7830ad379b854c2182e17815e270b499\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b9bc66206ee4b3d996c8040028560c3 : SUCCESS in 2m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5420562622c64953b8e0aa9f5bfc3809 : SUCCESS in 53m 41s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7e31ed5ae2bc4bea95c82258a935eb3e : FAILURE in 2h 23m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5a38aa60a2de446b8345e9e75a12b2ad : FAILURE in 1h 30m 33s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9ea03497665e49dcb4b90c63795732c1 : SUCCESS in 55m 24s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/4facb71b3c074dbe979d91e9523470b6 : SUCCESS in 44m 20s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a6def27224fd46c48aa9841247f871cb : FAILURE in 1h 16m 07s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c04147c01de1438e88efa732169ac580 : SUCCESS in 18m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ddd70f3937444309a3fdaee544dec8e4 : SUCCESS in 23m 56s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/87c7ffc1fc5d4fe7ba5437175f2889b1 : SUCCESS in 28m 58s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/b6b8ac81a0ed4160a8b1a44f7781ceb5 : SUCCESS in 29m 32s\n- manila-tempest-plugin-lvm-2023.2 https://zuul.opendev.org/t/openstack/build/2d90779d5cfd4dc799e3aa132f41f102 : SUCCESS in 30m 54s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/5928608d0542450ab0d28f9ee8bc91cf : SUCCESS in 17m 20s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0d239e0dc8ba4dd299f628d28f66b240 : FAILURE in 19m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9cb1ca7c16cc4ca4fe75f565f2959bbda714f0fb","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-09 03:54:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"2f32ded7d77ad82a3a927c97faa493c10d18b549","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2025-04-16 15:30:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"403b46cc11643e6908d900bc990e06b8b2b919ac","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-04-17 17:41:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb961c0a4c8ab7e6fb2ba5cd7c557e7befe76768","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 17:41:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dea6e02aa5c4422d6bfc106df953e11a22bf0389","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 18:50:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cf4914b46714dc0b2863a42887f2408\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d2a7c1a73db47979e4187d1b564e031 : SUCCESS in 4m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5fd034e2a83d43c181973fcb7bb501fc : SUCCESS in 1h 04m 01s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a743512c12bd495eb5c5d6964c85947c : SUCCESS in 29m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/89039a08eb6349429cbf30e5150a9e9f : SUCCESS in 32m 13s","accounts_in_message":[],"_revision_number":1},{"id":"bfc4c72a8b3fd0a6ee1d8a622f1f82dbfbd8276e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 18:50:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"147028febfba339b4ab6f80425fb00aed0136f40","revisions":{"147028febfba339b4ab6f80425fb00aed0136f40":{"kind":"REWORK","_number":1,"created":"2025-04-08 21:07:09.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/14/946714/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/14/946714/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/946714/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/946714/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/946714/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/946714/1"}}},"commit":{"parents":[{"commit":"65ca50c79641214d982239ff59c942bec6f92c31","subject":"Merge \"Disable Ingress\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/65ca50c79641214d982239ff59c942bec6f92c31"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-04-08 21:05:42.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-04-08 21:05:42.000000000","tz":-420},"subject":"Stop testing the LVM driver with Jammy","message":"Stop testing the LVM driver with Jammy\n\nWe duplicated the LVM driver job to have an integration\ntest job on Ubuntu 22.04 (Jammy Jellyfish). We don\u0027t\nneed to run it against master any longer, because\nthe runtime guidance for 2025.2 doesn\u0027t require it.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.2.html\n\nChange-Id: Iafe99618d96e28b4dcfbcf10692abd21b20ad63c\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/147028febfba339b4ab6f80425fb00aed0136f40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/147028febfba339b4ab6f80425fb00aed0136f40"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
