)]}'
{"id":"openstack%2Fmanila-tempest-plugin~951234","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I934a60865a504f1a3d9af8a1f8df99bf6a7acdf5","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-05-30 01:08:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I934a60865a504f1a3d9af8a1f8df99bf6a7acdf5","subject":"Remove stable/2023.2 job","status":"MERGED","created":"2025-05-29 05:54:53.000000000","updated":"2025-05-30 01:08:14.000000000","submitted":"2025-05-30 01:08:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951234","meta_rev_id":"3721cdc7c01e8529472253eba97132e51ec005b1","_number":951234,"virtual_id_number":951234,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-30 01:08:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2025-05-30 00:05:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2025-05-30 00:05:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-29 09:09:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-29 12:10:44.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-05-30 00:05:49.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"3b67b7fee3b45300b8e62cae972f08729dd5debf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-29 05:54:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5fc1b0cbd87f76c9ef88ff8443c4a1790b64757","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 09:09:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbf82bfad3ce436891c1972139110d9c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b894dc10e49e4a4b8901f09dd67e80c8 : SUCCESS in 2m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cdb9a792307540faab31ebb70f8245e1 : SUCCESS in 1h 21m 27s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8c54455c274248ddb4b5032251d8b90b : FAILURE in 2h 32m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5a9007ff05ad4e7fa59f7168613f6c2d : FAILURE in 1h 28m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/6feaf4b56fbe40a5a365ef31fee78eb8 : SUCCESS in 44m 34s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c9ee100eacf941b586269382f57e3c82 : SUCCESS in 1h 33m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fea56541b614451ca853b45737c8a248 : SUCCESS in 1h 17m 07s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a5148ed0c297434b938111b8cba36408 : SUCCESS in 26m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ed497e18694f4a7982c3990a5fe27de7 : SUCCESS in 39m 11s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/b564e836d3aa45e0a6346c04d25cd225 : SUCCESS in 51m 42s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/7a5d206b92b54d4a8fdc41bcd72c4bc5 : SUCCESS in 1h 06m 45s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/c27efc64bfc944478ef95ba3343010cb : SUCCESS in 1h 09m 50s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/e017807ae4ba4013a9c7e114646c02cd : SUCCESS in 25m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b8292f68e2be4385b5e48b96a1c1e3b7 : FAILURE in 32m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"178f2bab884def910ac6a6aee7b72b808aa38f1e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-05-29 12:10:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"2103de2016882e7a8c1692ae9d6bf3f002e4ba2a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-05-30 00:05:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d22653434b4ba52b896613d053ed1f9ec67e44a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 00:06:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a57e4f3714e2a627f8fde1e04a1f5703af54a2d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 01:08:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bfe3217cde44249a30559228d9765c9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a796d47b849b4b3f93ee83ad2911cd49 : SUCCESS in 3m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/81c0c1005a444bb5bce04d6aa6277dbb : SUCCESS in 1h 01m 53s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4a8ed7a0c45340b3b7a31eb89abd0b4f : SUCCESS in 31m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7e9a58d0886c4a5f97068787dda91d00 : SUCCESS in 33m 16s","accounts_in_message":[],"_revision_number":1},{"id":"3721cdc7c01e8529472253eba97132e51ec005b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 01:08:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"315eb287379ce419e117c71a6fa74148ca166ca9","revisions":{"315eb287379ce419e117c71a6fa74148ca166ca9":{"kind":"REWORK","_number":1,"created":"2025-05-29 05:54:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/951234/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/951234/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/951234/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/951234/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/951234/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/951234/1"}}},"commit":{"parents":[{"commit":"a3430289de4f68c9ff7b9027ea4dcc44f0f28ee2","subject":"tox: Drop envdir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a3430289de4f68c9ff7b9027ea4dcc44f0f28ee2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-29 05:54:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-29 05:54:45.000000000","tz":540},"subject":"Remove stable/2023.2 job","message":"Remove stable/2023.2 job\n\n... because stable/2023.2 reached its EOL.\n\nChange-Id: I934a60865a504f1a3d9af8a1f8df99bf6a7acdf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/315eb287379ce419e117c71a6fa74148ca166ca9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/315eb287379ce419e117c71a6fa74148ca166ca9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
