)]}'
{"id":"openstack%2Fmanila-tempest-plugin~952368","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I3145f9cd6847464b2920f1b0a35e6c211e45b26e","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bp/share-encryption","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2025-11-11 19:37:20.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2025-11-11 20:36:00.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-10-20 18:16:55.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2025-11-11 20:36:00.000000000","reason":"Change was submitted"},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2025-11-11 20:36:00.000000000","reason":"Change was submitted"},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2025-11-11 20:36:00.000000000","reason":"Change was submitted"},"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2025-11-11 20:36:00.000000000","reason":"Change was submitted"},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2025-11-11 20:36:00.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-10-14 20:59:30.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I3145f9cd6847464b2920f1b0a35e6c211e45b26e","subject":"Add tests for encryption_key_ref","status":"MERGED","created":"2025-06-11 13:34:54.000000000","updated":"2025-11-11 20:36:00.000000000","submitted":"2025-11-11 20:36:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":69,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"952368-bp/share-encryption","meta_rev_id":"7050da92ed138eb63a3bc88cf7ccde8479829e34","_number":952368,"virtual_id_number":952368,"owner":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-11 20:35:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2025-11-11 19:37:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2025-10-20 18:16:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"value":2,"date":"2025-11-06 13:20:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2025-11-11 19:37:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"reviewers":{"CC":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-11 16:38:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-08 20:05:24.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-07-10 23:35:18.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2025-07-17 09:32:52.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"CC"},{"updated":"2025-07-17 09:56:59.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"REVIEWER"},{"updated":"2025-07-24 09:44:25.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2025-07-24 09:44:25.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2025-07-24 09:44:25.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2025-10-02 15:40:07.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2025-10-02 15:40:07.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2025-10-02 21:42:32.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"CC"},{"updated":"2025-10-14 21:43:07.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"},{"updated":"2025-11-06 13:20:00.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"4ff027370de3bb3570e0bc81eebc05bfc9fe0c59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-06-11 13:34:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2db1e4b9387f548d4b41ca9dc8a82812e3edcbe5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 16:38:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4c1944bfdba4285a12ef1920d9a1ebe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c474b225cd2343a7878d6dad05843e80 : SUCCESS in 1m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/588c854dd34c4f338b0872cbbfaa05a7 : SUCCESS in 1h 11m 43s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8e3aa4e217044c0292632202389b6551 : FAILURE in 2h 56m 17s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/129c5f02809c4578b953e624b2cc514a : FAILURE in 1h 01m 14s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/beb18fee87a4455db323a745755e36b5 : FAILURE in 1h 07m 57s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/b15d4b80cf224034b1978dc5699ad549 : FAILURE in 1h 22m 20s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/43ab2f41ac1e45b39061d8d73ae5f14d : SUCCESS in 1h 17m 28s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9b256f4e0e2b47b8ad288238f45f82ca : FAILURE in 20m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2fa0f0346e6743a99e473e7f0ea97e1a : FAILURE in 34m 00s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/e6983c2644c64f47bb033173a130ef31 : SUCCESS in 1h 00m 47s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/9349077e97814d01b5c17efc1d8472a8 : SUCCESS in 1h 23m 22s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/f4a89939d6cf4644a6f658d49400bfec : SUCCESS in 1h 18m 57s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/1a0a4fd156384aad83d9c937d84bc7d4 : SUCCESS in 30m 21s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b8e0aafe05c940fa87ed67fcb1c4b627 : FAILURE in 31m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"50e79d742b6196e35fe2ff5ab50c4f11e884ecdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-06-12 07:14:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"991b6dda4a22428c88cc85d613ade288d49da92a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 08:41:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c28ffa1c4e304ec08db3469830e848e8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e706aa4e60744560887cd70e9e71b1c0 : SUCCESS in 3m 23s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e5e19fafd93a405a8aed84f7a6fdc21e : SUCCESS in 1h 02m 45s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/348005f84bb041938c5f189b1ba95f9d : FAILURE in 13m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3cf0c6a37c4b4066a68dc7a1a1be9197 : FAILURE in 1h 05m 07s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/aee0d6891ab84dad9d18fbb1565e0556 : SUCCESS in 1h 14m 49s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/eddedfca3da5469496a58daa439ff50e : SUCCESS in 1h 22m 10s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e9ef01c6bab543f5b89c4e8d557cd361 : FAILURE in 1h 19m 02s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fa1a09c5f41148259f99ea55cadb5e9e : SUCCESS in 30m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2de7cfcd72b64dff8030a9b69ad999fb : SUCCESS in 41m 06s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/221f4536c9cb477eae901e67c4555846 : SUCCESS in 1h 16m 58s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/376745ac906a44e683500d9f902c1d68 : SUCCESS in 37m 31s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/856d66d9e9d1408688afff3d92fc56b4 : SUCCESS in 41m 59s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/919179fa39154045aca66f4104b8de2c : SUCCESS in 20m 32s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d0515646234d4bb2a6dd843185cc33fe : FAILURE in 22m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8cfdbb7dfccca9439c7b4865a763a6901f928608","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-07-08 14:19:57.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"91979e9c92c01e8448e13a63a0acda085a10ad8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 17:03:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6eab69c4b3e049d694daae6c6f1efc1e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/54c1b508679a4e4c97db68e7a19afc42 : SUCCESS in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddab50edcf6c4d3b845c5df0e144e16c : SUCCESS in 3m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/66885b8e05664532a6206e847c7e83d3 : SUCCESS in 1h 08m 26s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/fdceda5100d348c7b8e19a79b8965cf8 : SUCCESS in 2h 23m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/49d25696bf854523bac38d4cfc8f9a4b : FAILURE in 29m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d7888af02f2349c58f06108024cb6a81 : SUCCESS in 25m 46s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/4d2fb0e32a694431bac8e165051e2b87 : FAILURE in 1h 32m 13s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/082526726f21411894707bb88a37ec6d : FAILURE in 21m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/eaecfaf3a1254d12ac991c6faffa12d0 : SUCCESS in 33m 39s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a8d0746f83654acf8fa480cdc009758e : SUCCESS in 37m 08s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/0678457b6677482fb1d689ac0a8d56cb : SUCCESS in 1h 21m 46s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/b5734adce7d14f979138cf0ecea08f8d : SUCCESS in 1h 07m 51s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/bb7cfdbdea4f4d17b9f2e0cb3b5de183 : SUCCESS in 1h 01m 14s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/260593c4a3c9470f9969ac1184057b8a : SUCCESS in 33m 08s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/91f4441e2ca24047ba354bb3c8e0c464 : FAILURE in 23m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"955e27852764bc92fc85b3ad38b2bb1d8e1f3647","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-08 20:05:24.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"b3a3391a566ed83a7dbead952c9cf3a3375e1ed7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-07-10 23:35:18.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cf93aa30ca54671e41bea3599ff6ae5f2a19df96","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-07-10 23:48:37.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8a73677e21421cf1e6772422f9e41b89a1890372","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-07-11 14:44:50.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0ada692cbc52d1fc7f8952b29863817804e0b054","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-07-11 14:44:51.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"75d0b9de3db61d084c5dce477e08087e893c3b74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 16:02:34.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/03ba45b0cdba4957994feea9e4255a21\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ba5022fe86f64c7389307ca92d634c13 : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0de96804589e464bb81ff8b4b4d15b2d : SUCCESS in 2m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/dda6ad7eeeea45fd903bf8914e816050 : FAILURE in 1h 10m 52s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a8db6c6540634267952d8cb5c983b3d1 : SUCCESS in 59m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6c67afce9d434682b4ff4ef22eadec20 : FAILURE in 18m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/2d67ef084ed8407ba8a9a05b7018ccd7 : SUCCESS in 58m 43s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/b9fa35c2931149cdac48c12fb87e9bfa : FAILURE in 1h 15m 12s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/123af9b4a4b14c8188c9ec34d5d86191 : FAILURE in 18m 32s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d4dd5637b1824f52b2a88ec2c3600453 : FAILURE in 33m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c4fcd2be9d7649e994740f42f5a0c1cb : FAILURE in 39m 57s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/0e5810cc5f314402b63a739a963e7809 : FAILURE in 32m 27s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/2905588550a6490c91cb0da86e48a960 : FAILURE in 37m 50s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/20ec3197ee0c463c91329dade5465558 : FAILURE in 32m 08s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ff51cf84f75e43a286b2dca8b7a05a38 : SUCCESS in 18m 23s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5195bcd4f32e45bf9f1f494980032d21 : FAILURE in 21m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4035e78552137817b0a43aec7f82b498c8846dca","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-07-14 08:23:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4d5399a4fb82ef4edf307ba3be4cead12470d1d8","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-17 09:32:52.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"255079d90537c8bb6c67239e86db72425ac6c632","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-07-17 09:33:50.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"93f24f4b4bb3bb0e876215dc9a05e578f47b37ff","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-07-17 09:34:00.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"511f869690826ee27eb62233bd51546f16bc56d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-07-17 09:41:17.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"04457f398c2691f2e69b0ce2f2508c683108c8e8","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-17 09:56:59.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e5f366926eeac742c187f66e1dd0ade30a51a0ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 11:02:50.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/54e5e1c3da8f42379108baaeeeea696b\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ea3806f110974291b54bb1345e7850ca : SUCCESS in 1m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dc5634a83be4295997c033359faf150 : SUCCESS in 2m 14s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d6424d1563c2465fb0ac15ddb8e930cf : FAILURE in 53m 10s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/437d529e67134c8f93b76722314acbe6 : SUCCESS in 1h 14m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/031c276c8be7406b839f42d285d315a5 : FAILURE in 1h 08m 29s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/63739d19f1d9476eab622e0aee05b376 : FAILURE in 39m 34s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/1653c4f916944fbfba029a4c773888fa : FAILURE in 1h 13m 56s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4786ef72d9de4ebfb5dc0e619a83910e : FAILURE in 16m 20s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/32003de0724b402d87b13ce3018cb7b9 : FAILURE in 21m 11s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e660963d6d3c43dab3f9bd83a60ee967 : FAILURE in 33m 17s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/11089cb3db4747a8aea20631305fb967 : FAILURE in 51m 37s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/a85a52888e9b42d7a89c3b0b57f0a226 : FAILURE in 1h 03m 09s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/b8c2b62616d5453499cf804840c3d2f1 : FAILURE in 40m 25s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/4de6a3fab74749b19c27a88b6367fe4e : SUCCESS in 22m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bfff761d04a1488c8b400f8d2f2c3ffe : FAILURE in 11m 19s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"72f0dfe68c5e2c4bbb93835ed0117b84eb5d0a17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-07-24 08:12:03.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"6e58766bc3c815aba7d240d16a6069e52f6291ab","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-07-24 08:13:45.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e2c1bb6ebc354bd0196c388012fc774320d494ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-07-24 09:52:58.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6bef879e3e3b863a9f23a0531a40f4c83dbbe290","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-07-24 10:08:59.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ec4b4bc132f087630d0440b743b77312b75a1e5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-24 10:58:55.000000000","message":"Patch Set 9: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/911089 is needed.","accounts_in_message":[],"_revision_number":9},{"id":"7762b0b0aa9b567828b838946cbef3eb6f38ccb9","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-07-24 10:59:54.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"50afab0db05b7fe4368cf34bbb0b89f93346387b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-24 13:46:49.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93815f70a6574850877d42fe709182fb\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b2920f4f740b4f69900f2cf3316c8749 : SUCCESS in 2m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed764cad338948b28d2b4a12d4dcbb28 : SUCCESS in 3m 16s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/98662ef14b264f53a16d0a08d232074c : SUCCESS in 1h 34m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/20efb83ea6dd4dd48eccd6d1ebf14b03 : FAILURE in 2h 41m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b06226c87e18438393a00f7bd80fb807 : FAILURE in 1h 32m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/335335cfdec54cf2b04f14392fb78ba4 : SUCCESS in 1h 03m 02s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/3f276cd99ccf4deabde2c7ad6992c866 : SUCCESS in 1h 47m 22s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2f3d1b5a6f2544d99f234ee32fd1b79c : FAILURE in 21m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/59957fde09ce4fa087baf97e4044adaf : SUCCESS in 31m 42s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3c97c5baf5ec4df8ab9c662d0d034fbc : SUCCESS in 35m 08s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/925c175f2d7a4c3a9ff9b330d026f954 : SUCCESS in 1h 28m 11s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/f3c3a89f30794bb3bcfcb473ec84ad69 : SUCCESS in 1h 06m 40s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/dda4fc39e76f45d48f3b2e5cfb4a0f4e : SUCCESS in 55m 52s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/e3dbed08ef8a47988e5ab51db6fcefd4 : SUCCESS in 25m 15s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4a05b806ca724dd98e452ebe730bf41c : FAILURE in 18m 35s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"acae823e1b277c7075af38f2ddbff88f8ebdc8ca","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-24 14:15:10.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"d6cdeeaba4dc7f4c20744422f8916d15e6272fa5","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-07-25 06:31:35.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e1dcff60166f0f636cfd0c179e723a63f9c4fd46","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-07-25 07:22:26.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"dbf713c8280577d26ae5a40c0d6c239642293cec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-25 09:24:49.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed15f4de81404c2aadd793304af5d736\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/432108cace234695928df1ca4d129f18 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9626abb4340f402da28478fa2a687825 : SUCCESS in 4m 15s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/34be94e00e0b4b98933f344d18c131f7 : SUCCESS in 1h 21m 36s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d36df6e7d2344e5cb80973d523dd5bf9 : FAILURE in 2h 47m 02s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b4c27af898d340a1b7fa8fae11241278 : FAILURE in 1h 29m 11s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/aebe656c33c547e486e6793d24b83850 : SUCCESS in 52m 29s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a5f204beeeb94032b41a144f4689e635 : SUCCESS in 43m 14s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ca54f68721144feaa98c8446602a2860 : FAILURE in 17m 33s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a5d8a2fed242450c988d8c90eda6540d : SUCCESS in 27m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/791dd06304b044a59b165e1c0cadc9a1 : SUCCESS in 37m 56s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/fb6976163e0e44a2a7d0b069d2bd1c84 : SUCCESS in 32m 37s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/354ab15364224a169bfcc044fbfecada : SUCCESS in 59m 30s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/b9d1478a831144539a93540c85186a08 : SUCCESS in 1h 02m 33s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/15600514d41143698f8820b9ce6b7dd5 : SUCCESS in 28m 49s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cca2a9a90c65468db46cde6d32ce5ea1 : FAILURE in 19m 20s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d8151e634f298ccd7b82c2bf07289c795e37b383","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-25 09:35:01.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"62e4b0193ceb18a9e92f4f194653d0f96d7534ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-08-06 10:19:40.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"a793cd31a193b23d67caf6eb71f371c88707bf6b","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-08-06 10:20:02.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"f3d1eb6ba625d755943f56da7564b2c75a2bdbfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 12:18:53.000000000","message":"Patch Set 10: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/911089 is needed.","accounts_in_message":[],"_revision_number":10},{"id":"a01c4c41588aafebc33ecb331e6e7b38dcee9810","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-08-19 10:33:53.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"f444abda3f05520c213f0457d7b3a4d67500ed77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 13:15:38.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b937fee0e8f14a4aa44bd04d94051513\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/41f471ab9e3042069106efb4d34fef08 : SUCCESS in 1m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fef2511ada1742ae8aebdeaf17d7f265 : SUCCESS in 4m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1741ca525b3b4732819f561e07f7a02b : SUCCESS in 1h 03m 58s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/aae3e3e7bd7b4429b7b079b4e6a75b7d : FAILURE in 2h 38m 43s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b0210022ab484655bf0cbe47ba705749 : FAILURE in 53m 43s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a556c2b91ab8489295999afe143d7c48 : FAILURE in 26m 16s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/470e87658a344dd1ada5b452a6054a2f : SUCCESS in 1h 17m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/00318daf659942a48c0d8d781962adb3 : FAILURE in 17m 29s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/01baac5043e3467faa6c482d5918d085 : SUCCESS in 36m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/73dd4bb1ccee43ad9e7c5da022017a99 : SUCCESS in 36m 37s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/1d4aaa22cf01422387387d45ba6e5697 : SUCCESS in 1h 00m 04s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/18bd40d7af7246e5a889929da53ec7a8 : SUCCESS in 55m 58s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/4959dddb3a014d75b728a9368ac1d2d3 : SUCCESS in 1h 18m 32s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/5d6760d44d7341148b22bba4c52af1e3 : SUCCESS in 25m 49s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/924fe6fb70164b20856ca36f92904f06 : FAILURE in 23m 12s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"079f314ddb766699265b0b88e1f0a82edb5e571e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-23 16:15:15.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"70a7090e3c935b9d2ab479c7811b687f563b51e7","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-09-24 07:39:01.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"aef68afa2456d105b2a21532f34db68bf3f84ff5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-09-24 07:39:28.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"3bb80b844f9e19ad6a75dcafc01dc0b99a73b637","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 08:54:41.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6090435099c46b79f92f4d65d13e79a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8b7dde744eaa4496b13e0bf722d40afd : SUCCESS in 2m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b46155e0641d4defa1c8c426069f97bd : SUCCESS in 2m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1c9b48397a1b4d259caaf86495dffca6 : SUCCESS in 39m 29s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7ca89c86c1be409e923adf22fe222f53 : SUCCESS in 1h 09m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5d4e76b44fc74307a77d9585d39fbbea : FAILURE in 1h 00m 33s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ab402b180385488abf32473a2b7bb9f9 : SUCCESS in 33m 21s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/4e5e1f850e5c4728a8c5e0917b34c04a : SUCCESS in 47m 51s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/11d0801417524d44b407519686d2cf6e : FAILURE in 14m 15s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/05a2071a6fe14afbb3a93ebe2df74a2a : SUCCESS in 28m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1ae15ea019f745f2ae320317f419f7ca : SUCCESS in 31m 49s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/b0de77a40dba48c1aa58a919037ae1ca : SUCCESS in 56m 21s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/f013ac75c65844d1a09b25a74d8600f6 : SUCCESS in 50m 31s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/3ba0fe149aac431b921a17e6f5f03367 : SUCCESS in 39m 18s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f27ddcc654214970bc003361220aef90 : SUCCESS in 27m 02s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f3afc636cb43493195c283f148b6875c : FAILURE in 16m 45s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"9cce169bf6442300bcf7984f5d063d6a37726614","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-02 17:03:31.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"ae3fc50662df2487c029325f3e67ad10f2f59bf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 18:51:09.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81fd5f409f2045d1bf12a7a3ef5c15d7\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d0e02bb272344e0fac47fb155bc4a7d1 : SUCCESS in 2m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a050e7e74e640d1ac76d03a7220c061 : SUCCESS in 1m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8cf0f13f32bf46c6aee7e1ef7f078179 : SUCCESS in 1h 16m 09s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/33de94f5ccac41d88254ad2c6412cf95 : FAILURE in 10m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/dfb6693389544f42a0b83bf06c4b16f3 : SUCCESS in 1h 42m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/5ae1c1585338438693c75f7983606e8b : SUCCESS in 1h 01m 55s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/048676af89d94e19927cda3f9ab0ff48 : SUCCESS in 1h 26m 06s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/338690a3718c4aa3a66561f754f30578 : FAILURE in 16m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1f1e0f3f5f694e23a242faeffea34e94 : SUCCESS in 36m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8ccc7b35220949ee967f775d909ede41 : SUCCESS in 45m 14s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/29ba36eff5134985a5f842c8f5a28bc0 : SUCCESS in 1h 08m 16s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/17439513a2254e28b041a59cda8ce53f : SUCCESS in 1h 01m 18s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/65a86efe758a4dc1a769eec8fccbf657 : SUCCESS in 1h 01m 11s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/1ec5300428d14407bd19c9c2b56787e0 : SUCCESS in 21m 52s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4b70a38fbfd64e0ebe4027d97262e4a8 : FAILURE in 13m 22s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"54b7a7a4242d1156b88da4b4bfe71b7cb3babe8a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-02 21:27:40.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"48cfb2d24749a5df5f98d612b6bc97b760cfa5d0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-02 21:42:32.000000000","message":"Patch Set 12: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":12},{"id":"d47d3c1dff979e60d32ec1ff48ea899ef24cac3a","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2025-10-02 22:56:13.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"d32b7eaef3617f7b2de24d0eb9080b4f785fe970","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-06 12:13:53.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"7f82bbfc2ccc06a46a485e1b5305f91dabe12e23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-06 12:58:23.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"4593b9f1ad6081caa7d0ed940ae54db09a6adacb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-06 14:21:18.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"b4e4845d9a746f32881fe8e862f442b131b45104","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-06 15:24:31.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"0a75e1ccfec9c12ec7e6a18151b17f060d61b747","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-06 15:49:15.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"ae985307604f3beb3cc90dbb1af8e67172442df6","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-06 15:51:53.000000000","message":"Patch Set 15:\n\n(8 comments)","accounts_in_message":[],"_revision_number":15},{"id":"4947d517f0f4b8f4b7916dc582781dac2867873c","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-06 16:26:00.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"16be37209d9c569269e85ab529b1f0a8a4fe18a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-06 16:26:06.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"1867258172cb8c22b83c18892ae10c917defc263","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-06 17:10:02.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"5dbcc37081a1fe316be803321cb0298de7cde6d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 18:46:54.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/453cb04651f146b4b28274857f39850b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10fdeab2777444efb2c1db5e088adb32 : SUCCESS in 1m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/10d2fc3d19a74190b39f2d8517ac0e7a : SUCCESS in 1h 22m 58s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/73efa21cb8414653bcca9b03fa0813cd : FAILURE in 9m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/40e69754140244cba5f25066c9829e59 : FAILURE in 25m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/360ddb52a3eb485da195243ff9a1030d : SUCCESS in 47m 45s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/4aa80efecb8b4482952c8825750a20f3 : SUCCESS in 44m 44s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8d55ae48c5964688a6ce759952fb072a : FAILURE in 13m 49s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7d15c432f00a4842a893399d4a4eb5f4 : SUCCESS in 27m 53s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e3f5e860707041fc8a08f76b592d7d2b : FAILURE in 28m 26s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/80c01e32eca048c287c98da23cc73d6e : SUCCESS in 59m 47s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/8b59329b9f95409c8d93fe28df9c828c : SUCCESS in 1h 30m 28s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/7268f8d5140f4807b717f57f73364a9b : SUCCESS in 57m 02s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/768eeed9a64347098c6e321299b66c73 : SUCCESS in 24m 04s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fcc322faa553441bbc722a8e7cff962b : FAILURE in 17m 04s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"eacc2cb92ed8549dd1e61d301f66df0beed1436c","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-07 07:17:58.000000000","message":"Patch Set 19:\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"2ef1ad99cbbc2cd988abf05893c7391337a6d38c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-09 22:36:28.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"7809498bea96a4673c207de435810957a99a5bed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-10 08:57:42.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"fdbab3d9420b3993c4ff64f7665e322389a8e4f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-10 09:57:26.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"50bbd432342033f40d1377dbc04bd9392e6339e9","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-10 09:58:04.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"3de2e427caded4af824e589931446db6ab50e500","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-10 10:52:40.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"ce78b2a62dda63204db65302d2c7b864450cf97b","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-10 11:22:42.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"6db57b65a0f2d577398193f29a16b2164506c4dd","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-10 11:23:54.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"7d509455dc08f2d37a687fc77e01c205c26ca35c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-10 12:17:12.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"e0fbf2c6774450021c551851ba75f62340aa82ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 12:54:04.000000000","message":"Patch Set 23: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/requirements/+/963685 is needed.","accounts_in_message":[],"_revision_number":23},{"id":"771e57c979f5befdab1e0ddda48634678acea98a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-10 13:04:00.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":24},{"id":"99ef913dcf0964354e4b660138a7e22fd7e2dac5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-10 13:50:19.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"dba7fc84f60a1e83fd8465180918016c4f7473b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-10 14:39:44.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"5202ceceee9ea20361e53bd2923759dd5aaee081","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 17:13:30.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a656e66a5444e6e8d8c28ebf83ceaf9\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c8c80881a85f4078999587abed0c8e0e : SUCCESS in 1m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7129c1278cc34e568f2236fe759e4a53 : SUCCESS in 2m 24s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f75b0abdc5d44c939ec4c8bab0d4e571 : SUCCESS in 52m 37s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/298771cd7c474cbe97e2cdcc3850e120 : SUCCESS in 2h 28m 34s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/234464b9bb88434bbd7b00692ae36c39 : SUCCESS in 1h 21m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/5fd1ae8b81b94a1b8cd3cbea0e3e8c76 : SUCCESS in 31m 15s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ba33f75f1016440daf1f63979e559859 : SUCCESS in 1h 35m 44s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f65eb28e096341e785980da4d1250808 : FAILURE in 10m 19s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/65982599fe224677a525227f8e25bbf6 : SUCCESS in 34m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6758f2112e0c4c7b84162ae75f76f60c : SUCCESS in 28m 27s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/daabdb3e4ada4a1dab11e40bcc0604b6 : SUCCESS in 54m 08s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/df811a6eac454dcd895356af3a1b5757 : SUCCESS in 1h 09m 37s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/ce136f2565a14b2c937523aba3194b6f : SUCCESS in 1h 16m 01s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/23a5f8b68d2844a7a7d2697939b17dc2 : SUCCESS in 21m 23s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/63aefc6d883a4c2aa5d0c528803e2e7f : FAILURE in 24m 44s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"3946f51041101ae017a2f913a1b82763f4e9bfa6","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-14 09:35:40.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"ce380f5e5044f86599d09e8e6430b1def959178c","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2025-10-14 10:33:16.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"0b86d93d12aa77ecf905435b36bc70985a69a31c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-10-14 20:59:30.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"dd62cb13e7a7d77fe884ed3b04898f83a3b3b2f6","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-14 21:43:07.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"4c6c49e471bc5a165e05ba87bb8837fcfb403742","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 00:19:45.000000000","message":"Patch Set 26:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ecd0980d7c446ca93b2f8c049ccdafb\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/10740a2ec36e467fa9815e5415363b0f : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e30016d8d86a410a8cc9e3c403db052e : SUCCESS in 2m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f43da63e71a2401ca8b77098ee61a524 : SUCCESS in 1h 11m 59s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/00b099fb35ad4065b0f67fa17a1adeec : SUCCESS in 2h 31m 12s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/41ef30155dde4ba4af9b332ce5a49e1b : FAILURE in 17m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8d8d754bf2f34c1f853b155f06311865 : SUCCESS in 1h 01m 57s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/cd0f99d34be3409eb71404ad5f49d29c : SUCCESS in 47m 34s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/dc6e16ecb3ed4bd8a4ffe328e342acbd : FAILURE in 10m 02s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1053f6a4e6a941b3b55df74314ec0b22 : SUCCESS in 21m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0c8d10702fae454e8c7f9a79d7e335f7 : SUCCESS in 43m 03s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/af8f2dacc30b4b01aaf1687a44fda6f4 : SUCCESS in 1h 18m 23s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/4f95dcd405e347d385a44b0bf945a669 : SUCCESS in 52m 37s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/59c8e3f204ce48deb11bc8baccb97e3b : SUCCESS in 1h 12m 31s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a5c9ca35146d4da9851dca06451b5fa3 : SUCCESS in 22m 45s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2942e8bba1264016875150df976fdffb : FAILURE in 10m 45s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"c68e7d1ad465d9ee1e3f4d7008e83a5af06b46ef","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-20 18:16:55.000000000","message":"Patch Set 26: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":26},{"id":"b72e7d1caf371fb6e87696f62ca254cb6e7145f6","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-24 06:44:19.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"c6b70c2a10d34eb33ce6b9ee288f3f00a714a2cb","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2025-11-06 13:20:00.000000000","message":"Patch Set 26: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":26},{"id":"1e8e69f50604ac0978dfd4368cf13bd1792a1e0e","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-11-06 13:22:12.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"30bddd8161168974240936250f8e00d9a1826ff3","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-11-11 19:37:20.000000000","message":"Patch Set 26: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"b29d28821884aa166c454d411692ac5115c982c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 19:37:58.000000000","message":"Patch Set 26: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":26},{"id":"f322405186e97a6b6ada829c8609d65cd04502a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 20:35:59.000000000","message":"Patch Set 26: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ddfb2ad862848d9bcffa9a9ce68a0e5\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/05b605f9dee24c19bd781db7f2c6d6ca : SUCCESS in 1m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e698280f4ab342d282b79cc9f0724ea4 : SUCCESS in 3m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/526919224463407c818d2951af6acae1 : SUCCESS in 38m 47s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5e3b961dee0f4bbbb9b1b867a21a725f : SUCCESS in 31m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4c0089a412e24528ab80fbb431d6305f : SUCCESS in 41m 14s","accounts_in_message":[],"_revision_number":26},{"id":"7050da92ed138eb63a3bc88cf7ccde8479829e34","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 20:36:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":26}],"current_revision_number":26,"current_revision":"fe868063d757ee4d4303472c753e320679eaed23","revisions":{"2bac8e5f595b9d8b57930611b1f9ed468f3a9e4f":{"kind":"REWORK","_number":1,"created":"2025-06-11 13:34:54.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/1"}}},"commit":{"parents":[{"commit":"326d5a399168c2a53ec01bf4f74dc6c4dad8eaf3","subject":"Merge \"Add CEPHFS filesystem metadata verifications\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/326d5a399168c2a53ec01bf4f74dc6c4dad8eaf3"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:34:43.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nOnly negative API tests are added since end-to-end testing needs\nbarbican and also barbican secret needs to create first and then\nshare will be created using that secret.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/911089\npartially-implements: blueprint share-encryption\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2bac8e5f595b9d8b57930611b1f9ed468f3a9e4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2bac8e5f595b9d8b57930611b1f9ed468f3a9e4f"}]},"branch":"refs/heads/master"},"fbdf440b4773b744b7da5338fd49c6c8dc37c50c":{"kind":"REWORK","_number":2,"created":"2025-06-12 07:14:27.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/2"}}},"commit":{"parents":[{"commit":"326d5a399168c2a53ec01bf4f74dc6c4dad8eaf3","subject":"Merge \"Add CEPHFS filesystem metadata verifications\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/326d5a399168c2a53ec01bf4f74dc6c4dad8eaf3"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-12 07:13:53.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nOnly negative API tests are added since end-to-end testing needs\nbarbican and also barbican secret needs to create first and then\nshare will be created using that secret.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/911089\npartially-implements: blueprint share-encryption\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fbdf440b4773b744b7da5338fd49c6c8dc37c50c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fbdf440b4773b744b7da5338fd49c6c8dc37c50c"}]},"branch":"refs/heads/master"},"6609029b6ae14cbdb203450fd56566f545d48fa3":{"kind":"REWORK","_number":3,"created":"2025-07-08 14:19:57.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/3"}}},"commit":{"parents":[{"commit":"ff055a60ad38063826e0169d6b346f4a573e92cd","subject":"Merge \"Add tests for Share Network Subnet Metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ff055a60ad38063826e0169d6b346f4a573e92cd"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-07-08 14:13:06.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nPositive test -\n1. create barbican connection, create secret and create share using\nsecret\nNegative test -\n1. invalid encryption key ref\n2. invalid share type extra-spec\n3. absent encryption key ref\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/911089\npartially-implements: blueprint share-encryption\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6609029b6ae14cbdb203450fd56566f545d48fa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6609029b6ae14cbdb203450fd56566f545d48fa3"}]},"branch":"refs/heads/master"},"b44daf308f4ac372810eb203e63c1c913b64d10e":{"kind":"REWORK","_number":4,"created":"2025-07-11 14:44:51.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/4"}}},"commit":{"parents":[{"commit":"ff055a60ad38063826e0169d6b346f4a573e92cd","subject":"Merge \"Add tests for Share Network Subnet Metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ff055a60ad38063826e0169d6b346f4a573e92cd"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-07-11 14:44:04.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nPositive test -\n 1. create barbican secret and create share using secret\n\nNegative test -\n 1. invalid encryption key ref\n 2. invalid share type extra-spec\n 3. absent encryption key ref\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/911089\npartially-implements: blueprint share-encryption\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b44daf308f4ac372810eb203e63c1c913b64d10e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b44daf308f4ac372810eb203e63c1c913b64d10e"}]},"branch":"refs/heads/master"},"eb0bf691333494d1f6af7195c8cf08e43acb24e9":{"kind":"REWORK","_number":5,"created":"2025-07-17 09:33:50.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/5"}}},"commit":{"parents":[{"commit":"ff055a60ad38063826e0169d6b346f4a573e92cd","subject":"Merge \"Add tests for Share Network Subnet Metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ff055a60ad38063826e0169d6b346f4a573e92cd"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-07-17 09:33:13.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nPositive test -\n 1. create barbican secret and create share using secret\n\nNegative test -\n 1. invalid encryption key ref\n 2. invalid share type extra-spec\n 3. absent encryption key ref\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/911089\npartially-implements: blueprint share-encryption\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/eb0bf691333494d1f6af7195c8cf08e43acb24e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/eb0bf691333494d1f6af7195c8cf08e43acb24e9"}]},"branch":"refs/heads/master"},"3110f6ed4616446c29c807094c43249955630efb":{"kind":"REWORK","_number":6,"created":"2025-07-17 09:41:17.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/6"}}},"commit":{"parents":[{"commit":"ff055a60ad38063826e0169d6b346f4a573e92cd","subject":"Merge \"Add tests for Share Network Subnet Metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ff055a60ad38063826e0169d6b346f4a573e92cd"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-07-17 09:41:09.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nPositive test -\n 1. create barbican secret and create share using secret\n\nNegative test -\n 1. invalid encryption key ref\n 2. invalid share type extra-spec\n 3. absent encryption key ref\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/911089\npartially-implements: blueprint share-encryption\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3110f6ed4616446c29c807094c43249955630efb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3110f6ed4616446c29c807094c43249955630efb"}]},"branch":"refs/heads/master"},"af196d7670aeebb5e5cbd64803122b318644fe4f":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":7,"created":"2025-07-24 08:12:03.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/7"}}},"commit":{"parents":[{"commit":"dd1fd2c4ef121eabf3984e6ccd2a7fb3aeae00ba","subject":"Merge \"Drop explicit dependency on python-subunit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dd1fd2c4ef121eabf3984e6ccd2a7fb3aeae00ba"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-07-24 08:11:36.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nPositive test -\n 1. create barbican secret and create share using secret\n\nNegative test -\n 1. invalid encryption key ref\n 2. invalid share type extra-spec\n 3. absent encryption key ref\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/911089\nDepends-On: https://review.opendev.org/c/openstack/manila-tempest-plugin/+/955394\npartially-implements: blueprint share-encryption\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/af196d7670aeebb5e5cbd64803122b318644fe4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/af196d7670aeebb5e5cbd64803122b318644fe4f"}]},"branch":"refs/heads/master"},"314076211e563823df71bcf786cf229c294450fe":{"kind":"REWORK","_number":8,"created":"2025-07-24 09:52:58.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/8"}}},"commit":{"parents":[{"commit":"dd1fd2c4ef121eabf3984e6ccd2a7fb3aeae00ba","subject":"Merge \"Drop explicit dependency on python-subunit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dd1fd2c4ef121eabf3984e6ccd2a7fb3aeae00ba"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-07-24 09:52:48.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nPositive test -\n 1. create barbican secret and create share using secret\n\nNegative test -\n 1. invalid encryption key ref\n 2. invalid share type extra-spec\n 3. absent encryption key ref\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/911089\nDepends-On: https://review.opendev.org/c/openstack/manila-tempest-plugin/+/955394\npartially-implements: blueprint share-encryption\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/314076211e563823df71bcf786cf229c294450fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/314076211e563823df71bcf786cf229c294450fe"}]},"branch":"refs/heads/master"},"55da5e1f82ee1351fe2b1078de55a022bf291916":{"kind":"REWORK","_number":9,"created":"2025-07-24 10:08:59.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/9"}}},"commit":{"parents":[{"commit":"dd1fd2c4ef121eabf3984e6ccd2a7fb3aeae00ba","subject":"Merge \"Drop explicit dependency on python-subunit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dd1fd2c4ef121eabf3984e6ccd2a7fb3aeae00ba"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-07-24 10:08:09.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nPositive test -\n 1. create barbican secret and create share using secret\n\nNegative test -\n 1. invalid encryption key ref\n 2. invalid share type extra-spec\n 3. absent encryption key ref\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/911089\nDepends-On: https://review.opendev.org/c/openstack/manila-tempest-plugin/+/955394\npartially-implements: blueprint share-encryption\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/55da5e1f82ee1351fe2b1078de55a022bf291916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/55da5e1f82ee1351fe2b1078de55a022bf291916"}]},"branch":"refs/heads/master"},"e10f0526229bd808528504abb64744985cede60e":{"kind":"REWORK","_number":10,"created":"2025-08-06 10:19:40.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/10"}}},"commit":{"parents":[{"commit":"82a06bdf2d7b9d5be58c59fd7b0271976dc518b3","subject":"Remove remaining unicode prefix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/82a06bdf2d7b9d5be58c59fd7b0271976dc518b3"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-08-06 09:52:43.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nPositive test -\n 1. create barbican secret and create share using secret UUID\n\nNegative test -\n 1. invalid encryption key ref\n 2. invalid share type extra-spec\n 3. absent encryption key ref\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/911089\nDepends-On: https://review.opendev.org/c/openstack/manila-tempest-plugin/+/955394\npartially-implements: blueprint share-encryption\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e10f0526229bd808528504abb64744985cede60e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e10f0526229bd808528504abb64744985cede60e"}]},"branch":"refs/heads/master"},"eeca33b15c0c6f88f2e81baa9855de004dd7e56f":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":11,"created":"2025-09-24 07:39:28.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/11"}}},"commit":{"parents":[{"commit":"5ca4e5190a7921f019b2c9101e1d5374a9f8ac80","subject":"Merge \"Fix share server migration with replica test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5ca4e5190a7921f019b2c9101e1d5374a9f8ac80"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-09-24 07:39:09.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nPositive test -\n 1. create barbican secret and create share using secret UUID\n\nNegative test -\n 1. invalid encryption key ref\n 2. invalid share type extra-spec\n 3. absent encryption key ref\n\npartially-implements: blueprint share-encryption\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/eeca33b15c0c6f88f2e81baa9855de004dd7e56f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/eeca33b15c0c6f88f2e81baa9855de004dd7e56f"}]},"branch":"refs/heads/master"},"621a9c9115c57d0f71588a1151e327bd2576d2c7":{"kind":"REWORK","_number":12,"created":"2025-10-02 17:03:31.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/12"}}},"commit":{"parents":[{"commit":"5ca4e5190a7921f019b2c9101e1d5374a9f8ac80","subject":"Merge \"Fix share server migration with replica test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5ca4e5190a7921f019b2c9101e1d5374a9f8ac80"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-10-02 17:02:44.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nPositive test -\n 1. create barbican secret and create share using secret UUID\n\nNegative test -\n 1. invalid encryption key ref\n 2. invalid share type extra-spec\n 3. absent encryption key ref\n\npartially-implements: blueprint share-encryption\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/621a9c9115c57d0f71588a1151e327bd2576d2c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/621a9c9115c57d0f71588a1151e327bd2576d2c7"}]},"branch":"refs/heads/master"},"d44260344e6cb182df5306ac62b700694056d4a9":{"kind":"REWORK","_number":13,"created":"2025-10-06 12:13:53.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/13"}}},"commit":{"parents":[{"commit":"5ca4e5190a7921f019b2c9101e1d5374a9f8ac80","subject":"Merge \"Fix share server migration with replica test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5ca4e5190a7921f019b2c9101e1d5374a9f8ac80"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-10-06 12:13:38.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nPositive test -\n 1. create barbican secret and create share using secret UUID\n\nNegative test -\n 1. invalid encryption key ref\n 2. invalid share type extra-spec\n 3. absent encryption key ref\n\npartially-implements: blueprint share-encryption\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d44260344e6cb182df5306ac62b700694056d4a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d44260344e6cb182df5306ac62b700694056d4a9"}]},"branch":"refs/heads/master"},"299ecaeb7c78f39ab7350e209e7bb9d63766b6c9":{"kind":"REWORK","_number":14,"created":"2025-10-06 12:58:23.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/14"}}},"commit":{"parents":[{"commit":"5ca4e5190a7921f019b2c9101e1d5374a9f8ac80","subject":"Merge \"Fix share server migration with replica test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5ca4e5190a7921f019b2c9101e1d5374a9f8ac80"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-10-06 12:58:17.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nPositive test -\n 1. create barbican secret and create share using secret UUID\n\nNegative test -\n 1. invalid encryption key ref\n 2. invalid share type extra-spec\n 3. absent encryption key ref\n\npartially-implements: blueprint share-encryption\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/299ecaeb7c78f39ab7350e209e7bb9d63766b6c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/299ecaeb7c78f39ab7350e209e7bb9d63766b6c9"}]},"branch":"refs/heads/master"},"45236fe6ef8356b8e1a79f30b9c2ccf5f8426dbc":{"kind":"REWORK","_number":15,"created":"2025-10-06 14:21:18.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/15"}}},"commit":{"parents":[{"commit":"5ca4e5190a7921f019b2c9101e1d5374a9f8ac80","subject":"Merge \"Fix share server migration with replica test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5ca4e5190a7921f019b2c9101e1d5374a9f8ac80"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-10-06 14:21:12.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nPositive test -\n 1. create barbican secret and create share using secret UUID\n\nNegative test -\n 1. invalid encryption key ref\n 2. invalid share type extra-spec\n 3. absent encryption key ref\n\npartially-implements: blueprint share-encryption\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/45236fe6ef8356b8e1a79f30b9c2ccf5f8426dbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/45236fe6ef8356b8e1a79f30b9c2ccf5f8426dbc"}]},"branch":"refs/heads/master"},"69a2c74741d0e2d645276cdb1a1c7389f6dcc4a8":{"kind":"REWORK","_number":16,"created":"2025-10-06 15:24:31.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/16"}}},"commit":{"parents":[{"commit":"5ca4e5190a7921f019b2c9101e1d5374a9f8ac80","subject":"Merge \"Fix share server migration with replica test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5ca4e5190a7921f019b2c9101e1d5374a9f8ac80"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-10-06 15:24:25.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nPositive test -\n 1. create barbican secret and create share using secret UUID\n\nNegative test -\n 1. invalid encryption key ref\n 2. invalid share type extra-spec\n 3. absent encryption key ref\n\npartially-implements: blueprint share-encryption\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/69a2c74741d0e2d645276cdb1a1c7389f6dcc4a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/69a2c74741d0e2d645276cdb1a1c7389f6dcc4a8"}]},"branch":"refs/heads/master"},"536c2c1df864a122559432c03553fdd341fda9c0":{"kind":"REWORK","_number":17,"created":"2025-10-06 15:49:15.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/17"}}},"commit":{"parents":[{"commit":"5ca4e5190a7921f019b2c9101e1d5374a9f8ac80","subject":"Merge \"Fix share server migration with replica test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5ca4e5190a7921f019b2c9101e1d5374a9f8ac80"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-10-06 15:48:26.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nPositive test -\n 1. create barbican secret and create share using secret UUID\n\nNegative test -\n 1. invalid encryption key ref\n 2. invalid share type extra-spec\n 3. absent encryption key ref\n\npartially-implements: blueprint share-encryption\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/536c2c1df864a122559432c03553fdd341fda9c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/536c2c1df864a122559432c03553fdd341fda9c0"}]},"branch":"refs/heads/master"},"8c6f185b0b3c4e4786069a948f78c9ffa4ea9e12":{"kind":"REWORK","_number":18,"created":"2025-10-06 16:26:06.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/18"}}},"commit":{"parents":[{"commit":"5ca4e5190a7921f019b2c9101e1d5374a9f8ac80","subject":"Merge \"Fix share server migration with replica test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5ca4e5190a7921f019b2c9101e1d5374a9f8ac80"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-10-06 16:25:24.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nPositive test -\n 1. create barbican secret and create share using secret UUID\n\nNegative test -\n 1. invalid encryption key ref\n 2. invalid share type extra-spec\n 3. absent encryption key ref\n\npartially-implements: blueprint share-encryption\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8c6f185b0b3c4e4786069a948f78c9ffa4ea9e12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8c6f185b0b3c4e4786069a948f78c9ffa4ea9e12"}]},"branch":"refs/heads/master"},"7d6a4180ce45a7995ba4dd99712ea21fbfab2ca9":{"kind":"REWORK","_number":19,"created":"2025-10-06 17:10:02.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/19"}}},"commit":{"parents":[{"commit":"5ca4e5190a7921f019b2c9101e1d5374a9f8ac80","subject":"Merge \"Fix share server migration with replica test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5ca4e5190a7921f019b2c9101e1d5374a9f8ac80"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-10-06 17:09:57.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nPositive test -\n 1. create barbican secret and create share using secret UUID\n\nNegative test -\n 1. invalid encryption key ref\n 2. invalid share type extra-spec\n 3. absent encryption key ref\n\npartially-implements: blueprint share-encryption\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7d6a4180ce45a7995ba4dd99712ea21fbfab2ca9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7d6a4180ce45a7995ba4dd99712ea21fbfab2ca9"}]},"branch":"refs/heads/master"},"d7e189a76c1d78a407b629f698d33cdd313bc3b4":{"kind":"REWORK","_number":20,"created":"2025-10-10 08:57:42.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/20"}}},"commit":{"parents":[{"commit":"496237a646a3cff808a52bce540356029f69c5b0","subject":"Merge \"Remove remaining job with Ubuntu Jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/496237a646a3cff808a52bce540356029f69c5b0"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-10-10 08:57:36.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nPositive test -\n 1. create barbican secret and create share using secret UUID\n\nNegative test -\n 1. invalid encryption key ref\n 2. invalid share type extra-spec\n 3. absent encryption key ref\n\npartially-implements: blueprint share-encryption\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d7e189a76c1d78a407b629f698d33cdd313bc3b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d7e189a76c1d78a407b629f698d33cdd313bc3b4"}]},"branch":"refs/heads/master"},"e933314cd9960dace3eeb52fa9a5f8234f917b85":{"kind":"REWORK","_number":21,"created":"2025-10-10 09:57:26.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/21"}}},"commit":{"parents":[{"commit":"496237a646a3cff808a52bce540356029f69c5b0","subject":"Merge \"Remove remaining job with Ubuntu Jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/496237a646a3cff808a52bce540356029f69c5b0"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-10-10 09:57:16.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nPositive test -\n 1. create barbican secret and create share using secret UUID\n\nNegative test -\n 1. invalid encryption key ref\n 2. invalid share type extra-spec\n 3. absent encryption key ref\n\npartially-implements: blueprint share-encryption\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e933314cd9960dace3eeb52fa9a5f8234f917b85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e933314cd9960dace3eeb52fa9a5f8234f917b85"}]},"branch":"refs/heads/master"},"f6bec50e044cebc02d0bcc83b3ecb5c851dbfb93":{"kind":"REWORK","_number":22,"created":"2025-10-10 10:52:40.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/22"}}},"commit":{"parents":[{"commit":"496237a646a3cff808a52bce540356029f69c5b0","subject":"Merge \"Remove remaining job with Ubuntu Jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/496237a646a3cff808a52bce540356029f69c5b0"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-10-10 10:52:22.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nPositive test -\n 1. create barbican secret and create share using secret UUID\n\nNegative test -\n 1. invalid encryption key ref\n 2. invalid share type extra-spec\n 3. absent encryption key ref\n\npartially-implements: blueprint share-encryption\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/963685\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f6bec50e044cebc02d0bcc83b3ecb5c851dbfb93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f6bec50e044cebc02d0bcc83b3ecb5c851dbfb93"}]},"branch":"refs/heads/master"},"ccde0062c53b23787b4cbe538c320a13cbeba8a8":{"kind":"REWORK","_number":23,"created":"2025-10-10 12:17:12.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/23"}}},"commit":{"parents":[{"commit":"496237a646a3cff808a52bce540356029f69c5b0","subject":"Merge \"Remove remaining job with Ubuntu Jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/496237a646a3cff808a52bce540356029f69c5b0"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-10-10 12:17:06.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nPositive test -\n 1. create barbican secret and create share using secret UUID\n\nNegative test -\n 1. invalid encryption key ref\n 2. invalid share type extra-spec\n 3. absent encryption key ref\n\npartially-implements: blueprint share-encryption\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/963685\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ccde0062c53b23787b4cbe538c320a13cbeba8a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ccde0062c53b23787b4cbe538c320a13cbeba8a8"}]},"branch":"refs/heads/master"},"c72fc1f7654f8a978358cff63b8886b99f0b8862":{"kind":"REWORK","_number":24,"created":"2025-10-10 13:04:00.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/24"}}},"commit":{"parents":[{"commit":"496237a646a3cff808a52bce540356029f69c5b0","subject":"Merge \"Remove remaining job with Ubuntu Jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/496237a646a3cff808a52bce540356029f69c5b0"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-10-10 12:57:13.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nPositive test -\n 1. create barbican secret and create share using secret UUID\n\nNegative test -\n 1. invalid encryption key ref\n 2. invalid share type extra-spec\n 3. absent encryption key ref\n\npartially-implements: blueprint share-encryption\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/963685\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c72fc1f7654f8a978358cff63b8886b99f0b8862"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c72fc1f7654f8a978358cff63b8886b99f0b8862"}]},"branch":"refs/heads/master"},"9174b92504faa4ca33526dec1ae828c9cbfa8767":{"kind":"REWORK","_number":25,"created":"2025-10-10 13:50:19.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/25"}}},"commit":{"parents":[{"commit":"496237a646a3cff808a52bce540356029f69c5b0","subject":"Merge \"Remove remaining job with Ubuntu Jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/496237a646a3cff808a52bce540356029f69c5b0"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-10-10 13:50:05.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nPositive test -\n 1. create barbican secret and create share using secret UUID\n\nNegative test -\n 1. invalid encryption key ref\n 2. invalid share type extra-spec\n 3. absent encryption key ref\n\npartially-implements: blueprint share-encryption\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/963685\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9174b92504faa4ca33526dec1ae828c9cbfa8767"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9174b92504faa4ca33526dec1ae828c9cbfa8767"}]},"branch":"refs/heads/master"},"fe868063d757ee4d4303472c753e320679eaed23":{"kind":"REWORK","_number":26,"created":"2025-10-10 14:39:44.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/68/952368/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/952368/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/952368/26"}}},"commit":{"parents":[{"commit":"496237a646a3cff808a52bce540356029f69c5b0","subject":"Merge \"Remove remaining job with Ubuntu Jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/496237a646a3cff808a52bce540356029f69c5b0"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-06-11 13:33:02.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-10-10 14:39:38.000000000","tz":0},"subject":"Add tests for encryption_key_ref","message":"Add tests for encryption_key_ref\n\nPositive test -\n 1. create barbican secret and create share using secret UUID\n\nNegative test -\n 1. invalid encryption key ref\n 2. invalid share type extra-spec\n 3. absent encryption key ref\n\npartially-implements: blueprint share-encryption\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/963685\nChange-Id: I3145f9cd6847464b2920f1b0a35e6c211e45b26e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fe868063d757ee4d4303472c753e320679eaed23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fe868063d757ee4d4303472c753e320679eaed23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
