)]}'
{"id":"openstack%2Fmanila-tempest-plugin~952708","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Iefc555162037b25e5b76d63b60f0ffb82896c3c9","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-01-07 22:47:30.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2026-01-07 21:37:31.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2026-01-07 22:47:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iefc555162037b25e5b76d63b60f0ffb82896c3c9","subject":"Refactor stable branch jobs","status":"MERGED","created":"2025-06-16 19:13:31.000000000","updated":"2026-01-07 22:47:30.000000000","submitted":"2026-01-07 22:47:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952708","meta_rev_id":"18a5031a34a6e7138447a40d964a99ac824423f4","_number":952708,"virtual_id_number":952708,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-07 22:47:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2026-01-07 18:33:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2026-01-07 21:37:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2026-01-07 21:37:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-16 19:13:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-07 18:33:55.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-01-07 19:39:03.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2026-01-07 19:39:03.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"842a1d5f451256663b0047098f61fd6612772ed8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-06-16 19:13:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5f6b6aa26467c359b1b03b1c56daa5f6a76ef60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 19:13:42.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/manila-tempest-plugin on branch master.  The\nproblem was:\n\n  found undefined alias\n    in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 508, column 15","accounts_in_message":[],"_revision_number":1},{"id":"247f5408d5b2d007642f20577f58c5d2b467c90c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-06-16 19:14:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a269db86b9b632bef05861fa72e6ed1d21960d79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 21:34:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c88bd65a48e54328a2f9e4f501d2c408\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df21229d640f41848472221ec7c0c681 : SUCCESS in 2m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/86f9ae6568be4122877bb00280153b36 : SUCCESS in 32m 31s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/598e3bfd66d641729961557b648271a9 : SUCCESS in 2h 06m 05s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/dffec33fed734d26ba1dbfdf85ab9a81 : FAILURE in 56m 43s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/7e30441d51d04e21870df66517e3ed47 : SUCCESS in 36m 00s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/165a3222bd8645da97f20c0360ee940d : SUCCESS in 1h 12m 03s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7910df2eadcc42269e7eaa7d0ab40691 : SUCCESS in 45m 06s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ab810d3245c9499db1e60f9330c87b00 : SUCCESS in 22m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2f7b71836421438ab121ddcb5a6b88ee : SUCCESS in 29m 42s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/313822e8cb9f4fc49e9c24033c45206e : SUCCESS in 40m 23s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/3bcd5188af0f489bbe2439ee83d633ed : SUCCESS in 54m 00s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/5d3a4bd7420f4b53a992e7385fa35023 : SUCCESS in 40m 02s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/c528426bf3a949d99c5e43724a862a36 : SUCCESS in 27m 52s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a8b8aba9c6c74233af8c353a522a4fdc : FAILURE in 21m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dabc4e145d335f52c3908a2fef95bb0cdbab1537","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-12-30 13:21:31.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"7a1aa7a381a0a5d366f37615266974482e0f4df2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 14:45:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/541d48341b034cb2aab2d9aedc880a1c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/feed05365dfc4f0c85f2af13da640cf4 : SUCCESS in 2m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f9a4e12e920445c1af834e32eb6075dc : SUCCESS in 41m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/72b06d4e30eb4415a98dec5dc98cb002 : SUCCESS in 1h 07m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c5aff5d91f2c4ec0bbaf2213c8b8d38b : FAILURE in 1h 22m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/636469664c50467a8a516ad644e1131d : SUCCESS in 1h 05m 09s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/5c94baf79d8e43c8aeee150a509f80bb : SUCCESS in 50m 45s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d167b54827b3478e892ac01867df1157 : SUCCESS in 1h 16m 23s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cb1ff2f6b2154db88f16e3859a309283 : SUCCESS in 32m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/931bd021633b4b90967f6331c0474bf6 : SUCCESS in 42m 57s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/d8aacb59fbf34cfea2265ce2590bf2f0 : SUCCESS in 53m 19s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/dda4bc7edf484e61a7d85f3b73846c02 : SUCCESS in 1h 12m 55s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/cde6f0f227b84507aa9b556f14364d79 : SUCCESS in 1h 05m 34s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d0a5f912a9a4481f9cf56428412c9cd8 : SUCCESS in 32m 32s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ac20dda1183d446f8f7edf2b6682644a : FAILURE in 41m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"41d0965c733465cf7046cdb55bdf90a056cde76c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-01-06 15:34:41.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"7cb47ae8bc4968b32d69859519d67407ade79773","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 18:04:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcb76c401d9a4d81b869e8cf07311f0a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf876205ab0e4cc896d1bf4f287b31f1 : SUCCESS in 1m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/00a15d75450c4db2bef83bc729947044 : SUCCESS in 1h 08m 33s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c6d3b86598ff4b6aaabb6d27a0bd0945 : FAILURE in 2h 21m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8569efc9e343446d9ea95a0d0c5ceb93 : FAILURE in 24m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f81dfb625a584449b4e61d9d186733ad : FAILURE in 1h 30m 22s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ef84bef0aebb486caa6b8e0bc910d6f6 : SUCCESS in 1h 33m 29s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/37581df6fea4435bb50bdddee247e07d : SUCCESS in 1h 12m 26s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fbd502220d5647d19fedab7068aeed93 : SUCCESS in 21m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1d08756876524226b19b17f0d9cb6ef7 : SUCCESS in 42m 24s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/e3e8f75aacb44492861759559e327a2a : SUCCESS in 1h 19m 26s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/4a3d15bbdb164b3a9ff4db18d2f3228b : SUCCESS in 1h 28m 41s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/c95c7e0b583b43c69846f03d627d0b49 : SUCCESS in 34m 17s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/bae3515856a640d699ec7ff09e6bd6d7 : SUCCESS in 20m 25s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0a47607971f247ecb1bd5bd078e9b9e8 : FAILURE in 27m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"413f0016e40216cee4c14f45a0bedce3011ee645","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-07 18:33:55.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e19d724b7e3aa30f3a74eecf2fb8f886b9835045","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2026-01-07 21:37:31.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a7aee0a24b06341c4294d4f628aee7a6480e2b94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 21:38:22.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0e98d37d61e49779059bf76a547ae07b304674ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 22:47:29.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5a770ce1ea8493aa025820dffd1b7aa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b43df9532a8d498280713cfa2de199b5 : SUCCESS in 2m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/832c3644531449bf9f4b33eb0d7f426a : SUCCESS in 56m 17s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/eb1f74ffcea944fda507feda768bec45 : SUCCESS in 33m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8ccbecd956e44aa2ad921c411d43bba9 : SUCCESS in 40m 09s","accounts_in_message":[],"_revision_number":4},{"id":"18a5031a34a6e7138447a40d964a99ac824423f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 22:47:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f0984572701b18b413de283ccecbe340aa048beb","revisions":{"36d7d8249ac913166f55eec0e95ee7562258716a":{"kind":"REWORK","_number":1,"created":"2025-06-16 19:13:31.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/08/952708/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/08/952708/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/08/952708/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/08/952708/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/08/952708/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/08/952708/1"}}},"commit":{"parents":[{"commit":"326d5a399168c2a53ec01bf4f74dc6c4dad8eaf3","subject":"Merge \"Add CEPHFS filesystem metadata verifications\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/326d5a399168c2a53ec01bf4f74dc6c4dad8eaf3"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-06-16 19:10:16.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-06-16 19:13:12.000000000","tz":-180},"subject":"[WIP] Refactor stable branch jobs","message":"[WIP] Refactor stable branch jobs\n\nChange-Id: Iefc555162037b25e5b76d63b60f0ffb82896c3c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/36d7d8249ac913166f55eec0e95ee7562258716a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/36d7d8249ac913166f55eec0e95ee7562258716a"}]},"branch":"refs/heads/master"},"e75478f1c1eaee032265a1b9dfce5bb2696a2d79":{"kind":"REWORK","_number":2,"created":"2025-06-16 19:14:43.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/08/952708/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/08/952708/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/08/952708/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/08/952708/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/08/952708/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/08/952708/2"}}},"commit":{"parents":[{"commit":"326d5a399168c2a53ec01bf4f74dc6c4dad8eaf3","subject":"Merge \"Add CEPHFS filesystem metadata verifications\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/326d5a399168c2a53ec01bf4f74dc6c4dad8eaf3"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-06-16 19:10:16.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-06-16 19:14:34.000000000","tz":-180},"subject":"[WIP] Refactor stable branch jobs","message":"[WIP] Refactor stable branch jobs\n\nChange-Id: Iefc555162037b25e5b76d63b60f0ffb82896c3c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e75478f1c1eaee032265a1b9dfce5bb2696a2d79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e75478f1c1eaee032265a1b9dfce5bb2696a2d79"}]},"branch":"refs/heads/master"},"7742b5d3a22c6157373339f0d1cdd0c991084b6c":{"kind":"REWORK","_number":3,"created":"2025-12-30 13:21:31.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/08/952708/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/08/952708/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/08/952708/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/08/952708/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/08/952708/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/08/952708/3"}}},"commit":{"parents":[{"commit":"a19b4387a9a52e8856c20fe39d906195c27bcc48","subject":"Merge \"Drop testing with PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a19b4387a9a52e8856c20fe39d906195c27bcc48"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-06-16 19:10:16.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-12-30 13:21:16.000000000","tz":-180},"subject":"[WIP] Refactor stable branch jobs","message":"[WIP] Refactor stable branch jobs\n\nChange-Id: Iefc555162037b25e5b76d63b60f0ffb82896c3c9\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7742b5d3a22c6157373339f0d1cdd0c991084b6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7742b5d3a22c6157373339f0d1cdd0c991084b6c"}]},"branch":"refs/heads/master"},"f0984572701b18b413de283ccecbe340aa048beb":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-01-06 15:34:41.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/08/952708/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/08/952708/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/08/952708/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/08/952708/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/08/952708/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/08/952708/4"}}},"commit":{"parents":[{"commit":"a19b4387a9a52e8856c20fe39d906195c27bcc48","subject":"Merge \"Drop testing with PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a19b4387a9a52e8856c20fe39d906195c27bcc48"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-06-16 19:10:16.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2026-01-06 15:34:41.000000000","tz":0},"subject":"Refactor stable branch jobs","message":"Refactor stable branch jobs\n\nRemove older stable branch jobs and variables that relied\non older Ubuntu images. Most of these branches are now EOL\u0027d,\nso we can remove the testing for them.\n\nChange-Id: Iefc555162037b25e5b76d63b60f0ffb82896c3c9\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f0984572701b18b413de283ccecbe340aa048beb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f0984572701b18b413de283ccecbe340aa048beb"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
