)]}'
{"id":"openstack%2Fmanila-tempest-plugin~954164","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ie62a19689529eb2de94096e5f757bfcfbe372040","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2025-07-08 19:17:56.000000000","reason":"Change was submitted"},"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2025-07-08 19:17:56.000000000","reason":"Change was submitted"},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2025-07-08 17:15:44.000000000","reason":"\u003cGERRIT_ACCOUNT_32919\u003e replied on the change","reason_account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2025-07-08 17:12:35.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-07-08 19:17:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie62a19689529eb2de94096e5f757bfcfbe372040","subject":"Remove unicode prefix from code","status":"MERGED","created":"2025-07-04 16:57:31.000000000","updated":"2025-07-08 21:34:00.000000000","submitted":"2025-07-08 19:17:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954164","meta_rev_id":"915a4d92ee833cb7afea47a9defd6c84801ea7ce","_number":954164,"virtual_id_number":954164,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-08 19:17:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2025-07-08 17:15:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":2,"date":"2025-07-08 17:12:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":1,"date":"2025-07-08 17:30:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-04 18:02:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-05 02:17:49.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-07-06 12:26:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2025-07-06 12:26:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2025-07-08 16:05:49.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2025-07-08 16:26:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"}],"messages":[{"id":"a02bc18b71825cd94caae36b4e44ac48c70fcf60","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-04 16:57:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"076baf06159d34de7ebbfe4b2bea8e6bafddafdb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-04 17:07:53.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4a791de9ed4ec8263f914c4e9f66d87e5f719679","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-04 17:07:56.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"a318f9d1bbecb5e6b2ad17e4dbb29f8cc8b3ac03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 18:02:22.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3d189f3071e420b83a9d9cc88f0f8dc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e606a3c4d93b434aabd8ab2076011f5c : FAILURE in 1m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7160ebab6cae4dea9a289481b0779812 : FAILURE in 29m 31s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/fd007c944b7246f6accb8d1bfe03f8c3 : FAILURE in 30m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d67655d39ed643deaa73121de4d5d19e : FAILURE in 32m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/4f7f937f1e53440f87108a3861b5e19c : FAILURE in 48m 35s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/83b90d51de6448a9a3a59e51000278c2 : FAILURE in 17m 45s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/41559608103a4049a7b12608e00f5bb2 : FAILURE in 14m 03s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/08589b5ead8c4635a82349f4b8d4e542 : FAILURE in 7m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/10db651dda394184b6c9681bedfdfc70 : FAILURE in 7m 43s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/04e05e60b2c849d6835afa87db02017f : FAILURE in 28m 33s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/1d7db6f419014270955edd2d4e351464 : FAILURE in 26m 02s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/1130b9669e164cbebfc61b0c2e45d8fd : FAILURE in 28m 36s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/5bc35063616f4a7fa11a1090f1eff04e : FAILURE in 19m 47s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5c76a54d6c2b48f5bcbd89ac2c5cd582 : FAILURE in 15m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"02df47a1550c261d8e22d280eb9f3645e7f6fa84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-04 20:48:39.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"109728c78186a123b78e917fec2d1543b7e9922b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 23:16:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3da1174de4e84d6cb072ac7530903891\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ed2b5878c434a518b3f438f788a8562 : SUCCESS in 3m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d4f2b9b607e14f30b732df32f9d7192d : SUCCESS in 1h 27m 12s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/764bc84cf09c4654ab26ef53e325c5f9 : SUCCESS in 2h 23m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2f34bb1b5a6c49338b5671222074b504 : FAILURE in 1h 21m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/54137670b57b42d78442c76831246928 : FAILURE in 1h 04m 06s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/477cc8b3abc34106a8b2c93bee5d3e6f : SUCCESS in 1h 38m 37s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d34bd47ec3834e80ab1ac11431efbe37 : FAILURE in 19m 23s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/dfe8679e51ef47c69f221185d9bdd617 : SUCCESS in 34m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c2914f63689a41e9a5f546fc0c16a9fa : SUCCESS in 35m 18s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/8c2657747de94eb1b281739e67b105e3 : SUCCESS in 1h 09m 25s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/f79b2d2f20de4e11b0c18c9c6f7f6f2e : SUCCESS in 1h 12m 33s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/b5508bbd00014ead86e429cf07994ce3 : SUCCESS in 1h 10m 29s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3fb97212090744b1abdbfad42facf5af : SUCCESS in 30m 30s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/55e8f1e4bf9d46e5acea02d6351cd6e1 : FAILURE in 24m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2fa87daae55cb07df81fc3869dff521620de8b3d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-05 02:17:49.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"12f6e57c46a80d071d5c415fc50fbb67e993fc8e","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-07-08 16:05:49.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"206627d278a50e0f9046a33883b0c78ffd29762b","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2025-07-08 16:15:53.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ae3b5ff9f37d3761e0dec604f51041708c2b16a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-08 16:19:04.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"efe27db917ad79bc8da2604cec5be93fb8674d8f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-08 16:19:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"765f692480b0936bf4185757fc7136de1306e31b","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2025-07-08 17:12:35.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"06a2e053c72fda8ad706201c4834831fd1b180f1","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-07-08 17:15:44.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"bc91d2964f6f2f862d8c0dfb63d23447fe9169a7","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2025-07-08 17:30:06.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"35318d545335b82b6f548bb490289fe0296b444b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 18:32:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cab408eec01a4e1f890e2e1de8bf83a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e64fc0bf2ca49bc922108ae9fff208c : SUCCESS in 4m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a132f73daf9f4bc8a21fea4ea6e74df7 : SUCCESS in 1h 13m 24s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d4b7c46497984502a5bf5dd3c5ff9f91 : SUCCESS in 2h 05m 42s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/343bff4a22044e9c8cf1c9ad2e89b57b : FAILURE in 1h 13m 20s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a051da8bc0ba45ef8de8541e0962d5f5 : SUCCESS in 37m 15s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a85b570c7fa54c548f03997570175e4f : SUCCESS in 1h 19m 28s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3e0af4f4bddb41f6ad20ec0143d00b04 : FAILURE in 10m 04s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/93b3577873714d28a1bbe44d88b05955 : SUCCESS in 30m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/abd024449bea473fbe377c41e26293a4 : SUCCESS in 38m 47s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/a009ced6bfb746918e1bc61c9470f432 : SUCCESS in 31m 26s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/ec17baec5d534c10ae491b9f37fcaf07 : SUCCESS in 1h 00m 37s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/e50ec4830a8d4c859448b0aa6a3eb066 : SUCCESS in 1h 04m 29s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/8faa9c4e87684cf8ad4123df77f60c39 : SUCCESS in 30m 21s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/32c54d70e02645d0b414f583106a04bc : FAILURE in 10m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"799567f55ddd5b55718bc4dc964446563051054e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 18:32:42.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e0a86dd220ae5c84de8f01369f43d3d00ca84ac3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 19:17:56.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7b0c2592d5e459c92881262dd02608c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b72ba5786ce453bb679ed43d123255a : SUCCESS in 2m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/08c0728c747a46498dbd64096fd90a7f : SUCCESS in 31m 33s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/97c0531ce10a410598090a496caebca2 : SUCCESS in 25m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/21574005dd964857903f3d6e62265a1d : SUCCESS in 42m 57s","accounts_in_message":[],"_revision_number":4},{"id":"83b057ed217e8e6caf0cde7e89ca29f347968349","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 19:17:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"915a4d92ee833cb7afea47a9defd6c84801ea7ce","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-08 21:34:00.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"885fd8a9e1609e469e502e0b1902e8bb591bc6f4","revisions":{"3cf5ecae1c80a92ed2c9bebdd8e8aa6906135dc4":{"kind":"REWORK","_number":1,"created":"2025-07-04 16:57:31.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/64/954164/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/64/954164/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/954164/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/954164/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/954164/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/954164/1"}}},"commit":{"parents":[{"commit":"ff055a60ad38063826e0169d6b346f4a573e92cd","subject":"Merge \"Add tests for Share Network Subnet Metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ff055a60ad38063826e0169d6b346f4a573e92cd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-04 16:57:31.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-04 16:57:31.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ie62a19689529eb2de94096e5f757bfcfbe372040\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3cf5ecae1c80a92ed2c9bebdd8e8aa6906135dc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3cf5ecae1c80a92ed2c9bebdd8e8aa6906135dc4"}]},"branch":"refs/heads/master"},"dea460fe917a480cb214791dfb55466031397511":{"kind":"REWORK","_number":2,"created":"2025-07-04 17:07:53.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/64/954164/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/64/954164/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/954164/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/954164/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/954164/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/954164/2"}}},"commit":{"parents":[{"commit":"ff055a60ad38063826e0169d6b346f4a573e92cd","subject":"Merge \"Add tests for Share Network Subnet Metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ff055a60ad38063826e0169d6b346f4a573e92cd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-04 16:57:31.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-04 17:07:53.000000000","tz":0},"subject":"Remove unicode prefix from code","message":"Remove unicode prefix from code\n\nAll strings are considered as unicode string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: Ie62a19689529eb2de94096e5f757bfcfbe372040\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dea460fe917a480cb214791dfb55466031397511"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dea460fe917a480cb214791dfb55466031397511"}]},"branch":"refs/heads/master"},"86b7ba0218ea0f9d5d5797c488b102b6aef27a8e":{"kind":"REWORK","_number":3,"created":"2025-07-04 20:48:39.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/64/954164/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/64/954164/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/954164/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/954164/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/954164/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/954164/3"}}},"commit":{"parents":[{"commit":"ff055a60ad38063826e0169d6b346f4a573e92cd","subject":"Merge \"Add tests for Share Network Subnet Metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ff055a60ad38063826e0169d6b346f4a573e92cd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-04 16:57:31.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-04 20:48:39.000000000","tz":0},"subject":"Remove unicode prefix from code","message":"Remove unicode prefix from code\n\nAll strings are considered as unicode string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: Ie62a19689529eb2de94096e5f757bfcfbe372040\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/86b7ba0218ea0f9d5d5797c488b102b6aef27a8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/86b7ba0218ea0f9d5d5797c488b102b6aef27a8e"}]},"branch":"refs/heads/master"},"885fd8a9e1609e469e502e0b1902e8bb591bc6f4":{"kind":"REWORK","_number":4,"created":"2025-07-08 16:19:04.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/64/954164/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/64/954164/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/954164/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/954164/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/954164/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/954164/4"}}},"commit":{"parents":[{"commit":"ff055a60ad38063826e0169d6b346f4a573e92cd","subject":"Merge \"Add tests for Share Network Subnet Metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ff055a60ad38063826e0169d6b346f4a573e92cd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-04 16:57:31.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-08 16:19:04.000000000","tz":0},"subject":"Remove unicode prefix from code","message":"Remove unicode prefix from code\n\nAll strings are considered as unicode string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: Ie62a19689529eb2de94096e5f757bfcfbe372040\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/885fd8a9e1609e469e502e0b1902e8bb591bc6f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/885fd8a9e1609e469e502e0b1902e8bb591bc6f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
