)]}'
{"id":"openstack%2Fmanila-tempest-plugin~955394","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I10bdd54482a1c0d145f49cc3a095e880242d43f8","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bp/share-encryption","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2025-08-04 17:26:51.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"38162":{"account":{"_account_id":38162,"deleted":true},"last_update":"2025-09-11 21:24:48.000000000","reason":"Change was submitted"},"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2025-09-11 21:24:48.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-08-04 18:36:41.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"35507":{"account":{"_account_id":35507,"name":"Kiran Pawar","email":"kiranpawarpict2010@gmail.com","username":"kpawar"},"last_update":"2025-07-24 08:11:11.000000000","reason":"\u003cGERRIT_ACCOUNT_36179\u003e replied on the change","reason_account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2025-09-11 21:24:48.000000000","reason":"Change was submitted"},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2025-08-04 17:12:02.000000000","reason":"\u003cGERRIT_ACCOUNT_32919\u003e replied on the change","reason_account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2025-08-13 15:40:49.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2025-09-11 21:24:48.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2025-09-11 21:24:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I10bdd54482a1c0d145f49cc3a095e880242d43f8","subject":"Add barbican to dummy driver job","status":"MERGED","created":"2025-07-18 17:24:21.000000000","updated":"2025-09-11 21:24:48.000000000","submitted":"2025-09-11 21:24:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"955394-bp/share-encryption","meta_rev_id":"63935cd4af88709379a17972bfd353a51ebf01f8","_number":955394,"virtual_id_number":955394,"owner":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-11 21:24:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2025-08-04 17:26:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2025-08-04 18:36:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":1,"date":"2025-08-04 17:12:02.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":2,"date":"2025-08-13 15:40:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2025-08-04 18:36:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":1,"date":"2025-08-13 15:40:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":38162,"deleted":true},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":38162,"deleted":true}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-18 17:24:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-18 21:01:33.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2025-07-22 13:55:13.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-07-24 07:56:06.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":35507,"name":"Kiran Pawar","email":"kiranpawarpict2010@gmail.com","username":"kpawar"},"state":"REVIEWER"},{"updated":"2025-07-24 07:56:06.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":38162,"deleted":true},"state":"REVIEWER"},{"updated":"2025-07-24 07:56:06.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2025-07-24 07:56:06.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2025-07-24 07:56:06.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2025-07-24 07:56:06.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2025-07-24 08:11:11.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":35507,"name":"Kiran Pawar","email":"kiranpawarpict2010@gmail.com","username":"kpawar"},"state":"REMOVED"},{"updated":"2025-07-24 08:11:11.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"},{"updated":"2025-07-28 17:37:21.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2025-07-31 04:43:22.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2025-08-04 18:36:41.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"28485f418a53b120650f0588b6641aab6f7f145f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-18 17:24:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54c61cc46a4fabc70f0147edff41c9830ee6cada","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 17:24:37.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/manila-tempest-plugin on branch master.  The\nproblem was:\n\n  extra keys not allowed @ data[\u0027devstack_plugins\u0027]\n\nThe problem appears in the the \"manila-tempest-plugin-dummy-dhss\" job stanza:\n\n  job:\n      name: manila-tempest-plugin-dummy-dhss\n      description: Test the Dummy driver with DHSS\u003dTrue\n      parent: manila-tempest-plugin-standalone-base\n      required-projects:\n        - openstack/barbican\n      devstack_plugins:\n        barbican: https://opendev.org/openstack/barbican\n      vars:\n  ...\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 701","accounts_in_message":[],"_revision_number":1},{"id":"8715774994cd3078530885cc387dcdbafb8c68a1","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-18 17:28:55.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"61242c92b5930330886b246fec5378f2c5ad5b44","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-18 17:39:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"960b11a761941170af689b1d9f8ec317afe35491","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 20:49:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1740b960f2c34eada17773cad5fb6527\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6e07275094f47bfa4b230bdaacff06b : SUCCESS in 2m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0245a9246f1449dc9500b968547ccf90 : SUCCESS in 1h 09m 03s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/6962aa0e4dec4585833a991e18f74f2a : TIMED_OUT in 3h 02m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/086268260e4d4d7f9113b6706f20f582 : FAILURE in 1h 08m 41s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/09c2b334a944448c80dcbc8772a41574 : SUCCESS in 50m 26s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a6b51f8c41484d0b9a387ef4e72324dd : SUCCESS in 42m 05s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/89a5e3ecfff445598498a2d142a3b1b2 : FAILURE in 15m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/980dcd73a129443ca90b0f20a8abdf58 : SUCCESS in 21m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2797199b122c4302bf1f1bfc0d47ad2c : SUCCESS in 28m 27s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/23ca630f7d1d493b9091aa18a380e4c2 : SUCCESS in 1h 24m 10s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/0204445aa6194db1b5e9c43340a38f3b : SUCCESS in 1h 10m 26s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/83aeb7259ad14a72b4d4f17e1045cb35 : SUCCESS in 31m 20s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/5c17210a79d04857848119f3d54f71a8 : SUCCESS in 20m 51s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5d0762b921dc424da8209487329604a6 : FAILURE in 17m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4ffdd91613e7f5f8b120f53451e076746e742bb4","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-07-18 21:01:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"333264f233e63e9f88d64ef04aa49f5493169af7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-21 06:30:31.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f75f6d4995c248a77e5b7732b3a761c80b8af04f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 06:43:05.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/955393 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"31f9e307148a5264a45cabe75168c8a7bfca08d0","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-21 12:36:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"be1bbac7893ee1be49480b01d2b934e1fbec9fe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 14:57:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b9f70f251c5497ba32ae9d0061fa3db\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51f97f5ea36941e0aa745744335ec2a5 : SUCCESS in 3m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ad7082fd9d8d4e3ca6ea1c326a5df08c : SUCCESS in 59m 30s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2318164d647f49208a6a934508a0eeab : SUCCESS in 2h 15m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2b05b73b05a4482b9519a05b55e24762 : FAILURE in 58m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/42c370f3fcba4b1f9aaf4479dbb7fa98 : SUCCESS in 1h 03m 30s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/498474412a804acc9a7f7bf147f2fd15 : SUCCESS in 1h 32m 03s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1bc0aa0e00e84fbfaa8b76d0eeafbe5e : FAILURE in 21m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/84f6c109416841b1904619dc93b254ba : SUCCESS in 33m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b08f2b157df34c8cbd5e86ac4ab91c7d : SUCCESS in 31m 51s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/f4b21123656345159c6695b505ad6e00 : SUCCESS in 1h 13m 24s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/eb2ff8a20e9547da9cce3f29836ad416 : SUCCESS in 1h 08m 25s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/6f59eaf5197245eeb4e4432c506912cd : SUCCESS in 1h 14m 57s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/66b30f18c680415bad20444af3f270aa : SUCCESS in 26m 21s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/216dac273d134324b390a24a54287dac : FAILURE in 22m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ef5b550e4303356d53a1f61f3e1ff66c3e379e8c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-22 13:55:13.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"26642b2e7162e48945cdec0c7f226bfed8c292b7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-23 04:06:24.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"78c00c493fd2822f88fd10bc0abbf34710d55eeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 05:55:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0d62768d3f14394b22ed1423cd8d5d3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c3535c7ec2943e4b9084d6c1c2b7666 : SUCCESS in 3m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9fbd3eed3b1d45128cc298d1106de707 : SUCCESS in 1h 03m 56s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a308703b462941f991f113c766cc2eed : FAILURE in 1h 47m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8c2a5d3e406345399f88af3abc145693 : FAILURE in 43m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3a61d9241e7046a0bc2ac4a7c31c8f04 : SUCCESS in 51m 46s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f5c8f5dcb72143db8e5d536b613c7ea6 : SUCCESS in 1h 15m 02s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ff0c3e31810544d0838b4aa20951e2ec : FAILURE in 9m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/65aa7b1a53914866b3b7001b8097d9ea : SUCCESS in 33m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cd226c33eb494eb2968dd582340ed4d6 : SUCCESS in 30m 53s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/3a2eb316d3854929b013528f0f5e0eb1 : SUCCESS in 1h 10m 26s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/825cc9d33fa04129a3e6b8fbe528e552 : SUCCESS in 51m 29s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/5109e868d85d4f0a8c01825398ecd784 : SUCCESS in 1h 03m 42s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d7979c6c23dc4f0ca0c14441bec75f1c : SUCCESS in 21m 04s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e7caa4eb68b14899a956a53be98af407 : FAILURE in 9m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7e94c77f3f2f9cfac6149ba48537a73c5fb3876f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-23 15:32:49.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"08f4d2e4430a33fa49059d0d5297cb234f039396","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-23 15:34:04.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"31cc4488efaf6ca106afd4ed7c7caddaa3e507e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 17:48:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cd01a9cc7e7425786672d6abaf77ae4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2db8f363d1cf44baa905ef99f15002c6 : SUCCESS in 2m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b5045e1c52524d839d52b879d905238b : SUCCESS in 1h 10m 02s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/eed38bd5b6d6478db7aa434275e94c8f : FAILURE in 2h 12m 43s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d60c51e593d047a38a9d1151670673a6 : FAILURE in 1h 14m 51s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/5bdc6e2c9a414ce4aa6d8ece60a1a476 : SUCCESS in 1h 00m 46s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ba62c4c64bd74a91ae801d2e7e4ee4f1 : SUCCESS in 1h 30m 46s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f8d1063fcee8460d933f2c729112e17c : FAILURE in 22m 49s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9e34d8df0cb04ddda5ed6bb0e8d36e1b : SUCCESS in 30m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7b6e5c8d9b3b4d6e971dbdc97c317775 : SUCCESS in 38m 17s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/452677cb51e7414db78f118df595f642 : SUCCESS in 45m 52s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/7bb8187582b747dda8089e7eca1ec2d2 : SUCCESS in 1h 06m 35s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/8f0139cf7b8444fa893688aaccfaa000 : SUCCESS in 1h 07m 28s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/569b115a8e034bc6875ba65df19c4cf2 : SUCCESS in 30m 54s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/dd73489c638f4bb59d8be5f90724bb26 : FAILURE in 11m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c23ddb29abd81876e9d852ee7022ebbc7f4cc0f0","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-24 07:41:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c90eb7df78e271cd8e168b965d1f4c52ae47af71","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-24 07:42:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1605b3ecbeb2c57c820a9283ffc0e94211d50e0c","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-24 07:56:06.000000000","message":"Patch Set 5:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"405f9b6ecd616e33f01b86e6aab82a25db40b220","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-24 08:11:11.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_35507\u003e.","accounts_in_message":[{"_account_id":35507,"name":"Kiran Pawar","email":"kiranpawarpict2010@gmail.com","username":"kpawar"}],"_revision_number":5},{"id":"e436095364f8de5ecf6f7905779610a396cf7cd9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-24 12:19:28.000000000","message":"Topic set to bp/share-encryption","accounts_in_message":[],"_revision_number":5},{"id":"88717914b8e8c2ae9ee2d9316e24bfad81477a00","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-07-24 12:21:55.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"931399a82ae80c33918a3ffec6276ba2023053ad","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-24 14:15:25.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"6a4fe04571ab9c2d34db105d9766440998511a7a","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2025-07-28 17:37:21.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2787cec1a8773ca9cc0164ad04a24cba821d4caf","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2025-07-29 03:39:44.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"86db49493aad407bf7b501bf48d36dfb1f977d5d","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-07-30 12:05:21.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7c506c943655de47a18d10ac0030f71bceffc483","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-07-31 04:43:22.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"08852e14882c1b437cb19087145a9025d79cdaa5","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-07-31 09:43:03.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2fce3e23426c6c57903f6835dd28fe50d7b4c319","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-31 11:39:18.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5ddf739ad40c13e56db2c376effc9c9442b203d2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-07-31 14:58:48.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"8f7c4dc5842de37891e3eb3ea78f16b6f93843cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-08-04 14:32:45.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"f0c3d191dfc61cb1cf5341d1caf48b5ddacbdd34","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-08-04 14:34:13.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"6abde06aa1b2d75ae64ac01454801ceafc62166a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 17:11:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a95f4a34467436297a59465040284d9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5e6c1fb9c25461a9f820c988cda718f : SUCCESS in 1m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bbd6d2732a6b4301b4673060d3dfed0d : SUCCESS in 1h 34m 34s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c676ecf58f624b3c8c1f5268d0ea7ea4 : FAILURE in 2h 33m 07s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a619e8989432416fa8bc171686cd9f74 : FAILURE in 1h 09m 05s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ea3db4d7dbff46288099e99b788d7eea : SUCCESS in 1h 00m 23s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/426a7c60372b4b6da0031ca75d58aab6 : SUCCESS in 1h 34m 05s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9de2a9a6722c4c139f309a540d43ce02 : FAILURE in 20m 27s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e865588b0f3647359d4c9b5d4d52507a : SUCCESS in 29m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fc6adb93549c44119e53e17f9e455a99 : SUCCESS in 41m 33s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/10f4db752e5946abbb86e966f5d8db27 : SUCCESS in 1h 13m 01s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/bd5a02bf4e294aa29729d5499cab9d1a : SUCCESS in 55m 09s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/05231c92df6a400d9fa06ade693df38c : SUCCESS in 1h 01m 02s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3b0a4e56e9f04c4eb631273c50b06dab : SUCCESS in 31m 00s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d1825d217082478fa942d737a51de2e1 : FAILURE in 10m 36s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fb63bd48a7f1e31c4d6395446456de0939c957dd","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-08-04 17:12:02.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"30444703768897a44e149a250e617a7ca9fbf2ca","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-08-04 17:21:32.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/94/955394/6/upstream-check/manila-tempest-plugin-ontap-dhss/5efd9bd/ : RETRY_LIMIT in 4m 51s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/94/955394/6/upstream-check/manila-tempest-plugin-ontap-no-dhss/2259c18/ : RETRY_LIMIT in 8m 51s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"897abc1431abd98b3e4ffc5834e4988f7bdb4b91","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-08-04 17:26:51.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0a16626f0b3639b5e3212eec0fd7d24f0962bcf3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-04 18:36:41.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"2280036aed3a12f0558dea1a4e96898b1affdffc","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2025-08-13 15:40:49.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"af4c2b521e3e33f65de4fda9aeb638d419851a95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 16:17:58.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"41b8b0aa3d5ba5cdd208ca30aad26c3df7801b8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 17:03:12.000000000","message":"Patch Set 6: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/955393 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"4c8b38784427c098b48507d61972e111fd4e14c1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-11 18:09:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e0d459818be5f59b84407282835291c5633c63bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 20:36:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b1cd97a60ac45d5aefb6dad6785c083\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96bf4d3536b144f5b405991a9842da6f : SUCCESS in 2m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/437922356a824c639a60a4ad58adf958 : SUCCESS in 1h 17m 38s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f67efe6aaaba47f3a0cfe209e6b833c4 : SUCCESS in 2h 21m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2f987704e2424337a4bbd560929964df : FAILURE in 33m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/257871b802d54d0c953cdb64e7802c67 : SUCCESS in 27m 22s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/2d0873db837147eba6ed0edb4cc13ae1 : SUCCESS in 44m 05s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/07db96b355df4ac88598b96d4af00cee : FAILURE in 21m 36s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/47af0b64ac2f4e55a9f97c9c7f9a1dbe : SUCCESS in 32m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9bbb547696184679890858937c147c37 : SUCCESS in 43m 10s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/6f74185399004980926a3df8cd802423 : SUCCESS in 1h 16m 41s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/eca29c44aacf4c46bf2d6f854e00f7d4 : SUCCESS in 29m 22s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/8d3190e55f494ef2ade2ce32dba16311 : SUCCESS in 1h 08m 34s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/283cdb08e847442d98338b060a00aa2c : SUCCESS in 29m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/149c0e801be143c58fe003e35973bd35 : FAILURE in 10m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b89dada9b888a15cd7db6c3eb6ae856ee26e8fda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 20:36:44.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1d8944c7751e8c5ce03a66e5347b3898c389fe31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 21:24:48.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f078f901f8049e88185569bf1af2a83\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dca3a3694db0434db3c32e8b32875587 : SUCCESS in 2m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3d1c791c9cb742a0a90beb03c0088f9f : SUCCESS in 30m 33s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9d7c788982bf4476950fe3a64edf4591 : SUCCESS in 28m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2f5e5dd210d547eea386ba73565d44be : SUCCESS in 43m 28s","accounts_in_message":[],"_revision_number":6},{"id":"63935cd4af88709379a17972bfd353a51ebf01f8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 21:24:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"7dba712f422e93897d52ce6bb77dd96c702d0730","revisions":{"057a16bc409806a88af8588d556c0b0151ee0ee4":{"kind":"REWORK","_number":1,"created":"2025-07-18 17:24:21.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/94/955394/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/94/955394/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/955394/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/955394/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/955394/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/955394/1"}}},"commit":{"parents":[{"commit":"dd1fd2c4ef121eabf3984e6ccd2a7fb3aeae00ba","subject":"Merge \"Drop explicit dependency on python-subunit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dd1fd2c4ef121eabf3984e6ccd2a7fb3aeae00ba"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-07-18 17:19:16.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-07-18 17:24:18.000000000","tz":-240},"subject":"Add barbican to dummy driver job","message":"Add barbican to dummy driver job\n\nAdding barbican plugin to dhss\u003dtrue dummy driver job\n\nDepends-on: https://review.opendev.org/c/openstack/manila/+/955393\nChange-Id: I10bdd54482a1c0d145f49cc3a095e880242d43f8\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/057a16bc409806a88af8588d556c0b0151ee0ee4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/057a16bc409806a88af8588d556c0b0151ee0ee4"}]},"branch":"refs/heads/master"},"eb29d9c7790e246457e6a98d9268dff9a96da3ed":{"kind":"REWORK","_number":2,"created":"2025-07-18 17:39:38.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/94/955394/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/94/955394/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/955394/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/955394/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/955394/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/955394/2"}}},"commit":{"parents":[{"commit":"dd1fd2c4ef121eabf3984e6ccd2a7fb3aeae00ba","subject":"Merge \"Drop explicit dependency on python-subunit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dd1fd2c4ef121eabf3984e6ccd2a7fb3aeae00ba"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-07-18 17:19:16.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-07-18 17:39:26.000000000","tz":-240},"subject":"Add barbican to dummy driver job","message":"Add barbican to dummy driver job\n\nAdding barbican plugin to dhss\u003dtrue dummy driver job\n\nDepends-on: https://review.opendev.org/c/openstack/manila/+/955393\nChange-Id: I10bdd54482a1c0d145f49cc3a095e880242d43f8\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/eb29d9c7790e246457e6a98d9268dff9a96da3ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/eb29d9c7790e246457e6a98d9268dff9a96da3ed"}]},"branch":"refs/heads/master"},"d0db1075d09552d7ddeac118553e02599dd0e29b":{"kind":"REWORK","_number":3,"created":"2025-07-21 06:30:31.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/94/955394/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/94/955394/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/955394/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/955394/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/955394/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/955394/3"}}},"commit":{"parents":[{"commit":"dd1fd2c4ef121eabf3984e6ccd2a7fb3aeae00ba","subject":"Merge \"Drop explicit dependency on python-subunit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dd1fd2c4ef121eabf3984e6ccd2a7fb3aeae00ba"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-07-18 17:19:16.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-07-21 06:30:18.000000000","tz":-240},"subject":"Add barbican to dummy driver job","message":"Add barbican to dummy driver job\n\nAdding barbican plugin to dhss\u003dtrue dummy driver job\n\nDepends-on: https://review.opendev.org/c/openstack/manila/+/955393\nChange-Id: I10bdd54482a1c0d145f49cc3a095e880242d43f8\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d0db1075d09552d7ddeac118553e02599dd0e29b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d0db1075d09552d7ddeac118553e02599dd0e29b"}]},"branch":"refs/heads/master"},"436d6ce589130d28ffaeb75e90b2a176a2711960":{"kind":"REWORK","_number":4,"created":"2025-07-23 04:06:24.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/94/955394/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/94/955394/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/955394/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/955394/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/955394/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/955394/4"}}},"commit":{"parents":[{"commit":"dd1fd2c4ef121eabf3984e6ccd2a7fb3aeae00ba","subject":"Merge \"Drop explicit dependency on python-subunit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dd1fd2c4ef121eabf3984e6ccd2a7fb3aeae00ba"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-07-18 17:19:16.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-07-23 04:06:11.000000000","tz":-240},"subject":"Add barbican to dummy driver job","message":"Add barbican to dummy driver job\n\nAdding barbican plugin to dhss\u003dtrue dummy driver job\n\nDepends-on: https://review.opendev.org/c/openstack/manila/+/955393\nChange-Id: I10bdd54482a1c0d145f49cc3a095e880242d43f8\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/436d6ce589130d28ffaeb75e90b2a176a2711960"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/436d6ce589130d28ffaeb75e90b2a176a2711960"}]},"branch":"refs/heads/master"},"dcae9879282d1a4817485057d7a90bb0076bbf4e":{"kind":"REWORK","_number":5,"created":"2025-07-23 15:34:04.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/94/955394/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/94/955394/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/955394/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/955394/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/955394/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/955394/5"}}},"commit":{"parents":[{"commit":"dd1fd2c4ef121eabf3984e6ccd2a7fb3aeae00ba","subject":"Merge \"Drop explicit dependency on python-subunit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dd1fd2c4ef121eabf3984e6ccd2a7fb3aeae00ba"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-07-18 17:19:16.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-07-23 15:33:52.000000000","tz":-240},"subject":"Add barbican to dummy driver job","message":"Add barbican to dummy driver job\n\nAdding barbican plugin to dhss\u003dtrue dummy driver job\n\nDepends-on: https://review.opendev.org/c/openstack/manila/+/955393\nChange-Id: I10bdd54482a1c0d145f49cc3a095e880242d43f8\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dcae9879282d1a4817485057d7a90bb0076bbf4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dcae9879282d1a4817485057d7a90bb0076bbf4e"}]},"branch":"refs/heads/master"},"7dba712f422e93897d52ce6bb77dd96c702d0730":{"kind":"REWORK","_number":6,"created":"2025-08-04 14:32:45.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/94/955394/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/94/955394/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/955394/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/955394/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/955394/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/955394/6"}}},"commit":{"parents":[{"commit":"dd1fd2c4ef121eabf3984e6ccd2a7fb3aeae00ba","subject":"Merge \"Drop explicit dependency on python-subunit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dd1fd2c4ef121eabf3984e6ccd2a7fb3aeae00ba"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-07-18 17:19:16.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-08-04 14:32:23.000000000","tz":-240},"subject":"Add barbican to dummy driver job","message":"Add barbican to dummy driver job\n\nAdding barbican plugin to dhss\u003dtrue dummy driver job\n\nDepends-on: https://review.opendev.org/c/openstack/manila/+/955393\nChange-Id: I10bdd54482a1c0d145f49cc3a095e880242d43f8\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7dba712f422e93897d52ce6bb77dd96c702d0730"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7dba712f422e93897d52ce6bb77dd96c702d0730"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
