)]}'
{"id":"openstack%2Fmanila-tempest-plugin~955732","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ic60f6a833369cb326d8956da5763c51d75d3c311","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2025-07-25 09:38:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic60f6a833369cb326d8956da5763c51d75d3c311","subject":"Introduce fips-py311 job","status":"MERGED","created":"2025-07-23 21:21:15.000000000","updated":"2025-07-25 10:25:00.000000000","submitted":"2025-07-25 09:38:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955732","meta_rev_id":"8e3b8a2bf335459dd02d8f554453ba7b62e8a0c2","_number":955732,"virtual_id_number":955732,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-25 09:38:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2025-07-25 10:25:00.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2025-07-24 15:11:48.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":2,"date":"2025-07-25 09:00:54.000000000","_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":1,"date":"2025-07-25 09:00:54.000000000","_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-23 23:07:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-24 14:15:07.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-07-24 15:11:48.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"},{"updated":"2025-07-25 09:00:54.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2025-07-25 10:24:54.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2025-07-25 10:25:00.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"000b1228b5e20545a2dea082c9b73d77a9572369","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-07-23 21:21:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b609586a9b5c1164577c91878c13a95f3582763f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 23:07:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7d69b42a3aa4d089d108311464ca41f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8f6b07102d8454791f7f268b810c9c6 : SUCCESS in 2m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bef1c0f22e3e4dee8fdf99a9458c7b04 : SUCCESS in 53m 22s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/32596b49ef5947b0883e4c38c279745f : SUCCESS in 1h 45m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b6841c909b5e4796b17f445c897f6b0c : SUCCESS in 1h 13m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/1ac7652384ce457997246342931777eb : SUCCESS in 44m 19s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ca3ea8a4c63c4c9ba1cd72e1267e826d : SUCCESS in 1h 09m 08s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3742a1ca03024b45928a849d78f2de64 : FAILURE in 20m 56s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c3e5095acebd42cbbf8be039d63e61e7 : SUCCESS in 26m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3021acbfea5849e3984f4749daff6fc1 : SUCCESS in 31m 46s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/7d9afb496ed94e9b8c04330eb898cd6c : SUCCESS in 1h 27m 03s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/f42aeb9311c34b14b11dc3d280dbdfd7 : SUCCESS in 51m 34s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/9d9eb58a672041a4b4f9161ecdabb353 : SUCCESS in 52m 21s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/60536b90bff54da8bf99356554cfe963 : SUCCESS in 29m 39s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3c91f11c470e4f91a7a1a27c1840d954 : FAILURE in 16m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a4313fa71a6ef8e5ef8a917317627f24c34fde8","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-24 14:15:07.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"324a66540280bfb6766021aaff741ef99fe50781","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-07-24 15:11:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ac596c4742078ae1c4f9f1768627d3ebb0eb712b","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2025-07-25 09:00:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"30e8c773cbf5b899d50066e0dbbf1bf957300c3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-25 09:01:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dc06df8f24ed2d8e4268cb5f644b6d942db7103e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-25 09:38:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15d59642d83b40ff81681eac1dfbd8a0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/390737c24240462ca0a0cc0ef692a4a4 : SUCCESS in 2m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0201379a01ca41e593555978d084b056 : SUCCESS in 36m 51s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/63018b7d048f4706ae1da8d3ff039884 : SUCCESS in 35m 11s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/741ebe76873f42b9af0a2cf027b88431 : SUCCESS in 25m 46s","accounts_in_message":[],"_revision_number":1},{"id":"085e9199a114685cc454e54854adbf14752fa2ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-25 09:38:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"846e901c6e4b4594619efe294cb0df01c95bcec4","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-07-25 10:24:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8e3b8a2bf335459dd02d8f554453ba7b62e8a0c2","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-07-25 10:25:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"432f6b122b88661568e50faa42f1ee0bc879e3a9","revisions":{"432f6b122b88661568e50faa42f1ee0bc879e3a9":{"kind":"REWORK","_number":1,"created":"2025-07-23 21:21:15.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/32/955732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/32/955732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/32/955732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/32/955732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/32/955732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/32/955732/1"}}},"commit":{"parents":[{"commit":"326d5a399168c2a53ec01bf4f74dc6c4dad8eaf3","subject":"Merge \"Add CEPHFS filesystem metadata verifications\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/326d5a399168c2a53ec01bf4f74dc6c4dad8eaf3"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-23 21:14:28.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-23 21:17:59.000000000","tz":-240},"subject":"Introduce fips-py311 job","message":"Introduce fips-py311 job\n\nThe centos9 node used to test fips fails to deploy\na few openstack services as of Flamingo.\n\nSwitch it to a python3.11 environment for FIPS jobs\nso that Flamingo services can run successfully there.\n\nChange-Id: Ic60f6a833369cb326d8956da5763c51d75d3c311\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/432f6b122b88661568e50faa42f1ee0bc879e3a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/432f6b122b88661568e50faa42f1ee0bc879e3a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
