)]}'
{"id":"openstack%2Fmanila-tempest-plugin~956046","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I5dae5e7daeb6b0b76a7a369714dce571a19cb000","project":"openstack/manila-tempest-plugin","branch":"master","topic":"negative-quota-test","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-07-31 18:38:43.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2025-07-31 15:09:45.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2025-07-31 16:07:40.000000000","reason":"\u003cGERRIT_ACCOUNT_19262\u003e replied on the change","reason_account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}},"hashtags":[],"change_id":"I5dae5e7daeb6b0b76a7a369714dce571a19cb000","subject":"Fix share type used in negative quota test","status":"MERGED","created":"2025-07-28 20:55:06.000000000","updated":"2025-07-31 22:47:25.000000000","submitted":"2025-07-31 22:37:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956046-negative-quota-test","meta_rev_id":"e59a6b3f2f52868ffa5890f6aa0bab2bc0b5fdcc","_number":956046,"virtual_id_number":956046,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-31 22:37:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":2,"date":"2025-07-31 15:09:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-31 16:07:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"all":[{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-31 16:07:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-28 23:47:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-29 03:10:47.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-07-30 16:44:49.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2025-07-30 16:44:49.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"}],"messages":[{"id":"e046e1de0b01b3c4f40206aeb0cd1dbac06e89ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-07-28 20:55:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45ab88f1fdc398b83a66585c4dcbe6d53612b890","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 23:47:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d178c96759554159b12d234bed9a9443\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfe4d15f303f4ad48fa4049d8fe988f3 : SUCCESS in 2m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bca73845cb004ac5809c27ecfdb64e98 : SUCCESS in 1h 00m 25s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/0ccbab398f0e4a038e1f4bd85b4f48a8 : SUCCESS in 2h 39m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5d6b16daa2044b8092d45a5331d7fe1b : FAILURE in 1h 21m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9fd6ceb9eb914033b65c8835260b4d4b : SUCCESS in 1h 07m 59s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e6f79733cdb24736be1c1b42ae393394 : SUCCESS in 1h 02m 46s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b5c7b81d0778469ebe88248480060033 : FAILURE in 21m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/579a3d1db20044138a11cd1d15ad147c : SUCCESS in 35m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/37eb21258c25445498dc39b39ea5b996 : SUCCESS in 47m 45s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/6b51aba3da134bc7877819b63f4dfa43 : SUCCESS in 1h 17m 27s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/6447b1c89a4e438c94f0d76961a1871b : SUCCESS in 1h 05m 30s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/22d6decb6bbd465f8a25c9689745444d : FAILURE in 1h 10m 24s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/7d9c0f561e2e4671a1af2b499904c19f : SUCCESS in 28m 45s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0cb04ecd100b453fa3262599be8218e0 : FAILURE in 29m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a843417cf6d0d6425f79940efbbe8dc120f3bb9d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-07-29 00:29:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2d2e05a1027d5186867aa0ae2a7569c5b2807dce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-29 03:10:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75b0d7b6855540e9b75b3206ee2d0c89\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8547b3d28194fed8829a448d0fb1007 : SUCCESS in 3m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/748087c95b8740b9954d0c4ea6e079e9 : SUCCESS in 32m 39s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c59ce3bb5b6248898ad91aba98d61ab0 : FAILURE in 2h 35m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/72c107583d394f9591edccf91050902c : FAILURE in 1h 20m 45s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b75efd066ac34297bf7fffa57a467db5 : SUCCESS in 1h 09m 24s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/288d08c55df04650bc8b97f37a3fdca2 : SUCCESS in 1h 06m 13s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6b850cf51e964a6eba2736333323c25e : FAILURE in 20m 31s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3f0e3e8758f14ec0988e3547252aa9eb : SUCCESS in 26m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5f33653d5c8349b98ca8738ff31000d0 : SUCCESS in 47m 38s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/03f94a686681482f9d28f460d1c8049c : SUCCESS in 31m 49s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/1240674966a34770a47e5c1b69a5b5d5 : SUCCESS in 1h 04m 24s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/235f6e9863be443faab02bf269b76fdb : SUCCESS in 1h 17m 00s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/5f6f4c307b004f02aa175260b7468b3f : SUCCESS in 32m 08s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3b82bcdd89b0422cba3e7d8c51a83984 : FAILURE in 11m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"113a938976e13695e64a4134e4bfab635fcc12f8","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-29 03:10:47.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"28740e39e648f8872756a081d5fb7c6a7c7129ff","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2025-07-31 15:09:45.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2033c2b8ef58b0ed2618ccda61003d4dd4eef4e4","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-07-31 16:07:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3259ded1a0f5fe317a5a1e422eeb4000e13b7c50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 16:08:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"168048142e096dce8973034a0e4ba19c9b48285d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 17:29:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/910de708981c4ccdb61b1e52b8e2a996\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29c23e95a63c4460aee1514a1d74f9e5 : SUCCESS in 4m 14s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/67531b701ec146cd81185a2cd0ac5b3e : FAILURE in 1h 17m 08s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a63d036025c94730aa3e8cb8320bb4f8 : SUCCESS in 28m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/493655bc516c449b8dda8a1f80ccac3b : SUCCESS in 31m 50s","accounts_in_message":[],"_revision_number":1},{"id":"66702d0b43b85a43c3831ff140033784136581e3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-07-31 18:38:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5f06d3abe440f94f2e8a59e51b511fdfd16562e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 21:16:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7f2bd2fc5bf4c71a327c933a74ce3e3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ecdf77ca0564925a9181523e97e6daf : SUCCESS in 2m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6cc5be1e010e42d0a22f8fc1d57509ac : SUCCESS in 1h 12m 06s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2c96bb9ad9434585b1295f0d2210fce9 : FAILURE in 2h 32m 03s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e913b7d9aae34b4d87e052110111bb41 : FAILURE in 1h 18m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/55f31bb3d09d4f569fb7faf89656b291 : SUCCESS in 1h 03m 18s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c4d16eaf6b2e4b45ab05ea9556762069 : SUCCESS in 1h 24m 52s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4da03a9a3da44230a908d2642e85cf5f : FAILURE in 20m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/813e3bd0556a4bb2b3c17ae86f80ec99 : SUCCESS in 28m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0f9214753458471b828612966429df66 : SUCCESS in 40m 37s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/ab25202cb8b24f9cb3f0ba430cf4625f : SUCCESS in 1h 07m 49s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/888d8ad5cfc141f0a45e2c36abbb7338 : SUCCESS in 1h 13m 03s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/20add364c1624a9db53faee000b46bf8 : SUCCESS in 1h 18m 15s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/17b50a47969c4d37a7c93b882cf14f8a : SUCCESS in 32m 32s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f920bd03b60849a8892474801e1edca8 : FAILURE in 21m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a217d1b1fc69bb7e3caeac7bf783cfcf1a1c5661","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 21:16:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dfa0ae7eb995f41d40fab835f4548197d2f8bf87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 22:37:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be187df636e247f686359254e0bb5adc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8057341312e14761a2e536447f9052fc : SUCCESS in 5m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c8ee588cd307476d9de409f9cf8db517 : SUCCESS in 1h 15m 55s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b1c9bf93408e41568ed1de6fa476b443 : SUCCESS in 38m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0aaca14940c8491f9278253936d0fe42 : SUCCESS in 36m 36s","accounts_in_message":[],"_revision_number":1},{"id":"22887f9c978a656d3e0382d16d23a31fc74c4055","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 22:37:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e59a6b3f2f52868ffa5890f6aa0bab2bc0b5fdcc","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-31 22:47:25.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be1bb4bf6fc4092b18a23ba4b86d327e43f4c527","revisions":{"be1bb4bf6fc4092b18a23ba4b86d327e43f4c527":{"kind":"REWORK","_number":1,"created":"2025-07-28 20:55:06.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/46/956046/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/46/956046/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/46/956046/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/46/956046/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/46/956046/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/46/956046/1"}}},"commit":{"parents":[{"commit":"972a1510a8bad4f58da8cb2187f8c8a0954a07f1","subject":"Merge \"Introduce fips-py311 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/972a1510a8bad4f58da8cb2187f8c8a0954a07f1"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-07-28 20:54:50.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-07-28 20:54:50.000000000","tz":-420},"subject":"Fix share type used in negative quota test","message":"Fix share type used in negative quota test\n\nIt would be useful for tests to always specify the\nshare type to be used so as to not rely on the default\nshare type that may be misconfigured, or not configured\nat all.\n\nChange-Id: I5dae5e7daeb6b0b76a7a369714dce571a19cb000\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/be1bb4bf6fc4092b18a23ba4b86d327e43f4c527"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/be1bb4bf6fc4092b18a23ba4b86d327e43f4c527"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
