)]}'
{"id":"openstack%2Fmanila-tempest-plugin~956494","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ie52139c31ce786c2c246eb7c62f8f8ec5aeb65fc","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bp/share-backup-out-of-place-restore","attention_set":{"36761":{"account":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"last_update":"2026-08-30 18:57:30.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ie52139c31ce786c2c246eb7c62f8f8ec5aeb65fc","subject":"add tests for targeted share restores","status":"NEW","created":"2025-08-04 15:24:07.000000000","updated":"2026-08-30 18:57:30.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"e5eb5527cdf4d42a9781ba4f5b06874435ed4dd7","_number":956494,"virtual_id_number":956494,"owner":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-30 18:57:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-04 18:07:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"300b095851490f71b09c4d37876e557246d0f32c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-08-04 15:24:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5348961eaac39fe8b7b8929c378c1de6d247d9b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-08-04 15:28:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b6d84eebc1744e55c623ec7d0a1ab5c33ef8cdca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-08-04 15:41:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5ed18c5d5a16abf2b626cd181dc03e200f300c1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-08-04 15:44:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"373e52120e990e6ca1cdd93cd16892b28b6c0c58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-08-04 15:49:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5fa1fc0ff6393d69c6f87d08f674c0e4942ee5e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 18:07:12.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3a8e660c72d45aa807fd6a026b14b04\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba771c6188a54f74985b798edd0f82a6 : SUCCESS in 2m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7755e89de3574cda84e3c86b222f99a8 : SUCCESS in 1h 06m 05s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8619b60a9bf3448eb9d0d241c28048f5 : SUCCESS in 1h 00m 05s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/258f756d31734e56a58034abe892fddf : FAILURE in 1h 11m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/12147ef02db64bf7bfd82b3e527030c9 : SUCCESS in 1h 01m 56s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/7b610a1d93b240a38b73ea9528921a4a : SUCCESS in 1h 42m 43s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/91f16bbbd67a46f8b25194adddb11168 : FAILURE in 20m 25s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a06d0e414e994f648c47e2e37f6a69a4 : FAILURE in 21m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8dcd68bdea3542eb92799ad2ab76165d : FAILURE in 9m 40s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/70aa17f5e1d24d26a3cb2ed8c8d9098b : SUCCESS in 1h 11m 21s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/e019614bdfea4dc390ec84619a4dc6b7 : SUCCESS in 1h 07m 56s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/b523214d6b974f5d867290f092a0ca6c : SUCCESS in 33m 20s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/dd7a8e37bcf2416c95c174c699afbb73 : SUCCESS in 32m 15s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/949f913d72264827b3912b5d2421efdf : FAILURE in 23m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8e6eac5474198465f069f696093484206e6f5f4c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-08-05 03:14:06.000000000","message":"Topic set to bp/share-backup-out-of-place-restore","accounts_in_message":[],"_revision_number":5},{"id":"cd91eb5f6dcc2495faf8dfd8f19bfa460e6da94e","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-10-05 13:25:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"39a96e7c96a24bae6ccd6e0109fc48de9d112c1b","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-10-24 08:30:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ca67fe5bb1fd9f4c9b0bdc9ca00d9eb1ed0cb914","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 10:21:39.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fcd5ffd856d1438bbaa6e1b700111c2e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6c6ddb3e44842ffb88e4c1b744ba03e : SUCCESS in 2m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7f6a24b5961d409da1b431f83845fd22 : SUCCESS in 59m 12s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/aaf38f9d075349e4947f52403cd8e4cd : SUCCESS in 1h 48m 26s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/93bbe6356e2a4867b3d36b3289192cc4 : FAILURE in 59m 33s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/fe267404d6324f9783cdaf0219ee3e75 : SUCCESS in 1h 04m 39s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/62f189836f0f400cab38b8b33143de4d : SUCCESS in 49m 27s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8a886ff4102d493cb743b1232b10e29d : FAILURE in 20m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d647733b16bd4c2c8b121f44d5f77d5d : FAILURE in 18m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0a0c3609d3b049af8ff82dfdbd9f4d85 : SUCCESS in 43m 23s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/93d58fcb006b4662b637d6ac5b2d7236 : SUCCESS in 1h 11m 26s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/22416366be884d6a8a081388a19f9c39 : SUCCESS in 59m 22s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/6d1cf1ef72314070840dcb78fa20149a : SUCCESS in 1h 21m 32s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f2c4168813f541c99ae24cb0d385e3ec : SUCCESS in 32m 21s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e41b5ac6bb2c469b8e4864b322599894 : FAILURE in 22m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"67625f9f1d7c0626e2d24bc8c65d852ff9c76669","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-11-27 16:20:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"64c69bf6abe6e806f987d3bf443cbe554fd693b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 19:29:02.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e518da0088343aeba26b0dbea4a83b0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d64236a187bd4ae1ba8fd0766ed97002 : SUCCESS in 2m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2220edb4a94049b7ae78e6a5893ffafc : SUCCESS in 1h 06m 20s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/dacfe32b100d48a2aa124f9d6d1bc994 : TIMED_OUT in 3h 02m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9b32d6cd185c4cf793ab8f71430d166d : FAILURE in 32m 24s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/2fe9c125b79047899f302aacf5f9fd66 : SUCCESS in 45m 46s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/9ae3944af83447788c4b0b6c0ff5c268 : SUCCESS in 1h 39m 27s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/db29da38ee0847ebace156a15d334cb4 : FAILURE in 15m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e387e993857b4d45b93a5a0dadb4a253 : FAILURE in 36m 47s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/18e2c399b93f424f8129b253d492f54a : SUCCESS in 50m 21s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/4a9593d7a48b475fb9040f6a4bec8a0b : SUCCESS in 1h 01m 36s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/2002172de2a44044ae958d4585fbb40e : SUCCESS in 59m 38s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/47fb1d41621f4da9858205cfb63ef5e5 : SUCCESS in 41m 22s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/51061ff81713411a8ebb1aa6f025ba72 : FAILURE in 21m 35s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0ef04228dcf54bf28167ec4532ef96f3 : FAILURE in 27m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"381012c76519191f429f3616814c9a03a6a1c0de","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-12-12 14:03:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a7b936495369e06a84a245f4dc3376fce81d4953","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-12-12 14:27:24.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"f9b51e3c6bfd85e5ca031dcd2a8e8a4f0a3ea7a8","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-12-12 14:29:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"176b7ee2d7b8dd186c5fed78bcaa95b03755e532","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 16:54:29.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/36e18304a5c946878def6bb8598b2c74\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b639c027e83e4b078d267d6ded4513c0 : FAILURE in 1m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c139daa2b5364af2bf30b52c634fc8db : SUCCESS in 1h 02m 23s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/662c86eb4e874da9a83d709d432aa7cf : FAILURE in 2h 17m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ce9f219a7f374d20a79c4a3b174e3f7f : SUCCESS in 1h 39m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/2deaefd063464384abe18652d8746744 : SUCCESS in 31m 15s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c2362b47f47f40b6bbc13ee0fd980311 : SUCCESS in 1h 53m 17s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0cbe3c1911ad47d1b3047f05c72ee865 : FAILURE in 22m 43s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5324a72699744f9e914737a1533d426f : FAILURE in 31m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fcc19da9c6a745fc864bb0094068a08b : SUCCESS in 47m 43s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/c04630ed6bb64961909aa6f6f82badb9 : SUCCESS in 1h 11m 03s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/f28f042cc3384ae5832f13c0b95bc1a3 : SUCCESS in 37m 51s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/418a7734c4434e88a3d8b7de65e92bfc : SUCCESS in 52m 41s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/7124c7c1fb55483fbdadc59a353ebd0a : FAILURE in 35m 35s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3407d217c97e4407a5a44b36ea5f2f8a : FAILURE in 25m 08s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1052ea6d030f72db0c093b4cb93bc7c318ceb2a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-12-12 18:06:54.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"2110d288802fd18d50194d204b3a01117de8004b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-12-12 18:16:30.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"67832df201915bfff940e69b451d92cae817d11f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 18:57:50.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8e12b13cbe1486a8107c5e43be6cf88\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85060281553c4736bed7a0082cbe4a01 : FAILURE in 2m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2733562a35dd42c883818c7210615bb0 : FAILURE in 23m 03s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ca14dbaf8a244af8bf5fda2fdffd234a : FAILURE in 22m 30s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6920e23aa90e4abeb9ff87ca8259ddaf : FAILURE in 36m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d4819d976f264ab3910ef941a7473b39 : FAILURE in 22m 40s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/d155682f2c4140bdab4b010843437f18 : FAILURE in 34m 13s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a47400ef03184517a058c49f74715ce0 : FAILURE in 23m 34s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f178579ac8b3456182fcb4cba8409403 : FAILURE in 21m 47s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/10e2548bd6f44a7590a6e2a935543143 : FAILURE in 19m 43s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/1e1dce54c2bd49a1b2c7df7507aa9ee9 : FAILURE in 21m 07s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/dec51ca6db624957a908bf50abc0a686 : FAILURE in 35m 35s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/c4f4ad6b40f2444994ee3cd35665184a : FAILURE in 12m 55s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/121ffaf8df1f4108a0b2438170006d90 : FAILURE in 18m 09s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cf59b54f1dfc4ed18a39511502e6b697 : FAILURE in 27m 08s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"490ba63afdc9b5713c475cde8538b352e6990dfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-12-12 19:31:41.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"7cee57a007e06ba93b4bda375648dc4000c49244","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 20:51:23.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/35bd9f2d094c44738346644680adb2f6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/319d0ccd83154c459a2f784940d9f2af : SUCCESS in 2m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/065b37f69bd84c17bb68d18fc8251675 : SUCCESS in 58m 41s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c6d6eeb3440d4fe3b2fa7bc346c20c94 : SUCCESS in 1h 00m 56s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/86e7a4fd697a4438ad1a0b4c52ff4b23 : FAILURE in 59m 48s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3438836e3b924edb894d425197bf8398 : SUCCESS in 37m 58s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/183dc383ea224fb3ae1c682ba143263f : SUCCESS in 1h 07m 39s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/43e8f67e0f4c451e83dade5d36732d56 : FAILURE in 23m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/db441a0db8304613a13eeead1e44ce5d : FAILURE in 26m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/15d95380e898497a9a113bcdcb7173c7 : SUCCESS in 40m 46s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/dc7fb0b1f54f491b9bb09f1279a393c5 : SUCCESS in 54m 17s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/14a4d718725a4bcb8192832cf3c2940b : SUCCESS in 1h 14m 30s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/dd4fed7d07fc4280a02b5b84e85a0191 : SUCCESS in 1h 09m 52s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/caf15577c9ac4b039918a44e2d7350e4 : FAILURE in 22m 40s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/81b91966451240d2b2e8b18e3b3d7321 : FAILURE in 16m 48s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e8c3b7031fe1c9a75d7c62f14e7e8d0de71c3e35","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-12-13 01:13:38.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a239c61030b981c8aef3fddc0972c594bcd6582c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-13 03:56:17.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/878aaaa65d0446189fb7a09602e10c9e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46bd668b26da47868d9f6f911800e608 : SUCCESS in 3m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/25d99a047c5747bbba41ede08db7cd02 : SUCCESS in 1h 15m 28s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/71bb97f9dfe847ca8e11ec9227130dda : FAILURE in 2h 37m 30s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fa961f338d0a4ee094d6c0cc5e9e27ca : FAILURE in 1h 03m 06s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8090fa36b70e4e9d83696209ddc4d2ce : SUCCESS in 45m 22s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/0479083afba54de19e41dbae06e6b6be : SUCCESS in 1h 08m 44s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f3dd235434214be09b9e3d09959cd5a7 : FAILURE in 23m 48s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/74f4ed87d7af4b0fbda8144d9d545b57 : FAILURE in 23m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c048d1942630441cbe3f788802b0dd0a : SUCCESS in 42m 16s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/db79ab48814d49398a646f122f61dde6 : SUCCESS in 34m 47s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/0c88161a026e4e4eaf9822f8819ad329 : SUCCESS in 33m 22s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/22d3aaff9aaa47fd907dc3c3a436c083 : SUCCESS in 1h 10m 08s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a80d79176ef1428bb0c6da7f6e388976 : FAILURE in 19m 17s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d5675fc9524a4b13a8b92421ed953ccb : FAILURE in 26m 27s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e55e236341bfde9a1777a567d5263f89194391ba","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-12-16 09:47:00.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"3ac076ae5438de26f2de6396021c5509d2e26ec2","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2026-08-30 16:45:19.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"48f41e72d0f07d8c25b5a86a5ab325c840cbc569","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2026-08-30 16:49:11.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"e5eb5527cdf4d42a9781ba4f5b06874435ed4dd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 18:57:30.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ff38b6ac1c5490e9483c32da8d56e4a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/143c23c84a1b46698179fdf32d1b8c7f : SUCCESS in 2m 23s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/333fb3a8674e47f29ef11dfd1878800b : SUCCESS in 54m 15s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/6dc2f273b7fb4bbfa1e0edb3f19c22df : SUCCESS in 2h 02m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/39bcc465237a4e638d61e6d115598207 : SUCCESS in 47m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c84c19dc83b0485a942e7b6b53c19416 : FAILURE in 44m 47s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c4b16ce8707d44cbb33f41ce83eae6a6 : FAILURE in 1h 19m 45s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8d00fe432b0a4da1889603171899467e : SUCCESS in 46m 10s (non-voting)\n- manila-tempest-plugin-lustre https://zuul.opendev.org/t/openstack/build/90c2cd1403364f55af4062ddfe35a2e4 : SUCCESS in 55m 18s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/61b4bc45b6404c2b9aef78ea627cb33d : FAILURE in 23m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/598680c2f45447e7b5160964e2eb6324 : SUCCESS in 38m 46s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/900ccb25018a481a8bff2c7ccf87a1ec : SUCCESS in 51m 16s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/f6341e43e1714d94aebe4abc70381b4b : SUCCESS in 1h 03m 06s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/af5bade91abb4798872c97e204ba18a4 : SUCCESS in 1h 02m 37s\n- manila-tempest-plugin-dummy-no-dhss-old-rbac https://zuul.opendev.org/t/openstack/build/2e7c51179d0c40c7be2da3410b4b023b : SUCCESS in 26m 40s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bce9f163ad8443b28e344e6e909b4538 : SUCCESS in 44m 11s (non-voting)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"0b3089578b735308e3174cd002f59146d7549e09","revisions":{"d9502c80e5c3fa5dbc8fcf748806a4351e301781":{"kind":"REWORK","_number":1,"created":"2025-08-04 15:24:07.000000000","uploader":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"ref":"refs/changes/94/956494/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/94/956494/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/1"}}},"commit":{"parents":[{"commit":"972a1510a8bad4f58da8cb2187f8c8a0954a07f1","subject":"Merge \"Introduce fips-py311 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/972a1510a8bad4f58da8cb2187f8c8a0954a07f1"}]}],"author":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-08-04 15:16:07.000000000","tz":0},"committer":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-08-04 15:21:24.000000000","tz":0},"subject":"add tests for targeted share restores for share backups","message":"add tests for targeted share restores for share backups\n\nadd tempest tests that exercise the ability of share backup\nto restore to a non source share.\n\nChange-Id: Ie52139c31ce786c2c246eb7c62f8f8ec5aeb65fc\nSigned-off-by: Zachary Goggin \u003czachary.goggin@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d9502c80e5c3fa5dbc8fcf748806a4351e301781"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d9502c80e5c3fa5dbc8fcf748806a4351e301781"}]},"branch":"refs/heads/master"},"7f98d1a974a2ca4e00acbdc9ab50c57d0c84daf8":{"kind":"REWORK","_number":2,"created":"2025-08-04 15:28:46.000000000","uploader":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"ref":"refs/changes/94/956494/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/94/956494/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/2"}}},"commit":{"parents":[{"commit":"972a1510a8bad4f58da8cb2187f8c8a0954a07f1","subject":"Merge \"Introduce fips-py311 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/972a1510a8bad4f58da8cb2187f8c8a0954a07f1"}]}],"author":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-08-04 15:16:07.000000000","tz":0},"committer":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-08-04 15:28:07.000000000","tz":0},"subject":"add tests for targeted share restores for share backups","message":"add tests for targeted share restores for share backups\n\nadd tempest tests that exercise the ability of share backup\nto restore to a non source share.\n\nChange-Id: Ie52139c31ce786c2c246eb7c62f8f8ec5aeb65fc\nSigned-off-by: Zachary Goggin \u003czachary.goggin@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7f98d1a974a2ca4e00acbdc9ab50c57d0c84daf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7f98d1a974a2ca4e00acbdc9ab50c57d0c84daf8"}]},"branch":"refs/heads/master"},"088b2a9ff80afb6d87f989f3f1f78545a105804b":{"kind":"REWORK","_number":3,"created":"2025-08-04 15:41:32.000000000","uploader":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"ref":"refs/changes/94/956494/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/94/956494/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/3"}}},"commit":{"parents":[{"commit":"972a1510a8bad4f58da8cb2187f8c8a0954a07f1","subject":"Merge \"Introduce fips-py311 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/972a1510a8bad4f58da8cb2187f8c8a0954a07f1"}]}],"author":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-08-04 15:16:07.000000000","tz":0},"committer":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-08-04 15:40:51.000000000","tz":0},"subject":"add tests for targeted share restores","message":"add tests for targeted share restores\n\nadd tempest tests that exercise the ability of share backup\nto restore to a non source share.\n\nChange-Id: Ie52139c31ce786c2c246eb7c62f8f8ec5aeb65fc\nSigned-off-by: Zachary Goggin \u003czachary.goggin@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/088b2a9ff80afb6d87f989f3f1f78545a105804b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/088b2a9ff80afb6d87f989f3f1f78545a105804b"}]},"branch":"refs/heads/master"},"d50813bb1e0f35877edc0359f835b2542c3eddd0":{"kind":"REWORK","_number":4,"created":"2025-08-04 15:44:37.000000000","uploader":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"ref":"refs/changes/94/956494/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/94/956494/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/4"}}},"commit":{"parents":[{"commit":"972a1510a8bad4f58da8cb2187f8c8a0954a07f1","subject":"Merge \"Introduce fips-py311 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/972a1510a8bad4f58da8cb2187f8c8a0954a07f1"}]}],"author":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-08-04 15:16:07.000000000","tz":0},"committer":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-08-04 15:44:27.000000000","tz":0},"subject":"add tests for targeted share restores","message":"add tests for targeted share restores\n\nadd tempest tests that exercise the ability of share backup\nto restore to a non source share.\n\nChange-Id: Ie52139c31ce786c2c246eb7c62f8f8ec5aeb65fc\nSigned-off-by: Zachary Goggin \u003czachary.goggin@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d50813bb1e0f35877edc0359f835b2542c3eddd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d50813bb1e0f35877edc0359f835b2542c3eddd0"}]},"branch":"refs/heads/master"},"f5f67835fbfdc54831ff4a703a82df1bfc81b8b0":{"kind":"REWORK","_number":5,"created":"2025-08-04 15:49:17.000000000","uploader":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"ref":"refs/changes/94/956494/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/94/956494/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/5"}}},"commit":{"parents":[{"commit":"972a1510a8bad4f58da8cb2187f8c8a0954a07f1","subject":"Merge \"Introduce fips-py311 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/972a1510a8bad4f58da8cb2187f8c8a0954a07f1"}]}],"author":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-08-04 15:16:07.000000000","tz":0},"committer":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-08-04 15:48:15.000000000","tz":0},"subject":"add tests for targeted share restores","message":"add tests for targeted share restores\n\nadd tempest tests that exercise the ability of share backup\nto restore to a non source share.\n\nChange-Id: Ie52139c31ce786c2c246eb7c62f8f8ec5aeb65fc\nSigned-off-by: Zachary Goggin \u003czachary.goggin@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f5f67835fbfdc54831ff4a703a82df1bfc81b8b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f5f67835fbfdc54831ff4a703a82df1bfc81b8b0"}]},"branch":"refs/heads/master"},"8c5b5c87af95546c539974e179c02dbb8f395189":{"kind":"REWORK","_number":6,"created":"2025-12-12 14:27:24.000000000","uploader":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"ref":"refs/changes/94/956494/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/94/956494/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/6"}}},"commit":{"parents":[{"commit":"972a1510a8bad4f58da8cb2187f8c8a0954a07f1","subject":"Merge \"Introduce fips-py311 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/972a1510a8bad4f58da8cb2187f8c8a0954a07f1"}]}],"author":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-08-04 15:16:07.000000000","tz":0},"committer":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-12-12 14:26:53.000000000","tz":0},"subject":"add tests for targeted share restores","message":"add tests for targeted share restores\n\nadd tempest tests that exercise the ability of share backup\nto restore to a non source share.\n\nChange-Id: Ie52139c31ce786c2c246eb7c62f8f8ec5aeb65fc\nSigned-off-by: Zachary Goggin \u003czachary.goggin@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8c5b5c87af95546c539974e179c02dbb8f395189"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8c5b5c87af95546c539974e179c02dbb8f395189"}]},"branch":"refs/heads/master"},"65cb6d6aa5a06d5147ef3e0d36e0d699ae019be0":{"kind":"REWORK","_number":7,"created":"2025-12-12 18:06:54.000000000","uploader":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"ref":"refs/changes/94/956494/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/94/956494/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/7"}}},"commit":{"parents":[{"commit":"972a1510a8bad4f58da8cb2187f8c8a0954a07f1","subject":"Merge \"Introduce fips-py311 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/972a1510a8bad4f58da8cb2187f8c8a0954a07f1"}]}],"author":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-08-04 15:16:07.000000000","tz":0},"committer":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-12-12 18:06:45.000000000","tz":0},"subject":"add tests for targeted share restores","message":"add tests for targeted share restores\n\nadd tempest tests that exercise the ability of share backup\nto restore to a non source share.\n\nChange-Id: Ie52139c31ce786c2c246eb7c62f8f8ec5aeb65fc\nSigned-off-by: Zachary Goggin \u003czachary.goggin@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/65cb6d6aa5a06d5147ef3e0d36e0d699ae019be0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/65cb6d6aa5a06d5147ef3e0d36e0d699ae019be0"}]},"branch":"refs/heads/master"},"c461ac429cc7905bc7f4bb11e7a06ab52d5994ce":{"kind":"REWORK","_number":8,"created":"2025-12-12 18:16:30.000000000","uploader":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"ref":"refs/changes/94/956494/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/94/956494/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/8"}}},"commit":{"parents":[{"commit":"283b664c8be0655bcf370bce10c851cf8dacd2c6","subject":"Merge \"Stop attempting share type deletions that\u0027d fail\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/283b664c8be0655bcf370bce10c851cf8dacd2c6"}]}],"author":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-08-04 15:16:07.000000000","tz":0},"committer":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-12-12 18:16:18.000000000","tz":0},"subject":"add tests for targeted share restores","message":"add tests for targeted share restores\n\nadd tempest tests that exercise the ability of share backup\nto restore to a non source share.\n\nChange-Id: Ie52139c31ce786c2c246eb7c62f8f8ec5aeb65fc\nSigned-off-by: Zachary Goggin \u003czachary.goggin@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c461ac429cc7905bc7f4bb11e7a06ab52d5994ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c461ac429cc7905bc7f4bb11e7a06ab52d5994ce"}]},"branch":"refs/heads/master"},"79d65b2a80c3718905c41fddcb1813e8ea1847a7":{"kind":"REWORK","_number":9,"created":"2025-12-12 19:31:41.000000000","uploader":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"ref":"refs/changes/94/956494/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/94/956494/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/9"}}},"commit":{"parents":[{"commit":"283b664c8be0655bcf370bce10c851cf8dacd2c6","subject":"Merge \"Stop attempting share type deletions that\u0027d fail\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/283b664c8be0655bcf370bce10c851cf8dacd2c6"}]}],"author":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-08-04 15:16:07.000000000","tz":0},"committer":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-12-12 19:31:31.000000000","tz":0},"subject":"add tests for targeted share restores","message":"add tests for targeted share restores\n\nadd tempest tests that exercise the ability of share backup\nto restore to a non source share.\n\nChange-Id: Ie52139c31ce786c2c246eb7c62f8f8ec5aeb65fc\nSigned-off-by: Zachary Goggin \u003czachary.goggin@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/79d65b2a80c3718905c41fddcb1813e8ea1847a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/79d65b2a80c3718905c41fddcb1813e8ea1847a7"}]},"branch":"refs/heads/master"},"0b3089578b735308e3174cd002f59146d7549e09":{"kind":"REWORK","_number":10,"created":"2026-08-30 16:49:11.000000000","uploader":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"ref":"refs/changes/94/956494/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/94/956494/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/94/956494/10"}}},"commit":{"parents":[{"commit":"82c6bcfcfe72fc43021f4bb125b11c811d918c9b","subject":"Merge \"Add LUSTRE protocol support and CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/82c6bcfcfe72fc43021f4bb125b11c811d918c9b"}]}],"author":{"name":"Zachary Goggin","email":"zachary.goggin@ecmwf.int","date":"2026-08-30 16:48:46.000000000","tz":0},"committer":{"name":"Zachary Goggin","email":"zachary.goggin@ecmwf.int","date":"2026-08-30 16:49:07.000000000","tz":0},"subject":"add tests for targeted share restores","message":"add tests for targeted share restores\n\nadd tempest tests that exercise the ability of share backup\nto restore to a non source share.\n\nChange-Id: Ie52139c31ce786c2c246eb7c62f8f8ec5aeb65fc\nSigned-off-by: Zachary Goggin \u003czachary.goggin@ecmwf.int\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0b3089578b735308e3174cd002f59146d7549e09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0b3089578b735308e3174cd002f59146d7549e09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
