)]}'
{"id":"openstack%2Fmanila-tempest-plugin~958516","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ia09b64e4b4afd4bb480a18e80bae46922a2210c7","project":"openstack/manila-tempest-plugin","branch":"master","topic":"fix-stable-branches-migration-replica","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-09-05 20:12:37.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"12404":{"account":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"last_update":"2025-09-12 01:07:12.000000000","reason":"Change was submitted"},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2025-09-11 15:33:07.000000000","reason":"\u003cGERRIT_ACCOUNT_32919\u003e replied on the change","reason_account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2025-09-11 17:52:35.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}},"hashtags":[],"change_id":"Ia09b64e4b4afd4bb480a18e80bae46922a2210c7","subject":"Fix share server migration with replica test","status":"MERGED","created":"2025-08-26 07:53:57.000000000","updated":"2025-09-12 01:07:12.000000000","submitted":"2025-09-12 01:07:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958516-fix-stable-branches-migration-replica","meta_rev_id":"109a9e34aa1f5a8a51db8ad6a9ab03be8b2432c9","_number":958516,"virtual_id_number":958516,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-12 01:07:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2025-09-05 20:12:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2025-09-11 15:33:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":2,"date":"2025-09-11 17:52:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-04 02:34:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2025-09-11 19:35:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-26 08:36:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-27 22:08:36.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2025-09-04 02:34:52.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2025-09-05 20:12:46.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2025-09-05 20:12:46.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"}],"messages":[{"id":"57a9611dcf674807c1d41e280497a8197b42569e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-08-26 07:53:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cc75d2d6a616a5d7b9c27d2e28fe4eaaf99dcce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 08:36:24.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/616af198322c48d5809b8777cffdde36\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d38412375e494c84903ee03cffd6a6ad : FAILURE in 1m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9dcf02db0bce4148a9e6fc45906d6b4a : FAILURE in 28m 12s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4649a740a69c4589961a0bb7638f559f : FAILURE in 37m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/751a0e0dda4748739bb7401ab08e2766 : FAILURE in 36m 12s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3966ac6830824523b417dae0001e7557 : FAILURE in 32m 52s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/5d2fc34663664f6ca1658eee79d245d9 : FAILURE in 21m 00s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9a9a5b87ae8c4277a8f2b595c07548a7 : FAILURE in 22m 54s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/493d1c0f3eb9485a9f51ff7ec50781b6 : FAILURE in 18m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c9e90e9e84304bc99bf128db4ddf3219 : FAILURE in 22m 52s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/333bda46ad504b57890c782aa0e2abff : FAILURE in 27m 04s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/0fde145a2ce545e9b02bb940c39e95c7 : FAILURE in 13m 13s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/31cd535c20a44a2da93cec01cc005c5c : FAILURE in 28m 04s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/acb7ee670a624eafb85c89c758cc5b31 : FAILURE in 10m 28s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2f97fe0ffaf14e6292aacf5cff5c4ef4 : FAILURE in 23m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b4e90720964b0b4a14c19d3962412b2433db3dfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-08-26 09:33:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cc57d230ba25e01854064cbab6afe98604932393","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 10:18:43.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0601fc64c58c418d9e7747effd135918\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e7edb4924ba4d2ca807a68344c3aa2e : FAILURE in 2m 07s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7ac4969f499442148509e94423ae45f8 : FAILURE in 14m 30s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/55690335d6c64c998e9c8e12177c982d : FAILURE in 15m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f7c60de21d5848299dcfded840ae432e : FAILURE in 16m 33s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d4b493498f524b84aa81a432abae0c0d : FAILURE in 14m 28s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/3646870eb2db4942b95b7210aad9a6a7 : FAILURE in 42m 43s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9e0feaa732564a02a2f44a1ae1cba0e3 : FAILURE in 12m 07s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9a812f6f0a3b4cfe8849e0b35563e713 : FAILURE in 11m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/911a632a44884c7d8fe698e3caae5448 : FAILURE in 9m 44s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/5889a2a210a64a3d9f4d0e7612c99b69 : FAILURE in 22m 06s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/3dfa80a9020344a8b5555e7d739d7d96 : FAILURE in 18m 40s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/adc95c9d67184ea996b00d496b2532ee : FAILURE in 28m 55s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3d7c5e1a3ba246d0927563e09bf02dc7 : FAILURE in 8m 11s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/43caaaf09fe64cef83761cca803d1041 : FAILURE in 11m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"43c7e6aa5fba66b479ab704d0736651db6f810c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-08-26 14:15:22.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3c422a83eb947c08bfd7f234e223eaca44d81640","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 15:48:40.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aea3a41b737848109d871473476eb4bd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79d82403cc0348dfa833052e2a67ec01 : FAILURE in 2m 03s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f41b21b28672438b8d4dc3c77a79197c : SUCCESS in 1h 25m 51s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7bf0849e6fc94f4488a76f481927716b : SUCCESS in 1h 02m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8727907733f645cd93ef06e459b8ac0f : FAILURE in 37m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/00f1951150774a08bc0471f10123ad53 : SUCCESS in 1h 00m 17s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e4fb7c8011dd43d7899f5538fbc6be4c : SUCCESS in 45m 31s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/de1a13789d09468390f8e09d69295bab : FAILURE in 14m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fe83c414b7dc4707b4e0a59fdee1a740 : SUCCESS in 38m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/71c87e47e3ed413eb00e7a15f4d009b5 : SUCCESS in 30m 41s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/acb76a54133e470f91681340043122f3 : SUCCESS in 43m 11s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/f1c379e56f5b41ff9405c9d372e2d131 : SUCCESS in 1h 04m 36s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/deaf7ae8ff1a49239660cd828403822e : SUCCESS in 1h 20m 44s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/087663c1330c4354a676ba4eac3447cf : SUCCESS in 24m 52s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e99e24f194d844c1b24526e011ed3fb8 : FAILURE in 13m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"418f0675a3ef2f819ad75e6f162f582aff74e0ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-08-27 02:48:07.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"f0347b9449c88eadf29d19ad26e4132765b45c5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 05:12:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89296d044e634a86827f145770ce7275\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c43ad7e85c246a1b481fbee30c83e0a : SUCCESS in 2m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/077492f97c014e728c738e5923bfba7d : SUCCESS in 1h 29m 31s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/92f5e48cef6b43c49b6984fcafff5422 : FAILURE in 2h 19m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4103ee29e77c42b0974936c9a8979f16 : FAILURE in 1h 26m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/675852fba3e648f8a748c4bc5106fa91 : SUCCESS in 1h 02m 27s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c287a396e84f447c98e6dc80761de3a6 : SUCCESS in 1h 30m 02s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/da77a5a24a1a4e889f58b3a8a33688a7 : FAILURE in 22m 17s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4e236e32ff2d429796636b993f46c7af : SUCCESS in 33m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d53d292b8050450b823ae1537dc6c594 : SUCCESS in 41m 06s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/8c9679dbb54b410585d04f227e9b519b : SUCCESS in 1h 15m 05s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/92dfc7e73fdf467ba364884e94ef3243 : SUCCESS in 56m 12s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/3bdedb4237734436b608fa575dc33f90 : SUCCESS in 1h 07m 18s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d5d22a1240f54e9083201524534aa15e : SUCCESS in 31m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/eac0dda92c204db79a83d7280e7913a7 : FAILURE in 24m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7fef4ad5936cf61bbba2c18f15a4bcfe2886b014","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-27 22:08:36.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a0b9487765bdaae5f23e2136c89b4255bba8b9e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-09-02 02:29:28.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"656f06d4c8ebf2ae0aa514b0a80dd2777b9d52af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 04:35:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4733980e7434982bfcff428dacc9555\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b6529b147204bcda37785584813c387 : SUCCESS in 4m 07s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/eb1930757bf34a50b0aeec5265cf5b89 : SUCCESS in 1h 08m 39s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2bed0d29ab9b4cb3bcd4eb19cc0d932f : SUCCESS in 2h 02m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e1487581e0f74423a2c28e54f72ca3ed : FAILURE in 49m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8e63d9838f3348e68d78936561a98345 : SUCCESS in 58m 56s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/039bb87d39084780a29a64c007a3e4b2 : SUCCESS in 1h 25m 36s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/35aaf7d3943b483ba4e208fa40bae687 : FAILURE in 20m 56s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ef5534dcad844bcb8949f332838ee9ca : SUCCESS in 34m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/755c0e14047c466db6a21a7ecf56e77d : SUCCESS in 44m 03s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/63eae04a0a3342b98e44168a297671da : SUCCESS in 1h 13m 50s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/580b74ea66dc4ccaa99e40ede3e1c142 : SUCCESS in 1h 01m 18s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/b92380c7dd7b45b58dd4f6f56fe3c11c : SUCCESS in 1h 06m 16s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/789e26c59d72443097c60786cee740e0 : SUCCESS in 19m 48s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d4bab22639e046779866c0c42fa81d69 : FAILURE in 22m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a06fb8ec0446f27231d78d53a5d2d389ac005b75","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-09-02 05:18:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8de1e6d121cf1fcd262e50ceeb87419ae9514d51","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2025-09-04 02:34:52.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"6606d0226efbc25ba2c71ec5d1e608c779f613c4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-05 20:12:37.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c0195d83bd72a94d62771875983219904bbb7e54","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-09-11 15:33:07.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"1361ea3694997bb5c3f797ff32bdd0cdd31f5f5b","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2025-09-11 17:52:35.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5b3e1c71708a25872a7b1b4b297f30e662ef3aa9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-11 19:35:10.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"7c04a72fcb8718b63469a4b04df23aeeb53b315f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 19:35:33.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c8ebb6da879d00018c10d9c950c75f801461e34a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 20:29:03.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4886eb1d855c4c1794a6b8c45e7662f1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11a86231685e4d30ae9510049ac7be4d : SUCCESS in 2m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/52b0547dc3aa43c4ab317a24c0e22aaf : FAILURE in 51m 00s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/08ccc4603edd42799fa451dd24dc143d : SUCCESS in 28m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3937b537e0e049cbabc22296fc76dbef : SUCCESS in 48m 37s","accounts_in_message":[],"_revision_number":5},{"id":"aed7637aacd458d9d04b0bfe9c0156184a1f5b43","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-11 21:50:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"45db9b181c2fd0811ea6ceed00649a71c810e5ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 23:45:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ffdc6a0f37b43ca95a2afada1cd23bd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e9babd728c34726963f9b50aaf20e30 : SUCCESS in 2m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7fa87acc52434331a6fb1a23877775ce : SUCCESS in 1h 12m 53s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/dd5fc0d527be4e30a909f0d65ca05ee9 : SUCCESS in 1h 51m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/cbf1bceb806742a0ac8dbc72fe0cb587 : SUCCESS in 1h 39m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a60eda6d64984731a544f6c5544de6e0 : SUCCESS in 1h 04m 56s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/4119a61dab5d4e619ed8f527206df907 : SUCCESS in 1h 10m 01s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/bcc3239868354f0d9713821f63b2143a : FAILURE in 22m 27s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e0c1a85c0a5b41d1add00c82b7ff9b82 : SUCCESS in 31m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9560cb2b8de747f79035b658d02add41 : SUCCESS in 31m 21s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/461e88bbbb764c56bb824381ced898e0 : SUCCESS in 57m 55s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/027c6d61dc7448d59b9b9984cdc637eb : SUCCESS in 1h 09m 53s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/797be24982dd49849c04b8902e470638 : SUCCESS in 1h 10m 11s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3bf6c46576844c7bb5b10ed7111eefa9 : SUCCESS in 29m 19s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f056aab6dde24512983566ba671d9254 : FAILURE in 11m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"aae870b709b7c3f5db36ccbe772c76e1728b1a51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 23:45:24.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d1ba3778e5c95be87592cbdc3ed6f56ceeecbdcb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 01:07:12.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fdb36ae583e498bb3886e2e63dd88f6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2a7e82dac0643eea95d6c8d1035f7a3 : SUCCESS in 2m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/49b2b69331b549c59747e551214a2878 : SUCCESS in 1h 17m 18s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e7a1c7e447b544abae7177e101aaf8c0 : SUCCESS in 29m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ed5bb9d76d2a4842b594e1b01c1316da : SUCCESS in 28m 05s","accounts_in_message":[],"_revision_number":5},{"id":"109a9e34aa1f5a8a51db8ad6a9ab03be8b2432c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 01:07:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"113ab29453a6b479d470f148f969724a1a35409d","revisions":{"37564c0bb2031d5e4ecc331dcceda996aa616a21":{"kind":"REWORK","_number":1,"created":"2025-08-26 07:53:57.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/16/958516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/16/958516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/958516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/958516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/958516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/958516/1"}}},"commit":{"parents":[{"commit":"b643b1237767132be466737e7c849a72b969d5e8","subject":"[CI] Accept Ceph release defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b643b1237767132be466737e7c849a72b969d5e8"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-08-26 06:13:41.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-08-26 07:53:55.000000000","tz":480},"subject":"Fix share server migration with replica test","message":"Fix share server migration with replica test\n\nOld stable branches before 2025.1 still expect migration failed with\nreplica set.\nThis revert the old behavor for stable branches untill all our\nstable branches are above stable/2025.1\n\nResume the negative test but add enable flag so it wouldn\u0027t affect\nbranches \u003e\u003d stable/2025.1\n\nThis reverts commit 55c075de689a220c316ef02719a47de66822c2a6.\n\nChange-Id: Ia09b64e4b4afd4bb480a18e80bae46922a2210c7\nSigned-off-by: ricolin \u003crlin@vexxhost.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/37564c0bb2031d5e4ecc331dcceda996aa616a21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/37564c0bb2031d5e4ecc331dcceda996aa616a21"}]},"branch":"refs/heads/master"},"efeeb2357303776de5956bc8b7c2809e484ed96d":{"kind":"REWORK","_number":2,"created":"2025-08-26 09:33:59.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/16/958516/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/16/958516/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/958516/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/958516/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/958516/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/958516/2"}}},"commit":{"parents":[{"commit":"b643b1237767132be466737e7c849a72b969d5e8","subject":"[CI] Accept Ceph release defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b643b1237767132be466737e7c849a72b969d5e8"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-08-26 06:13:41.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-08-26 09:33:49.000000000","tz":480},"subject":"Fix share server migration with replica test","message":"Fix share server migration with replica test\n\nOld stable branches before 2025.1 still expect migration failed with\nreplica set.\nThis revert the old behavor for stable branches untill all our\nstable branches are above stable/2025.1\n\nResume the negative test but add enable flag so it wouldn\u0027t affect\nbranches \u003e\u003d stable/2025.1\n\nThis reverts commit 55c075de689a220c316ef02719a47de66822c2a6.\n\nChange-Id: Ia09b64e4b4afd4bb480a18e80bae46922a2210c7\nSigned-off-by: ricolin \u003crlin@vexxhost.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/efeeb2357303776de5956bc8b7c2809e484ed96d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/efeeb2357303776de5956bc8b7c2809e484ed96d"}]},"branch":"refs/heads/master"},"3caa57fe5429c3175789b5271224fa6dd30dcbde":{"kind":"REWORK","_number":3,"created":"2025-08-26 14:15:22.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/16/958516/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/16/958516/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/958516/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/958516/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/958516/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/958516/3"}}},"commit":{"parents":[{"commit":"b643b1237767132be466737e7c849a72b969d5e8","subject":"[CI] Accept Ceph release defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b643b1237767132be466737e7c849a72b969d5e8"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-08-26 06:13:41.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-08-26 14:15:14.000000000","tz":480},"subject":"Fix share server migration with replica test","message":"Fix share server migration with replica test\n\nOld stable branches before 2025.1 still expect migration failed with\nreplica set.\nThis revert the old behavor for stable branches untill all our\nstable branches are above stable/2025.1\n\nResume the negative test but add enable flag so it wouldn\u0027t affect\nbranches \u003e\u003d stable/2025.1\n\nThis reverts commit 55c075de689a220c316ef02719a47de66822c2a6.\n\nChange-Id: Ia09b64e4b4afd4bb480a18e80bae46922a2210c7\nSigned-off-by: ricolin \u003crlin@vexxhost.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3caa57fe5429c3175789b5271224fa6dd30dcbde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3caa57fe5429c3175789b5271224fa6dd30dcbde"}]},"branch":"refs/heads/master"},"7f29777076f79d7831a77c3fac6e6bd212575b34":{"kind":"REWORK","_number":4,"created":"2025-08-27 02:48:07.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/16/958516/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/16/958516/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/958516/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/958516/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/958516/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/958516/4"}}},"commit":{"parents":[{"commit":"b643b1237767132be466737e7c849a72b969d5e8","subject":"[CI] Accept Ceph release defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b643b1237767132be466737e7c849a72b969d5e8"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-08-26 06:13:41.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-08-27 02:47:58.000000000","tz":480},"subject":"Fix share server migration with replica test","message":"Fix share server migration with replica test\n\nOld stable branches before 2025.1 still expect migration failed with\nreplica set.\nThis revert the old behavor for stable branches untill all our\nstable branches are above stable/2025.1\n\nResume the negative test but add enable flag so it wouldn\u0027t affect\nbranches \u003e\u003d stable/2025.1\n\nThis reverts commit 55c075de689a220c316ef02719a47de66822c2a6.\n\nChange-Id: Ia09b64e4b4afd4bb480a18e80bae46922a2210c7\nSigned-off-by: ricolin \u003crlin@vexxhost.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7f29777076f79d7831a77c3fac6e6bd212575b34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7f29777076f79d7831a77c3fac6e6bd212575b34"}]},"branch":"refs/heads/master"},"113ab29453a6b479d470f148f969724a1a35409d":{"kind":"REWORK","_number":5,"created":"2025-09-02 02:29:28.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/16/958516/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/16/958516/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/958516/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/958516/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/958516/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/958516/5"}}},"commit":{"parents":[{"commit":"b643b1237767132be466737e7c849a72b969d5e8","subject":"[CI] Accept Ceph release defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b643b1237767132be466737e7c849a72b969d5e8"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-08-26 06:13:41.000000000","tz":480},"committer":{"name":"Rico Lin","email":"rlin@vexxhost.com","date":"2025-09-02 02:29:13.000000000","tz":0},"subject":"Fix share server migration with replica test","message":"Fix share server migration with replica test\n\nOld stable branches before 2025.1 still expect migration to fail with\nreplica set.\nThis reverts the old behavior for stable branches until all our\nstable branches are above stable/2025.1\n\nResume the negative test but add enable flag so it wouldn\u0027t affect\nbranches \u003e\u003d stable/2025.1\n\nThis reverts commit 55c075de689a220c316ef02719a47de66822c2a6.\n\nChange-Id: Ia09b64e4b4afd4bb480a18e80bae46922a2210c7\nSigned-off-by: ricolin \u003crlin@vexxhost.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/113ab29453a6b479d470f148f969724a1a35409d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/113ab29453a6b479d470f148f969724a1a35409d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
