)]}'
{"id":"openstack%2Fmanila-tempest-plugin~959685","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Id90044bca44387864aee8984bc2366c089c87cfc","project":"openstack/manila-tempest-plugin","branch":"master","topic":"remove-v1","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-02-12 21:20:04.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-02-12 21:20:04.000000000","reason":"Change was submitted"},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2026-02-12 16:13:13.000000000","reason":"\u003cGERRIT_ACCOUNT_32919\u003e replied on the change","reason_account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2026-02-09 20:22:30.000000000","reason":"removed on reply"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2026-02-12 21:20:04.000000000","reason":"Change was submitted"}},"hashtags":["remove-manila-api-v1"],"change_id":"Id90044bca44387864aee8984bc2366c089c87cfc","subject":"Stop testing the V1 API","status":"MERGED","created":"2025-09-05 08:45:35.000000000","updated":"2026-02-12 21:20:04.000000000","submitted":"2026-02-12 21:20:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959685-remove-v1","meta_rev_id":"0969778598952b3922feb36de0f83c4413bedad7","_number":959685,"virtual_id_number":959685,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-12 21:20:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2026-02-12 20:02:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-02-12 16:13:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":2,"date":"2026-02-12 01:57:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2026-02-12 20:02:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-05 11:39:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-30 23:23:52.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-01-26 21:47:11.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"},{"updated":"2026-01-26 21:47:11.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2026-01-26 21:47:11.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2026-02-12 20:02:36.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"947d2169ddfbeac3710b90f427c4069a0c1a773d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-05 08:45:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09688d75cf198898f0fb4b28e4c0f3c84600206a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-05 10:21:15.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a9b3cb141b8970c2289514e309ae8f9745c5c192","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 11:39:54.000000000","message":"Patch Set 2: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/79b06e038667402cb12cdf5d8967253d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6333ea864394425b47bbd18c5baceb1 : FAILURE in 1m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5c1c81c3ff914309851c55afadf1ceae : FAILURE in 51m 23s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/540dd6b1217345fe89b915177d7ee354 : FAILURE in 1h 05m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/18c18b45076e47f39bf0675322e5992f : FAILURE in 36m 28s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/01bac04585b3480487aff98c226da197 : FAILURE in 30m 34s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8b1cc9c181654f759a5d1bec42bf1672 : FAILURE in 41m 02s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4beab886c3084cd0b034666697a3ca16 : FAILURE in 10m 13s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d295488f59c94fabbe504ee933c54b0c : FAILURE in 30m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3d92e1d06f364ffea5ce907d56026359 : FAILURE in 23m 58s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/a1e74d872cbc4adf92a86d9838bee6af : FAILURE in 1h 13m 49s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/46245da8d63f4521a300091eaeee7e3e : FAILURE in 58m 18s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/a0a8484e284c4a178f06f005678f5b4d : FAILURE in 1h 08m 04s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/41a3929548504f0898bb50b822367cb1 : SUCCESS in 20m 55s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6e8519f16e944854a3234b6b1c6e94c0 : FAILURE in 10m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6c139a9130d917444fa5a01e59f97c5a0426d39f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-05 12:30:09.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f7161aedee6d160215ec260d55b5dc9870a8ec2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 14:57:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/79c98a091cb247c19f8e83e2b771aea0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/231662a804ee498a8fdd9bab82584f3c : SUCCESS in 3m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2f489ed2b7d64b7d97da9f3ab81c7d5a : SUCCESS in 1h 31m 12s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d908146e7ff748d2867a4a5e4222c80b : SUCCESS in 2h 19m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f8451dea43ba46c0b34ab7c8cb4f893a : FAILURE in 35m 06s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9b86f78498044489b44e38a2a92a7824 : FAILURE in 35m 52s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/0d7f3f8541984067a7ed99ee4cc6c0b5 : FAILURE in 49m 09s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c909e98e15ff4f318b7da7771db0f0c6 : FAILURE in 12m 17s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cfd36c36182049e795b56a9e58f5fd03 : FAILURE in 35m 11s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7315f99b45034edea2140f21045179e4 : FAILURE in 31m 03s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/f65a21fa9ca5431db15fe2b7ac19e2c9 : SUCCESS in 1h 13m 27s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/98f4d0c132334211bd8defbe66075fbb : SUCCESS in 35m 19s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/2dcf7095e1e64aaaaef55c140ebbbe3a : SUCCESS in 1h 04m 49s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/90aec5b0016e415d9c7cc3e5e373a473 : SUCCESS in 27m 15s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/dc4d3879ea5840faa2d171904c3ea835 : FAILURE in 19m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5c760f2d1681b21ce24fb9d274a9dc6eecb8154d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-30 16:27:38.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"be85c4a45ca358bdfa822288c8d834deb97b1ec4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-30 16:27:43.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"8649224a53dc0457593e527eff115718164d39b9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-30 16:27:50.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"b4219fca46a16d4c6109451f6d6d8fec4c33b63e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 19:08:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb3712122c4c4a46b5cbd42834499da6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e5abf565262443aaf7d22995dc74564 : SUCCESS in 2m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f88962aeafdc472ab76eccfb3c3337aa : SUCCESS in 1h 15m 04s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e1421648482f4c79be4886d91fa1f8fb : FAILURE in 2h 35m 14s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8de2989681ca4657af1e169d98b911f9 : FAILURE in 1h 09m 38s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/02ef12e83f854e83bfaa6564f2d8406a : FAILURE in 1h 14m 22s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/5ba2280e382c4db7b8305f81fef45d32 : FAILURE in 1h 34m 41s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/40ba63acbaec495a97d02806d7a82dd2 : FAILURE in 12m 46s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d68fdbefd6a44850aceb0128c87fa566 : FAILURE in 36m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4acd2e30e163474b86264daf37d5352c : FAILURE in 45m 51s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/d04fad89cdef4999968d64c35a7c8567 : SUCCESS in 1h 23m 29s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/da0f9206b871482db2f5414d313d6f61 : SUCCESS in 1h 15m 14s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/0dcb6339a5cd48d995613881b512263d : SUCCESS in 1h 10m 39s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3fb47be680df4a1e9df001307f2703ba : SUCCESS in 28m 44s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/551c34b6ce79444a83e094ee4c7bda8a : FAILURE in 12m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"452d643b971969c67964dbdd347596a0b440f9d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-30 23:23:52.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":5},{"id":"cc86e88e874dff4f5244f62a92d6c6d804134cc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 01:53:33.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66a9ddf3cd8a4bb6aa3dda8ced626b9c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/767072e8615646f8a8e6d6936cbc6185 : FAILURE in 2m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5a71ab9017c44d48bd5f010c043030d2 : SUCCESS in 50m 31s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3991e3daf384403f83f56e8763713b13 : SUCCESS in 2h 24m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c64007e8437044139b43f5c14ed613f5 : FAILURE in 1h 02m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/71ea34ff55a241f2ac774913e0a70f7f : FAILURE in 1h 03m 23s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/20af88a6ad47482bbfcc19bad880fb4a : FAILURE in 50m 13s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cfa746a1a6a243e5a79ca5aac3db6193 : FAILURE in 10m 39s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4546fbb236304a83aa6caf5179c307c5 : FAILURE in 27m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f7388d169f5b417c8fa8d21d35f98820 : FAILURE in 36m 23s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/76cdb02f78e04242a55d438397b684b7 : SUCCESS in 1h 20m 17s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/6c61342b0c354ff4b7736570373aa256 : SUCCESS in 47m 23s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/fa855e84294e44aaad675d415bf46b41 : FAILURE in 1h 18m 02s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/5a11d74a3879400da32f6afde2a4d92e : SUCCESS in 28m 54s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fbd949673ab0410baf715914a4fbc6e5 : FAILURE in 23m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a56d2d7d6e613a0588c1ed413e23a319bd13808b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-03 12:13:44.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"87459279fec9585b4d5fd7eea361b13efe88af12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-03 12:39:11.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a44218e682fad4740b6956102b9663ee4808c9a2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-03 12:40:24.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2a90a2e0a077df340e38929e075262a743f7589b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 14:39:14.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2bc08cdebd2847acafd2181d80a8ebb4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3bcd69602d9485985a9972b61fdb252 : SUCCESS in 1m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/960540a57c674060aa60b88c389b65cc : SUCCESS in 54m 57s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c372449df38a446188ced25cff512a63 : SUCCESS in 1h 57m 42s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f34712d1e3454b0b9bed589e93b235a6 : FAILURE in 25m 10s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/6d1ed45b0caa45ad9e2b6196105f5d6d : FAILURE in 1h 10m 19s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/005511dcced54b8f98e10b946c3b0048 : FAILURE in 1h 33m 01s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2ceb18610be04156ac839d94fd7b20fe : FAILURE in 15m 02s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2cf75dedcba546d498b12d938f8628fd : FAILURE in 34m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/641ca0b68b994c14af05ffec7183dfa7 : FAILURE in 24m 06s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/1321ceb48bf0450db8b7ed9455947f49 : SUCCESS in 1h 12m 24s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/e1dc81e003944243946d5acc7974f311 : SUCCESS in 1h 18m 45s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/2cf209c9f38f4e22aa6ec6c975d18ed0 : SUCCESS in 55m 06s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ee1d4e00317945658aa22a858355157d : SUCCESS in 32m 11s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0c33c274b3474735aa9b7bf0b68be699 : FAILURE in 19m 00s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b37505db7ab9bc702f1e73ab22ad15930a1eab30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"real_author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-20 21:40:40.000000000","message":"Patch Set 8: Patch Set 7 was rebased on behalf of \u003cGERRIT_ACCOUNT_15334\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"_revision_number":8},{"id":"94fded5027a23cca9a5e25aaa2789aa0d8e15663","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 23:10:29.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ade0a4469594157950810137f6e80f2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/224d108bd9db49949fb98db601412ef9 : SUCCESS in 3m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5259aed5bd394395a71d79b33b5651b5 : SUCCESS in 34m 36s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/0a7f4c4b3895415f93e9b89ce9d3fdf3 : SUCCESS in 51m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1855b1a98dbc4f47b64662d1764eb266 : FAILURE in 1h 24m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/48c7052511ba470b93765bd87040e84a : FAILURE in 35m 51s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/31f87952414d4d738511e7dbf0225b4b : FAILURE in 1h 09m 11s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/13a5b39f32f649ed8c73f42d68313625 : FAILURE in 1h 14m 13s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/076359a7d4be475aadb18b7ae33027d1 : FAILURE in 28m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e1227afcf06343fc94ae70f3c098c1ef : FAILURE in 42m 55s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/bab4ea24812b49a29647d015e3372a1c : SUCCESS in 1h 12m 54s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/2a8441e862464d34a9665706f93c3061 : SUCCESS in 38m 33s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/530db35773a0495faafbf0a4e9896a94 : SUCCESS in 33m 15s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/62455946152e4c96877b1931dc8d73d6 : SUCCESS in 34m 48s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1c4d5e3506344173a8ec3bafe9af7d34 : FAILURE in 1h 00m 59s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c44cdbe4f3108b811e33ee18af2559c57e242c0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-23 20:35:24.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"4896b9f7958aff443b1231eacd5c885d1a33d960","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 22:57:52.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffdb0a9ded0c439983c85654d38e81e8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fff10cee20444a01958f97784760e807 : SUCCESS in 1m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a3bf1feed0cf4f7c986156c694e23b97 : SUCCESS in 1h 09m 42s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/6f6fc8c193254c78a98b379b46fdd02f : FAILURE in 2h 16m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f2279899ff0747e1a5930841a3bb4037 : FAILURE in 1h 06m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/7fc75c7549084400a60e8223984b1c12 : FAILURE in 59m 44s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/3acc42fd7a114f9eaf0bd19669a7adc9 : FAILURE in 1h 36m 50s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4f95bf8faf8b4d32951f655003743178 : FAILURE in 1h 08m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cdea77bc1320483dbea841e619855c4f : FAILURE in 20m 42s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8e917b54d6204973b63056d0fe857ed9 : FAILURE in 38m 28s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/fccbd515827844178fe10f06ba184f3c : SUCCESS in 1h 03m 22s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/b22d8cd4aefa47c096a9ba5a7c7e40df : SUCCESS in 1h 07m 48s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/d35266921ed6492fb667edd8ad6889da : SUCCESS in 1h 15m 48s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/e5266d38106b4964b88609b14eee066f : SUCCESS in 31m 29s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a05082ad743b417b95062a2d4570a021 : FAILURE in 52m 40s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c2bdb60aec88e1a30cf5b87676e2bf233a711b14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-26 18:44:49.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"8baf8a074e7bde8d1d160ade2c9f6760e27a8569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 21:15:31.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79628e90d9344820810e999cff87e973\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a09277584894004b1694ae2b081d707 : SUCCESS in 2m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/84dabbe1da3a41e0a49e948771875323 : SUCCESS in 1h 13m 06s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e2056feb20594ed6b93aea188c53ec8c : FAILURE in 2h 16m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/08b36db24e9e43f3b8127e4a7c7d57a3 : FAILURE in 30m 22s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d20627b5bd184661bf01e1cc08567d96 : FAILURE in 58m 57s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/1dae1228a97041ca87788fe3ae408613 : SUCCESS in 1h 27m 08s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ae6b534545694dec98c6ab7e6773b331 : FAILURE in 1h 23m 34s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/298b0f355ce640eba1f01a2d0e2a0de9 : SUCCESS in 34m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ae399ead313c49499d683b1f9e987e78 : SUCCESS in 40m 33s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/a26689ff94784620aa2d79545e82750b : SUCCESS in 1h 23m 33s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/abcc4df50f344c0699a9163dc16cb1b0 : SUCCESS in 1h 18m 25s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/bae23c58bbb04026b71117bee963df78 : SUCCESS in 31m 38s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3709432c8c49437e98e7fdd6a24c17f1 : SUCCESS in 31m 26s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/28cfda872e284bbfbe0871394d965801 : FAILURE in 49m 15s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"2812c31c1dc51dd88ebd0ed6eeb04fd4c66ab268","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-29 19:57:01.000000000","message":"Topic remove-v1-base removed","accounts_in_message":[],"_revision_number":10},{"id":"e8f19202efc360c726278952bb3c35de519095e4","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-29 19:57:12.000000000","message":"Hashtag added: remove-manila-api-v1","accounts_in_message":[],"_revision_number":10},{"id":"0a014b41e3f789bf66d6b84c5cf1e5b5604791d4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-29 19:57:22.000000000","message":"Topic set to remove-v1","accounts_in_message":[],"_revision_number":10},{"id":"db874dab7d0966dc1206af7f23f46a5c6c347c12","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2026-02-09 20:22:30.000000000","message":"Patch Set 10: Code-Review-1\n\nThis change is not needed.","accounts_in_message":[],"_revision_number":10},{"id":"185b212d4a6b554c0e8db93ee4c644909d2c8b4b","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2026-02-09 20:25:41.000000000","message":"Patch Set 10: -Code-Review","accounts_in_message":[],"_revision_number":10},{"id":"b6377618f4bcfb896f10fb6dea6e18f8716cb9b0","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2026-02-09 21:02:40.000000000","message":"Patch Set 10:\n\nReverting previous vote/comment.","accounts_in_message":[],"_revision_number":10},{"id":"37fb15e4da1afeb06e57e993f6f63d1b8012edc6","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2026-02-09 21:47:38.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"2cbaecf81d9ca90083c3538ed241e2e585567ff1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 23:24:51.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/17d2e428d82f405dbe0e662a1061b352\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0868cda6e30e44e9a131475ae74f5f06 : FAILURE in 2m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6e1c50765655431f94d2205a27255a29 : SUCCESS in 35m 08s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b514b73eda324d8fa4a4430a8626f5bd : SUCCESS in 48m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d80ad736ff444f81bdeef03915a1bf2b : FAILURE in 38m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d9afcc77aacb469092c9bea992f1f489 : FAILURE in 59m 10s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/3eba4dbf156c4abca49a02afe252dfc2 : SUCCESS in 49m 13s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/371cc32971e341d594f60dece3d3e8a2 : FAILURE in 1h 04m 56s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fe615cd449c0427386b648305c570367 : SUCCESS in 38m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/19b4dd03579c4715a861dda6d387d09c : SUCCESS in 33m 23s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/845fa655c5b4432f87c4f9b493e425fc : FAILURE in 5m 59s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/b870d4351a2a497096d5403f84375e52 : FAILURE in 13m 34s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/c3b964f64fa642649564c23a7489d5a3 : SUCCESS in 1h 26m 09s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/109b5d480ad94840a6800437b19d509f : SUCCESS in 23m 48s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/386d340a50fe4b119a51eb2095923327 : FAILURE in 41m 40s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"199727a9652057c01c218437cbbedf7701299c01","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-10 23:53:56.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d5bda70d390a807bc30877c8bdaa3f143170e12e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 02:20:11.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0cddf7e01a514a3bba9abcff93ea4b3e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf1404988b1747c98b9a6ac4cdeb6f60 : FAILURE in 2m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a4568fa6586d4936a326d97551a82d4d : SUCCESS in 1h 15m 58s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/acc9112c8a724c38a764cd9cf5b77f8d : SUCCESS in 1h 42m 07s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8d37fedcd1254ca79b72659730973f99 : FAILURE in 2h 19m 33s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c582abe8b55e4a5a9a7b2517d7775f13 : FAILURE in 58m 55s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/d39c63ec7637444da6c10146ebb228ba : SUCCESS in 1h 31m 44s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f56f61ba7ecf4bc9b015a9aa12e08b47 : FAILURE in 1h 05m 31s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1e48daecf62547a8aaaf4449f0398fa1 : SUCCESS in 30m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/df2710bdbca642c58af8eda51ea17947 : SUCCESS in 40m 10s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/5919218a54c1445ca673af57b454459f : SUCCESS in 1h 11m 22s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/5f95637fa8ac4cc385d53c1d85305f4a : SUCCESS in 1h 11m 11s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/c851bca40b984ab4969d81b506e1cde6 : SUCCESS in 1h 03m 05s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/72b4431718c44c6c8136c7104ef26539 : SUCCESS in 32m 47s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/182a3cb9999241b396cfee3f22cadded : FAILURE in 31m 48s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"14c4fa23ed2b27bafb763c4fa1fbfd94a9f44526","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"real_author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-11 15:23:27.000000000","message":"Patch Set 11: Patch Set 10 was rebased on behalf of \u003cGERRIT_ACCOUNT_16643\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"_revision_number":11},{"id":"f19ae6bb7deff2762ef029ad282fd5e741de5733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 17:45:17.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b45c0c47431463c806411a7c476a27a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a31ca357d1d346fda8cd7ab898d10ff7 : SUCCESS in 3m 03s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/19bb870301c54095992655e1b3508f90 : SUCCESS in 33m 51s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/fee9f849c8ec409d9c340c2a42002f36 : SUCCESS in 2h 12m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/479df7ed71a04f17b4944bc732133e6f : FAILURE in 1h 18m 11s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/1305f841e28c463bbeff5577864242e0 : FAILURE in 1h 08m 08s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/48197965ad4f46a190c199d8e3ce6f85 : SUCCESS in 46m 16s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5b61051663904d83bdc1fa76a3b5a0a0 : FAILURE in 1h 21m 48s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4fed1a52fb8648a5a423540cfaadbf8f : SUCCESS in 34m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/36fa03db1c734d74ac1c582f8e2a9d4c : SUCCESS in 48m 35s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/f8dd5c71cc114847b16d4040da4b0054 : SUCCESS in 1h 08m 04s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/6e9b97cfc46a4988adb6b87bf5ad1d35 : SUCCESS in 1h 21m 43s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/4a6ad59b1f1f4a19bba8129be316aed2 : SUCCESS in 34m 20s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a2db602a85764dd7b7cf30110f1d1f97 : SUCCESS in 19m 01s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9eb766d61a72410eb97bf8125eefb302 : FAILURE in 29m 03s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"02d3bdc122166b12445cc8ab2d64ef7be87f6ccc","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2026-02-12 01:57:17.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"0a7ee00a559dfc21124de612c7aad1b20eb23a90","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-02-12 16:13:13.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"66f708bef7aa5aa7691b86d499a8eb8201f39f90","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-12 20:02:36.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"f733d734996bec2c1c0e8853a72c03a29cbd046b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 20:03:03.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"1e90d62e926b0725e62bb79677d894442eb82476","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2026-02-12 20:06:34.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"5a6ab9c8a503cbc1a35355c2ef12b455dc88dcd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 21:20:03.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6992cbb6934c47a48544e9d1dc12c8fa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9348533054ad40cd9f4b1cd59c27548b : SUCCESS in 4m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/906230f21f534106bb3589f0e3403e8c : SUCCESS in 1h 12m 19s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c1d8a5f21caa46c59f6666a86e5afea3 : SUCCESS in 30m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/dd1099b25e6c4cc29bf9b914db7982c5 : SUCCESS in 41m 16s","accounts_in_message":[],"_revision_number":11},{"id":"0969778598952b3922feb36de0f83c4413bedad7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 21:20:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"878716d715e307fe3406d20d9817eb4ba3b86c0e","revisions":{"03e58129758bd3e890e7279d645ad0dcbce1a1d0":{"kind":"REWORK","_number":1,"created":"2025-09-05 08:45:35.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/85/959685/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/959685/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/1"}}},"commit":{"parents":[{"commit":"53cf5e810bec87fc7acfee60709b94508ccecd41","subject":"Merge \"Drop empty documentation sections\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/53cf5e810bec87fc7acfee60709b94508ccecd41"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 07:18:03.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 08:45:27.000000000","tz":60},"subject":"Remove use of a v1 client","message":"Remove use of a v1 client\n\nExcept where specifically being tests in a v1 legacy test (that\nwe can skip).\n\nChange-Id: Id90044bca44387864aee8984bc2366c089c87cfc\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/03e58129758bd3e890e7279d645ad0dcbce1a1d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/03e58129758bd3e890e7279d645ad0dcbce1a1d0"}]},"branch":"refs/heads/master"},"37db2b85a69af9e6e696944a81d35d425a4e288e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-09-05 10:21:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/85/959685/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/959685/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/2"}}},"commit":{"parents":[{"commit":"53cf5e810bec87fc7acfee60709b94508ccecd41","subject":"Merge \"Drop empty documentation sections\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/53cf5e810bec87fc7acfee60709b94508ccecd41"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 07:18:03.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 10:21:15.000000000","tz":0},"subject":"Remove use of a v1 client","message":"Remove use of a v1 client\n\nChange-Id: Id90044bca44387864aee8984bc2366c089c87cfc\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/37db2b85a69af9e6e696944a81d35d425a4e288e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/37db2b85a69af9e6e696944a81d35d425a4e288e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8f43f1a4252ac72477617a2433b92bf8208e9ae7":{"kind":"REWORK","_number":3,"created":"2025-09-05 12:30:09.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/85/959685/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/959685/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/3"}}},"commit":{"parents":[{"commit":"53cf5e810bec87fc7acfee60709b94508ccecd41","subject":"Merge \"Drop empty documentation sections\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/53cf5e810bec87fc7acfee60709b94508ccecd41"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 07:18:03.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 12:30:01.000000000","tz":60},"subject":"Remove use of a v1 client","message":"Remove use of a v1 client\n\nChange-Id: Id90044bca44387864aee8984bc2366c089c87cfc\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8f43f1a4252ac72477617a2433b92bf8208e9ae7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8f43f1a4252ac72477617a2433b92bf8208e9ae7"}]},"branch":"refs/heads/master"},"08a69e6944abff73b15f538d0b903dba156720fb":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-10-30 16:27:38.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/85/959685/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/959685/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/4"}}},"commit":{"parents":[{"commit":"496237a646a3cff808a52bce540356029f69c5b0","subject":"Merge \"Remove remaining job with Ubuntu Jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/496237a646a3cff808a52bce540356029f69c5b0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 07:18:03.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-30 16:27:15.000000000","tz":0},"subject":"Remove use of a v1 client","message":"Remove use of a v1 client\n\nChange-Id: Id90044bca44387864aee8984bc2366c089c87cfc\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/08a69e6944abff73b15f538d0b903dba156720fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/08a69e6944abff73b15f538d0b903dba156720fb"}]},"branch":"refs/heads/master"},"17981e3dd83009c2a1d1f7aef402ba1f1fa7ef7c":{"kind":"REWORK","_number":5,"created":"2025-10-30 23:23:52.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/85/959685/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/959685/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/5"}}},"commit":{"parents":[{"commit":"496237a646a3cff808a52bce540356029f69c5b0","subject":"Merge \"Remove remaining job with Ubuntu Jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/496237a646a3cff808a52bce540356029f69c5b0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 07:18:03.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-10-30 23:23:41.000000000","tz":-420},"subject":"Stop testing the V1 API","message":"Stop testing the V1 API\n\nThe /v1 API has been deprecated for a long time [1];\nthe tempest plugin will stop testing it so it can\nbe removed in the Manila service.\n\nRemoval of this code will affect testing in older\nmaintained stable branches; however, the long\ndeprecation is reason enough to not go lengths to\npreserve testing in these older branches.\n\nChange-Id: Id90044bca44387864aee8984bc2366c089c87cfc\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n[1] https://lists.openstack.org/pipermail/openstack-dev/2016-June/096517.html\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/17981e3dd83009c2a1d1f7aef402ba1f1fa7ef7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/17981e3dd83009c2a1d1f7aef402ba1f1fa7ef7c"}]},"branch":"refs/heads/master"},"6ef462a9666c728d3f7d3e4765461b282059cde9":{"kind":"REWORK","_number":6,"created":"2025-11-03 12:13:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/85/959685/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/959685/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/6"}}},"commit":{"parents":[{"commit":"496237a646a3cff808a52bce540356029f69c5b0","subject":"Merge \"Remove remaining job with Ubuntu Jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/496237a646a3cff808a52bce540356029f69c5b0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 07:18:03.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-03 12:08:25.000000000","tz":0},"subject":"Stop testing the V1 API","message":"Stop testing the V1 API\n\nThe /v1 API has been deprecated for a long time [1]; the tempest plugin\nwill stop testing it so it can be removed in the Manila service.\n\nRemoval of this code will affect testing in older maintained stable\nbranches; however, the long deprecation is reason enough to not go\nlengths to preserve testing in these older branches.\n\n[1] https://lists.openstack.org/pipermail/openstack-dev/2016-June/096517.html\n\nChange-Id: Id90044bca44387864aee8984bc2366c089c87cfc\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6ef462a9666c728d3f7d3e4765461b282059cde9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6ef462a9666c728d3f7d3e4765461b282059cde9"}]},"branch":"refs/heads/master"},"e6cdf0e2e5ec0206ae4558e481daa6a8cb8f0783":{"kind":"REWORK","_number":7,"created":"2025-11-03 12:39:11.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/85/959685/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/959685/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/7"}}},"commit":{"parents":[{"commit":"496237a646a3cff808a52bce540356029f69c5b0","subject":"Merge \"Remove remaining job with Ubuntu Jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/496237a646a3cff808a52bce540356029f69c5b0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 07:18:03.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-03 12:36:08.000000000","tz":0},"subject":"Stop testing the V1 API","message":"Stop testing the V1 API\n\nThe /v1 API has been deprecated for a long time [1]; the tempest plugin\nwill stop testing it so it can be removed in the Manila service.\n\nRemoval of this code will affect testing in older maintained stable\nbranches; however, the long deprecation is reason enough to not go\nlengths to preserve testing in these older branches.\n\n[1] https://lists.openstack.org/pipermail/openstack-dev/2016-June/096517.html\n\nChange-Id: Id90044bca44387864aee8984bc2366c089c87cfc\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e6cdf0e2e5ec0206ae4558e481daa6a8cb8f0783"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e6cdf0e2e5ec0206ae4558e481daa6a8cb8f0783"}]},"branch":"refs/heads/master"},"cab5291a9d891620a2ceacb229fe1f6a3a0256cd":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-01-20 21:40:40.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"real_uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/85/959685/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/959685/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/8"}}},"commit":{"parents":[{"commit":"3f76739525905fc34792d9c2490f3546c3af75b2","subject":"Merge \"Refactor stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3f76739525905fc34792d9c2490f3546c3af75b2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 07:18:03.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-20 21:40:40.000000000","tz":0},"subject":"Stop testing the V1 API","message":"Stop testing the V1 API\n\nThe /v1 API has been deprecated for a long time [1]; the tempest plugin\nwill stop testing it so it can be removed in the Manila service.\n\nRemoval of this code will affect testing in older maintained stable\nbranches; however, the long deprecation is reason enough to not go\nlengths to preserve testing in these older branches.\n\n[1] https://lists.openstack.org/pipermail/openstack-dev/2016-June/096517.html\n\nChange-Id: Id90044bca44387864aee8984bc2366c089c87cfc\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cab5291a9d891620a2ceacb229fe1f6a3a0256cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cab5291a9d891620a2ceacb229fe1f6a3a0256cd"}]},"branch":"refs/heads/master","description":"Rebase"},"a0d066573e194f4abf7000be81f45dc5a3f0af51":{"kind":"REWORK","_number":9,"created":"2026-01-23 20:35:24.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/85/959685/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/959685/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/9"}}},"commit":{"parents":[{"commit":"3f76739525905fc34792d9c2490f3546c3af75b2","subject":"Merge \"Refactor stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3f76739525905fc34792d9c2490f3546c3af75b2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 07:18:03.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-01-23 20:35:17.000000000","tz":-480},"subject":"Stop testing the V1 API","message":"Stop testing the V1 API\n\nThe /v1 API has been deprecated for a long time [1]; the tempest plugin\nwill stop testing it so it can be removed in the Manila service.\n\nRemoval of this code will affect testing in older maintained stable\nbranches; however, the long deprecation is reason enough to not go\nlengths to preserve testing in these older branches.\n\n[1] https://lists.openstack.org/pipermail/openstack-dev/2016-June/096517.html\n\nChange-Id: Id90044bca44387864aee8984bc2366c089c87cfc\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a0d066573e194f4abf7000be81f45dc5a3f0af51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a0d066573e194f4abf7000be81f45dc5a3f0af51"}]},"branch":"refs/heads/master"},"7be7cb0b4295f9a94b933377932aaf537b726878":{"kind":"REWORK","_number":10,"created":"2026-01-26 18:44:49.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/85/959685/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/959685/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/10"}}},"commit":{"parents":[{"commit":"3f76739525905fc34792d9c2490f3546c3af75b2","subject":"Merge \"Refactor stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3f76739525905fc34792d9c2490f3546c3af75b2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 07:18:03.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-01-26 18:40:50.000000000","tz":-480},"subject":"Stop testing the V1 API","message":"Stop testing the V1 API\n\nThe /v1 API has been deprecated for a long time [1]; the tempest plugin\nwill stop testing it so it can be removed in the Manila service.\n\nRemoval of this code will affect testing in older maintained stable\nbranches; however, the long deprecation is reason enough to not go\nlengths to preserve testing in these older branches.\n\n[1] https://lists.openstack.org/pipermail/openstack-dev/2016-June/096517.html\n\nChange-Id: Id90044bca44387864aee8984bc2366c089c87cfc\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7be7cb0b4295f9a94b933377932aaf537b726878"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7be7cb0b4295f9a94b933377932aaf537b726878"}]},"branch":"refs/heads/master"},"878716d715e307fe3406d20d9817eb4ba3b86c0e":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-02-11 15:23:27.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"real_uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/85/959685/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/959685/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/959685/11"}}},"commit":{"parents":[{"commit":"febe94d18e28298f4f476df9c55725860924658b","subject":"Bump flake8-import-order version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/febe94d18e28298f4f476df9c55725860924658b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 07:18:03.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-02-11 15:23:27.000000000","tz":0},"subject":"Stop testing the V1 API","message":"Stop testing the V1 API\n\nThe /v1 API has been deprecated for a long time [1]; the tempest plugin\nwill stop testing it so it can be removed in the Manila service.\n\nRemoval of this code will affect testing in older maintained stable\nbranches; however, the long deprecation is reason enough to not go\nlengths to preserve testing in these older branches.\n\n[1] https://lists.openstack.org/pipermail/openstack-dev/2016-June/096517.html\n\nChange-Id: Id90044bca44387864aee8984bc2366c089c87cfc\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/878716d715e307fe3406d20d9817eb4ba3b86c0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/878716d715e307fe3406d20d9817eb4ba3b86c0e"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
