)]}'
{"id":"openstack%2Fmanila-tempest-plugin~959686","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I7be74fc22594d3a59932b8bc469f9209cf985fe6","project":"openstack/manila-tempest-plugin","branch":"master","topic":"remove-v1","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-02-13 14:48:18.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-02-13 14:48:18.000000000","reason":"Change was submitted"}},"hashtags":["remove-manila-api-v1"],"change_id":"I7be74fc22594d3a59932b8bc469f9209cf985fe6","subject":"Remove Manila V1 API client","status":"MERGED","created":"2025-09-05 08:45:35.000000000","updated":"2026-02-13 14:48:18.000000000","submitted":"2026-02-13 14:48:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959686-remove-v1","meta_rev_id":"cffa22f99b3a169deaf5e6d9b34a860e135b63ac","_number":959686,"virtual_id_number":959686,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-13 14:48:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2026-02-13 13:38:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2026-02-13 13:10:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":1,"date":"2026-02-13 06:09:25.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2026-02-13 13:38:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-05 12:08:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-29 19:33:54.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-02-13 06:09:25.000000000","updated_by":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-02-13 13:10:58.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2026-02-13 13:38:47.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"ec7f9c7feef456651c04c0c89a86c0b66f49b61d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-05 08:45:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b239a3e61ddfc22c7f2ede38145e349a9354f0ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-05 10:21:19.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a8682e7a45c08fcfd9d14597afb8caa01397eed6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 12:08:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/af8fc611a5174ed6bb6cc22a6d65d34f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72ac146458594d6ba1e62f61d7b9b8e8 : FAILURE in 3m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3b198f246e32404499f605f9cb180cb6 : FAILURE in 55m 43s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/6a4e5206a55c4e5e810402afe6fb9c4b : FAILURE in 1h 44m 07s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/03a528fb95b44ad9b7089e13cc50ba40 : FAILURE in 1h 16m 16s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/4958c406432449c997ef10c57ee0929a : FAILURE in 51m 42s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/3eedc53ea36c4e1294af4c11a8b8f7de : FAILURE in 1h 12m 19s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/abfc276dbd884630ad18d662a9d19182 : FAILURE in 16m 19s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/76711f60f4c54f3eb52f2b61bb3f98d6 : FAILURE in 33m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/dd3cd2512bf34610b8f15a8c0289c07c : FAILURE in 26m 15s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/26f8163f2f6145698f0a6a34d1889dfe : FAILURE in 55m 15s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/c0787eb9c0e54a32aff5c7613833cd39 : FAILURE in 1h 13m 07s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/a7c95e4ac5a94c13b50c30156a1965d8 : FAILURE in 1h 08m 51s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/46aaf8289224441db02cf10de8e388af : SUCCESS in 25m 58s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/255e52862f6341b19a3bc2f7a104feaa : FAILURE in 23m 25s (non-voting)\n\nWarning:\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_shares_actions.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_security_services_mapping.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_share_networks_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_shares.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_shares_actions.py","accounts_in_message":[],"_revision_number":2},{"id":"607857e5ff391a8ff22b5f129b3420359278d6d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-05 12:30:09.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"703becbdccb6eddb315a640efe55eed4f5b2eca9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 14:39:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/07dabe18337e48389de3865af1c964be\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91a5610886084614a63b8f0b5971e9c7 : SUCCESS in 3m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4e468abc59bc42ae9866170dec2b5803 : SUCCESS in 39m 33s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4f042cde3485404b85ef34254996ac7c : FAILURE in 1h 58m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/967dce244d20414ca1cdd45cc584ba2b : FAILURE in 31m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f2eb72cdbb8844688a3d7b2117a9eab3 : FAILURE in 58m 23s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/32a6919818be4b20af9d3f0005a89fa3 : FAILURE in 1h 34m 47s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/462b38a43894449b940262935ebdbd58 : FAILURE in 22m 02s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/15376d7f3021424bb9052a3e9d1c933e : FAILURE in 33m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c49cb07477da4b90a6903944ec87e399 : FAILURE in 36m 25s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/bee151e9dc034fc195bc6a902cc45a71 : SUCCESS in 35m 24s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/68c0f26923c540199c7deed48de76c87 : SUCCESS in 49m 39s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/a875832ccc9249c7b6cebac9bbad7619 : SUCCESS in 1h 11m 13s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/33bebb048476471389c3cf74a971c1b3 : SUCCESS in 32m 09s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7c0986985d4c4bf09389827d08512dc8 : FAILURE in 25m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2b4641edba5fe6e4a18d6919af1e7fa025e3907f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-30 16:27:38.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9f0318edd5351527e1d9f056200b25d064589259","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 18:24:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d244257dece547e4afd6b12901bdd796\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2813ac0d701d441e80da297e85a22e36 : SUCCESS in 2m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/86b6582911ce43e29092ab2f23353157 : SUCCESS in 1h 12m 32s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8fb45376604c49a78e4b775d692a8fb0 : FAILURE in 1h 38m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0ec6cb05a7e243079a8e785dbfb8d2df : FAILURE in 58m 41s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b328411a23b2491ca513311519516f12 : FAILURE in 30m 20s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/9a9b9e2d7eeb46908bd6742495bb3e78 : FAILURE in 1h 20m 40s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/19373de267ab45098afa9f038729344b : FAILURE in 21m 41s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ccadfd20011d4190af37abe62f4956fa : FAILURE in 36m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/48e12befd36e4c1c851c82f22eb1d910 : FAILURE in 38m 11s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/db243e38df1c4fd4ae46119735f153de : SUCCESS in 36m 08s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/42b0bfa92f6641e29620803e5cbf8a85 : SUCCESS in 56m 21s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/4be4bac5f81a42c5b3ccf69904b5eaee : SUCCESS in 1h 22m 30s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f1cf3f11001d423a94541cfb0d2c29ba : SUCCESS in 34m 22s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/018cb82c2a224bc1b5ebd439bceaed85 : FAILURE in 17m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e23459ea584097fa1fb991262589e3ecb50fb36c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"real_author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-20 21:40:46.000000000","message":"Patch Set 5: Patch Set 4 was rebased on behalf of \u003cGERRIT_ACCOUNT_15334\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"_revision_number":5},{"id":"b3d42e4a69a390e03dc247ad787db0c50217a96d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 23:28:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b19061a0476a401e99baf3177c478d96\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff95900ee3f544bfa8f551e55577df68 : SUCCESS in 4m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f3c3615a21c04bf8a1d9820cbe4e9fcb : SUCCESS in 53m 38s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3f5610d74f35422ebb9f43276fa46161 : FAILURE in 1h 44m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e7de3c5daf94412da247392ddf6ef0bf : FAILURE in 23m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/222c6e92f8644864961685f972f056f5 : FAILURE in 55m 00s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/678b8dc16a8d4243b4b76a578a9951b6 : FAILURE in 1h 40m 16s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/448fc1430f54496bbccada5896bc0868 : FAILURE in 1h 20m 10s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f9802c78bdf94bb5810db594d08883bb : FAILURE in 30m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a598ea71c35a48678889d99ad2e9d6f3 : FAILURE in 30m 35s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/bf91419715a14178a74b466e99b7acc1 : SUCCESS in 1h 02m 00s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/fb99d972918d4c7ab5fc359449a6eb12 : SUCCESS in 1h 06m 30s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/a217dfde33e8413cab4baa6d9826183d : SUCCESS in 1h 11m 54s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ee790724954a4d44922179f37b305b51 : SUCCESS in 30m 25s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6c6b408872ee4d35ad39e1883d0ff8fb : FAILURE in 35m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7c512b582aacbbec5e9fd6f5d13616c3d3578918","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"real_author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-23 20:37:51.000000000","message":"Patch Set 6: Patch Set 5 was rebased on behalf of \u003cGERRIT_ACCOUNT_15334\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"_revision_number":6},{"id":"356f76e913d19534f6ed174b8ad25197a15beef0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 22:57:16.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/52ea2253039a47e6be855af2ac9cefa5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46d13e135590408db7e76794c7bb9cba : SUCCESS in 2m 37s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f59245cd50b1442e9890d253ad8792f3 : SUCCESS in 1h 02m 35s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/620582e583da426189270128f80349cf : FAILURE in 2h 14m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ded2b8ba4a6c49ca869af70194ae62a0 : FAILURE in 49m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/034a307af80f469e870b5fe273058908 : FAILURE in 1h 00m 26s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/6a045c1d7a1a4132a61b899e33327199 : FAILURE in 1h 14m 28s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/17ac4689259241e294afa9bc98b441d5 : FAILURE in 1h 00m 18s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6558fb32af3a488ab749a85d6c9474f9 : FAILURE in 33m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/15bee2802ccb4b24abd8f6026de01e40 : FAILURE in 43m 01s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/9a52f717974141a69efba7f9f03482ee : SUCCESS in 53m 11s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/5a39f980edc44b68b393d214ccf3cfde : SUCCESS in 52m 45s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/11baf95597794ca2b822fb46654768de : SUCCESS in 52m 25s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/bd91ecfeb1b643eaa257ae2d4b481b60 : SUCCESS in 34m 27s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9ab860b54b4d403ab462ca96c3932161 : FAILURE in 59m 54s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"83be065aa8147682fb0c492a148670db23048daf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-26 18:49:39.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"bec7988449613556a44002e5fc2859d2db475877","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-26 18:49:48.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"fc4ed74eb88600ddc7317699c9cb7a9693ebc9ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 21:25:33.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/27a2d93b5fea4b4097bdae7341ced3d5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/940b1e8ba902451b89fec90c54a912ad : SUCCESS in 2m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/680edac594d64a5bab6a145c7e80e8a0 : SUCCESS in 1h 13m 24s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/bba5eff4df104a56b26df2b6b56f3471 : FAILURE in 2h 29m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b8bee5d1c61c478b833d62cf801a712a : FAILURE in 58m 51s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/26fb953fb3b94eedb1bb1421143b644e : FAILURE in 59m 18s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/6b352f3820ae4fbebc336a0af45f28a7 : FAILURE in 1h 35m 38s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9602f2de89644008b9331963eef7cc28 : FAILURE in 51m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d622a17cfe5940748cfab3a244eb895b : FAILURE in 37m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a8290c9203ee4d00b324a4cd49dade41 : FAILURE in 42m 05s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/5e5ca43b3bb744269935d92a644c2d90 : SUCCESS in 1h 12m 02s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/26e336f45a2a4e8fae940bf77d2d1774 : SUCCESS in 1h 08m 08s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/579c0747f7e74b2ba6719562d82f5529 : SUCCESS in 1h 04m 11s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/917a7cad5f474abea20dd35160e133d5 : SUCCESS in 38m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/49661bf17b1a45e09754f0563b67b570 : FAILURE in 58m 43s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ac26764781d7879c73a0ac6025335ccdbc24c2c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-29 19:33:54.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"ad497cd7f37b0865959d4a256f9a739e5378a457","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-29 19:35:17.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ea956649171e195a1c89ad184c7574b559251d2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-29 19:44:09.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"11e70b954c6d0293ca1ab0e564df444836178cdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-29 19:44:20.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"ed0c03889543b7912cca54f8f0e73eff80eb6ad4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-29 19:56:41.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"c4fbd6974d31c07cb24e09406554397f3000a433","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-29 19:57:27.000000000","message":"Hashtag added: remove-manila-api-v1","accounts_in_message":[],"_revision_number":12},{"id":"d210fb8267a02b019fbe8b8cf8571996825fc17c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 22:17:25.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc05a92640ff463a9d4ffec78fbaf0e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fee720d8cae473388f87c753a667427 : SUCCESS in 2m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/95609287e5ec422a9e7af907a23caba2 : SUCCESS in 29m 05s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/0bdadc7a76da410c812a43bce0760b70 : FAILURE in 2h 20m 07s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9dbae4f7490a4a1f9c8d19c0d45243eb : FAILURE in 1h 30m 57s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/4f5eb53d4a2d48baaa7b377edee5ab34 : FAILURE in 1h 13m 14s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/77c98c2a45204c8dbd8fc2726620e81a : SUCCESS in 1h 14m 30s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b353d51b6896432ebeb977deefe1f300 : FAILURE in 34m 01s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b22bf1bf48ae454e9ea8985c44a0f03b : SUCCESS in 23m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f1c1c5e7a32c462d8817d7c1ba52dd5e : SUCCESS in 30m 52s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/a6c3b2a04249433cbea68e750473ece1 : SUCCESS in 39m 11s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/8761cc446f0e4a9d9b0e020fa3f9f6a9 : SUCCESS in 1h 11m 53s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/1ecd6aa2fdc44890a67c19aa084020e3 : SUCCESS in 33m 23s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/c6ab2c95b68744f7ab21d698b088ac74 : SUCCESS in 28m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3d502baec924436b813ad41393a8b7fd : FAILURE in 31m 46s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"de66577b7e496f7fa42ffb0cba336e17b09eb981","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"real_author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-11 15:23:32.000000000","message":"Patch Set 13: Patch Set 12 was rebased on behalf of \u003cGERRIT_ACCOUNT_16643\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"_revision_number":13},{"id":"3c31a06344eebec12b7bcfc33c16a5d615edef60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 17:38:34.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/998c772f070d40e3ac1a501322a94336\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39766fd7de0c4b05bcdcd0e3207d458d : SUCCESS in 3m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/863ea4e137044cd68ef952db37713690 : SUCCESS in 53m 57s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/bf71250d9b4741fe99e06e533ab0d033 : SUCCESS in 2h 07m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a18cf00267c041a4b661a25f03694865 : FAILURE in 37m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9954ddade0ac4aa09ab759264c12245d : FAILURE in 56m 55s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e1a51638664c4bf5bbcb71f902349cb3 : SUCCESS in 1h 13m 52s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7f8456bdbca84d6099fc49b39ea559ea : FAILURE in 42m 18s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/50c3fae9657f4dc38e0da279e5a9d7d4 : SUCCESS in 26m 42s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4a3096508618472197b428a3e113e0d7 : SUCCESS in 39m 53s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/c2879c7032464571a40c77eb6e8279b8 : SUCCESS in 29m 51s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/18f4911ffa4847318431ba0c31a8e76a : SUCCESS in 1h 12m 41s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/63d6fbcbcb494d1286422c99fd45cfd0 : SUCCESS in 1h 15m 11s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/01407aa786fd4f4ab3920a1a4c283b9a : SUCCESS in 33m 35s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6fbfa7ef712d4863b917df1353182e4e : FAILURE in 1h 02m 05s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"00036399f62fec676f21052594b74475578fe331","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-02-13 06:09:25.000000000","message":"Patch Set 13: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"038cd96c84934f84e0f4a2d53e4f8c611326b23d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-13 12:19:27.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"d46e4bca87a69aa94b34e74a5d0f22372032c674","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2026-02-13 13:10:58.000000000","message":"Patch Set 13: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"ae5d2a4854e3990c74ba382bace60994fb4641e5","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-13 13:38:47.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"689fcef5326e3ec7e04facca5880c3b0fd94a57d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 13:39:11.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"3c2ff5e167e3efdb160390f2b66046ab1f7808ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 14:48:18.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02393ab5ee1a47a3aca1ea0882f814de\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40cec01b87ae4f59a9922ae73522c9b6 : SUCCESS in 3m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/78a0a0c1f45e416dbf019af8568cae73 : SUCCESS in 1h 07m 19s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c9a3867f3bf14fc09b6894d004ae2cb1 : SUCCESS in 31m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c2da41fc19cf429dba3968de9047d42f : SUCCESS in 44m 34s","accounts_in_message":[],"_revision_number":13},{"id":"cffa22f99b3a169deaf5e6d9b34a860e135b63ac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 14:48:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"6406026b51e9c282902dbc3987f1edb9879821b8","revisions":{"ad9d9ae5388596eae6919de321cf4842707ae56a":{"kind":"REWORK","_number":1,"created":"2025-09-05 08:45:35.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/86/959686/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/86/959686/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/1"}}},"commit":{"parents":[{"commit":"03e58129758bd3e890e7279d645ad0dcbce1a1d0","subject":"Remove use of a v1 client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/03e58129758bd3e890e7279d645ad0dcbce1a1d0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 08:40:45.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 08:45:29.000000000","tz":60},"subject":"Remove v1 client","message":"Remove v1 client\n\nChange-Id: I7be74fc22594d3a59932b8bc469f9209cf985fe6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad9d9ae5388596eae6919de321cf4842707ae56a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad9d9ae5388596eae6919de321cf4842707ae56a"}]},"branch":"refs/heads/master"},"25c2b41a403786fd8d187fb375f62b18a90ff945":{"kind":"NO_CHANGE","_number":2,"created":"2025-09-05 10:21:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/86/959686/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/86/959686/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/2"}}},"commit":{"parents":[{"commit":"37db2b85a69af9e6e696944a81d35d425a4e288e","subject":"Remove use of a v1 client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/37db2b85a69af9e6e696944a81d35d425a4e288e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 08:40:45.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 10:21:19.000000000","tz":0},"subject":"Remove v1 client","message":"Remove v1 client\n\nChange-Id: I7be74fc22594d3a59932b8bc469f9209cf985fe6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/25c2b41a403786fd8d187fb375f62b18a90ff945"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/25c2b41a403786fd8d187fb375f62b18a90ff945"}]},"branch":"refs/heads/master","description":"Rebase"},"8e93baa98d5ea1c8e2e50dae5b1ae56681816063":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-09-05 12:30:09.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/86/959686/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/86/959686/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/3"}}},"commit":{"parents":[{"commit":"8f43f1a4252ac72477617a2433b92bf8208e9ae7","subject":"Remove use of a v1 client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8f43f1a4252ac72477617a2433b92bf8208e9ae7"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 08:40:45.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 12:30:04.000000000","tz":60},"subject":"Remove v1 client","message":"Remove v1 client\n\nChange-Id: I7be74fc22594d3a59932b8bc469f9209cf985fe6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8e93baa98d5ea1c8e2e50dae5b1ae56681816063"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8e93baa98d5ea1c8e2e50dae5b1ae56681816063"}]},"branch":"refs/heads/master"},"bc6b496cf2f5ea8a1b0c4269f2e500cc58eb3d2b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-10-30 16:27:38.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/86/959686/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/86/959686/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/4"}}},"commit":{"parents":[{"commit":"08a69e6944abff73b15f538d0b903dba156720fb","subject":"Remove use of a v1 client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/08a69e6944abff73b15f538d0b903dba156720fb"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 08:40:45.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-30 16:27:15.000000000","tz":0},"subject":"Remove v1 client","message":"Remove v1 client\n\nChange-Id: I7be74fc22594d3a59932b8bc469f9209cf985fe6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bc6b496cf2f5ea8a1b0c4269f2e500cc58eb3d2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bc6b496cf2f5ea8a1b0c4269f2e500cc58eb3d2b"}]},"branch":"refs/heads/master"},"82461151f36042fadceda29395422b7a7557abe8":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-01-20 21:40:46.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"real_uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/86/959686/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/86/959686/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/5"}}},"commit":{"parents":[{"commit":"cab5291a9d891620a2ceacb229fe1f6a3a0256cd","subject":"Stop testing the V1 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cab5291a9d891620a2ceacb229fe1f6a3a0256cd"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 08:40:45.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-20 21:40:46.000000000","tz":0},"subject":"Remove v1 client","message":"Remove v1 client\n\nChange-Id: I7be74fc22594d3a59932b8bc469f9209cf985fe6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/82461151f36042fadceda29395422b7a7557abe8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/82461151f36042fadceda29395422b7a7557abe8"}]},"branch":"refs/heads/master","description":"Rebase"},"48d4e08da0326adcd1caa39a5aa4b30cba96ca8c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-01-23 20:37:51.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"real_uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/86/959686/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/86/959686/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/6"}}},"commit":{"parents":[{"commit":"a0d066573e194f4abf7000be81f45dc5a3f0af51","subject":"Stop testing the V1 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a0d066573e194f4abf7000be81f45dc5a3f0af51"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 08:40:45.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-23 20:37:51.000000000","tz":0},"subject":"Remove v1 client","message":"Remove v1 client\n\nChange-Id: I7be74fc22594d3a59932b8bc469f9209cf985fe6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/48d4e08da0326adcd1caa39a5aa4b30cba96ca8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/48d4e08da0326adcd1caa39a5aa4b30cba96ca8c"}]},"branch":"refs/heads/master","description":"Rebase"},"8109494ad94f145d5721610808413c696d899faf":{"kind":"REWORK","_number":7,"created":"2026-01-26 18:49:39.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/86/959686/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/86/959686/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/7"}}},"commit":{"parents":[{"commit":"a0d066573e194f4abf7000be81f45dc5a3f0af51","subject":"Stop testing the V1 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a0d066573e194f4abf7000be81f45dc5a3f0af51"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 08:40:45.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-01-26 18:47:38.000000000","tz":-480},"subject":"Remove Manila V1 API client","message":"Remove Manila V1 API client\n\nThe Manila V1 API has been deprecated since 2016 and tests\nstopped using it in change Id90044bca44387864aee8984bc2366c089c87cfc.\nComplete this removal by:\n\n1. Making SharesV2Client inherit directly from RestClient\n   instead of the V1 SharesClient\n2. Add microversion support added to all methods\n3. Removing V1 client registration from plugin.py\n\nChange-Id: I7be74fc22594d3a59932b8bc469f9209cf985fe6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8109494ad94f145d5721610808413c696d899faf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8109494ad94f145d5721610808413c696d899faf"}]},"branch":"refs/heads/master"},"c401ed7887de53d1ea49b6949b35e8570daf1cad":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-01-26 18:49:48.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/86/959686/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/86/959686/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/8"}}},"commit":{"parents":[{"commit":"7be7cb0b4295f9a94b933377932aaf537b726878","subject":"Stop testing the V1 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7be7cb0b4295f9a94b933377932aaf537b726878"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-05 08:40:45.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-01-26 18:49:48.000000000","tz":0},"subject":"Remove Manila V1 API client","message":"Remove Manila V1 API client\n\nThe Manila V1 API has been deprecated since 2016 and tests\nstopped using it in change Id90044bca44387864aee8984bc2366c089c87cfc.\nComplete this removal by:\n\n1. Making SharesV2Client inherit directly from RestClient\n   instead of the V1 SharesClient\n2. Add microversion support added to all methods\n3. Removing V1 client registration from plugin.py\n\nChange-Id: I7be74fc22594d3a59932b8bc469f9209cf985fe6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c401ed7887de53d1ea49b6949b35e8570daf1cad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c401ed7887de53d1ea49b6949b35e8570daf1cad"}]},"branch":"refs/heads/master","description":"Rebase"},"7cda2f5c5c894e483118a294f8844e3f0005889c":{"kind":"REWORK","_number":9,"created":"2026-01-29 19:33:54.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/86/959686/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/86/959686/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/9"}}},"commit":{"parents":[{"commit":"7be7cb0b4295f9a94b933377932aaf537b726878","subject":"Stop testing the V1 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7be7cb0b4295f9a94b933377932aaf537b726878"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-09-05 08:40:45.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-01-29 19:33:28.000000000","tz":-480},"subject":"Remove Manila V1 API client","message":"Remove Manila V1 API client\n\nThe Manila V1 API has been deprecated since 2016 and tests\nstopped using it in change Id90044bca44387864aee8984bc2366c089c87cfc.\nComplete this removal by:\n\n1. Making SharesV2Client inherit directly from RestClient\n   instead of the V1 SharesClient\n2. Add microversion support added to all methods\n3. Removing V1 client registration from plugin.py\n\nChange-Id: I7be74fc22594d3a59932b8bc469f9209cf985fe6\nCo-Authored-By: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7cda2f5c5c894e483118a294f8844e3f0005889c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7cda2f5c5c894e483118a294f8844e3f0005889c"}]},"branch":"refs/heads/master"},"12c4c54273bc648baa00fc17915edb22e0e72256":{"kind":"REWORK","_number":10,"created":"2026-01-29 19:44:09.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/86/959686/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/86/959686/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/10"}}},"commit":{"parents":[{"commit":"7be7cb0b4295f9a94b933377932aaf537b726878","subject":"Stop testing the V1 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7be7cb0b4295f9a94b933377932aaf537b726878"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-09-05 08:40:45.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-01-29 19:43:56.000000000","tz":-480},"subject":"Remove Manila V1 API client","message":"Remove Manila V1 API client\n\nThe Manila V1 API has been deprecated since 2016 and tests\nstopped using it in change Id90044bca44387864aee8984bc2366c089c87cfc.\nComplete this removal by:\n\n1. Making SharesV2Client inherit directly from RestClient\n   instead of the V1 SharesClient\n2. Add microversion support added to all methods\n3. Removing V1 client registration from plugin.py\n\nChange-Id: I7be74fc22594d3a59932b8bc469f9209cf985fe6\nCo-Authored-By: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/12c4c54273bc648baa00fc17915edb22e0e72256"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/12c4c54273bc648baa00fc17915edb22e0e72256"}]},"branch":"refs/heads/master"},"b7f6bda52b566b5e0fb07610614dd1c238ee4d7c":{"kind":"REWORK","_number":11,"created":"2026-01-29 19:44:20.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/86/959686/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/86/959686/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/11"}}},"commit":{"parents":[{"commit":"7be7cb0b4295f9a94b933377932aaf537b726878","subject":"Stop testing the V1 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7be7cb0b4295f9a94b933377932aaf537b726878"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-09-05 08:40:45.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-01-29 19:44:20.000000000","tz":0},"subject":"Remove Manila V1 API client","message":"Remove Manila V1 API client\n\nThe Manila V1 API has been deprecated since 2016 and tests\nstopped using it in change Id90044bca44387864aee8984bc2366c089c87cfc.\nComplete this removal by:\n\n1. Making SharesV2Client inherit directly from RestClient\n   instead of the V1 SharesClient\n2. Add microversion support added to all methods\n3. Removing V1 client registration from plugin.py\n\nChange-Id: I7be74fc22594d3a59932b8bc469f9209cf985fe6\nCo-Authored-By: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b7f6bda52b566b5e0fb07610614dd1c238ee4d7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b7f6bda52b566b5e0fb07610614dd1c238ee4d7c"}]},"branch":"refs/heads/master"},"05245ac3da21cff8db5cc380c8d7b18a1dc8396b":{"kind":"REWORK","_number":12,"created":"2026-01-29 19:56:41.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/86/959686/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/86/959686/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/12"}}},"commit":{"parents":[{"commit":"7be7cb0b4295f9a94b933377932aaf537b726878","subject":"Stop testing the V1 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7be7cb0b4295f9a94b933377932aaf537b726878"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-09-05 08:40:45.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-01-29 19:55:51.000000000","tz":-480},"subject":"Remove Manila V1 API client","message":"Remove Manila V1 API client\n\nThe Manila V1 API has been deprecated since 2016 and tests\nstopped using it in change Id90044bca44387864aee8984bc2366c089c87cfc.\nComplete this removal by:\n\n1. Making SharesV2Client inherit directly from RestClient\n   instead of the V1 SharesClient\n2. Add microversion support added to all methods\n3. Removing V1 client registration from plugin.py\n\nChange-Id: I7be74fc22594d3a59932b8bc469f9209cf985fe6\nCo-Authored-By: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/05245ac3da21cff8db5cc380c8d7b18a1dc8396b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/05245ac3da21cff8db5cc380c8d7b18a1dc8396b"}]},"branch":"refs/heads/master"},"6406026b51e9c282902dbc3987f1edb9879821b8":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-02-11 15:23:32.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"real_uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/86/959686/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/86/959686/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/959686/13"}}},"commit":{"parents":[{"commit":"878716d715e307fe3406d20d9817eb4ba3b86c0e","subject":"Stop testing the V1 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/878716d715e307fe3406d20d9817eb4ba3b86c0e"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-09-05 08:40:45.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-02-11 15:23:32.000000000","tz":0},"subject":"Remove Manila V1 API client","message":"Remove Manila V1 API client\n\nThe Manila V1 API has been deprecated since 2016 and tests\nstopped using it in change Id90044bca44387864aee8984bc2366c089c87cfc.\nComplete this removal by:\n\n1. Making SharesV2Client inherit directly from RestClient\n   instead of the V1 SharesClient\n2. Add microversion support added to all methods\n3. Removing V1 client registration from plugin.py\n\nChange-Id: I7be74fc22594d3a59932b8bc469f9209cf985fe6\nCo-Authored-By: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6406026b51e9c282902dbc3987f1edb9879821b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6406026b51e9c282902dbc3987f1edb9879821b8"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
