)]}'
{"id":"openstack%2Fmanila-tempest-plugin~963050","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I9bde4b20212d6111a75d91059f8e983d86ba9b59","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2025-12-30 11:38:30.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-12-30 13:13:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9bde4b20212d6111a75d91059f8e983d86ba9b59","subject":"Drop testing with PostgreSQL","status":"MERGED","created":"2025-10-04 14:37:43.000000000","updated":"2025-12-30 13:13:59.000000000","submitted":"2025-12-30 13:13:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963050","meta_rev_id":"03fa75b34fca772230f1a207b3af368829df20d4","_number":963050,"virtual_id_number":963050,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-30 13:13:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2025-12-30 11:38:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2025-10-06 17:53:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2025-10-07 10:41:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-12-27 23:50:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2025-12-30 11:38:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2025-12-17 14:02:49.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-04 16:26:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-06 17:53:31.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2025-10-07 10:41:47.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2025-12-24 04:44:19.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-12-27 23:50:33.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-12-27 23:50:33.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"0c8d068be6c0aa8c1c34dac84378c8bc5daa295f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-04 14:37:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c31d582cb5bce45d800b7dc8b599c6e5e3b68d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-04 14:39:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"86ed6b7f303970f88b9c9604ca68a0d1cc723935","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-04 16:26:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0289ea429c0b4288899ac412f964558c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad29a124afec461e8ff60e687509a76a : SUCCESS in 3m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9e2a0d4b558c454aaf318c486df46ad3 : SUCCESS in 31m 45s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5b800029eb6d4e7598fd1e0da943c22c : SUCCESS in 1h 43m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b6f241111cbe4192abd3b6a709ae26f9 : FAILURE in 58m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/7cda7238a64e48cd9095d110371c2b98 : SUCCESS in 1h 03m 44s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/dbb4c6a1758442b7a531e7bee9cfb5bc : SUCCESS in 1h 14m 09s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5fc7230b64444fb3932cdea150800a17 : SUCCESS in 1h 21m 20s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d42322f3cb0c4f29aceb31498bc81187 : SUCCESS in 24m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e3979f2f9ab041a7aff3acb5e6ef28c5 : SUCCESS in 28m 07s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/8395daf8cbde4c2db909f97a61a599d3 : SUCCESS in 1h 15m 40s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/2527477dd1dc41c3b899d74af5071b6b : SUCCESS in 37m 14s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/1991b87007494cb393f857ca9e660c6a : SUCCESS in 1h 17m 27s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3234f0e93218404a92acc76a8857a995 : SUCCESS in 33m 22s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f4468e95c5a24ed8b5c6ccd1de4da013 : FAILURE in 56m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fead250b8f1dfe850d837b6810c1ac8087bb30ac","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-06 17:53:31.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"478678c9c73dd202a1df41cdecd66bbd668dc572","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2025-10-07 10:41:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e5b37be240590381bf343e7ac8bc6bd88009f31d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 19:09:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4ce03aef875153de7cf8b35b437e8a45704e1a1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 20:31:56.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/23884c3cd8b148a7834a6ad3bece0e11\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b65ada8035924211b6cbda273ee4dd8b : SUCCESS in 1m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d6eb2b13e15248c98668eb7821db2794 : SUCCESS in 1h 12m 09s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0ec0c3d09bd94f7b98ea23e63003cec8 : SUCCESS in 23m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b551f4bfdf42401092723888e43a3429 : FAILURE in 41m 45s","accounts_in_message":[],"_revision_number":2},{"id":"3a5d11df440bf7f491d676aa77468af38bf0a587","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2025-12-17 14:02:49.000000000","message":"Patch Set 2: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0e752457e10aff85d0dd117793d3f199c1db7bbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 15:31:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7489f887717f4314bb6de019a5fe3da9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56225701fa7f47e5aa2ac1141ae79b6d : SUCCESS in 1m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d923e9825fd34a03a6520ef582360cc8 : SUCCESS in 36m 38s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/0e7e26c2194b43178bb086b0a3cea785 : FAILURE in 1h 07m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7bb1aef7410d4ee7b3c4aae68735a9a4 : FAILURE in 1h 01m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9cde712cd28b4fe698dadae1f73888ef : SUCCESS in 53m 00s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/2968a96dd0d547a6b0ff7f0bfb973b0a : SUCCESS in 48m 18s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4fcc2fbfbe6543c59d75c3f5d0ceb868 : SUCCESS in 1h 21m 07s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e30cb76499534f8a93348bdac84a7221 : SUCCESS in 35m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d2f7f9b7d75a459e8e8a6c9a7c66c2ac : SUCCESS in 26m 30s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/bf9c93965a8646b68db2e944d1b85911 : SUCCESS in 1h 23m 16s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/e0c141e6fc0449fbb34d13de170d8eb5 : SUCCESS in 50m 21s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/0771c94e8465440db07b6247f3ef06c9 : SUCCESS in 1h 06m 08s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/1a12acf4384b4463823d9a4147d9bfa9 : FAILURE in 31m 00s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/97c262ac288042569d82280275eff457 : FAILURE in 31m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"73fd72e800c14cc67ad74a083c63abb5911e50ac","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-23 04:33:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2ef38dc2bbab0a5f4825ab4a0021d5251e18becd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-23 06:07:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b236941655364f939835fb82cf2c7477\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/857d27dc1b824608a01c37ff4ac652d4 : SUCCESS in 2m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ff2abf3b10f24768b3c6aac1a7cb90f7 : SUCCESS in 1h 16m 58s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5a6976e52673407f81592c423930c475 : FAILURE in 1h 05m 58s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/363ea682884d4fdd923a83a5053ae3b1 : FAILURE in 1h 10m 49s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/1a0d48f25775472ca935a8cf009ce758 : SUCCESS in 1h 07m 38s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/348a820367514a128f6bd0e1b83fb752 : SUCCESS in 1h 27m 01s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0e99a4961329457ea64021999e5e48de : SUCCESS in 1h 17m 12s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/95686d186ece47e7b2b451ef016de38e : SUCCESS in 34m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4167812fd28049bea950986cffe50dc7 : SUCCESS in 44m 05s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/30212a155e8947aaa99604fe0a32ece2 : SUCCESS in 1h 11m 14s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/a279169d239f421ca079b42fc143b818 : SUCCESS in 1h 13m 47s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/662d0078b8984ae3be36424b1b40ec0f : SUCCESS in 1h 10m 58s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3fe93db5ba324008927f1e18eb8c732d : FAILURE in 19m 44s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d847ff1475af4f10b878ea2dc4e20b85 : FAILURE in 39m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2dcd569537ea5ff5ccb35031a0ab336042a3a2b1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-24 00:25:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"251051abff6999afea7eedf448e5e6b77ec65f4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-24 02:52:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6460bc6651204eb391c58aa7582c27ab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9258e507c3fb45c5b99f2c55fd69eee9 : SUCCESS in 2m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ee7e9da4626c4b8a85df8ca92964da06 : SUCCESS in 48m 39s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d8d6e416394b433685f909539c7b0e88 : SUCCESS in 2h 21m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/cacfe4cc743e444090a9f6985b9361d8 : FAILURE in 1h 13m 15s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e342cad9a8fc4b07b7bd71c7186ff42d : SUCCESS in 30m 27s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/b3e425bcc59740d49427c01fb3977de9 : SUCCESS in 1h 28m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9446c5b7f23843ab886ebb81d4d4aa8e : SUCCESS in 1h 16m 22s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6120eca3429146e4819b66e7a40cd02d : SUCCESS in 24m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2b300b128460438ab6547b86a105a219 : SUCCESS in 41m 53s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/b256f780c5c94108a9a10d5088ee9a97 : SUCCESS in 1h 08m 59s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/059880cb98864c189b1d32022960a4bf : SUCCESS in 1h 09m 33s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/bbdc97d1d2694aa68950f7d28263fe34 : SUCCESS in 1h 13m 29s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/df70ffc7666d4b87acf18f93311902a3 : SUCCESS in 29m 03s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/97cc15af17184e93902c89988d3411dd : FAILURE in 1h 03m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ca73947fd4bfcb8377f2bece753d634ba9eeea1b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-12-24 04:44:19.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/963050/2/upstream-check-manila/manila-tempest-plugin-ontap-dhss/e467c2a/ : SUCCESS in 1h 50m 28s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"454cdcd1347c5c9d8f8c1385a67914318323461f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-27 23:50:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c1ae48f74c0bacab9875a3b36e9da6498ce1b1d5","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-12-30 11:38:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e811acd0baa710f806ff68a3ed7a2bcb7b308d88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 11:39:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6eb4236b5394a5f15ae408ed32b9ab0fa272e500","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 13:13:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fee014da12e84c8888dc6257d9aa6c80\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e56d41e70fc14dd6ac0d802b6c128136 : SUCCESS in 2m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1f9afee3c90a43d6acce1a4044b13688 : SUCCESS in 1h 26m 30s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2ec4654778bf4716b5be43a2b1e030ee : SUCCESS in 25m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0ee8d8dabb82400d80bf877d20c51f39 : SUCCESS in 29m 32s","accounts_in_message":[],"_revision_number":2},{"id":"03fa75b34fca772230f1a207b3af368829df20d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 13:13:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dfb7d677bc9c443884af85b638696e73b861f144","revisions":{"265a4bd815a453df78704751471c29ed8c677984":{"kind":"REWORK","_number":1,"created":"2025-10-04 14:37:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/963050/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/50/963050/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/50/963050/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/50/963050/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/50/963050/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/50/963050/1"}}},"commit":{"parents":[{"commit":"70dca36c99ff0cc274750ccbfbbae94b3b93594f","subject":"Drop branch regex for retired branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/70dca36c99ff0cc274750ccbfbbae94b3b93594f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-04 14:36:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-04 14:37:35.000000000","tz":540},"subject":"Drop testing with PostgreSQL","message":"Drop testing with PostgreSQL\n\nneutron completely removed its code to support PostgreSQL by [1].\nSwitch the testing coverage to use mysql instead, even in stable\nbranches, because any problems are not likely to be fixed.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/934171\n\nChange-Id: I9bde4b20212d6111a75d91059f8e983d86ba9b59\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/265a4bd815a453df78704751471c29ed8c677984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/265a4bd815a453df78704751471c29ed8c677984"}]},"branch":"refs/heads/master"},"dfb7d677bc9c443884af85b638696e73b861f144":{"kind":"REWORK","_number":2,"created":"2025-10-04 14:39:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/963050/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/50/963050/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/50/963050/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/50/963050/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/50/963050/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/50/963050/2"}}},"commit":{"parents":[{"commit":"70dca36c99ff0cc274750ccbfbbae94b3b93594f","subject":"Drop branch regex for retired branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/70dca36c99ff0cc274750ccbfbbae94b3b93594f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-04 14:36:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-04 14:38:51.000000000","tz":540},"subject":"Drop testing with PostgreSQL","message":"Drop testing with PostgreSQL\n\nneutron completely removed its code to support PostgreSQL by [1].\nSwitch the testing coverage to use mysql instead, even in stable\nbranches, because any problems are not likely to be fixed.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/934171\n\nChange-Id: I9bde4b20212d6111a75d91059f8e983d86ba9b59\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dfb7d677bc9c443884af85b638696e73b861f144"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dfb7d677bc9c443884af85b638696e73b861f144"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
