)]}'
{"id":"openstack%2Fmanila-tempest-plugin~964328","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Iabb3bdcf0689d57b5f49bca8cfe77937e6257a8c","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1824442","attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-03-05 21:34:36.000000000","reason":"A robot voted negatively on a label"},"38038":{"account":{"_account_id":38038,"name":"Riane TOrres","display_name":"Riane Torres","email":"riane.torres@gmail.com","username":"realrtorres"},"last_update":"2026-03-05 21:34:36.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Iabb3bdcf0689d57b5f49bca8cfe77937e6257a8c","subject":"Add share network tests for visibility protected fields","status":"NEW","created":"2025-10-17 18:55:33.000000000","updated":"2026-03-05 21:34:36.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2cf4e3ed1832a32c7904cd0fcf734d0691e91167","_number":964328,"virtual_id_number":964328,"owner":{"_account_id":38038,"name":"Riane TOrres","display_name":"Riane Torres","email":"riane.torres@gmail.com","username":"realrtorres"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-03-05 21:34:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-17 20:45:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-05 19:34:07.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"a5636a46d076fa66237b63b4e1c262a96485a342","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38038,"name":"Riane TOrres","display_name":"Riane Torres","email":"riane.torres@gmail.com","username":"realrtorres"},"date":"2025-10-17 18:55:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f331818d61f499fe761f7eaebfb52952594221b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 20:45:11.000000000","message":"Patch Set 1: Verified-1\n\n(22 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ceca7ff07b24742926b1ad5ba7a076f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f23d8181ab0d4ba6902204659efa5957 : FAILURE in 2m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b0ab3d71b0414d2d92883768fbc5b9cf : SUCCESS in 54m 33s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5a26c55a13d84697be447415e6071e6d : FAILURE in 1h 47m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f7e1e33a6845451fbae14af17901a815 : SUCCESS in 1h 10m 12s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a015d0e9cfc44deca32d7105bfabb377 : FAILURE in 55m 52s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/9d2c72ad6b574c20a210591338b522e6 : FAILURE in 1h 31m 57s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fef1901f749e41b081fc415122f251f5 : FAILURE in 21m 24s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/afee1abc35fd4c878c1b2d0946a4862b : FAILURE in 34m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0671aed7f1894195a89f59e817b07f8a : FAILURE in 35m 29s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/5b61aad2be94405eaf16978cd0d33070 : SUCCESS in 34m 49s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/21ae77dbc9914596a730f7f2c3a0113e : SUCCESS in 1h 17m 35s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/00b1eccd69f64b57bb84074c85ef6a25 : SUCCESS in 1h 03m 16s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/0373537ce9314c09afc9ebd0a42ead1d : SUCCESS in 24m 06s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fb38c314866c4948af33bfcb5ffd5444 : FAILURE in 16m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"116a8e256f832305959c1d119459456d7131b6a7","author":{"_account_id":38038,"name":"Riane TOrres","display_name":"Riane Torres","email":"riane.torres@gmail.com","username":"realrtorres"},"date":"2025-11-14 18:51:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a2f2325444de961dbb7747e5ebd1e708e61bbc3b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-11-21 20:02:02.000000000","message":"Topic set to bug/1824442","accounts_in_message":[],"_revision_number":1},{"id":"8c2324f0777098c5ab011d21a7264010b486138f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-03-05 19:34:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2cf4e3ed1832a32c7904cd0fcf734d0691e91167","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 21:34:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e1ab01f3b454d828094632194a84757\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6c14a4df6a149a1bceaeff0511ded9d : SUCCESS in 4m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5e28929d96f6490a8342863c4361ea85 : SUCCESS in 1h 11m 50s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8b9db0f65e4e40d5beca9211f1ea3983 : SUCCESS in 1h 58m 24s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/15ca8b82a40749839668769ba2ccf40e : FAILURE in 1h 33m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/772a50e60ec547ff8eb76b43c072bb84 : FAILURE in 27m 42s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/bb1b054a0d524f2381a4894fb8c3ee4a : FAILURE in 1h 34m 26s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/298e2a46b70e43dbbc6819141fe63d66 : FAILURE in 41m 41s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/930fdd77264d409e90111ecdf2c97eda : FAILURE in 31m 47s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d26f9d66479d4e0aa06104f7ab2e314a : FAILURE in 26m 53s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/923165858f484fe88431f28403c37e79 : SUCCESS in 1h 09m 39s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/f417843398944d60b11799d90457dd85 : SUCCESS in 39m 01s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/4e72cac533d84ae08a04626ba9521463 : SUCCESS in 1h 09m 01s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/6f41ea99639c4559adbbae72afc37347 : SUCCESS in 32m 11s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f4f60041af4d4fe8881fafb2c7f25aab : FAILURE in 1h 03m 54s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cc541c1925e916ef28b58d4b1d4a1a659df9ae57","revisions":{"fe88a7e06012cdc7f1942297c8edfccb4933a31b":{"kind":"REWORK","_number":1,"created":"2025-10-17 18:55:33.000000000","uploader":{"_account_id":38038,"name":"Riane TOrres","display_name":"Riane Torres","email":"riane.torres@gmail.com","username":"realrtorres"},"ref":"refs/changes/28/964328/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/28/964328/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/28/964328/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/28/964328/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/28/964328/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/28/964328/1"}}},"commit":{"parents":[{"commit":"496237a646a3cff808a52bce540356029f69c5b0","subject":"Merge \"Remove remaining job with Ubuntu Jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/496237a646a3cff808a52bce540356029f69c5b0"}]}],"author":{"name":"Riane Torres","email":"riane.torres@gmail.com","date":"2025-10-17 01:17:30.000000000","tz":0},"committer":{"name":"Riane Torres","email":"riane.torres@gmail.com","date":"2025-10-17 18:55:26.000000000","tz":0},"subject":"Created tests for admin and user privileges for segmentation_id and network_type","message":"Created tests for admin and user privileges for segmentation_id and network_type\n\nChange-Id: Iabb3bdcf0689d57b5f49bca8cfe77937e6257a8c\nSigned-off-by: Riane Torres \u003criane.torres@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fe88a7e06012cdc7f1942297c8edfccb4933a31b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fe88a7e06012cdc7f1942297c8edfccb4933a31b"}]},"branch":"refs/heads/master"},"cc541c1925e916ef28b58d4b1d4a1a659df9ae57":{"kind":"REWORK","_number":2,"created":"2026-03-05 19:34:07.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/28/964328/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/28/964328/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/28/964328/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/28/964328/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/28/964328/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/28/964328/2"}}},"commit":{"parents":[{"commit":"496237a646a3cff808a52bce540356029f69c5b0","subject":"Merge \"Remove remaining job with Ubuntu Jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/496237a646a3cff808a52bce540356029f69c5b0"}]}],"author":{"name":"Riane Torres","email":"riane.torres@gmail.com","date":"2025-10-17 01:17:30.000000000","tz":0},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-03-05 19:32:45.000000000","tz":-180},"subject":"Add share network tests for visibility protected fields","message":"Add share network tests for visibility protected fields\n\nThis change introduces two new tests for ensuring the share\nnetwork subnet\u0027s segmentation ID and network type fields are only\nvisible for admins.\n\nChange-Id: Iabb3bdcf0689d57b5f49bca8cfe77937e6257a8c\nSigned-off-by: Riane Torres \u003criane.torres@gmail.com\u003e\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cc541c1925e916ef28b58d4b1d4a1a659df9ae57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cc541c1925e916ef28b58d4b1d4a1a659df9ae57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
