)]}'
{"id":"openstack%2Fmanila-tempest-plugin~965878","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I38e761c412f634b43a34af15b0f407b8d2b0bcf6","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2025-11-10 13:01:49.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-11-10 14:18:31.000000000","reason":"Change was submitted"},"38371":{"account":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"last_update":"2025-11-10 14:18:31.000000000","reason":"Change was submitted"},"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2025-11-10 10:53:22.000000000","reason":"\u003cGERRIT_ACCOUNT_36179\u003e replied on the change","reason_account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2025-11-10 14:18:31.000000000","reason":"Change was submitted"},"38373":{"account":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"last_update":"2025-11-10 14:18:31.000000000","reason":"Change was submitted"},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2025-11-10 14:18:31.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2025-11-10 14:18:31.000000000","reason":"Change was submitted"},"38479":{"account":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"last_update":"2025-11-10 14:18:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I38e761c412f634b43a34af15b0f407b8d2b0bcf6","subject":"Fix NetApp Manila Replication test TearDown Failures","status":"MERGED","created":"2025-11-01 19:23:55.000000000","updated":"2025-11-10 14:18:31.000000000","submitted":"2025-11-10 14:18:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"965878","meta_rev_id":"6b3db128bf0de23e5aaee595682ad3899dcc2084","_number":965878,"virtual_id_number":965878,"owner":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-10 14:18:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2025-11-10 13:01:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2025-11-10 10:53:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2025-11-10 13:01:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"}],"reviewers":{"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-01 21:19:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-02 11:45:10.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-11-03 17:52:57.000000000","updated_by":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2025-11-03 17:52:57.000000000","updated_by":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2025-11-03 17:52:57.000000000","updated_by":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"reviewer":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"state":"REVIEWER"},{"updated":"2025-11-03 17:52:57.000000000","updated_by":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"reviewer":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"state":"REVIEWER"},{"updated":"2025-11-03 17:52:57.000000000","updated_by":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"REVIEWER"},{"updated":"2025-11-03 17:52:57.000000000","updated_by":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2025-11-03 17:52:57.000000000","updated_by":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2025-11-03 17:52:57.000000000","updated_by":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"}],"messages":[{"id":"9710cd8733339e86da39caa503d4fc64e6010613","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2025-11-01 19:23:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35104c9277a6511eb48036e9f6d4c920984ea4aa","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2025-11-01 19:24:19.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"9cba8cb3f2813db415d804a91267d90755cf404d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-01 21:19:36.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5f042a359a44ab0bfb2ef74e7cb870a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d52013a056874bbfa2a717aaf7413be7 : FAILURE in 3m 14s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6e7dc40b5de149f99761435298c7b470 : SUCCESS in 1h 05m 09s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/27e649a2d5c04677963a25f8145a95b2 : FAILURE in 1h 49m 56s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0453e3005e1945ea94361342680e6482 : FAILURE in 1h 00m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/85e018e7f3c8475c81a529a6614a181d : FAILURE in 45m 11s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/3c42e29728a94998b84c6d4246152508 : SUCCESS in 1h 06m 43s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b9f338e3482348cb9674dfae077ffc6b : FAILURE in 23m 49s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7be5d8598ad34e19a521eebe4159c9cf : SUCCESS in 27m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/62d7afcc37dd42f1a49d7e970b7f7de3 : SUCCESS in 43m 39s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/3d87aca00d59413fa4b1bd56d78fb446 : SUCCESS in 1h 14m 15s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/a55e28d39260488b83ff4bd193f52d8a : SUCCESS in 1h 03m 31s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/c0d38c045c7943e18f12464f59d9c6ef : SUCCESS in 1h 10m 08s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/08ddb6daae3a4e69bac2784da20105f9 : SUCCESS in 25m 43s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/605fe74f0c3d44a6a9e92967587c6bb9 : FAILURE in 14m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2650061f536646e75149167e94638713ad6ad15a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2025-11-02 09:48:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"15ac27fab6adc1154971c12daad91122011e30e8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2025-11-02 09:48:58.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"983edf314bf40e1469ef5f31e5046c48194aa447","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2025-11-02 09:53:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"835b337929d6de181431f3d5a7b3ea6cedc866f8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2025-11-02 09:54:12.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"f683f06f58e5e9cc755496d30f3c250304564df5","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2025-11-02 09:54:33.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cef8c6c1bfbcc37441df87b37f4293f18fc8db8e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-11-02 11:45:10.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/78/965878/5/upstream-check/manila-tempest-plugin-ontap-dhss/72ef86c/ : FAILURE in 1h 42m 50s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"21d747081a2346a04d0544ffbe10be86aedf85ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-02 12:23:09.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e61e915d996747ec955a60179a9a854f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91a3195f663f4e919b18077f2065f307 : SUCCESS in 3m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7110d6ef574941e78fec422e9dfbde47 : SUCCESS in 1h 14m 01s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/66b8c0be7ea243b5a644ec7969fb39a4 : FAILURE in 2h 28m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/49eb21bde0564764a110e22bca081961 : FAILURE in 44m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b8347e40bb1740f7a6ce4760cbeee884 : SUCCESS in 44m 19s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/87c4fbd99d494a6ab5487b14024295c5 : FAILURE in 1h 31m 17s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/adeb66987b1e4e9c9cd13e931ca9c563 : FAILURE in 21m 58s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1ba55c58cf96484aaf7cc00075114454 : SUCCESS in 33m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2731b586185e48258ad27a5417d6330f : FAILURE in 39m 50s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/20fd81aa4f0b4a2e9862a2999ed5971f : SUCCESS in 1h 06m 46s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/4af58124c2cf47e88fafc64293a16c06 : SUCCESS in 55m 25s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/bf84e1190c994f649808f18c0ebdd793 : SUCCESS in 47m 19s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a3ee3670d2314b3388d67e938d7db369 : SUCCESS in 29m 15s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7e3dcfac930c4511b91fdf8e726d8139 : FAILURE in 16m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f6d23e35c32d3aefb8c8926dd94dd53e44492c1c","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2025-11-02 12:27:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d2b583f53206058e7bb01d8c4cbac2355fa4f981","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-11-02 13:25:27.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/78/965878/5/upstream-check/manila-tempest-plugin-ontap-dhss/ba09f9e/ : FAILURE in 50m 10s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"71aac4cb699a079c2906b5ca2a65bcb6adf821cd","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2025-11-02 16:52:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"52d807d9315e92105e3e8d11ff558a1103e77735","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-11-02 17:49:36.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/78/965878/5/upstream-check/manila-tempest-plugin-ontap-dhss/0cd8f7f/ : FAILURE in 49m 08s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"df2330fced6873d58bf9c54b655bf49f65fd2262","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2025-11-03 13:42:23.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"3b562fd26db5d0169f0dc4a170409d48f60c8865","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2025-11-03 14:50:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5e21fc557babc3fa6bf7376acda85eff757cfd61","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-11-03 15:07:30.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/78/965878/6/upstream-check/manila-tempest-plugin-ontap-dhss/dc3cd07/ : RETRY_LIMIT in 10m 02s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"23dfe6194b6a200d8a84ba018facdc0f7ce4a9a9","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2025-11-03 15:16:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2054d2714559d69cb5325e85c2565af881f7b498","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 15:24:45.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2570c2e38a048ffbbd55dd8cb6c3dcf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99ad68ad6ebf4f85addad6a55d931395 : SUCCESS in 3m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5c94511b9dde4add8b64402c9f496239 : SUCCESS in 51m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/879d790bd5774231b92b298d678d4b0a : SUCCESS in 1h 41m 35s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d0e3162098394f8d93f0131231358dfb : FAILURE in 1h 21m 53s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/1c4d4f5ab29f420e9cf576f9c76a6e6f : SUCCESS in 57m 06s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f726f2f07ecf40e28d30fe486cfb8565 : SUCCESS in 1h 06m 58s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0900dd5eaa7c4440b9200d4ad53d3997 : FAILURE in 26m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c135bf1443ea4704b654e056acac1458 : SUCCESS in 28m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6d8a4811bac648e59c3636d6172ad67a : FAILURE in 43m 06s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/1f01e61f83e0401ea436d9a330a73090 : SUCCESS in 1h 22m 17s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/b191fdc1ce544ba8b25e7e9d422172e7 : SUCCESS in 58m 11s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/b8b9384a362346a292001ed10770af28 : SUCCESS in 55m 24s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/7d6607f41b334924b36f4b05076eca15 : SUCCESS in 24m 03s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1fa15619774d4557a669ed1690943728 : FAILURE in 25m 58s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"54bb22d8fdbece9861599309649144a2974af2d3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2025-11-03 16:10:42.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"5ceb7fc82a7b6991ccad7a943ad871dcaa05927a","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2025-11-03 16:17:12.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"dc96367f3a3fab9c4e6442c2506d2571790c4258","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-11-03 17:17:13.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/78/965878/7/upstream-check/manila-tempest-plugin-ontap-dhss/44e70b8/ : POST_FAILURE in 51m 53s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"459988aa3f046d583bae6832fe4e098db40bf330","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2025-11-03 17:48:15.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"90b78f5ef2d50c81ae19095626b6b57379ed2151","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2025-11-03 17:50:36.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":8},{"id":"f3a71b21248d0db89a0082f9dbc74b485dad028d","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2025-11-03 17:50:48.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6ed0cb0b629f39bc0c9b3e61064a0518b78d230c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2025-11-03 17:53:07.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"54200ac178ca5323a2fac8dc9e7091a6cb09f6d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 20:11:54.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cba8506052394e8ab3441c0a0fd144df\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/453666ef0e37441d9363a6df815902a1 : SUCCESS in 4m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1557ed7b3f574918bf8e022ff127e63f : SUCCESS in 1h 06m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d3a3635b48584631aff53366544df977 : SUCCESS in 1h 58m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/941ecddb13f24e50b23947b644b88a01 : RETRY_LIMIT in 1m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/2e4372f01ec545ceb8feadbf862f275d : SUCCESS in 1h 05m 49s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/87458609cbb348329bbe7a059032610a : SUCCESS in 1h 35m 23s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fdac346514214d4ebd045e814f4d6f80 : FAILURE in 23m 49s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0bfcee560f4841908ee0f49db7b94484 : SUCCESS in 41m 12s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e403a8a48ae540d4a525f58b50674d04 : SUCCESS in 39m 13s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/e397be69430b456e8ce39039bf8a28a4 : SUCCESS in 1h 17m 07s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/79eee50795bb456e8aa91895cd450d98 : RETRY_LIMIT in 1m 19s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/1a829435da144e789a0e37b10779729b : SUCCESS in 1h 09m 11s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/8d0c64a870dd42ee9120004f269a9209 : SUCCESS in 30m 07s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/25e4df5faac44ddfb0e9f685b0d6da46 : FAILURE in 25m 16s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a983defad2b06db2d77b4f00daad06987f65080b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-11-03 21:09:07.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"2fbf852cd861c2136e9bb3fb4cca1fc762e4b6d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 23:02:15.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03c5ccb5d1764ad8b1325bc4e3171b00\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50b6b552509f473abb68a34121cf75e8 : SUCCESS in 4m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/17d9efd9e8c24eb89cc2ca41868a7219 : SUCCESS in 1h 09m 13s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/bff55928d54344c7b3f80059917fbcfa : FAILURE in 1h 51m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ca84d9d95eeb478491a38c50aecfa690 : FAILURE in 1h 09m 38s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/44befdd69aa3464988b6e6e99045eea8 : SUCCESS in 57m 24s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/0537bd21f84449b09ca1ea4236bd9b21 : SUCCESS in 1h 32m 45s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a27df52a84a342478f247118ce1f18d3 : FAILURE in 21m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a9e1b20faf644a6b8ca5377baa4994c5 : SUCCESS in 33m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4b638a68cc5746608f899b652a482517 : SUCCESS in 44m 51s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/7b62535e7ebf45a6af2046f4b4ffe367 : SUCCESS in 49m 28s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/775cb0b83e4748fb9ca7c2f23a9b31ab : SUCCESS in 1h 14m 45s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/6da961de3cba4ab8801a6d365684f6e0 : SUCCESS in 1h 21m 27s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/c4f48fc9476b4eba91cfc5085da4fd8c : SUCCESS in 32m 16s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f96dcdff68f94cd3968a6cd95238f164 : FAILURE in 24m 32s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"49e623c3d7b3683fd9e0208a63d833f2b4ec21e2","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-11-06 14:51:41.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"a93a02a00363709d09aadf2f77021c28638062ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 05:05:48.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"aa48732bce696bc311177c36a7ce23d20cdd3c81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 06:08:51.000000000","message":"Patch Set 9: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/manila/+/966397 is needed.","accounts_in_message":[],"_revision_number":9},{"id":"8916f814918168b160835d2bd756b812979a3ae4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2025-11-07 06:22:29.000000000","message":"Patch Set 10: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":10},{"id":"3baa58caac0cf49184891805f232cc5549917a92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 08:41:53.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d1f6b4206794bdba1105dc446f83665\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/065142b006ea49df92eab9cb64ee417f : SUCCESS in 2m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cc1c60b236ef4fa493bda59fc4d31244 : SUCCESS in 47m 57s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ce0671275614478b8885987968b8ee11 : SUCCESS in 2h 18m 56s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/97083d1526f942ed99b229e58c677ee4 : FAILURE in 1h 14m 15s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/28dbe3f71fa74eb589491fab86e3c19c : SUCCESS in 1h 05m 37s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/23d944e2ec574211bf8edb8800db03f5 : SUCCESS in 1h 03m 12s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ff28f6c9140143e68feaf5cca4f19dfb : FAILURE in 19m 50s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d874e63a881f46618cba554b9519d2fc : SUCCESS in 31m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/536390343a1642cb9d9d436afed417cf : SUCCESS in 48m 23s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/c24f4b6c125b433283970964847574b2 : SUCCESS in 1h 18m 54s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/713726960048407c93075e2ce1164b2a : SUCCESS in 1h 10m 16s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/3456e6ddb9e6443e98df9e4d61718fd2 : FAILURE in 17m 19s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/67b47a2f19004472a1a819e157a3ad1d : SUCCESS in 29m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b531f63265fe46e08269abc23eb6a561 : FAILURE in 24m 13s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f0adba2c62f7d8dec20eadc2cd7cceb60a6ee60c","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2025-11-07 09:18:13.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b34c84a9a5c648e07e8e5c4f1e93ce55f60dcaa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 12:19:28.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7d259a39b98425eb519e9fb687efd00\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/885c4f8bdc6045d5be2ce4f8b1f796d7 : SUCCESS in 2m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/47d4f3e2fd1f41c5ab0f9d77e5f289ae : SUCCESS in 1h 12m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/091b464820d14c0fa278d7e703537a91 : FAILURE in 2h 47m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3dfd3147c81c4e179fadff1ca6c392b1 : FAILURE in 1h 02m 36s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ef5568e2ccf14507885495606b137767 : SUCCESS in 1h 17m 47s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/84fee306a34447ff8f2be78a8df8d52c : SUCCESS in 1h 40m 03s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/175a47494a96401386151c8c6c27a5e8 : FAILURE in 23m 00s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e994eb44c1ef43ed9485fb124b9454d7 : SUCCESS in 33m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3dded56c408d471494ecc162cc0c6092 : SUCCESS in 28m 39s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/3d47a2f7304f4d4dbb91f781e2b85a58 : SUCCESS in 1h 23m 20s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/b8a1b36260174085a07be1dc41ffcec0 : SUCCESS in 1h 05m 49s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/a31f7d6b764f42ba9228b1b2bdaa9375 : FAILURE in 17m 17s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d521cc84cbda43bfbd865e7d7454431d : SUCCESS in 34m 53s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/356f425b22e943d4ad109700e94b1542 : FAILURE in 24m 32s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"975f7d058d2a18ae50c5fa78f785612635e4ff64","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2025-11-07 18:13:29.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"24efc83ecc74b3b48eddc42e7384f44fa5d6a07e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 20:25:57.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/388df1b3b16d4ed8843a5c6939ae8b32\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20608ddfdae24160bc2fa9ff091548ff : SUCCESS in 2m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5364c7ef0cd94d0b9b456eb86cc122fb : SUCCESS in 58m 07s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a22017aae33545a59ffe45599025d872 : FAILURE in 2h 04m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7c7d03c188b24190874201e522a15001 : FAILURE in 1h 36m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c6e81006e3b44762a74238faedafde63 : SUCCESS in 1h 02m 17s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/cd361de3024840dbabc029d4c5072153 : SUCCESS in 46m 02s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c36066019ead4d28afdaf15e18c9be2c : FAILURE in 14m 50s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/91508ded9f77459d8452a716f9cff28a : SUCCESS in 34m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ec20a56ab7244ef2b64b698eea2b2574 : SUCCESS in 33m 58s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/2882524fd15a492b84e82a02d25d3bdf : SUCCESS in 1h 08m 49s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/6aa5060e72ef40e0835727be243d35ec : SUCCESS in 1h 11m 40s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/83af52dbd6e8416582d0e1a28095a360 : SUCCESS in 1h 10m 09s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/7868665ac91a490e8656d0a1896c0d8c : SUCCESS in 33m 20s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8889fd8400c24eb49ef7fcdd0503b24c : FAILURE in 15m 09s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a99456f7d4ff64912cc885288bb1a12d81cadb31","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-11-10 10:53:22.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"9b2dac7acfb044f1be3e4c3322558d0092d10902","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-11-10 13:01:49.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"6ff886d6792006c1f6c6d72fb20c89c552f84fe2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 13:02:12.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"fce55ecd65dc99302017960425585992ceea2ad4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 14:18:31.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71fced0bbde04c8ca9a1610ab960a6f6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/331225c3156b43f7bc4200b4f05298d8 : SUCCESS in 2m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a9f454d50883459ca60d7a6396c74d9b : SUCCESS in 1h 12m 03s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/993acc1ff27f476db9a7c9fc4c3f580d : SUCCESS in 38m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/27b476f20ad74a2e9cd54760ffcfc286 : SUCCESS in 41m 03s","accounts_in_message":[],"_revision_number":10},{"id":"6b3db128bf0de23e5aaee595682ad3899dcc2084","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 14:18:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"da2bba08a133b3c79f7dabbde7d4bfaa9ab8a2c8","revisions":{"6e30219c5e7f5b400b4f2ef7fb60a988167158c5":{"kind":"REWORK","_number":1,"created":"2025-11-01 19:23:55.000000000","uploader":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"ref":"refs/changes/78/965878/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/965878/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/1"}}},"commit":{"parents":[{"commit":"496237a646a3cff808a52bce540356029f69c5b0","subject":"Merge \"Remove remaining job with Ubuntu Jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/496237a646a3cff808a52bce540356029f69c5b0"}]}],"author":{"name":"Manideep","email":"manideep.openstack@gmail.com","date":"2025-11-01 19:23:03.000000000","tz":-240},"committer":{"name":"Manideep","email":"manideep.openstack@gmail.com","date":"2025-11-01 19:23:44.000000000","tz":-240},"subject":"[WIP] Triage NetApp Manila Replication test Failures","message":"[WIP] Triage NetApp Manila Replication test Failures\n\nChange-Id: I38e761c412f634b43a34af15b0f407b8d2b0bcf6\nSigned-off-by: Manideep \u003cmanideep.openstack@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6e30219c5e7f5b400b4f2ef7fb60a988167158c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6e30219c5e7f5b400b4f2ef7fb60a988167158c5"}]},"branch":"refs/heads/master"},"1d7ce048e701d253bcd73a160683221634e96f48":{"kind":"REWORK","_number":2,"created":"2025-11-02 09:48:18.000000000","uploader":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"ref":"refs/changes/78/965878/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/965878/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/2"}}},"commit":{"parents":[{"commit":"496237a646a3cff808a52bce540356029f69c5b0","subject":"Merge \"Remove remaining job with Ubuntu Jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/496237a646a3cff808a52bce540356029f69c5b0"}]}],"author":{"name":"Manideep","email":"manideep.openstack@gmail.com","date":"2025-11-01 19:23:03.000000000","tz":-240},"committer":{"name":"Manideep","email":"manideep.openstack@gmail.com","date":"2025-11-02 09:47:49.000000000","tz":-300},"subject":"[WIP] Fix NetApp Manila Replication test TearDown Failures","message":"[WIP] Fix NetApp Manila Replication test TearDown Failures\n\nChange-Id: I38e761c412f634b43a34af15b0f407b8d2b0bcf6\nSigned-off-by: Manideep \u003cmanideep.openstack@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1d7ce048e701d253bcd73a160683221634e96f48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1d7ce048e701d253bcd73a160683221634e96f48"}]},"branch":"refs/heads/master"},"a51b3e850c6342e31239cb73886b19e5f90ebbbe":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-11-02 09:48:58.000000000","uploader":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"ref":"refs/changes/78/965878/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/965878/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/3"}}},"commit":{"parents":[{"commit":"496237a646a3cff808a52bce540356029f69c5b0","subject":"Merge \"Remove remaining job with Ubuntu Jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/496237a646a3cff808a52bce540356029f69c5b0"}]}],"author":{"name":"Manideep","email":"manideep.openstack@gmail.com","date":"2025-11-01 19:23:03.000000000","tz":-240},"committer":{"name":"Manideep","email":"manideep.openstack@gmail.com","date":"2025-11-02 09:48:58.000000000","tz":0},"subject":"[WIP] Fix NetApp Manila Replication test TearDown Failures","message":"[WIP] Fix NetApp Manila Replication test TearDown Failures\n\nDepends-On: I8e16e1622e975dec2a42850f0e06159875d70f39\nChange-Id: I38e761c412f634b43a34af15b0f407b8d2b0bcf6\nSigned-off-by: Manideep \u003cmanideep.openstack@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a51b3e850c6342e31239cb73886b19e5f90ebbbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a51b3e850c6342e31239cb73886b19e5f90ebbbe"}]},"branch":"refs/heads/master","description":"Edit commit message"},"645ccec916cf9224d08d4a99124e4e36ffb1d2fb":{"kind":"REWORK","_number":4,"created":"2025-11-02 09:53:40.000000000","uploader":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"ref":"refs/changes/78/965878/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/965878/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/4"}}},"commit":{"parents":[{"commit":"496237a646a3cff808a52bce540356029f69c5b0","subject":"Merge \"Remove remaining job with Ubuntu Jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/496237a646a3cff808a52bce540356029f69c5b0"}]}],"author":{"name":"Manideep","email":"manideep.openstack@gmail.com","date":"2025-11-01 19:23:03.000000000","tz":-240},"committer":{"name":"Manideep","email":"manideep.openstack@gmail.com","date":"2025-11-02 09:53:29.000000000","tz":-300},"subject":"[WIP] Fix NetApp Manila Replication test TearDown Failures","message":"[WIP] Fix NetApp Manila Replication test TearDown Failures\n\nChange-Id: I38e761c412f634b43a34af15b0f407b8d2b0bcf6\nSigned-off-by: Manideep \u003cmanideep.openstack@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/645ccec916cf9224d08d4a99124e4e36ffb1d2fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/645ccec916cf9224d08d4a99124e4e36ffb1d2fb"}]},"branch":"refs/heads/master"},"9e2cd7a06747c2584c79f09c8fef7233fee37ab8":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-11-02 09:54:12.000000000","uploader":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"ref":"refs/changes/78/965878/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/965878/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/5"}}},"commit":{"parents":[{"commit":"496237a646a3cff808a52bce540356029f69c5b0","subject":"Merge \"Remove remaining job with Ubuntu Jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/496237a646a3cff808a52bce540356029f69c5b0"}]}],"author":{"name":"Manideep","email":"manideep.openstack@gmail.com","date":"2025-11-01 19:23:03.000000000","tz":-240},"committer":{"name":"Manideep","email":"manideep.openstack@gmail.com","date":"2025-11-02 09:54:12.000000000","tz":0},"subject":"[WIP] Fix NetApp Manila Replication test TearDown Failures","message":"[WIP] Fix NetApp Manila Replication test TearDown Failures\n\nDepends-On: I8e16e1622e975dec2a42850f0e06159875d70f39\nChange-Id: I38e761c412f634b43a34af15b0f407b8d2b0bcf6\nSigned-off-by: Manideep \u003cmanideep.openstack@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9e2cd7a06747c2584c79f09c8fef7233fee37ab8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9e2cd7a06747c2584c79f09c8fef7233fee37ab8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"bc5b503298422ba10b57010e497934b0b0853615":{"kind":"REWORK","_number":6,"created":"2025-11-03 13:42:23.000000000","uploader":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"ref":"refs/changes/78/965878/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/965878/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/6"}}},"commit":{"parents":[{"commit":"496237a646a3cff808a52bce540356029f69c5b0","subject":"Merge \"Remove remaining job with Ubuntu Jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/496237a646a3cff808a52bce540356029f69c5b0"}]}],"author":{"name":"Manideep","email":"manideep.openstack@gmail.com","date":"2025-11-01 19:23:03.000000000","tz":-240},"committer":{"name":"Manideep","email":"manideep.openstack@gmail.com","date":"2025-11-03 13:42:08.000000000","tz":-300},"subject":"[WIP] Fix NetApp Manila Replication test TearDown Failures","message":"[WIP] Fix NetApp Manila Replication test TearDown Failures\n\nDepends-On: I8e16e1622e975dec2a42850f0e06159875d70f39\nChange-Id: I38e761c412f634b43a34af15b0f407b8d2b0bcf6\nSigned-off-by: Manideep \u003cmanideep.openstack@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bc5b503298422ba10b57010e497934b0b0853615"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bc5b503298422ba10b57010e497934b0b0853615"}]},"branch":"refs/heads/master"},"14bb1d84b8997fc1432e3f6c6d267987b4e3d497":{"kind":"REWORK","_number":7,"created":"2025-11-03 16:10:42.000000000","uploader":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"ref":"refs/changes/78/965878/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/965878/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/7"}}},"commit":{"parents":[{"commit":"496237a646a3cff808a52bce540356029f69c5b0","subject":"Merge \"Remove remaining job with Ubuntu Jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/496237a646a3cff808a52bce540356029f69c5b0"}]}],"author":{"name":"Manideep","email":"manideep.openstack@gmail.com","date":"2025-11-01 19:23:03.000000000","tz":-240},"committer":{"name":"Manideep","email":"manideep.openstack@gmail.com","date":"2025-11-03 16:10:32.000000000","tz":-300},"subject":"[WIP] Fix NetApp Manila Replication test TearDown Failures","message":"[WIP] Fix NetApp Manila Replication test TearDown Failures\n\nDepends-On: I8e16e1622e975dec2a42850f0e06159875d70f39\nChange-Id: I38e761c412f634b43a34af15b0f407b8d2b0bcf6\nSigned-off-by: Manideep \u003cmanideep.openstack@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/14bb1d84b8997fc1432e3f6c6d267987b4e3d497"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/14bb1d84b8997fc1432e3f6c6d267987b4e3d497"}]},"branch":"refs/heads/master"},"b8d7404dc4e4deb082897f29e598c2a1190e98fd":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2025-11-03 17:48:15.000000000","uploader":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"ref":"refs/changes/78/965878/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/965878/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/8"}}},"commit":{"parents":[{"commit":"496237a646a3cff808a52bce540356029f69c5b0","subject":"Merge \"Remove remaining job with Ubuntu Jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/496237a646a3cff808a52bce540356029f69c5b0"}]}],"author":{"name":"Manideep","email":"manideep.openstack@gmail.com","date":"2025-11-01 19:23:03.000000000","tz":-240},"committer":{"name":"Manideep","email":"manideep.openstack@gmail.com","date":"2025-11-03 17:48:15.000000000","tz":0},"subject":"[WIP] Fix NetApp Manila Replication test TearDown Failures","message":"[WIP] Fix NetApp Manila Replication test TearDown Failures\n\n[test_replication_negative.py:test_try_add_replica_nonexistent_subnet]  Changed to use self.create_share_network from the base class instead of self.shares_v2_client.create_share_network, so that the created share network is automatically tracked and cleaned up during test cleanup. The direct client call did not handle cleanup, causing resource leaks.\n\n\n\nDepends-On: I8e16e1622e975dec2a42850f0e06159875d70f39\nChange-Id: I38e761c412f634b43a34af15b0f407b8d2b0bcf6\nSigned-off-by: Manideep \u003cmanideep.openstack@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b8d7404dc4e4deb082897f29e598c2a1190e98fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b8d7404dc4e4deb082897f29e598c2a1190e98fd"}]},"branch":"refs/heads/master","description":"Edit commit message"},"fda9fa198f54ed6f81b8a89196d8bb8a97df32bb":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2025-11-03 17:53:07.000000000","uploader":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"ref":"refs/changes/78/965878/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/965878/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/9"}}},"commit":{"parents":[{"commit":"496237a646a3cff808a52bce540356029f69c5b0","subject":"Merge \"Remove remaining job with Ubuntu Jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/496237a646a3cff808a52bce540356029f69c5b0"}]}],"author":{"name":"Manideep","email":"manideep.openstack@gmail.com","date":"2025-11-01 19:23:03.000000000","tz":-240},"committer":{"name":"Manideep","email":"manideep.openstack@gmail.com","date":"2025-11-03 17:53:07.000000000","tz":0},"subject":"Fix NetApp Manila Replication test TearDown Failures","message":"Fix NetApp Manila Replication test TearDown Failures\n\n[test_replication_negative.py:test_try_add_replica_nonexistent_subnet]  Changed to use self.create_share_network from the base class instead of self.shares_v2_client.create_share_network, so that the created share network is automatically tracked and cleaned up during test cleanup. The direct client call did not handle cleanup, causing resource leaks.\n\n\n\nDepends-On: I8e16e1622e975dec2a42850f0e06159875d70f39\nChange-Id: I38e761c412f634b43a34af15b0f407b8d2b0bcf6\nSigned-off-by: Manideep \u003cmanideep.openstack@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fda9fa198f54ed6f81b8a89196d8bb8a97df32bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fda9fa198f54ed6f81b8a89196d8bb8a97df32bb"}]},"branch":"refs/heads/master","description":"Edit commit message"},"da2bba08a133b3c79f7dabbde7d4bfaa9ab8a2c8":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2025-11-07 06:22:29.000000000","uploader":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"ref":"refs/changes/78/965878/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/965878/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/965878/10"}}},"commit":{"parents":[{"commit":"496237a646a3cff808a52bce540356029f69c5b0","subject":"Merge \"Remove remaining job with Ubuntu Jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/496237a646a3cff808a52bce540356029f69c5b0"}]}],"author":{"name":"Manideep","email":"manideep.openstack@gmail.com","date":"2025-11-01 19:23:03.000000000","tz":-240},"committer":{"name":"Manideep","email":"manideep.openstack@gmail.com","date":"2025-11-07 06:22:29.000000000","tz":0},"subject":"Fix NetApp Manila Replication test TearDown Failures","message":"Fix NetApp Manila Replication test TearDown Failures\n\n[test_replication_negative.py:test_try_add_replica_nonexistent_subnet]  Changed to use self.create_share_network from the base class instead of self.shares_v2_client.create_share_network, so that the created share network is automatically tracked and cleaned up during test cleanup. The direct client call did not handle cleanup, causing resource leaks.\n\n\n\nChange-Id: I38e761c412f634b43a34af15b0f407b8d2b0bcf6\nSigned-off-by: Manideep \u003cmanideep.openstack@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da2bba08a133b3c79f7dabbde7d4bfaa9ab8a2c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da2bba08a133b3c79f7dabbde7d4bfaa9ab8a2c8"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
