)]}'
{"id":"openstack%2Fmanila-tempest-plugin~966368","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I3d13627e71d3fc324eb0b020fc83920ad3fd18a7","project":"openstack/manila-tempest-plugin","branch":"master","topic":"2024-1-last","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-11-14 21:01:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3d13627e71d3fc324eb0b020fc83920ad3fd18a7","subject":"Update stable branch jobs","status":"MERGED","created":"2025-11-06 22:15:49.000000000","updated":"2025-11-14 21:01:30.000000000","submitted":"2025-11-14 21:01:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"966368-2024-1-last","meta_rev_id":"1a194d44eba128c94ed879c04d71971fb64db34b","_number":966368,"virtual_id_number":966368,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-14 21:01:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2025-11-07 19:08:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2025-11-07 19:08:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-07 00:13:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-07 19:08:14.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"fca31c4396c6cbccbc24fb2f5006d8a72c3e7abe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-11-06 22:15:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"730222ff275cf9704770cc126f726a5c05ed571f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 00:13:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2adb7ef5b585419eb388ed30c3832bb2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be6f780481fb40839729288823b5c791 : SUCCESS in 2m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5c82a7548e0548aba02cf65199788e5a : SUCCESS in 1h 22m 15s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3c575d9b0e2746ae874a29a8d8c9efd0 : SUCCESS in 1h 44m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/de0e267f46ca472386ab00c14d609572 : FAILURE in 1h 13m 55s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/96e00add1cf34cdd9c32895e89e37406 : SUCCESS in 39m 53s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/82d625f548b74ebfadf1fa441113b275 : SUCCESS in 1h 17m 09s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a14e746c8b244d8b886f054011c45222 : FAILURE in 15m 37s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/21f063183c1640d3bb37a7ac72c8a354 : SUCCESS in 37m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d69c1c034554402d8f6b2b31ec1a1388 : SUCCESS in 43m 52s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/85a57e693ad840f6b909b9802611cb4d : SUCCESS in 1h 02m 08s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/c98ab0923627490b8aa7b8d4a6378bd3 : FAILURE in 29m 12s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/9d2432f0a56b4b6bb758ab18694825ed : FAILURE in 19m 45s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/5e15b0db3bab4e048a7f8dd0bef64f58 : SUCCESS in 31m 32s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/46f6f94aff72468990eeca12fb346a59 : FAILURE in 19m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9bc2b641c4bd15ab841d571c34dbafdca6b51ecf","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-11-07 03:43:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e4ab4b23de9e5f653af613a1e3ace3f0568fddf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 06:16:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05a6d2a9882346a183a612d2bf34ce25\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccea7598eb54408db5492b40117c9fc3 : SUCCESS in 2m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2df147f8d065452fbd994b5ea4335643 : SUCCESS in 1h 07m 23s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4ecdc43c25924266ab8d40300bea22bc : SUCCESS in 2h 17m 02s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/921bc357d1b54b1b8f7ad698d4ab0b7b : FAILURE in 1h 02m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/205b4cbc07ef4b8cbe734da336b0c6ef : SUCCESS in 1h 07m 37s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/773136081d1b419a9925906a347fe123 : SUCCESS in 1h 31m 05s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fc9eb4e6319a452c9b60f440fdbde32b : FAILURE in 15m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/90cc5f27dd3943eaa3ad1e9cd3009dab : SUCCESS in 33m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1f12b4b627794e638cc67a5c62c9f064 : SUCCESS in 37m 42s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/7782c2c40e9d4b838b9350595037d2a9 : SUCCESS in 1h 08m 39s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/0a91f8c246e445129d140309f022ea9c : SUCCESS in 1h 22m 06s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/44d587ebb7954208bddb9696a4c15378 : SUCCESS in 1h 09m 41s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/0d25313efb1a498daa378da23dc5723c : SUCCESS in 30m 51s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a3b09b830046487b9b3e347d7d6c5b21 : FAILURE in 24m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ba4a7165feeaeac059c4de9f0b543fb897e2b5ea","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-11-07 19:08:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"06d59e56ba69b5d82371942699763614741ae856","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 19:35:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"44fbb036d33c1bc6438824bfa993733f8e2d2af6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 21:01:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91ec6c71437b44179c5c78fd7f41eb8c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbe9ce137a4544d9bad5f9fcbaf6b826 : SUCCESS in 2m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8f789aaec0a040229828787e46f7a9e3 : SUCCESS in 1h 21m 10s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2d61e3527ee54c78a5ef2682e4cfcfee : SUCCESS in 35m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/72d8e43f385345358444ba7870ae972d : SUCCESS in 42m 28s","accounts_in_message":[],"_revision_number":1},{"id":"1a194d44eba128c94ed879c04d71971fb64db34b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 21:01:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9dae3a4053c334619680a934c2bdc7300f0a5d19","revisions":{"9dae3a4053c334619680a934c2bdc7300f0a5d19":{"kind":"REWORK","_number":1,"created":"2025-11-06 22:15:49.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/68/966368/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/966368/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/966368/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/966368/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/966368/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/966368/1"}}},"commit":{"parents":[{"commit":"fe1c29d2724d12f478d8cc7483beabb3fec0e0c6","subject":"Add stable/2025.2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fe1c29d2724d12f478d8cc7483beabb3fec0e0c6"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-06 21:44:52.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-06 21:44:52.000000000","tz":0},"subject":"Update stable branch jobs","message":"Update stable branch jobs\n\nAdd new stable/2025.2 job, remove unmaintained/2024.1\njobs.\n\nChange-Id: I3d13627e71d3fc324eb0b020fc83920ad3fd18a7\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9dae3a4053c334619680a934c2bdc7300f0a5d19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9dae3a4053c334619680a934c2bdc7300f0a5d19"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
