)]}'
{"id":"openstack%2Fmanila-tempest-plugin~967139","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I1a5ae086b68f34f3761b61f7b4cfdcbf717ab1da","project":"openstack/manila-tempest-plugin","branch":"master","topic":"cleanup-fixes","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-11-18 12:35:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1a5ae086b68f34f3761b61f7b4cfdcbf717ab1da","subject":"Stop attempting share type deletions that\u0027d fail","status":"MERGED","created":"2025-11-13 22:34:04.000000000","updated":"2025-11-18 12:35:49.000000000","submitted":"2025-11-18 12:35:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967139-cleanup-fixes","meta_rev_id":"f59abccb219ad4c59ff6d2ae4d3d1b745e17b2b3","_number":967139,"virtual_id_number":967139,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-18 12:35:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2025-11-17 14:18:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2025-11-18 11:54:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2025-11-18 11:54:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-14 00:56:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-17 14:18:24.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2025-11-18 11:54:46.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"3005b63530eed9c2470f5f5202285ea9a4f42acc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-11-13 22:34:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55ff542df08a248e33ccade70f60d61089c883b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-11-13 22:36:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e99e04e01ffbef03cbc8a6fb40525babc7d0e17a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 00:56:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dc0ffa1cac444e7b88f72c515a9cea4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d51650ff4bf4d29a5a427311b4f207a : SUCCESS in 1m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/20af5f588fb74a9ea570677ee840069d : SUCCESS in 39m 56s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7ed0ffbac36f47639bb1ee57cba3aac6 : SUCCESS in 2h 17m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f6633179dc5f4398a7322ad7cc4500fe : FAILURE in 46m 11s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d8e86aeea01944f3812a958d9ae9799a : SUCCESS in 32m 57s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/b440dd3989c141079030a5151e8b01d9 : SUCCESS in 45m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/13492719b0114f6599faf8905a68f098 : FAILURE in 12m 37s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bc6c598c7c764ff1af17557c03e887db : SUCCESS in 22m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/73bb106b23e44bd3933add7bfd6bc78d : SUCCESS in 43m 02s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/de40a451753c4df8acd2662307fc840f : SUCCESS in 1h 28m 50s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/cf7c394fad6c42829a6636a1318dbce7 : SUCCESS in 51m 07s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/55622b556d724e098805b79f9b7d4021 : SUCCESS in 1h 19m 17s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/75be0aee64be41429d28e8e49c443501 : SUCCESS in 19m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/64e4e956785f46c78fcb86d513db9a24 : FAILURE in 24m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ec5c74afb362b4ce9a598127ba5b020bbfa61aed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-11-14 18:53:04.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ab3cf0df86043279d3189ab0f933e1f76383203d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-11-14 18:54:36.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"a4f6c8c78ff1395298c6bd67f61ea6fb88da45b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-11-14 18:57:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"31fec5215dae6492ef788f69eb612712f2a5660f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 21:20:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51149fd23dd047cbb25437472a2ec4d6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4981b6205f446eb993c2b2187b9ce40 : SUCCESS in 2m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/001d7d82a71a4cfcb28908630a788b3d : SUCCESS in 1h 13m 52s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d25f7ead22a246be86ceb59584d53de4 : SUCCESS in 2h 22m 36s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c48faa92e5ab47daa1244ab43cdd2308 : FAILURE in 1h 22m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/6ac99ddb3b72424988cf28014cf8858e : SUCCESS in 44m 18s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/d55ee5406ea64adabe43911912b5fa08 : SUCCESS in 1h 28m 52s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2dec8bb241e0498c93d030e21e9018b1 : FAILURE in 15m 07s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d7accd325f5e46f4bf22669c9a1f81ec : SUCCESS in 34m 42s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1bb12200bab944319199acd3f5db5146 : SUCCESS in 30m 58s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/fa0ef390c233483aa2ea16d84e059c42 : SUCCESS in 1h 16m 23s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/8b3819b066d348289ac5600ebd804bfd : SUCCESS in 1h 19m 31s\n- manila-tempest-plugin-lvm-2024.1 https://zuul.opendev.org/t/openstack/build/a848db56d82c4d95a622955ebef2edd0 : SUCCESS in 55m 30s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/dafdd7e4c4854720a54f840cf7f23d3d : SUCCESS in 28m 45s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0817fb9442d94ad69c9a097fbf50a647 : FAILURE in 25m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"48b5a99e8b08355e5ed42a2ddd0583a056c07916","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-11-17 14:18:24.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e521b6f30eaed0f47edc068744e6f3a2503ec2a8","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2025-11-18 11:54:46.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"426f5d485ca1fa9bcf325a1fba04d2790006e8c8","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2025-11-18 11:54:58.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a4217c3cc2e0ead81d9ba1c4b847b0d54f1422fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 11:55:22.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f44e6ea0e232b8ba569fedeb6e7f36649feef0c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 12:35:49.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa43089f2e61493dbddaf4da9f1a5b47\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/647ecdbc5e12476b95cab01d298d1a3e : SUCCESS in 2m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f32d681c31074679be89df3ab0912150 : SUCCESS in 39m 58s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b794a4a87a834bd6917a0a7376da5680 : SUCCESS in 26m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5ab2c15fc56a444cbf05051e8c3cf214 : SUCCESS in 29m 27s","accounts_in_message":[],"_revision_number":5},{"id":"f59abccb219ad4c59ff6d2ae4d3d1b745e17b2b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 12:35:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"78daf10b5b430b6d43bfcf9c3f0eb96d8cf013ea","revisions":{"abcc2dc2d3710d36584b2b0ef3fc883e67f030de":{"kind":"REWORK","_number":1,"created":"2025-11-13 22:34:04.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/39/967139/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/39/967139/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/967139/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/967139/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/967139/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/967139/1"}}},"commit":{"parents":[{"commit":"5ca4e5190a7921f019b2c9101e1d5374a9f8ac80","subject":"Merge \"Fix share server migration with replica test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5ca4e5190a7921f019b2c9101e1d5374a9f8ac80"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-11-13 22:30:16.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-11-13 22:30:16.000000000","tz":-480},"subject":"Stop attempting share type deletions that\u0027d fail","message":"Stop attempting share type deletions that\u0027d fail\n\nYou can\u0027t delete a share type that\u0027s in use\nby shares. Add this logic to the cleanup script\nhandler so we stop seeing unnecessary cleanup\nattempts. Failed cleanups are always, unfortunately,\nthe test runner\u0027s responsibility since we can\u0027t\nunderstand why something failed to be deleted.\n\nChange-Id: I1a5ae086b68f34f3761b61f7b4cfdcbf717ab1da\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/abcc2dc2d3710d36584b2b0ef3fc883e67f030de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/abcc2dc2d3710d36584b2b0ef3fc883e67f030de"}]},"branch":"refs/heads/master"},"b6efa913ed7f2bb7550d8ff5cfa4fdc608f01903":{"kind":"REWORK","_number":2,"created":"2025-11-13 22:36:21.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/39/967139/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/39/967139/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/967139/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/967139/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/967139/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/967139/2"}}},"commit":{"parents":[{"commit":"5ca4e5190a7921f019b2c9101e1d5374a9f8ac80","subject":"Merge \"Fix share server migration with replica test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5ca4e5190a7921f019b2c9101e1d5374a9f8ac80"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-11-13 22:30:16.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-11-13 22:35:45.000000000","tz":-480},"subject":"Stop attempting share type deletions that\u0027d fail","message":"Stop attempting share type deletions that\u0027d fail\n\nYou can\u0027t delete a share type that\u0027s in use\nby shares. Add this logic to the cleanup script\nhandler so we stop seeing unnecessary cleanup\nattempts. Failed cleanups are always, unfortunately,\nthe test runner\u0027s responsibility since we can\u0027t\nunderstand why something failed to be deleted.\n\nChange-Id: I1a5ae086b68f34f3761b61f7b4cfdcbf717ab1da\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b6efa913ed7f2bb7550d8ff5cfa4fdc608f01903"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b6efa913ed7f2bb7550d8ff5cfa4fdc608f01903"}]},"branch":"refs/heads/master"},"83faf45d50dca120685ecf3d6c6a648dcbc62d4d":{"kind":"REWORK","_number":3,"created":"2025-11-14 18:53:04.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/39/967139/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/39/967139/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/967139/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/967139/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/967139/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/967139/3"}}},"commit":{"parents":[{"commit":"5ca4e5190a7921f019b2c9101e1d5374a9f8ac80","subject":"Merge \"Fix share server migration with replica test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5ca4e5190a7921f019b2c9101e1d5374a9f8ac80"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-11-13 22:30:16.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-11-14 18:53:04.000000000","tz":0},"subject":"Stop attempting share type deletions that\u0027d fail","message":"Stop attempting share type deletions that\u0027d fail\n\nYou can\u0027t delete a share type that\u0027s in use\nby shares. Add this logic to the cleanup script\nhandler so we stop seeing unnecessary cleanup\nattempts. Failed cleanups are always, unfortunately,\nthe test runner\u0027s responsibility since we can\u0027t\nunderstand why something failed to be deleted.\n\nChange-Id: I1a5ae086b68f34f3761b61f7b4cfdcbf717ab1da\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/83faf45d50dca120685ecf3d6c6a648dcbc62d4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/83faf45d50dca120685ecf3d6c6a648dcbc62d4d"}]},"branch":"refs/heads/master"},"801147346185a659213223e4742e54ae1e719aa1":{"kind":"REWORK","_number":4,"created":"2025-11-14 18:54:36.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/39/967139/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/39/967139/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/967139/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/967139/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/967139/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/967139/4"}}},"commit":{"parents":[{"commit":"5ca4e5190a7921f019b2c9101e1d5374a9f8ac80","subject":"Merge \"Fix share server migration with replica test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5ca4e5190a7921f019b2c9101e1d5374a9f8ac80"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-11-13 22:30:16.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-11-14 18:54:35.000000000","tz":0},"subject":"Stop attempting share type deletions that\u0027d fail","message":"Stop attempting share type deletions that\u0027d fail\n\nYou can\u0027t delete a share type that\u0027s in use\nby shares. Add this logic to the cleanup script\nhandler so we stop seeing unnecessary cleanup\nattempts. Failed cleanups are always, unfortunately,\nthe test runner\u0027s responsibility since we can\u0027t\nunderstand why something failed to be deleted.\n\nChange-Id: I1a5ae086b68f34f3761b61f7b4cfdcbf717ab1da\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/801147346185a659213223e4742e54ae1e719aa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/801147346185a659213223e4742e54ae1e719aa1"}]},"branch":"refs/heads/master"},"78daf10b5b430b6d43bfcf9c3f0eb96d8cf013ea":{"kind":"REWORK","_number":5,"created":"2025-11-14 18:57:21.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/39/967139/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/39/967139/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/967139/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/967139/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/967139/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/967139/5"}}},"commit":{"parents":[{"commit":"5ca4e5190a7921f019b2c9101e1d5374a9f8ac80","subject":"Merge \"Fix share server migration with replica test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5ca4e5190a7921f019b2c9101e1d5374a9f8ac80"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-11-13 22:30:16.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-11-14 18:57:14.000000000","tz":-480},"subject":"Stop attempting share type deletions that\u0027d fail","message":"Stop attempting share type deletions that\u0027d fail\n\nYou can\u0027t delete a share type that\u0027s in use\nby shares. Add this logic to the cleanup script\nhandler so we stop seeing unnecessary cleanup\nattempts. Failed cleanups are always, unfortunately,\nthe test runner\u0027s responsibility since we can\u0027t\nunderstand why something failed to be deleted.\n\nChange-Id: I1a5ae086b68f34f3761b61f7b4cfdcbf717ab1da\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/78daf10b5b430b6d43bfcf9c3f0eb96d8cf013ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/78daf10b5b430b6d43bfcf9c3f0eb96d8cf013ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
