)]}'
{"id":"openstack%2Fmanila-tempest-plugin~967509","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I743ef4bdb315abf91e7eb89a5f8f1983400ded65","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"38479":{"account":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"last_update":"2025-12-19 06:52:39.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I743ef4bdb315abf91e7eb89a5f8f1983400ded65","subject":"Intermittent ReplicationTest failure fix","status":"ABANDONED","created":"2025-11-18 11:26:58.000000000","updated":"2025-12-19 06:52:39.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"73472114be0acfbe52e920acab5b3ec8d90f2451","_number":967509,"virtual_id_number":967509,"owner":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-11-25 07:49:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-18 13:33:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-18 16:35:20.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"a7b9ec1026462b6c0ff92a5d851a067c46628c02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2025-11-18 11:26:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c2d0744e272cb6e4e7e54b08d3e327927d7cb5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 13:33:56.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/47a9013548764994956975d46c6ba67c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b250d5ef379e4c6c95e26ed2322279e8 : FAILURE in 3m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/665661ad37c64e368a98a2cc8140eb54 : SUCCESS in 1h 13m 54s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/07197684fff142199aa21d1699694bc2 : SUCCESS in 2h 05m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2c6ff41b511142a19ec5246ac67e4567 : FAILURE in 1h 19m 07s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e14da728b2e24a5db97dd01f201ac3ca : FAILURE in 37m 00s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/491e81667b5a441b9af9c7b46b63ab4b : SUCCESS in 40m 55s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c64e1663dd7b4adfb61dbf345f4fc414 : FAILURE in 22m 41s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e5bb01cf6fdc4171b5b1a247f36641dc : SUCCESS in 38m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/14a9099ec95d4bb3aeefc1a415214724 : SUCCESS in 46m 21s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/5b7a74d81e1c44869e147292e2701f2f : SUCCESS in 58m 37s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/7ebc170441c449468036f4de1907384d : SUCCESS in 1h 28m 05s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/2d62b28dea834f6e8bcdd108d4cc3c58 : SUCCESS in 36m 48s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/1917adb710654dc2a92ec2f3bc5d7551 : SUCCESS in 27m 52s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/56c50480bb734f75a27b81bdadb694c5 : FAILURE in 28m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6bbc02ee8704c742dd91dbbac7db582c52d8a37c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2025-11-18 14:04:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"94c7c5a9b6d5d41a61a2181f46643eb6dbfab4ec","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2025-11-18 15:01:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"300e9ae58c66991db694c39c73069b58cf4cf382","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 16:19:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9ab2964d3b245a19bff032cc32d3130\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e8220499fe64723a6f5c9443e1768b6 : SUCCESS in 3m 24s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3aa8b8f8eb6647d485ec81b923a77975 : SUCCESS in 1h 12m 58s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/29beece7134b4726bf829351c5444cfd : SUCCESS in 2h 12m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/accd87a46c1544aab48254ab2d2870b0 : FAILURE in 1h 39m 16s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/197813ed4b214ec6bad5aac96e7ba838 : SUCCESS in 42m 55s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8bcd0b2490ed427c95016ba1d14b83ce : SUCCESS in 1h 35m 27s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f64be2e2736044f089ea3b77cb4eadb4 : FAILURE in 22m 58s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/083cc65adeed4c5a9edc8ba3b6c59f81 : SUCCESS in 24m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f7ac1a88f0ab44318b3a3ff988f6cbc6 : SUCCESS in 44m 33s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/82452a2655aa4eaf9c4608718c14fba8 : SUCCESS in 40m 03s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/bb03be7baea94238b794cd48369e5580 : SUCCESS in 1h 17m 48s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/8c45e3964e1646238fae756dccdbf6af : SUCCESS in 55m 44s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/c7a13167ba6a4fc4be92955093c49974 : SUCCESS in 18m 43s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5ca7b54af4284feebad212537fb77bcc : FAILURE in 22m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"80fd450e0f340afaf7135b52d30c5bf6999c6881","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-11-18 16:35:20.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/09/967509/2/upstream-check/manila-tempest-plugin-ontap-dhss/492c6f7/ : SUCCESS in 1h 32m 31s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"90e9c8bd8135e9022b1253a760fd5e84c28b48ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2025-11-19 11:21:43.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"6a4f16ea5520629b5193ed1ca25fe4a9268a4f42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 14:02:15.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c90cc47c609a4a5eb813b08eaf521632\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ba4489407db4ffebba0fe8808b52bda : FAILURE in 1m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9a57fb744c31474fa4dc3c1f8347bf4b : FAILURE in 45m 08s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3b11abf301e24ff4b5fceafef88ed0fa : SUCCESS in 2h 34m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d1c01eb7b26d4d94bbfe03325ca649ef : FAILURE in 1h 28m 30s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d80a9e25df534d4f97674678b5cda205 : SUCCESS in 1h 01m 28s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ed12094f93ee45d793e26862455d261f : SUCCESS in 1h 50m 37s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/49281b4ffcc74c21890de3a10fa86e96 : FAILURE in 24m 22s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b22234481b62415c8ab29b546d34c377 : SUCCESS in 35m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e8211df3af26456694d21a59496ff496 : SUCCESS in 48m 53s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/5c9b0e5f220847098e04e550a629cd14 : SUCCESS in 36m 09s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/913f0c8aa07743b2bc5d77ba4b7d697d : SUCCESS in 1h 14m 11s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/a055964fa5a8466e952e4c2597e60d72 : SUCCESS in 38m 25s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/930f5dee156f40a9ba879f12cdfe2d89 : SUCCESS in 35m 09s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/910144e779a240e6bcc88af9dde62ad7 : FAILURE in 23m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"516652ad33d4c869619aa6576b4ec0ed594f83ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2025-11-19 14:28:49.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"641fee2101d702b9203aafff11645b457e080660","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 16:58:09.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a38d83fe3d2649499def08eb05ce962d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc6de522c7ee4c00ade9e6a315500902 : FAILURE in 2m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/00cab8531e2c4c5dafdfe8b1b5f23f32 : SUCCESS in 1h 16m 44s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/be616720e3304976b543af7a2a111cb5 : FAILURE in 2h 24m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d9dca56f2cc548db867e5df0bf8b54f0 : FAILURE in 1h 26m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/4f6425d1c19448f6a9df14905b0c6b7c : SUCCESS in 44m 48s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/69b945ed34ef428ab0fa2bad1f5f69e4 : SUCCESS in 49m 53s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e0144e0768414717b43118a502d949ab : FAILURE in 11m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/288b258a960b4d7684d89b1387266867 : SUCCESS in 23m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d72a6096c8b4456599c301fcfa65ec1f : SUCCESS in 41m 54s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/5a54e01cee5c49f0b3be2c347fe95f86 : SUCCESS in 42m 49s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/66f90eb1ba25498c8e955b3c3843554f : SUCCESS in 1h 17m 12s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/83a98017df004b008c5f3b70caf1d5ee : SUCCESS in 1h 00m 46s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/026667e53ae6498fad1d54eca073f40d : SUCCESS in 31m 02s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/436ea22906b34b9890a5de1412e3909c : FAILURE in 12m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b82d3809024a20e50f0a4797294ff3391524cda2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2025-11-24 05:14:50.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"3b100a180e1df7d5375c1bf5cf33de5bd8e3b747","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2025-11-24 05:15:56.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"09dfe4273646a8619707eb2975f5b9d6703085ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2025-11-24 05:16:50.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"5d1fd452bc312adf6b2855dd32984e467d3b4eb7","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2025-11-24 05:17:38.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a673a8aef5ccbb4dbe559f341124591c8122bf4e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-11-24 07:16:14.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/09/967509/7/upstream-check/manila-tempest-plugin-ontap-dhss/e93bfdf/ : SUCCESS in 1h 50m 59s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"79859c65f971819e67443c60efb6450d7cb85c22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 07:40:36.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e00d54127404e18b80a19813e784ae3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5404d12e115d4d9093d4c15ab341ea64 : SUCCESS in 2m 16s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/869bfb01878b493ba64077e476446377 : SUCCESS in 1h 16m 22s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f6009b42f7eb4f7b90a3826021056697 : SUCCESS in 2h 19m 55s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/100b1e50a8014ba6b723303ba3002132 : FAILURE in 1h 26m 55s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ca6ff86055064e22b3ded19e813c59d1 : SUCCESS in 49m 17s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/595486222e0e4c16b598c72e7d82bb7a : SUCCESS in 1h 40m 48s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c6433642218a40f3ac4f358e2388a9b0 : FAILURE in 12m 38s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/187d57442dea46ec96e9cd420c1690e6 : SUCCESS in 33m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3569896eca4a4d2fb391bc8cc09be5e5 : SUCCESS in 37m 21s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/fa0c4fa5039f405f83d440f159d9a4bb : SUCCESS in 48m 10s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/662802389d1a4b0cb7f2d333c890bb8e : SUCCESS in 52m 57s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/6c56a999ed4b4af59bf3a1efc3486c53 : SUCCESS in 46m 05s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/364d2fd7db2f42af95e4df7ff58dd887 : FAILURE in 19m 42s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b543fc68fd3e4aed96d38efacca8be42 : FAILURE in 13m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f128593bf52a505f5e2568008117756d8a93e3e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2025-11-25 05:12:55.000000000","message":"Patch Set 8: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"e2d6fc4aa5a66bcfed5e64f3081f25d174924c96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 07:49:58.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cfe6cd0013cf48b8b943d7b3a87ad348\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56d0654a6b564439a0fcca4a5ea0425e : SUCCESS in 2m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f8bf1b26b2b9493291b2cf69b8fd4c4a : SUCCESS in 52m 57s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d4e38a4629ec4a2e9b599e99bd4aaf68 : FAILURE in 2h 32m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2ba1794fbc7b44b884986f01c830da5b : FAILURE in 50m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f39d53e3d4ad4cdeb0da84fab05662ed : SUCCESS in 1h 09m 40s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/4e79eecf83e54afba4ac00e75cbbd18f : SUCCESS in 1h 35m 00s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3098dee125d44b72994181436ea26914 : FAILURE in 22m 45s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e896dfaa37bb40268e09e65fa9ccc540 : SUCCESS in 35m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/51c4557c95e44389bf0fb79a68499411 : FAILURE in 24m 10s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/95475586f775497e9064e6245d5508e4 : SUCCESS in 46m 41s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/d6a015e80d984f13bf5408f7d1d9fe5a : SUCCESS in 42m 17s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/3d132fe448be436bb306138a06f60a1b : SUCCESS in 54m 49s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/514c04978f444899a9d1fc2787c065d5 : FAILURE in 31m 29s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f081f3c8706a430aa521be5386095123 : FAILURE in 15m 02s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f86f99f308d45d8fc4b9fd39ba47e1b9d5e86f74","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2025-12-01 17:15:16.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6d9941af56b32b16fc413797365078507b71cbf4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-12-01 19:05:32.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/09/967509/8/upstream-check/manila-tempest-plugin-ontap-dhss/668987c/ : FAILURE in 1h 42m 21s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"73472114be0acfbe52e920acab5b3ec8d90f2451","tag":"autogenerated:gerrit:abandon","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2025-12-19 06:52:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"98f343a0d46a76806db17fc3cd3874614b3487c1","revisions":{"f176034ef09ce5cf36ac598bb03b256a227f5ee8":{"kind":"REWORK","_number":1,"created":"2025-11-18 11:26:58.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/09/967509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/09/967509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/1"}}},"commit":{"parents":[{"commit":"f1a041b3612b9a2ea4407a0719acb51db0036a7b","subject":"Merge \"Update stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f1a041b3612b9a2ea4407a0719acb51db0036a7b"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2025-11-18 11:26:19.000000000","tz":-300},"committer":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2025-11-18 11:26:19.000000000","tz":-300},"subject":"WIP: ReplicationTest failure fix","message":"WIP: ReplicationTest failure fix\n\nChange-Id: I743ef4bdb315abf91e7eb89a5f8f1983400ded65\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f176034ef09ce5cf36ac598bb03b256a227f5ee8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f176034ef09ce5cf36ac598bb03b256a227f5ee8"}]},"branch":"refs/heads/master"},"68c4f1862912298f82f5cf126d74cec5e9e75b57":{"kind":"REWORK","_number":2,"created":"2025-11-18 14:04:26.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/09/967509/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/09/967509/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/2"}}},"commit":{"parents":[{"commit":"f1a041b3612b9a2ea4407a0719acb51db0036a7b","subject":"Merge \"Update stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f1a041b3612b9a2ea4407a0719acb51db0036a7b"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2025-11-18 11:26:19.000000000","tz":-300},"committer":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2025-11-18 14:03:26.000000000","tz":-300},"subject":"WIP: ReplicationTest failure fix","message":"WIP: ReplicationTest failure fix\n\nChange-Id: I743ef4bdb315abf91e7eb89a5f8f1983400ded65\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/68c4f1862912298f82f5cf126d74cec5e9e75b57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/68c4f1862912298f82f5cf126d74cec5e9e75b57"}]},"branch":"refs/heads/master"},"8590aa43128f345f4ae8b07c206006b895dafeb2":{"kind":"REWORK","_number":3,"created":"2025-11-19 11:21:43.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/09/967509/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/09/967509/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/3"}}},"commit":{"parents":[{"commit":"f1a041b3612b9a2ea4407a0719acb51db0036a7b","subject":"Merge \"Update stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f1a041b3612b9a2ea4407a0719acb51db0036a7b"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2025-11-18 11:26:19.000000000","tz":-300},"committer":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2025-11-19 11:21:00.000000000","tz":-300},"subject":"ReplicationTest failure fix","message":"ReplicationTest failure fix\n\nChange-Id: I743ef4bdb315abf91e7eb89a5f8f1983400ded65\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8590aa43128f345f4ae8b07c206006b895dafeb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8590aa43128f345f4ae8b07c206006b895dafeb2"}]},"branch":"refs/heads/master"},"e45513c37c668a974246c8306229e954ea6c57c6":{"kind":"REWORK","_number":4,"created":"2025-11-19 14:28:49.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/09/967509/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/09/967509/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/4"}}},"commit":{"parents":[{"commit":"f1a041b3612b9a2ea4407a0719acb51db0036a7b","subject":"Merge \"Update stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f1a041b3612b9a2ea4407a0719acb51db0036a7b"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2025-11-18 11:26:19.000000000","tz":-300},"committer":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2025-11-19 14:28:34.000000000","tz":-300},"subject":"ReplicationTest failure fix","message":"ReplicationTest failure fix\n\nChange-Id: I743ef4bdb315abf91e7eb89a5f8f1983400ded65\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e45513c37c668a974246c8306229e954ea6c57c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e45513c37c668a974246c8306229e954ea6c57c6"}]},"branch":"refs/heads/master"},"acc8fd05119cf1ec9d56df1ee996ad30aa5767aa":{"kind":"REWORK","_number":5,"created":"2025-11-24 05:14:50.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/09/967509/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/09/967509/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/5"}}},"commit":{"parents":[{"commit":"f1a041b3612b9a2ea4407a0719acb51db0036a7b","subject":"Merge \"Update stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f1a041b3612b9a2ea4407a0719acb51db0036a7b"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2025-11-18 11:26:19.000000000","tz":-300},"committer":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2025-11-24 05:14:36.000000000","tz":-300},"subject":"ReplicationTest failure fix","message":"ReplicationTest failure fix\n\nChange-Id: I743ef4bdb315abf91e7eb89a5f8f1983400ded65\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/acc8fd05119cf1ec9d56df1ee996ad30aa5767aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/acc8fd05119cf1ec9d56df1ee996ad30aa5767aa"}]},"branch":"refs/heads/master"},"97b3f66778a880db2f7d1fb1dfee08f068e76b7c":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-11-24 05:15:56.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/09/967509/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/09/967509/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/6"}}},"commit":{"parents":[{"commit":"f1a041b3612b9a2ea4407a0719acb51db0036a7b","subject":"Merge \"Update stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f1a041b3612b9a2ea4407a0719acb51db0036a7b"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2025-11-18 11:26:19.000000000","tz":-300},"committer":{"name":"Medha Choudhary","email":"medhaj06@gmail.com","date":"2025-11-24 05:15:56.000000000","tz":0},"subject":"ReplicationTest failure fix","message":"ReplicationTest failure fix\n\nDepends-on: I121e4d30294f41f14ee865005aa3e7aea83177a4\nChange-Id: I743ef4bdb315abf91e7eb89a5f8f1983400ded65\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/97b3f66778a880db2f7d1fb1dfee08f068e76b7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/97b3f66778a880db2f7d1fb1dfee08f068e76b7c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"49ce6e1da5416d950c22c2150088e6a7be606797":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2025-11-24 05:16:50.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/09/967509/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/09/967509/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/7"}}},"commit":{"parents":[{"commit":"f1a041b3612b9a2ea4407a0719acb51db0036a7b","subject":"Merge \"Update stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f1a041b3612b9a2ea4407a0719acb51db0036a7b"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2025-11-18 11:26:19.000000000","tz":-300},"committer":{"name":"Medha Choudhary","email":"medhaj06@gmail.com","date":"2025-11-24 05:16:50.000000000","tz":0},"subject":"ReplicationTest failure fix","message":"ReplicationTest failure fix\n\nDepends-On: I121e4d30294f41f14ee865005aa3e7aea83177a4\nChange-Id: I743ef4bdb315abf91e7eb89a5f8f1983400ded65\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/49ce6e1da5416d950c22c2150088e6a7be606797"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/49ce6e1da5416d950c22c2150088e6a7be606797"}]},"branch":"refs/heads/master","description":"Edit commit message"},"98f343a0d46a76806db17fc3cd3874614b3487c1":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2025-11-25 05:12:55.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/09/967509/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/09/967509/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/09/967509/8"}}},"commit":{"parents":[{"commit":"f1a041b3612b9a2ea4407a0719acb51db0036a7b","subject":"Merge \"Update stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f1a041b3612b9a2ea4407a0719acb51db0036a7b"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2025-11-18 11:26:19.000000000","tz":-300},"committer":{"name":"Medha Choudhary","email":"medhaj06@gmail.com","date":"2025-11-25 05:12:55.000000000","tz":0},"subject":"Intermittent ReplicationTest failure fix","message":"Intermittent ReplicationTest failure fix\n\nDepends-On: I121e4d30294f41f14ee865005aa3e7aea83177a4\nChange-Id: I743ef4bdb315abf91e7eb89a5f8f1983400ded65\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/98f343a0d46a76806db17fc3cd3874614b3487c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/98f343a0d46a76806db17fc3cd3874614b3487c1"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
