)]}'
{"id":"openstack%2Fmanila-tempest-plugin~973489","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I3775a6bd164b9ddeb19a8b8369b84cf22f4ffd92","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bp/qos-types","attention_set":{},"removed_from_attention_set":{"18816":{"account":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"last_update":"2026-08-24 06:11:58.000000000","reason":"Change was submitted"},"24993":{"account":{"_account_id":24993,"name":"Carlos","email":"carlos.carrero@veritas.com"},"last_update":"2026-08-24 06:11:58.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-08-24 05:21:09.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"35507":{"account":{"_account_id":35507,"name":"Kiran Pawar","email":"kiranpawarpict2010@gmail.com","username":"kpawar"},"last_update":"2026-06-13 05:38:07.000000000","reason":"Reviewer/Cc was removed"},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2026-08-24 06:11:58.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2026-08-11 12:11:03.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-08-24 06:11:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3775a6bd164b9ddeb19a8b8369b84cf22f4ffd92","subject":"Add tests for Qos Type API.","status":"MERGED","created":"2026-01-15 14:30:06.000000000","updated":"2026-08-24 06:11:58.000000000","submitted":"2026-08-24 06:11:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":62,"unresolved_comment_count":9,"has_review_started":true,"submission_id":"973489-bp/qos-types","meta_rev_id":"d335d6e7f866de41388ff6499a4f83629f0ee62f","_number":973489,"virtual_id_number":973489,"owner":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-24 06:11:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2026-08-24 05:21:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2026-08-11 12:11:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2026-08-24 05:21:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":24993,"name":"Carlos","email":"carlos.carrero@veritas.com"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24993,"name":"Carlos","email":"carlos.carrero@veritas.com"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-15 16:19:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-05 08:59:58.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":24993,"name":"Carlos","email":"carlos.carrero@veritas.com"},"state":"REVIEWER"},{"updated":"2026-02-05 08:59:58.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-02-17 17:06:31.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-02-23 04:56:06.000000000","updated_by":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"reviewer":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"state":"CC"},{"updated":"2026-04-27 08:12:14.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"state":"REVIEWER"},{"updated":"2026-04-27 08:12:14.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-06-12 18:27:06.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":35507,"name":"Kiran Pawar","email":"kiranpawarpict2010@gmail.com","username":"kpawar"},"state":"CC"},{"updated":"2026-06-12 18:27:06.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2026-06-13 05:38:07.000000000","updated_by":{"_account_id":35507,"name":"Kiran Pawar","email":"kiranpawarpict2010@gmail.com","username":"kpawar"},"reviewer":{"_account_id":35507,"name":"Kiran Pawar","email":"kiranpawarpict2010@gmail.com","username":"kpawar"},"state":"REMOVED"}],"messages":[{"id":"49e1c97130e159cc794ad28d87203f27b723aff1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-01-15 14:30:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2882dc5d4a9402186ed5f870088805e9569bc83f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 16:19:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ffe9746ba3b4c30a11a1a0084aaae75\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd4d1cadeb1d4435af5ec814b985f427 : SUCCESS in 2m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/50a79b11ca714cee8e8389d43a130759 : SUCCESS in 53m 50s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/480d1c3a738b43599692729855984bac : SUCCESS in 52m 24s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d3e326532e87461baa178f9b86c79b06 : FAILURE in 23m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/7ab507c14a4941f8a3dae71416a06605 : FAILURE in 43m 32s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/43372a07afee4e7683e23d8ede9a57d4 : FAILURE in 53m 44s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/01c3368b371f42b58d0669e89dd945f2 : FAILURE in 43m 01s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8a18bee98c024c0d8aa73d463a5a28bb : FAILURE in 36m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b6748939f29249feba9724141c808be2 : FAILURE in 30m 29s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/3ce3b6118cd64996990cc6d2a1ab8620 : SUCCESS in 49m 30s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/e62d923d599a4ee18ebf4b22e2fd4087 : SUCCESS in 1h 42m 54s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/6acbac0da4074fec875657db511ef4ac : FAILURE in 47m 47s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/cae8ed46264a46298dc115ea27432a8a : SUCCESS in 34m 08s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/533d4188b0044748923f00f939b1593c : FAILURE in 1h 01m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e59073b8858fac585b436e28e1662f8a929a773","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-01-17 12:22:12.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8fb936c7b9d620580edc06c42a40dfcb1a28b627","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-17 14:10:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0edac1ef1ab4a26ae9226e2907c9fd5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a2bd32367844ae59b31a2678aeafeb9 : SUCCESS in 2m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d6553a385bb9439c989fb3cd7444159d : SUCCESS in 1h 02m 27s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a60aa0e0c5d34f829949b50053b9e786 : SUCCESS in 1h 46m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/49c51940429e4b419ba94f6d0c83189b : FAILURE in 1h 15m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9cf62ca10ad647f89abb6afc630581de : FAILURE in 32m 45s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/30d2b78bb209462e94608e72d1eb10a5 : FAILURE in 1h 20m 16s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/74894f67cfeb447dbfd5e4c8205c6c88 : FAILURE in 1h 23m 42s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1a35eab08107419991caab530ea60fa6 : FAILURE in 21m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1e836d2320134a4e9c2ac1f29d5ab1fd : FAILURE in 29m 21s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/499a4998d28b4bc09c106c7bd0b8e161 : SUCCESS in 52m 24s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/84188669d21f40f1b4ae2c55b1593729 : FAILURE in 38m 25s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/2d69c825361649aaa67919663507a9a1 : SUCCESS in 1h 08m 48s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/2b42db2be0684e928c89bdcd5220a94d : SUCCESS in 32m 43s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c205d81730bb4515b3665343fa9c3580 : FAILURE in 28m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"97760ac447255b1a698cf632c66c9e19c6a043f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-01-19 08:59:08.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4a1f75be3e65985a61aa6b3f88153d967f944255","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 10:34:41.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/manila/+/967822 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"f9ac45899457f09fceceb60b1873f307ce64179d","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-01-19 10:39:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5dae559e9e30cb69250791a4f19ba4a69b8b1700","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 11:27:15.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/manila/+/967822 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"9394e9719bd4c77c97366e881422c3ab00b0d48d","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-01-19 12:24:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6a3da528103892a86c21fb976c8c4382ecb36069","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 14:43:03.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/782d125f97754c138cec510a7fe8c096\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2317ed2a42b043af99682702f6e243c4 : SUCCESS in 3m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1eeff43fd08442b896fb1b7dcfa02567 : SUCCESS in 1h 16m 23s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4e242a7e42ea44f28415a8bf63b89296 : FAILURE in 2h 13m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ee09a3a5972c4342bdf50bce6320cdf3 : FAILURE in 31m 07s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c89b158317374bf0a587d48eaf7918e8 : FAILURE in 1h 05m 30s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/db16f3cd92fa40c79839bcb0faa5b4c8 : FAILURE in 1h 31m 06s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d4838831bf4446a6aad6999a002fb64a : FAILURE in 1h 18m 31s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5ec381e229c24524aedb37ae244870ca : FAILURE in 36m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/525566acd51d4c69a71f2ddaf95b4601 : FAILURE in 26m 24s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/6ced40139b014ad0850191340876c3d4 : SUCCESS in 55m 59s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/3a5f9176167745e49bb9f81dfe1a8d4b : SUCCESS in 1h 09m 50s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/e6d62001d15344c7a39b65087a8be784 : SUCCESS in 1h 07m 43s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/dbc18c771fdf4810b1a485afc212c826 : SUCCESS in 34m 54s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/da4c304f455d45a6a5ca68e7495b2a32 : FAILURE in 51m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bc940aa3f344f73655a61ec2aff7a8411086e20b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-01-19 14:45:50.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a3fd4a2220298b448df563ca20dabe9bcaa5d221","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-01-19 16:08:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5ebd261abc68d5cd966cb40afbc1ab19d96d4f12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-01-19 16:47:04.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"76f5fb2410147080bf7cf5de6e5ee9c65edd5c7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 18:26:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c537440e08244ef9a246141a02560a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4731b794f364da7921142d92290d27d : SUCCESS in 3m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/661aaeecf8334e74b24024c14d3ff76f : SUCCESS in 1h 01m 07s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/bf3245dd7f6343f28f8466b9550ee0ac : SUCCESS in 55m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9d40e81dc9aa461bba71968caf5b5c80 : SUCCESS in 1h 25m 22s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/0bd30a0cd4f84edb8e4704aa45e03cc6 : SUCCESS in 1h 00m 20s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/b8ba6fd1e080431c890afecf847a3de3 : SUCCESS in 1h 37m 12s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0504f2a133764e28a45d8cb6e304cf4e : FAILURE in 38m 36s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f19a2523907548c88cf6dfdec6e891e4 : SUCCESS in 29m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ff718c3a17fc4c9eafad30b7d8cab3c0 : SUCCESS in 43m 19s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/b3ffe206b2364637846638838b65746b : SUCCESS in 34m 42s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/1bc51a8fcbd44d2bb37d85c9a01d67b5 : SUCCESS in 52m 34s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/e7f729fb73314470a1ac486bf1613acd : SUCCESS in 31m 11s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3b8bb6001b14409aa5b3aaa27d26c0a3 : SUCCESS in 34m 20s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f9275c8247b6421488cf528d72c2ea1c : FAILURE in 29m 17s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0e3f14e5c2fdec6dc34668996644c13fe55e16e0","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-02-05 09:00:06.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c73e8bc106916c5500250932eea420d36d3f60ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 11:37:51.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7812496e1f44451fb58bdfbeafc81999\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fadc4d5489d4a9b90684f1da426c76e : SUCCESS in 3m 07s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1147cf2695304d77b8591f64d90b0983 : SUCCESS in 1h 16m 40s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d811485a37404fa9a8cdfbe4b2dbb18c : FAILURE in 2h 31m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/04f79e14d70a4661b9550118b45fc727 : FAILURE in 1h 17m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f8d28a885226420a80f1a90cc33a2752 : FAILURE in 28m 47s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f0befc0a1e34490599888b0ca26ee2dc : FAILURE in 21m 23s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2fd6600acbc247fba90dc895ab348e66 : FAILURE in 1h 11m 04s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/97447fec6d764b549b43ff6bfad89939 : FAILURE in 32m 12s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ab7dc87dcf1c436ba37329e6fa0d7f29 : FAILURE in 41m 07s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/ad7ade5a96804646a57bd1a58a391b3e : FAILURE in 12m 02s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/364d3f33718b4a28955cc2d5d15be2d7 : FAILURE in 1h 11m 04s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/b1821067b8b143b884bb3fd30ca3030e : SUCCESS in 31m 36s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/b46e2ffc32ec4eafbaabfa549e69f684 : SUCCESS in 21m 25s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a5acd9f0a9c4430ba24a45e1d638187e : FAILURE in 58m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b0fb06fb949197220ec45360ff542f165e56b59f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-02-05 12:07:32.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"ad41a8defeed26d27b9b1d9f33029ed4fb5aa8e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 14:16:52.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c43409ba3d8b4246894c7ce7e4e9c9cf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d041741469c4b0f8a535bb8363a70ec : SUCCESS in 2m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/25262f655f6c4e85b849636cc1a558ff : SUCCESS in 1h 10m 10s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/fa0a45b678e0453893ce11ee17de8558 : SUCCESS in 2h 01m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fd7722fc77104383a995e37e872b1d83 : FAILURE in 36m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/495645b8953e4fbc8af6366eac531353 : SUCCESS in 1h 01m 10s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/7b16dd415c494bf48a481599a354f60e : SUCCESS in 1h 36m 15s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/727f77524fdf4529a1d4095996a2cd6f : FAILURE in 1h 17m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/22d23c2f6517434a951adad65d66a574 : SUCCESS in 35m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0fe289baeaad4de1aa240ff7dbd4f0ab : SUCCESS in 43m 47s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/b67f35adbdf04a868dcfdf9d4b129e89 : SUCCESS in 56m 49s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/0f972f1488e342f386d20f2c79196ba3 : SUCCESS in 1h 09m 42s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/59d1080b0e4b4ac5bdbf187f464bac3a : SUCCESS in 1h 25m 01s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/9da5b5a0aa9e407ea18a96af4fcd7996 : SUCCESS in 27m 15s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e33d740f31d14e17bf76efec365e5a7e : FAILURE in 1h 16m 30s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e27c7ef889945f3ee0df3310d2cbb4a52f70ecb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-02-17 14:07:01.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"df47b2dec7fafbb01c4ac6ab10209b350bec6e5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 16:51:52.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a6da2506e63471491040db74af19b6f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eafcda9e5bc4daea3f60c24ca8b9285 : SUCCESS in 3m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4380b4407903428f99a021924e85baa0 : SUCCESS in 1h 24m 40s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/04a21facba3f4c7f8cd7454dbe7b62e4 : FAILURE in 2h 37m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b1dab6344c8f4db1aa5b6e0f25a2bed2 : FAILURE in 42m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9e8c476c44c741f6a6b65d4e5b28297f : FAILURE in 1h 02m 47s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/1d77aa7543274239bc1417ce30910e94 : SUCCESS in 1h 24m 17s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/62e58c9dceac4425838a67117a55ddd9 : FAILURE in 1h 30m 22s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/08b3fef2d28142bfaf26645c6ab55046 : SUCCESS in 19m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/bf1c66fc357d45ef80b3abd8caee4888 : SUCCESS in 44m 44s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/2eda0e4f4e104eb79bbd589f41a46187 : SUCCESS in 1h 26m 25s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/2a33018627e1452bba16a88c063f4f21 : SUCCESS in 56m 32s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/46778aba84644b218451ea719c2018de : SUCCESS in 1h 03m 35s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/93403cca2fd648819f6c51753a65437c : SUCCESS in 32m 48s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/209baae3d75e41a58b44ea653ab72684 : FAILURE in 1h 11m 54s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d1738da1acc16c0346c4ce26fc37fc4fa3d9ee1b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-02-17 17:06:31.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/973489/8/upstream-check-manila/manila-tempest-plugin-ontap-dhss/a288cca/ : FAILURE in 12m 20s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"7320486397b32354a302cecfe0ac0d2a7c0c9d74","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-02-19 13:11:57.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3395809a0dd80322171b3337d9d7455189446e44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 15:20:47.000000000","message":"Patch Set 8: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/manila/+/967822 is needed.","accounts_in_message":[],"_revision_number":8},{"id":"8d9497ce3806d6dcac1e234d4dbf4b764efcc4bd","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-02-19 15:44:53.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a3e1ddfa2d16f86e6fb9139311207390b1098320","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 17:39:29.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e15a2a2aaf4544589c83b6dded4e747a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/862e96afd6f04598912f148d7f51693b : SUCCESS in 3m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/69bbde23a31a46e3bdef994bb1ae33ec : SUCCESS in 53m 27s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9c48b25b2b7046249f5a225b4a2e2b25 : FAILURE in 1h 50m 35s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/aae9301e7ba24e658e54be3531a306cf : FAILURE in 1h 27m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/acdfec282a824aa0a44f0936f6a7961c : FAILURE in 29m 12s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/1b454dbc2f284337999913f2fc2d9a16 : SUCCESS in 1h 37m 03s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f7ba7356d3f145d1a4cf17e21ff02e9a : FAILURE in 1h 04m 20s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a7ea9f40505c4880bb968828e12d5f45 : SUCCESS in 24m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/33a508ad387643fd9e219a211969eb7f : SUCCESS in 30m 13s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/37df838285f144079a987a04d1c973dd : SUCCESS in 1h 10m 18s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/19b5246a73274ab5847eef10d367d98d : SUCCESS in 32m 40s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/c95cea6ddf874ffe8c86dec8b0876a4e : SUCCESS in 1h 06m 44s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/bfeafd99f34c4b2ba847ae480625f419 : SUCCESS in 31m 20s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f99d106f065c4dcf80305e444f467424 : FAILURE in 28m 39s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c51dbaba80cd83b98ed38224435ea1f652c68923","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-02-19 17:53:13.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/973489/8/upstream-check-manila/manila-tempest-plugin-ontap-dhss/ce9a0c2/ : FAILURE in 11m 24s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"5b7598745d6c3f9d0dab7ad0bb023acff4645e40","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-02-20 11:28:44.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c2583a81e222a4197c8e52bd9081a9e94c76f2ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 13:48:21.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5b1f2b495e44a51a915bcd8e028a5fa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d46d85a6c4346e2828637e9705f56e5 : SUCCESS in 4m 19s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/584f6ac8b8354eb0a494697149f4f3dd : SUCCESS in 1h 12m 54s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/93c6a9bc1ba54a7792c0ff5ec155e85d : SUCCESS in 54m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/559e6e5be231414c82d7d1192fea967b : FAILURE in 1h 17m 43s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9dfe2907ce4a42a8807ebdf0e7e82fcf : FAILURE in 43m 00s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/0fda838946d740f0a4794fa4c4093e43 : SUCCESS in 1h 34m 58s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/241269c43ffd4e5a9b06a57c0edba5f3 : FAILURE in 2h 17m 23s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/328da6dd2a1141bbb0ee1d99679e9eb8 : SUCCESS in 22m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/da0d427ce68a44898b98d17221ee9810 : SUCCESS in 34m 35s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/b661885d2efe4b429b3e67f3c5d9abe8 : SUCCESS in 1h 12m 54s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/b73fa78c200c4be9bb84e5aabc1f1f46 : SUCCESS in 1h 01m 55s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/ab8913c535f24a108f6efb2d3c5811fb : SUCCESS in 55m 35s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/4b8c44cf0ba14ea09b2a69d2629563cb : SUCCESS in 24m 49s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e56715a6b67241f3b069763d436a54d9 : FAILURE in 59m 48s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"32179d6e307ab84209137157484131a1b57ed49b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-02-20 14:01:49.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/973489/8/upstream-check-manila/manila-tempest-plugin-ontap-dhss/8c1e918/ : FAILURE in 12m 27s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"938424c38b5a19cbcc7565dccfcfd4ca6cb3d05b","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2026-02-23 04:56:06.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"99171c9504ee29e804fbecc288c7be2665684304","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-02-23 04:56:24.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"6a1d50e0ec79f2be71d39ab7eae1dc2d8f78aaef","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2026-02-23 05:03:56.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a1c6c773e82dcda43db1faf52cf5efcba9eb151c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-02-23 05:08:51.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"470913fed36a92d364a11922eb234d06f2bb1809","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-02-23 08:55:52.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"2534dfd72fcbb890d03e8ce3f911de10c38c1a39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 10:17:38.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b2c8abe3a694e31a8302c2ebdbe6a69\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9acea8acf4944da9adfa8dcdc29a59d : FAILURE in 2m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/aa0df0d7506e4bdb8ac5ad526da70352 : SUCCESS in 1h 02m 30s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/deba91aa2110409d8b22bba7411793f2 : SUCCESS in 53m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/cb3af24ea36641019fcc82b48dcb9481 : FAILURE in 1h 00m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/faa03d5ff4e643dc864558d260789c87 : FAILURE in 1h 10m 45s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/445a34e0cc3e4d78b3d2ce9892cb89a4 : FAILURE in 1h 15m 39s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a777d51bba334c04b0b937172678add9 : FAILURE in 40m 28s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ad5a4c36c43c4f168cd6f76ea17afc8c : FAILURE in 33m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/499439010e364cbcbb8edc34956fad5a : FAILURE in 32m 30s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/83a2bfb03482490bb12a4a19c61d09da : SUCCESS in 1h 01m 12s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/9370c0b268cd44ecb09bdbe5f76e220c : SUCCESS in 34m 00s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/b1b93f082eb149b2910067f7f6e46cb1 : SUCCESS in 33m 55s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/5cd4a254966149bfbcab9cc8b2d8b26c : SUCCESS in 21m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c8794ef858084ddba392aaccef8eb8dd : FAILURE in 56m 54s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4cddf2048aab73333803027d3fa6609b4ec71778","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-02-23 10:36:52.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"5d54483920333b9dc3d18fb965d146b65cd21a61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 13:21:40.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/47765ade691c4210b935a462ab1f4b8d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2409db244c254dcfb8bbef24f7eca37e : FAILURE in 1m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f028adfc46c94987a72929c38677466b : SUCCESS in 1h 09m 30s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/0c4c71d235204c2c8f56a409ce581660 : FAILURE in 2h 07m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ddbe2bdc1dc24a209bea1e14ac4fcbf6 : FAILURE in 56m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d75b4cf9946e4190ac1aec4126cd5be2 : FAILURE in 33m 18s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/d6272ff916fc4ffe80e3ceba0008c41b : SUCCESS in 39m 36s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2535b002f6904f1e90ec13cd6777f527 : FAILURE in 1h 00m 32s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e169908835e54488b7bacd6408db50c0 : SUCCESS in 24m 11s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6983037f982f45a1b5a23578c0a1c239 : SUCCESS in 28m 23s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/0ca377afd74641d8a3e3c601c2c76a65 : SUCCESS in 36m 52s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/6a806a17e6f34562ad052a328d49d0e5 : SUCCESS in 2h 43m 08s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/8faaf964809c4de3a6aa3bfab96cd9a7 : SUCCESS in 48m 58s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/029f6d5565744c2e88dad5a02cd261aa : SUCCESS in 21m 05s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/588dbe35b3dc44bbb50956d781a1b198 : FAILURE in 1h 04m 30s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"be9333a22b5335d6bb360e0d264cb100a1fcf3ad","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-02-26 18:32:26.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"530c8953db481a8b21aed2c754278dfc578ae647","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 19:56:09.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/284cf192ed104f2499706dc80592835e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cefecd3abad4f249446944f273e1ed0 : SUCCESS in 4m 37s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/61e600972fbd421aa9d8e2848ca7a709 : SUCCESS in 37m 10s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7cf22fca6cbf4d4f96d2007c04db80ca : FAILURE in 1h 05m 58s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/29b7f9a4e8664745aa398179e6bd1861 : FAILURE in 34m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f6e3cd02e5614544bbc4e8e2819fee47 : FAILURE in 42m 55s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/b599b5c6b21548ea9207bd914f18e468 : FAILURE in 44m 33s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/82032200c76b4bdba28284fc8929d705 : FAILURE in 59m 43s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7d58a6ddb6c84d16b3a8fda35c0623e7 : FAILURE in 20m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1abf265b15994caca9f200e0442f430a : FAILURE in 37m 37s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/6a845c440a2645e1a18afd1762c75fc8 : SUCCESS in 1h 18m 12s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/7fff0d755d0a42cb9904dd05659cd09a : SUCCESS in 49m 46s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/6e50d00b0ccc48ebb476a22ca7137d52 : SUCCESS in 1h 07m 25s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/efbf8e4466d848c1b10f0a199957e727 : SUCCESS in 21m 21s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/15ffab8944304c729e378e1005cb2e2e : FAILURE in 1h 02m 09s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c74a41f509a628a188d4692ada95b86099827516","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-03-02 08:13:39.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7acf1cf82c955fc77a948f1637614aebc5c01dd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 10:56:33.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8000b937a204ed095c9a4c7e24429ee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/115aaa3ee3474d8896d0a321869f785e : SUCCESS in 3m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a0d30bb22f9b4ff3a01533445bb5d7eb : SUCCESS in 29m 19s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c89083d27c354a5685caaec976aadd4b : FAILURE in 2h 36m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b0cc3afe6dda4bf68724b300647e50ec : FAILURE in 53m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c15318e7f6cf486b975d42e565dbdc67 : FAILURE in 36m 49s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e878c47d241d404b8106498122505979 : FAILURE in 1h 18m 30s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ff626658c129465993d5bb1926f71344 : FAILURE in 1h 12m 15s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/04e1dcf26c934ce3a97fe30680910200 : FAILURE in 36m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/10afc4102db34ad0b4cc624a7ae5ce85 : FAILURE in 50m 32s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/4d4a87020fc748329a9b0fc638a388b5 : SUCCESS in 1h 08m 42s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/3025255ee9ab47ee94e6d8a443df6828 : SUCCESS in 29m 58s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/d583ac3cf3094d0c8c44cd75939e869d : SUCCESS in 32m 27s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/b1d900df2b2442c882b4812ad92c2684 : SUCCESS in 33m 37s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/45bc36e9e1a548839bfdbf0d31a931c6 : FAILURE in 1h 01m 46s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"735ab9ed7f3a5241ebe0fd7e2f9921a84d829d19","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-03-10 14:32:13.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"05677e515dc2e1b3e5ec5d3ce4cf9837bfe43eab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 16:23:25.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/03aa9971fbb442ebacd528de7a259b6d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78358f6ac19342ce92885ec82a798a68 : SUCCESS in 2m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a7d8a766040c42b396451ee41e90c16e : SUCCESS in 1h 08m 58s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7f8b6dc9ecba4caeb363f542a2bb36ca : SUCCESS in 1h 46m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/02d36da8498a4c37a73bfbc53eaec66d : FAILURE in 1h 05m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/91620cc76f4e49d5ab54d9e4d7f853b2 : FAILURE in 37m 39s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/5643000b9770423e9802d93197150cf9 : FAILURE in 43m 35s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/83c5d4c58fbb48a78286c96c1f027a2e : SUCCESS in 35m 25s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/43a28e403d9641f3b94ee9844c9dd3eb : FAILURE in 35m 39s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/988df96e030249df9266f9345fd5a1eb : FAILURE in 46m 38s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/6dada5898def4bcaa86cf307bf74d006 : SUCCESS in 1h 18m 21s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/58c7f5818349494382040a4b57773988 : SUCCESS in 37m 10s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/4e0c2cb667da4d94a063192c906f0726 : SUCCESS in 58m 01s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/c142188eaace4d4587e24a1b8e99cc6a : SUCCESS in 33m 49s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1b8d23cb493b4709b63e20fb26ca54d7 : FAILURE in 1h 11m 10s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d43f3675930b3bc86fa1a9a2411074aec78822be","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-03-12 15:35:04.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"16e5fd3a7a3e4235c68f564e9f0f03eb2d73fd06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 17:52:56.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa47fa9c3e734998a5a99e149ecc8b55\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6450f476b67429fb31b10eef797965d : SUCCESS in 3m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9304a436f5154212b13db2939d6e129e : SUCCESS in 1h 06m 41s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d009a9cc33334ff2ab8996baf0602854 : SUCCESS in 2h 09m 30s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2969200f1093430fb0618287be520f7d : FAILURE in 1h 09m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/80514b0066e3450da766efc0f3c900d0 : FAILURE in 55m 01s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ca1553db54354e3c91766da2950d2aac : FAILURE in 15m 47s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5f5874fe10024ae28b0234e3ce90e123 : FAILURE in 1h 39m 54s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e4964e8a203b45ee95375eb0cfb058fe : FAILURE in 38m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c17c63adc1f84b439caff4c48d6ee799 : FAILURE in 26m 51s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/14984e97fad5497b83521b13ca0c0965 : SUCCESS in 1h 16m 04s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/09a076325bf54f318c868596a4de48fb : SUCCESS in 1h 17m 31s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/d02b7ae93c4c4962bd992335dd038083 : SUCCESS in 1h 00m 24s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/1e262c498d2d454194dc7cc8d08d2b20 : SUCCESS in 33m 03s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b0b5b24869054d489472599791163837 : FAILURE in 30m 01s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"7139b47528e3cfc47df7c760a9d1c9c40f19f348","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-03-16 13:00:18.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"0a5646af07a89e44363690ee67e1922be2b10c53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 15:01:54.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3a806a7036f407ab5099e59e73a32ad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44db040386544b8daa2cd3c6ce00fb5e : SUCCESS in 3m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ec1609b6a41f44c683c2ca30b618af6f : SUCCESS in 1h 32m 04s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/934aed2e5f354c4ea0dd77dd5f69fa26 : FAILURE in 1h 18m 42s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e1034c776fc148e3897c8608fb830dbd : FAILURE in 57m 42s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/61454451a86a4ef48ba160914460c4f5 : FAILURE in 1h 03m 50s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/491917803ada4c7395cc669f6a88f1ae : FAILURE in 1h 34m 22s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b308dcbbec464af682eacb17ae6ed0f1 : FAILURE in 1h 28m 20s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/40baf08b59894baf8c748ec07519e2b4 : FAILURE in 29m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c99cf388fe304082b5fc2a0543b772ac : FAILURE in 43m 47s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/722f79f04a5842eeae17954f6f0acf3b : FAILURE in 1h 32m 26s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/f91399c021e34726a4e771604901467f : SUCCESS in 1h 17m 39s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/fa20d48c17ef49ec88a7f68643f81d77 : SUCCESS in 1h 18m 26s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/16fb65b922914ca0943e00c6726aefdf : SUCCESS in 32m 34s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/05169c2f7bf242d197a08ce962b1d31f : FAILURE in 1h 08m 37s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"abce72fc58b31f7316c871133b1a317ff9bc452a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-03-16 15:33:37.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"3cd87ba377b2833fe9c1797283fd6783695892ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 18:34:41.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c67974c86bcb41adb9903442a9d81524\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4998b8178bcf4219b97d6aef204dfa23 : SUCCESS in 3m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3733142fce884888baa21a96d6853d14 : SUCCESS in 2h 01m 21s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5d822406f9474e6f93dc73be8ee9f1dd : SUCCESS in 2h 50m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/91771102fb9b4912bc91d6e446b66289 : FAILURE in 1h 48m 45s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c5452179a64042b3ae720325522a80e0 : SUCCESS in 1h 06m 41s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ed7eeb609ed348d3b18b048e57219457 : FAILURE in 45m 28s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9f2a258a04f94bb49aaf96af60bcbbd5 : FAILURE in 2h 10m 07s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3a15f61848064e2d9784cebf057c9c51 : SUCCESS in 36m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/00c85e39f467415bb933218dd1475a58 : SUCCESS in 34m 21s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/5232e5582689483f8d6ab6f3eebe0ad1 : FAILURE in 41m 54s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/ea8bb3d4c8f544bb950831236d652dd4 : SUCCESS in 1h 31m 43s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/8b648f444240425d968376c9d932852f : SUCCESS in 1h 44m 09s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/4fdad2d3ce74488889356f5d8a6d513f : SUCCESS in 36m 04s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0c1ad0fc01cf4828a6576afe288284ec : FAILURE in 1h 02m 51s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"a2fad88686ca1bd799beb88d70363353c480ca23","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-03-17 03:46:29.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"fdaddb3afe5b2dcb1d6db16053372b52d628fc0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 05:07:52.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/126c3eeee9564fe586bffb5383a59b4f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faf3b308303241a6a20fcb08bdd1639b : SUCCESS in 1m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7b9680999506469e8277010c557f6023 : SUCCESS in 36m 55s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/fe6e96504f2140cd9025b98af2509398 : SUCCESS in 52m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1982a72c93344486a0a0c1deeb89f322 : FAILURE in 20m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a4663bf479c24f1cafffb53dc18422d5 : FAILURE in 12m 04s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/11e2f08882ad47f5958ac6172292c438 : SUCCESS in 55m 26s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f6ecb9e11c244468aeb7a6e83b9456fc : FAILURE in 1h 16m 08s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ce50cf07c22c4abbb5e468a185acb270 : SUCCESS in 23m 53s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/07552166fb2d46ffb27faf6d4962f65b : SUCCESS in 28m 42s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/1d6dcc6940b04c06a727040b9cca42a7 : SUCCESS in 1h 10m 56s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/8b2b1fec1dd145f7a7da31ec58dfc7f0 : SUCCESS in 1h 07m 43s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/583871fdbad6488a8817b6889ef18545 : SUCCESS in 32m 33s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/5fbb77c890f042adb7d9bf2ef0678f76 : SUCCESS in 18m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d42511cde2ed49bd84c6f47cf3b0f2f8 : FAILURE in 47m 03s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"8aca42cdb5d74de27a4d655575f10d8e77a90089","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-17 05:34:51.000000000","message":"Patch Set 12:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"b8050726c6b451519a440fccaa39a720053339bb","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-03-31 15:58:05.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"dae49d0ba53386f3910c9ad6805bf5f1bcf2a851","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 17:42:20.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9ec4e8947f94ce8b5878b3472fb64c6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5ab973908f34b2d9f62f258511956b2 : SUCCESS in 2m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e6d7478a54654afd9f7ac055c8c8a6ba : SUCCESS in 1h 13m 31s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9d1fdc0c5de847ff906dcf41bd9a9edf : SUCCESS in 1h 42m 30s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/050eb23ca76e41a8a5471c144d716bed : FAILURE in 59m 33s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c1361668f3e14ff88a8696fec6c8e312 : FAILURE in 1h 04m 13s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/edc498ca4baa40459e5f538e3385fd48 : FAILURE in 53m 13s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/98fff89e4c584d41bbf66dd53ee09e5d : SUCCESS in 1h 24m 55s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/22746d9fb0d54b4590ade84220a4bd65 : FAILURE in 35m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ec70a1fb59044c369f1ed8e8582ca805 : FAILURE in 26m 18s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/084bb2135fe04474ae70a56e2ec4443b : SUCCESS in 31m 57s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/eeeb5d1972b74c0e826ebb09db2f5c4c : SUCCESS in 1h 08m 14s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/518469aa79404bbcb19a4feb6206e181 : SUCCESS in 1h 15m 52s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/a7c63b211d2d4f8b841f46cde5aa41e1 : FAILURE in 1h 27m 18s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/c02a9612b6ed4fe6948725c445164826 : SUCCESS in 32m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4b52a3302ae74fa0824377ae80729187 : FAILURE in 1h 02m 49s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"70e5d6e608d2772623b4d24ada00bc397c71a346","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-04-02 11:56:08.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"e43d641f5ccf562fe654cb7b40b2ed084f5a3757","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 14:07:58.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3d0d5ceb24b45d8ab598572d8c86d40\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/652fe7e0e47348ec96c41a4cf15abb90 : SUCCESS in 2m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/172f01b0db9b4e588abe8c7fd8e7de93 : SUCCESS in 1h 11m 37s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/253b5beb14b24044966824706e99ff2e : FAILURE in 2h 01m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4b7104565ac24f0686f0f5c1a350ecfc : FAILURE in 59m 38s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/49981f318257452aa88715dcfa61c2fc : SUCCESS in 1h 18m 39s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e97977b941bd404da15c0d4f3bd3dcb3 : FAILURE in 22m 30s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7ac8f9c7fcdb45279d803c9b3656dbd7 : FAILURE in 1h 14m 37s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5400080144af4ee7ae996204320017f3 : SUCCESS in 28m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3320b4aef11b45e79fdd0d9dd959567d : SUCCESS in 40m 31s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/c497322d1f5c4690b22b6d377fa2c23d : SUCCESS in 1h 09m 57s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/c2a17be10a894f2aa977a302c064ce82 : SUCCESS in 1h 09m 30s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/819abd928216469290f4f0d50337800a : SUCCESS in 1h 11m 42s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/2a5f8f409b484c7b8c64e0228a898cbb : SUCCESS in 1h 20m 56s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/0f9da49e0b9c44d18fa11671db1c53ec : SUCCESS in 32m 19s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/81d9a08be7384090a47d597750392d7c : FAILURE in 1h 04m 46s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"8dd2986b9f8d7221947a574a49847636ff642ccd","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-04-02 14:18:13.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/973489/13/upstream-check-manila/manila-tempest-plugin-ontap-dhss/545499c/ : RETRY_LIMIT in 7m 22s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/973489/13/upstream-check-manila/manila-tempest-plugin-ontap-no-dhss/61962cd/ : RETRY_LIMIT in 7m 30s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"0723ec20e5f24701499a78badbaaf793856d1f0d","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-06-11 15:23:03.000000000","message":"Patch Set 13: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"16744cacbc4e17f4e113733809d4a846029cd83e","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-06-11 15:27:57.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"b43f10c344367061f47b7128bb712d69caf4cb34","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2026-06-12 18:27:06.000000000","message":"Patch Set 13: Code-Review-1\n\n(16 comments)","accounts_in_message":[],"_revision_number":13},{"id":"12469df0935799ab2a9e332cd4168d24d5afc7f0","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":35507,"name":"Kiran Pawar","email":"kiranpawarpict2010@gmail.com","username":"kpawar"},"date":"2026-06-13 05:38:07.000000000","message":"Removed cc \u003cGERRIT_ACCOUNT_35507\u003e.","accounts_in_message":[{"_account_id":35507,"name":"Kiran Pawar","email":"kiranpawarpict2010@gmail.com","username":"kpawar"}],"_revision_number":13},{"id":"5c0e557af2669cf52ba0b7edf520c0aa51169559","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-06-16 08:50:40.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"b84376e8d9bb770ddb0c47bcd0dcdc16f2d46a67","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-06-16 08:50:43.000000000","message":"Patch Set 13:\n\n(16 comments)","accounts_in_message":[],"_revision_number":13},{"id":"07f4de7da83b3186ed8002d62a4c6c411a00e1d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 11:29:24.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/699e9639158e40eca342fb47ef8018f2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/281b482019ea475e9219130b9b9c0755 : SUCCESS in 3m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/99454be11de6457181d2c52115f89eab : SUCCESS in 1h 20m 02s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f49e242095fb478cacb0a4808c79fa48 : FAILURE in 2h 24m 00s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a7d04460cd0949bb977b5359c46beccc : SUCCESS in 57m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/2a523cef9ebb4a9b86ba9985d9d3aef2 : SUCCESS in 1h 06m 50s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/1b4bb7debc3944f9bb1277eaf9ab5750 : SUCCESS in 1h 31m 21s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8798f7584c024e799b953d9d4900a393 : FAILURE in 1h 28m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2ea00df1302c49e594c45c03a021e945 : SUCCESS in 26m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2243b2c5cf724c57ab942e4dd98bafe8 : SUCCESS in 35m 56s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/8233edd31438430e9d71e695bc1f1a96 : SUCCESS in 34m 25s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/58bd48fe80ea46c29a2710c96734f389 : SUCCESS in 1h 20m 13s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/a684baf591484dfba7f831d9238d7d7d : SUCCESS in 1h 19m 45s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/1d8cf2f61cdb43ef95fc79f1dbf4588b : SUCCESS in 24m 45s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/78ae4af8447f401298d68d7c0888b799 : FAILURE in 12m 55s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"c1fc6706395857948a2e9c46701c292107faf20e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-06-16 12:55:02.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/973489/14/upstream-check-manila/manila-tempest-plugin-ontap-dhss/9babe52/ : SUCCESS in 1h 23m 01s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":14},{"id":"ecdf8604fc75d755e87211737b1697380ad63c9b","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2026-06-22 10:58:53.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"226222617ac4f352e6648ed47102709919708546","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-06-25 14:49:44.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"740bbcedd1b9bf48dd93b34faa15640b861772f6","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-07-14 14:06:56.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"0a5452945b7f7c3b0fb4aa8754b05dcd1a5fa29f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-08-11 11:29:05.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"f2e8972b953f2ce4bb385399411f98d5784be4c3","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-08-11 11:29:59.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"9d209b2be1296e66eaeb90a77c9746a904ca5508","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2026-08-11 12:11:03.000000000","message":"Patch Set 15: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"58a28c6a21aa086d4b5e9eb761a568029bb8a274","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 12:46:17.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cd47cc1d7144a24b901d33337b5282a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26c99448d2ef4e1587e06343e74aa957 : SUCCESS in 2m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/93782c636da8458686ced5a47b82c51a : SUCCESS in 39m 27s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/fbdb3f97841648baa0cbd0ba20eb43fe : SUCCESS in 55m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/42460e9a7314413abfa9bd620b5b9c05 : SUCCESS in 32m 30s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/befdd314145042a1addc57369640c75b : SUCCESS in 37m 11s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c5416959e953498e8c6cabfbe8fb5b8a : FAILURE in 1h 15m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0ff9f946f8a148a2aa70f1c04331f535 : SUCCESS in 40m 25s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/39959e184297441db9d768ea96d9ecbe : SUCCESS in 25m 39s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/15e00f320bce4a2993ae83fe6e214b4b : SUCCESS in 27m 54s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/7a083bbc18244beaaaaaff5314cc4efe : SUCCESS in 1h 09m 32s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/c141ad2698034d8b8848832551b7bcf8 : SUCCESS in 36m 51s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/3efb94554ed84f4fb72a436be3c0485a : SUCCESS in 36m 58s\n- manila-tempest-plugin-dummy-no-dhss-old-rbac https://zuul.opendev.org/t/openstack/build/4cd3c78ff98b428a98441eacb8ca247d : SUCCESS in 30m 13s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/047bf6dd52be4952850e3d0d28ceda15 : FAILURE in 26m 41s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"3de1857f36f9fcdda77baf1944413db13d9bb0b2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-24 05:21:09.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1\n\n(10 comments)","accounts_in_message":[],"_revision_number":15},{"id":"7091c0a87c41e4d64bf14b0cb8b62b21b2dad8cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 05:21:32.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"8d98c1c9f8f0d3aae303e29042026403e7c6692b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 06:11:57.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/783ab89fa5ba49ef87497c6544e6e113\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddf2dcf69dd6445aa9ee403e21210ac6 : SUCCESS in 3m 08s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d1ceabd5b1314024a6505c1c034520c5 : SUCCESS in 49m 19s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3b0bde0ae2dc4d798bdc016295b297ff : SUCCESS in 37m 39s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7963b5fa764b4bab9af55ebf2813f4ca : SUCCESS in 29m 55s","accounts_in_message":[],"_revision_number":15},{"id":"d335d6e7f866de41388ff6499a4f83629f0ee62f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 06:11:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"4508542cd633c5dc00bbbacf986be7b1b3b4bdbc","revisions":{"a3ee1caed5cc73feb82ed39b69a30b33227116d8":{"kind":"REWORK","_number":1,"created":"2026-01-15 14:30:06.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/89/973489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/973489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/1"}}},"commit":{"parents":[{"commit":"3f76739525905fc34792d9c2490f3546c3af75b2","subject":"Merge \"Refactor stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3f76739525905fc34792d9c2490f3546c3af75b2"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-01-15 14:17:52.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-01-15 14:22:42.000000000","tz":0},"subject":"Add tests for Qos Type API.","message":"Add tests for Qos Type API.\n\nPositive and negative tests for qos type and qos type specs are added\nfor API.\n\npartially-implements: blueprint qos-types\nDepends-On: https://review.opendev.org/c/openstack/manila/+/967822\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\nChange-Id: I3775a6bd164b9ddeb19a8b8369b84cf22f4ffd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a3ee1caed5cc73feb82ed39b69a30b33227116d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a3ee1caed5cc73feb82ed39b69a30b33227116d8"}]},"branch":"refs/heads/master"},"553a66598cace2a1226bbde556a60c5dc1e419ed":{"kind":"REWORK","_number":2,"created":"2026-01-17 12:22:12.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/89/973489/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/973489/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/2"}}},"commit":{"parents":[{"commit":"3f76739525905fc34792d9c2490f3546c3af75b2","subject":"Merge \"Refactor stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3f76739525905fc34792d9c2490f3546c3af75b2"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-01-15 14:17:52.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-01-17 12:22:06.000000000","tz":0},"subject":"Add tests for Qos Type API.","message":"Add tests for Qos Type API.\n\nPositive and negative tests for qos type and qos type specs are added\nfor API.\n\npartially-implements: blueprint qos-types\nDepends-On: https://review.opendev.org/c/openstack/manila/+/967822\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\nChange-Id: I3775a6bd164b9ddeb19a8b8369b84cf22f4ffd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/553a66598cace2a1226bbde556a60c5dc1e419ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/553a66598cace2a1226bbde556a60c5dc1e419ed"}]},"branch":"refs/heads/master"},"73d557d646083bc3b517bf75ea88ad73a8bc4dac":{"kind":"REWORK","_number":3,"created":"2026-01-19 08:59:08.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/89/973489/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/973489/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/3"}}},"commit":{"parents":[{"commit":"3f76739525905fc34792d9c2490f3546c3af75b2","subject":"Merge \"Refactor stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3f76739525905fc34792d9c2490f3546c3af75b2"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-01-15 14:17:52.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-01-19 08:59:01.000000000","tz":0},"subject":"Add tests for Qos Type API.","message":"Add tests for Qos Type API.\n\nPositive and negative tests for qos type and qos type specs are added\nfor API.\n\npartially-implements: blueprint qos-types\nDepends-On: https://review.opendev.org/c/openstack/manila/+/967822\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\nChange-Id: I3775a6bd164b9ddeb19a8b8369b84cf22f4ffd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/73d557d646083bc3b517bf75ea88ad73a8bc4dac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/73d557d646083bc3b517bf75ea88ad73a8bc4dac"}]},"branch":"refs/heads/master"},"108fdeb7ea906b69ed7cd2ba18e14617145992ba":{"kind":"REWORK","_number":4,"created":"2026-01-19 14:45:50.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/89/973489/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/973489/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/4"}}},"commit":{"parents":[{"commit":"3f76739525905fc34792d9c2490f3546c3af75b2","subject":"Merge \"Refactor stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3f76739525905fc34792d9c2490f3546c3af75b2"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-01-15 14:17:52.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-01-19 14:45:36.000000000","tz":0},"subject":"Add tests for Qos Type API.","message":"Add tests for Qos Type API.\n\nPositive and negative tests for qos type and qos type specs are added\nfor API.\n\npartially-implements: blueprint qos-types\nDepends-On: https://review.opendev.org/c/openstack/manila/+/967822\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\nChange-Id: I3775a6bd164b9ddeb19a8b8369b84cf22f4ffd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/108fdeb7ea906b69ed7cd2ba18e14617145992ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/108fdeb7ea906b69ed7cd2ba18e14617145992ba"}]},"branch":"refs/heads/master"},"076103afcc36d08b406291ed495effa66b74297e":{"kind":"REWORK","_number":5,"created":"2026-01-19 16:08:35.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/89/973489/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/973489/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/5"}}},"commit":{"parents":[{"commit":"3f76739525905fc34792d9c2490f3546c3af75b2","subject":"Merge \"Refactor stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3f76739525905fc34792d9c2490f3546c3af75b2"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-01-15 14:17:52.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-01-19 16:08:27.000000000","tz":0},"subject":"Add tests for Qos Type API.","message":"Add tests for Qos Type API.\n\nPositive and negative tests for qos type and qos type specs are added\nfor API.\n\npartially-implements: blueprint qos-types\nDepends-On: https://review.opendev.org/c/openstack/manila/+/967822\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\nChange-Id: I3775a6bd164b9ddeb19a8b8369b84cf22f4ffd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/076103afcc36d08b406291ed495effa66b74297e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/076103afcc36d08b406291ed495effa66b74297e"}]},"branch":"refs/heads/master"},"bf4f1169f72df6b4a7f4deb0aae7272444ab3395":{"kind":"REWORK","_number":6,"created":"2026-01-19 16:47:04.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/89/973489/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/973489/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/6"}}},"commit":{"parents":[{"commit":"3f76739525905fc34792d9c2490f3546c3af75b2","subject":"Merge \"Refactor stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3f76739525905fc34792d9c2490f3546c3af75b2"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-01-15 14:17:52.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-01-19 16:46:18.000000000","tz":0},"subject":"Add tests for Qos Type API.","message":"Add tests for Qos Type API.\n\nPositive and negative tests for qos type and qos type specs are added\nfor API.\n\npartially-implements: blueprint qos-types\nDepends-On: https://review.opendev.org/c/openstack/manila/+/967822\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\nChange-Id: I3775a6bd164b9ddeb19a8b8369b84cf22f4ffd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bf4f1169f72df6b4a7f4deb0aae7272444ab3395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bf4f1169f72df6b4a7f4deb0aae7272444ab3395"}]},"branch":"refs/heads/master"},"9f95dd8de268ea34648a326d89e26fed2f866f2f":{"kind":"REWORK","_number":7,"created":"2026-02-05 12:07:32.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/89/973489/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/973489/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/7"}}},"commit":{"parents":[{"commit":"3f76739525905fc34792d9c2490f3546c3af75b2","subject":"Merge \"Refactor stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3f76739525905fc34792d9c2490f3546c3af75b2"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-01-15 14:17:52.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-02-05 12:06:30.000000000","tz":0},"subject":"Add tests for Qos Type API.","message":"Add tests for Qos Type API.\n\nPositive and negative tests for qos type and qos type specs are added\nfor API.\n\npartially-implements: blueprint qos-types\nDepends-On: https://review.opendev.org/c/openstack/manila/+/967822\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\nChange-Id: I3775a6bd164b9ddeb19a8b8369b84cf22f4ffd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9f95dd8de268ea34648a326d89e26fed2f866f2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9f95dd8de268ea34648a326d89e26fed2f866f2f"}]},"branch":"refs/heads/master"},"7dec90115dc95ae38e898d26bd3acb1998972268":{"kind":"REWORK","_number":8,"created":"2026-02-17 14:07:01.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/89/973489/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/973489/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/8"}}},"commit":{"parents":[{"commit":"6406026b51e9c282902dbc3987f1edb9879821b8","subject":"Remove Manila V1 API client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6406026b51e9c282902dbc3987f1edb9879821b8"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-01-15 14:17:52.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-02-17 12:25:26.000000000","tz":0},"subject":"Add tests for Qos Type API.","message":"Add tests for Qos Type API.\n\nPositive and negative tests for qos type and qos type specs are added\nfor API.\n\npartially-implements: blueprint qos-types\nDepends-On: https://review.opendev.org/c/openstack/manila/+/967822\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\nChange-Id: I3775a6bd164b9ddeb19a8b8369b84cf22f4ffd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7dec90115dc95ae38e898d26bd3acb1998972268"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7dec90115dc95ae38e898d26bd3acb1998972268"}]},"branch":"refs/heads/master"},"4fc3ba89a76609be925a2bdfb040c01793bc9221":{"kind":"REWORK","_number":9,"created":"2026-02-23 08:55:52.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/89/973489/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/973489/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/9"}}},"commit":{"parents":[{"commit":"f99c487d8d1e78184d4c85e226c04346c7c75cf2","subject":"Use official service type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f99c487d8d1e78184d4c85e226c04346c7c75cf2"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-01-15 14:17:52.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-02-23 08:14:38.000000000","tz":0},"subject":"Add tests for Qos Type API.","message":"Add tests for Qos Type API.\n\nPositive and negative tests for qos type and qos type specs are added\nfor API (2.94).\n\npartially-implements: blueprint qos-types\nDepends-On: https://review.opendev.org/c/openstack/manila/+/967822\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\nChange-Id: I3775a6bd164b9ddeb19a8b8369b84cf22f4ffd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4fc3ba89a76609be925a2bdfb040c01793bc9221"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4fc3ba89a76609be925a2bdfb040c01793bc9221"}]},"branch":"refs/heads/master"},"b6a4483c64d0bec1ab8a0406cd2078d1bc1fe1c1":{"kind":"REWORK","_number":10,"created":"2026-02-23 10:36:52.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/89/973489/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/973489/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/10"}}},"commit":{"parents":[{"commit":"f99c487d8d1e78184d4c85e226c04346c7c75cf2","subject":"Use official service type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f99c487d8d1e78184d4c85e226c04346c7c75cf2"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-01-15 14:17:52.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-02-23 10:36:44.000000000","tz":0},"subject":"Add tests for Qos Type API.","message":"Add tests for Qos Type API.\n\nPositive and negative tests for qos type and qos type specs are added\nfor API (2.94).\n\npartially-implements: blueprint qos-types\nDepends-On: https://review.opendev.org/c/openstack/manila/+/967822\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\nChange-Id: I3775a6bd164b9ddeb19a8b8369b84cf22f4ffd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b6a4483c64d0bec1ab8a0406cd2078d1bc1fe1c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b6a4483c64d0bec1ab8a0406cd2078d1bc1fe1c1"}]},"branch":"refs/heads/master"},"011d982c7e6dd18810c1c94dcba736aac4a92a0f":{"kind":"REWORK","_number":11,"created":"2026-03-16 13:00:18.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/89/973489/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/973489/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/11"}}},"commit":{"parents":[{"commit":"327b5a3abd882f28235c9120faef096e41688c55","subject":"Merge \"Test share network privileged fields differently\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/327b5a3abd882f28235c9120faef096e41688c55"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-01-15 14:17:52.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-03-16 12:59:14.000000000","tz":0},"subject":"Add tests for Qos Type API.","message":"Add tests for Qos Type API.\n\nPositive and negative tests for qos type and qos type specs are added\nfor API (2.94).\n\npartially-implements: blueprint qos-types\nDepends-On: https://review.opendev.org/c/openstack/manila/+/967822\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\nChange-Id: I3775a6bd164b9ddeb19a8b8369b84cf22f4ffd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/011d982c7e6dd18810c1c94dcba736aac4a92a0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/011d982c7e6dd18810c1c94dcba736aac4a92a0f"}]},"branch":"refs/heads/master"},"2f3cd5dcc125977bc24bbac8bcb6c73fbc89774a":{"kind":"REWORK","_number":12,"created":"2026-03-16 15:33:37.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/89/973489/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/973489/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/12"}}},"commit":{"parents":[{"commit":"327b5a3abd882f28235c9120faef096e41688c55","subject":"Merge \"Test share network privileged fields differently\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/327b5a3abd882f28235c9120faef096e41688c55"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-01-15 14:17:52.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-03-16 15:33:28.000000000","tz":0},"subject":"Add tests for Qos Type API.","message":"Add tests for Qos Type API.\n\nPositive and negative tests for qos type and qos type specs are added\nfor API (2.94).\n\npartially-implements: blueprint qos-types\nDepends-On: https://review.opendev.org/c/openstack/manila/+/967822\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\nChange-Id: I3775a6bd164b9ddeb19a8b8369b84cf22f4ffd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2f3cd5dcc125977bc24bbac8bcb6c73fbc89774a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2f3cd5dcc125977bc24bbac8bcb6c73fbc89774a"}]},"branch":"refs/heads/master"},"40db696af9e539382e7983ec6dcc1dd26019baa9":{"kind":"REWORK","_number":13,"created":"2026-04-02 11:56:08.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/89/973489/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/973489/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/13"}}},"commit":{"parents":[{"commit":"a959f20dfe1e745579b608d4edd4a5cf3655bbc4","subject":"Bump API testing version to 2.95 and fix replica tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a959f20dfe1e745579b608d4edd4a5cf3655bbc4"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-01-15 14:17:52.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-04-02 11:51:22.000000000","tz":0},"subject":"Add tests for Qos Type API.","message":"Add tests for Qos Type API.\n\nPositive and negative tests for qos type and qos type specs are added\nfor API (2.94).\n\npartially-implements: blueprint qos-types\nDepends-On: https://review.opendev.org/c/openstack/manila/+/967822\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\nChange-Id: I3775a6bd164b9ddeb19a8b8369b84cf22f4ffd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/40db696af9e539382e7983ec6dcc1dd26019baa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/40db696af9e539382e7983ec6dcc1dd26019baa9"}]},"branch":"refs/heads/master"},"12a903957307a58c9525416744e6283b9ba1842c":{"kind":"REWORK","_number":14,"created":"2026-06-16 08:50:40.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/89/973489/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/973489/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/14"}}},"commit":{"parents":[{"commit":"fc809ff749625f28a276d2f5ab7c1cb12f992569","subject":"Merge \"Replication: verify access rules in detail\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fc809ff749625f28a276d2f5ab7c1cb12f992569"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-01-15 14:17:52.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-06-16 08:49:20.000000000","tz":0},"subject":"Add tests for Qos Type API.","message":"Add tests for Qos Type API.\n\nPositive and negative tests for qos type and qos type specs are added\nfor API (2.94).\n\npartially-implements: blueprint qos-types\nDepends-On: https://review.opendev.org/c/openstack/manila/+/967822\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\nChange-Id: I3775a6bd164b9ddeb19a8b8369b84cf22f4ffd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/12a903957307a58c9525416744e6283b9ba1842c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/12a903957307a58c9525416744e6283b9ba1842c"}]},"branch":"refs/heads/master"},"4508542cd633c5dc00bbbacf986be7b1b3b4bdbc":{"kind":"REWORK","_number":15,"created":"2026-08-11 11:29:05.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/89/973489/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/89/973489/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/89/973489/15"}}},"commit":{"parents":[{"commit":"69e8ab6d76489149aa59b4f86330991a2d6943de","subject":"Merge \"Use POSIX-compatible ls in migration test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/69e8ab6d76489149aa59b4f86330991a2d6943de"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-01-15 14:17:52.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-08-11 11:28:57.000000000","tz":0},"subject":"Add tests for Qos Type API.","message":"Add tests for Qos Type API.\n\nPositive and negative tests for qos type and qos type specs are added\nfor API (2.94).\n\npartially-implements: blueprint qos-types\nDepends-On: https://review.opendev.org/c/openstack/manila/+/967822\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\nChange-Id: I3775a6bd164b9ddeb19a8b8369b84cf22f4ffd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4508542cd633c5dc00bbbacf986be7b1b3b4bdbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4508542cd633c5dc00bbbacf986be7b1b3b4bdbc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
