)]}'
{"id":"openstack%2Fmanila-tempest-plugin~975860","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ifcc0e2308de80ad20d817d07330ebd9568cbea62","project":"openstack/manila-tempest-plugin","branch":"master","topic":"remove-service-list-test","hashtags":[],"change_id":"Ifcc0e2308de80ad20d817d07330ebd9568cbea62","subject":"Remove service list by invalid zone test temporarily","status":"ABANDONED","created":"2026-02-05 20:56:37.000000000","updated":"2026-02-06 19:12:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b20847215fdad8cb7b2586fdbf079c0f36a217c9","_number":975860,"virtual_id_number":975860,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-06 00:09:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-06 00:09:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2e6eb457f24fb07df97145472793ad722e19b45a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-05 20:56:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e600d416b0aa8a6a471f145461fc04e8df3b9ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 00:09:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/414587b0e6b742d2a3d1b209691bebda\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b75abf358df44cd1a3ba55350dad2904 : SUCCESS in 2m 19s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7f2b60548b93443aa39a74e63f39fd65 : SUCCESS in 51m 00s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9fc5b5f3406b4b8582ccfe325fc00da2 : TIMED_OUT in 3h 06m 35s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/658f43e8a1ed4c15acce6db25e8594ec : FAILURE in 49m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ce28d3b64ec34518bb202aa5a7834b21 : SUCCESS in 1h 04m 15s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/7406002d2d2f41a1b32d428157b12309 : SUCCESS in 1h 25m 50s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c5d44e37fabe43daac05c64d9050f26a : FAILURE in 1h 16m 41s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3bd24d1e0731437a976bb4ba43451b2b : SUCCESS in 32m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5f73b3272eb845059360dea215e571dc : SUCCESS in 42m 25s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/9d49323dc85b46b5ba86efc2d88ba399 : SUCCESS in 1h 17m 26s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/5023feb763e84307907632cf40a12866 : SUCCESS in 1h 12m 20s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/1105e00c992649d7bb48cb9bf520e450 : SUCCESS in 1h 08m 13s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/e0358a23ecea48dcb5a7a41e78abf1dc : SUCCESS in 18m 17s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e50995f073d54871a912b21f425cef1a : FAILURE in 51m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b20847215fdad8cb7b2586fdbf079c0f36a217c9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-06 19:12:37.000000000","message":"Abandoned\n\nNot needed anymore, as https://review.opendev.org/c/openstack/manila/+/928081 is no longer changing the behavior.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c615e4d776d7a0ff3129173392924d2e226cfc63","revisions":{"c615e4d776d7a0ff3129173392924d2e226cfc63":{"kind":"REWORK","_number":1,"created":"2026-02-05 20:56:37.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/60/975860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/975860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/975860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/975860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/975860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/975860/1"}}},"commit":{"parents":[{"commit":"3f76739525905fc34792d9c2490f3546c3af75b2","subject":"Merge \"Refactor stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3f76739525905fc34792d9c2490f3546c3af75b2"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-05 20:53:08.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-05 20:56:28.000000000","tz":-180},"subject":"Remove service list by invalid zone test temporarily","message":"Remove service list by invalid zone test temporarily\n\nWe are introducing the possibility to filter shares by \u0027ensuring\u0027\nstatus in manila [1]. The new API bump change also moves the\nservice filtering to the database, making it more efficient.\n\nHowever, there is one test failing: test_get_service_by_invalid_zone\nThis test is failing because the previous implementation did not\nlook up the availability zone in the database, so if an invalid\nvalue was specified, it would be ignored and an empty list would\nbe returned anyway.\n\nThis change removes that test temporarily, so that we can merge the\ncore change and update the test to fit the new behavior, where a\nwrong availability zone ID will produce a 404.\n\nChange-Id: Ifcc0e2308de80ad20d817d07330ebd9568cbea62\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c615e4d776d7a0ff3129173392924d2e226cfc63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c615e4d776d7a0ff3129173392924d2e226cfc63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
