)]}'
{"id":"openstack%2Fmanila-tempest-plugin~975861","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ib75b93216c55f223cbfcd0f21a81852a3a2b646f","project":"openstack/manila-tempest-plugin","branch":"master","topic":"re-enable-service-list-az-filter-test","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-02-06 19:12:26.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib75b93216c55f223cbfcd0f21a81852a3a2b646f","subject":"Add test_get_service_by_invalid_zone","status":"ABANDONED","created":"2026-02-05 21:00:27.000000000","updated":"2026-02-06 19:12:26.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"5566c6b301e23ca0dcbd886ce696de472462865e","_number":975861,"virtual_id_number":975861,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":-1,"date":"2026-02-06 19:03:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-05 22:33:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-06 19:03:17.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2026-02-06 19:03:21.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"0eb66453a5106b979641e11abdbf57130e9a85b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-05 21:00:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7aa000cce54aeda92af85bc642fa79fc5c9159f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 22:33:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7cf8220ccc640fd8905231c882d8d99\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77e0929c65514464903b8f7089a7413d : SUCCESS in 2m 19s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/62adf781160c4cb1aa15e96c440960e3 : SUCCESS in 1h 08m 54s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9508e778a64945679ed39eda0eba3f6e : SUCCESS in 50m 00s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3dd12bb04a934413b001f9326d0fcfbd : FAILURE in 1h 25m 42s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b751bb9566814f908acfb1e53108117f : FAILURE in 36m 46s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/2fbb93e49c1d4d9bb332a2635464de57 : FAILURE in 45m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/eba0788d950d499c813041babb38b94e : FAILURE in 1h 18m 19s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ac859410dabe4d82a1c95485f86e64e4 : FAILURE in 36m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1539d9e6b07b401eba5f2a9415066010 : FAILURE in 38m 08s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/823582593e5d4510863897ba4645eca5 : SUCCESS in 1h 05m 21s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/03950904b25442ab813d66f08c1edba2 : SUCCESS in 1h 14m 32s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/f5800d6a79674aec9ca8f051ca9a281a : SUCCESS in 32m 54s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/2ebb6163e3064aed93695f8cda2a324e : SUCCESS in 32m 26s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/aa402acc301d474380470cf8d39a97c7 : FAILURE in 1h 06m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"83f71fe2b6824ffdb4a940548a356f159c3951c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-06 17:34:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"17f6677d85ac4c55a7a0059b894ffdedd85561c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 17:34:21.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/manila-tempest-plugin for 975861,2","accounts_in_message":[],"_revision_number":2},{"id":"9c0c83d1fa3fd6cae45c8779e5469d68349c0d68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-06 17:38:03.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a748b4d8581f44e4cf75b803627bf85dd51dd0f0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-06 19:03:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4cf4960fa83c73e424988ab6848260e68ea1f325","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-06 19:03:21.000000000","message":"Patch Set 3: Code-Review-1","accounts_in_message":[],"_revision_number":3},{"id":"5566c6b301e23ca0dcbd886ce696de472462865e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-06 19:12:26.000000000","message":"Abandoned\n\nNot needed anymore, as https://review.opendev.org/c/openstack/manila/+/928081 is no longer changing the behavior.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dd68cdb3bcdf0695b6630bd46a47a48e67697470","revisions":{"547c427f2bbda118469ea964618cdd9a6ec619f6":{"kind":"REWORK","_number":1,"created":"2026-02-05 21:00:27.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/61/975861/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/975861/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/975861/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/975861/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/975861/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/975861/1"}}},"commit":{"parents":[{"commit":"5f585fb639ac54b906dbd14cc6fe1620406b6fe3","subject":"Update for using endpoint_type to create client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5f585fb639ac54b906dbd14cc6fe1620406b6fe3"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-05 20:58:46.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-05 21:00:19.000000000","tz":-180},"subject":"Add test_get_service_by_invalid_zone","message":"Add test_get_service_by_invalid_zone\n\nThis test was removed in a previous change, as it was expecting a\nbehavior that doesn\u0027t exist in the API anymore.\n\nThis change re-introduces this test, but now expecting that the\nservice lookup will fail with a 404 in case an unexsistent\navailability zone is provided as a filter.\n\nChange-Id: Ib75b93216c55f223cbfcd0f21a81852a3a2b646f\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/547c427f2bbda118469ea964618cdd9a6ec619f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/547c427f2bbda118469ea964618cdd9a6ec619f6"}]},"branch":"refs/heads/master"},"f45d575d3aa12fa488c470a171d9cb547bb529f0":{"kind":"REWORK","_number":2,"created":"2026-02-06 17:34:03.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/61/975861/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/975861/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/975861/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/975861/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/975861/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/975861/2"}}},"commit":{"parents":[{"commit":"5f585fb639ac54b906dbd14cc6fe1620406b6fe3","subject":"Update for using endpoint_type to create client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5f585fb639ac54b906dbd14cc6fe1620406b6fe3"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-05 20:58:46.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-06 17:33:39.000000000","tz":-180},"subject":"Add test_get_service_by_invalid_zone","message":"Add test_get_service_by_invalid_zone\n\nThis test was removed in a previous change, as it was expecting a\nbehavior that doesn\u0027t exist in the API anymore.\n\nThis change re-introduces this test, but now expecting that the\nservice lookup will fail with a 404 in case an unexsistent\navailability zone is provided as a filter.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/928081\nChange-Id: Ib75b93216c55f223cbfcd0f21a81852a3a2b646f\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f45d575d3aa12fa488c470a171d9cb547bb529f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f45d575d3aa12fa488c470a171d9cb547bb529f0"}]},"branch":"refs/heads/master"},"dd68cdb3bcdf0695b6630bd46a47a48e67697470":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-02-06 17:38:03.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/61/975861/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/975861/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/975861/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/975861/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/975861/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/975861/3"}}},"commit":{"parents":[{"commit":"5f585fb639ac54b906dbd14cc6fe1620406b6fe3","subject":"Update for using endpoint_type to create client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5f585fb639ac54b906dbd14cc6fe1620406b6fe3"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-05 20:58:46.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-06 17:37:54.000000000","tz":-180},"subject":"Add test_get_service_by_invalid_zone","message":"Add test_get_service_by_invalid_zone\n\nThis test was removed in a previous change, as it was expecting a\nbehavior that doesn\u0027t exist in the API anymore.\n\nThis change re-introduces this test, but now expecting that the\nservice lookup will fail with a 404 in case an unexsistent\navailability zone is provided as a filter.\n\nChange-Id: Ib75b93216c55f223cbfcd0f21a81852a3a2b646f\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dd68cdb3bcdf0695b6630bd46a47a48e67697470"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dd68cdb3bcdf0695b6630bd46a47a48e67697470"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
