)]}'
{"id":"openstack%2Fmanila-tempest-plugin~976374","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ie38e75e7fad279f75809f6e83d21e2227e80a894","project":"openstack/manila-tempest-plugin","branch":"master","topic":"pyproject","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-05-04 16:02:13.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-05-04 16:02:13.000000000","reason":"Change was submitted"},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2026-05-04 04:18:14.000000000","reason":"\u003cGERRIT_ACCOUNT_32919\u003e replied on the change","reason_account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2026-03-03 21:35:13.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}},"hashtags":[],"change_id":"Ie38e75e7fad279f75809f6e83d21e2227e80a894","subject":"Migrate setup configuration to pyproject.toml","status":"MERGED","created":"2026-02-11 12:12:40.000000000","updated":"2026-05-04 16:02:13.000000000","submitted":"2026-05-04 16:02:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"976374-pyproject","meta_rev_id":"6893526f31098cb0008f06d87d19026df2df38d3","_number":976374,"virtual_id_number":976374,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-04 16:02:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2026-02-11 22:51:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2026-05-04 04:18:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":2,"date":"2026-03-03 21:35:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-02 11:41:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":1,"date":"2026-05-04 15:09:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-11 14:25:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-11 22:51:19.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-02-19 05:38:55.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"},{"updated":"2026-02-19 05:38:55.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2026-05-02 11:41:16.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"9f3dc7c5e4826f56fa2bcc6fb02cba6483e25fe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-11 12:12:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2cd15a8547d3cd9b72a0fe756ebc1ea5f551985","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 14:25:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a5716364712458492fdd1a86b142e63\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/44c556b8a5ba42c29e4be8cb6ad1eaef : SUCCESS in 2m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d642e9c812a84945b637560707209a45 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ce15bea2e84461faa09dfb7dd2591ed : FAILURE in 3m 03s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/22a35a3ca47742bcb17525026aaf4587 : FAILURE in 23m 05s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/6e13b3e306d64dbb841d34f2e52a930d : SUCCESS in 2h 07m 43s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/eb5ef457259143b89da0c3c7a95aba16 : FAILURE in 59m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3a5119f222e54ca393acc268c75148a3 : SUCCESS in 1h 07m 28s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/015d01e24ea345dfb1be74e22e5c2c53 : SUCCESS in 1h 36m 41s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1b908e905db04e0f91c5d8a6e0ffb8d9 : FAILURE in 1h 20m 54s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d4efcc574c7049d8bc8e2f90c48c3927 : SUCCESS in 36m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a0013c180a634729902d8261e6177f87 : SUCCESS in 40m 56s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/01f713ff4623447ba9b5abe0d83a14ff : SUCCESS in 1h 22m 35s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/08f866d2e80942fe87ecfdfd7764a198 : SUCCESS in 1h 37m 10s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/3ac0017f63b74502ba26a5ce9585c626 : SUCCESS in 35m 47s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a2d64276931b4bd4bb5bd7ad65bf31f8 : SUCCESS in 30m 29s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b108c76e0c3f461a9c08f9fff8f64ebc : FAILURE in 1h 23m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd3a40d49276f469d24caad8b75917c3c14f30ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-11 15:09:40.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d7aceb64ddd56809b69d16f8ba59004698cb2424","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 17:42:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eecb02b670764b4da9a980c19b82db44\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fbd86a19852e4c2f8923ddb6a12e0216 : SUCCESS in 2m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/38843728f918447e95bf35dc7e26d81a : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b174f034a41a4117a8ff0f4bd70590ac : SUCCESS in 2m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d84dd94a318e4970bc14a55998101559 : SUCCESS in 1h 15m 24s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8ee4f810ef7247e385b43ab7e92cc4af : SUCCESS in 1h 31m 00s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/18f2ec836e0947d18f0600c7717fbd55 : FAILURE in 1h 28m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f30b281892c447ceb0c42b154e28fd90 : FAILURE in 35m 46s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/0d0ca41b3393447ea3607fe696cc4b2f : SUCCESS in 1h 07m 11s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2cb0ab929e024742b1e7794be9a68bc7 : FAILURE in 1h 21m 42s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/603c19d02ba64f47a1508c357137c903 : SUCCESS in 24m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3fd828ce790d4cfbb8f1148b0863baf5 : SUCCESS in 44m 44s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/d048b476f00a4c529a6e8d4d1b911753 : SUCCESS in 1h 14m 00s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/b7a02de9e1354516988bc8681e482894 : SUCCESS in 2h 30m 57s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/b590c64ae6c14c52b0da2ff1d0442966 : SUCCESS in 1h 11m 02s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f352fc466c284e579239ae6f9c754bba : SUCCESS in 28m 04s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2d6df5736dce494391764f8dadb1aae3 : FAILURE in 1h 02m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6864e592eb6978cf6662cf65877a197a6865a3cd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-11 22:51:19.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"bd2eb50a64a944d7831f9d3e857b74a227d9c5e8","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2026-03-03 21:35:13.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ad5e3bdd753e3efe333f3076bae28c356c6d9b22","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-02 11:41:16.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9aa6aaf7d9490cce8b3f95b8959964446a3d5872","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-05-04 04:18:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"872ea7127c55926d0346a88cc2f8843fdfbc9aec","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2026-05-04 15:09:50.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9e6c619b1c3c70c3f5673d14ac49b7985c9eed3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 15:10:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7d36d8a7e3101e6595b5f54935b31b8f082850f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 16:02:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cc449e620e94d21857be07f07c32c34\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bca446229ba24ab1b2cd02636406cc8e : SUCCESS in 3m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/11bca0addf2e4daaa979b947d2712fd6 : SUCCESS in 2m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/458fa2825964421294f1268a8604c40c : SUCCESS in 3m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/74d9c20814e745309297ab3d37bd635f : SUCCESS in 51m 03s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fb216884d5c648f2b5324b115464a1fc : SUCCESS in 40m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e62cddeaa905442b9fe21de14d69f6b1 : SUCCESS in 35m 03s","accounts_in_message":[],"_revision_number":2},{"id":"6893526f31098cb0008f06d87d19026df2df38d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 16:02:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"892af0fe751582d11a325c1f0b2031c654124527","revisions":{"e8446b0b027db8e164a40ef5512c643ec39147a7":{"kind":"REWORK","_number":1,"created":"2026-02-11 12:12:40.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/74/976374/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/74/976374/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/976374/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/976374/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/976374/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/976374/1"}}},"commit":{"parents":[{"commit":"3f76739525905fc34792d9c2490f3546c3af75b2","subject":"Merge \"Refactor stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3f76739525905fc34792d9c2490f3546c3af75b2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-11 12:12:07.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-11 12:12:07.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Ie38e75e7fad279f75809f6e83d21e2227e80a894\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e8446b0b027db8e164a40ef5512c643ec39147a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e8446b0b027db8e164a40ef5512c643ec39147a7"}]},"branch":"refs/heads/master"},"892af0fe751582d11a325c1f0b2031c654124527":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-02-11 15:09:40.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/74/976374/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/74/976374/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/976374/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/976374/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/976374/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/976374/2"}}},"commit":{"parents":[{"commit":"febe94d18e28298f4f476df9c55725860924658b","subject":"Bump flake8-import-order version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/febe94d18e28298f4f476df9c55725860924658b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-11 12:12:07.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-11 15:09:40.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Ie38e75e7fad279f75809f6e83d21e2227e80a894\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/892af0fe751582d11a325c1f0b2031c654124527"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/892af0fe751582d11a325c1f0b2031c654124527"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
