)]}'
{"id":"openstack%2Fmanila-tempest-plugin~976938","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I4e769eb0b2838a9d5c69cfcab20fd7295ab32875","project":"openstack/manila-tempest-plugin","branch":"master","topic":"remove-v1","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-02-19 18:56:22.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-02-19 06:14:55.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-02-19 19:40:50.000000000","reason":"Change was submitted"}},"hashtags":["remove-manila-api-v1"],"change_id":"I4e769eb0b2838a9d5c69cfcab20fd7295ab32875","subject":"Use official service type","status":"MERGED","created":"2026-02-16 12:57:09.000000000","updated":"2026-02-19 19:40:50.000000000","submitted":"2026-02-19 19:40:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976938-remove-v1","meta_rev_id":"587f189ba56e5d5035bc3240391acfd26e809d59","_number":976938,"virtual_id_number":976938,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-19 19:40:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2026-02-19 18:56:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2026-02-19 06:14:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2026-02-19 18:56:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-16 15:21:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-16 16:10:08.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-02-16 16:10:08.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2026-02-16 20:15:37.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"edfe45d77dfb5ada20a3495dd63b3f5d0f1c54b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-16 12:57:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"048f0106665127d263cc0bde76dce2a0f71f44f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 15:21:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f4c9e73858842698ec4d4183add263c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7548b066ccf42d880722854b8e14c54 : SUCCESS in 1m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cb5711973a9a4553a81b30059106b05a : FAILURE in 1h 05m 21s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/efb97276b7bd4f10ba10523a9fc4f712 : SUCCESS in 51m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6803afdc1d144c8e9c89477a6c6384f6 : FAILURE in 31m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a1f1aff37c88430cb97f53a59dda3251 : FAILURE in 53m 11s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/aaf4ceb2fa7247058b85e978cfbb2863 : SUCCESS in 1h 23m 56s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/12e00255a905484591d71b86f0f0538c : FAILURE in 32m 32s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/baf9718fd11d4b2e98a7209a5a1190d0 : SUCCESS in 30m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/28e18f0f43e94bc7baf398176d9207cf : SUCCESS in 41m 49s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/cb95b91363b447798f2f268bfdcb3cb5 : SUCCESS in 2h 21m 29s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/dbe6461f17ea43ad8badb4dfb0789239 : SUCCESS in 33m 42s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/3326b627a505412982d3e4120ecea222 : SUCCESS in 1h 09m 15s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/e6d772ebbbb84828990e944d89322161 : SUCCESS in 23m 56s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/636d7286eac74f539baeb00e10c9e9c3 : FAILURE in 27m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"83e0354d6891a1ad7a6ba5e4b278923b929f705d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-16 16:09:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6cf0df79fa824f7160a73d55734b039bbd92b37d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-16 16:11:33.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"498f4318b1138d613b6bf36410b10b6147605484","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 18:51:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3019b05cd1e41628be3b93b921cb04a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a80d201865a94a72811d3fc575188c3d : SUCCESS in 2m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d4f24c7addfb4318bd98d819e7813fe4 : SUCCESS in 1h 11m 54s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a42e447acf7e4bdd8af868f7c9953477 : FAILURE in 2h 33m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7cab0f603c294dc8a210203e325a2af5 : FAILURE in 1h 08m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/74b4dead713845ceb75551a27a7e168f : FAILURE in 53m 45s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/13d17ddcf4e24275a13f937c2834e6e7 : SUCCESS in 1h 25m 21s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/277b7375b6e84eef9c9f80c2369819de : SUCCESS in 1h 14m 57s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ef22edf0b9cf4498afc39f582a68022c : SUCCESS in 40m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b4f5d5e0d6d94635b842693a5c1ca251 : SUCCESS in 51m 57s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/abb1fd61538e429ab94e4c8b75cc2645 : SUCCESS in 1h 04m 00s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/db646f56c48544f8a61d1aaadff196a4 : SUCCESS in 1h 23m 32s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/8eedd015366e4406877615543b6c82b2 : SUCCESS in 1h 18m 40s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/b6598f92efc44adf83b12d9c3b37216a : SUCCESS in 19m 49s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/aa149185c9024cfcb6255b562b1fc9f4 : FAILURE in 41m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8828a45977384d219c8a9453ce36c8bce32bde7d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-02-16 20:15:37.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/976938/2/upstream-check-manila/manila-tempest-plugin-ontap-dhss/b39807f/ : SUCCESS in 1h 22m 21s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"fe2a6fc0636244b2e1c25bc3ce4316b08571b7a7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-17 12:25:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"43986f3e04f935b3aad83164a37e743c43b4a19a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-19 06:14:55.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a982e342a4f25b5e68d9cec8b2374995bd879c48","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-19 18:05:12.000000000","message":"Hashtag added: remove-manila-api-v1","accounts_in_message":[],"_revision_number":2},{"id":"6d247cfb8fd9c674fc64ff2c09a1622565fba673","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-19 18:56:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"adb4c50360a1f65c2047c5935a689c9df2e582be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 18:56:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c2bf027806bc54822b6836800c27f9f3df3cf6e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 19:40:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a74a1d2e2854c38b569b7ab86b6d0af\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b0e671fcbfb415da4dc89e6cead499d : SUCCESS in 2m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cb9efcf43da14d76b27abfb367fafd78 : SUCCESS in 35m 49s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a96841aa37994bf49b3f8e714e80e46a : SUCCESS in 39m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1dea1a03a580490eab9edfa44576605b : SUCCESS in 37m 44s","accounts_in_message":[],"_revision_number":2},{"id":"587f189ba56e5d5035bc3240391acfd26e809d59","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 19:40:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f99c487d8d1e78184d4c85e226c04346c7c75cf2","revisions":{"d3f4a93dc0cde722069824ce88736f262bd99dc2":{"kind":"REWORK","_number":1,"created":"2026-02-16 12:57:09.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/38/976938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/976938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/976938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/976938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/976938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/976938/1"}}},"commit":{"parents":[{"commit":"6406026b51e9c282902dbc3987f1edb9879821b8","subject":"Remove Manila V1 API client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6406026b51e9c282902dbc3987f1edb9879821b8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-16 12:56:43.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-16 12:56:43.000000000","tz":0},"subject":"Use official service type","message":"Use official service type\n\nChange-Id: I4e769eb0b2838a9d5c69cfcab20fd7295ab32875\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d3f4a93dc0cde722069824ce88736f262bd99dc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d3f4a93dc0cde722069824ce88736f262bd99dc2"}]},"branch":"refs/heads/master"},"f99c487d8d1e78184d4c85e226c04346c7c75cf2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-02-16 16:11:33.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/38/976938/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/976938/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/976938/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/976938/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/976938/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/976938/2"}}},"commit":{"parents":[{"commit":"6406026b51e9c282902dbc3987f1edb9879821b8","subject":"Remove Manila V1 API client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6406026b51e9c282902dbc3987f1edb9879821b8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-16 12:56:43.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-16 16:11:33.000000000","tz":0},"subject":"Use official service type","message":"Use official service type\n\nBuild on change Id90044bca44387864aee8984bc2366c089c87cfc and\nswitch to the official service type rather than the unofficial\none that we are removing in [1].\n\n[1] https://review.opendev.org/c/openstack/manila/+/959554\n\nChange-Id: I4e769eb0b2838a9d5c69cfcab20fd7295ab32875\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f99c487d8d1e78184d4c85e226c04346c7c75cf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f99c487d8d1e78184d4c85e226c04346c7c75cf2"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
