)]}'
{"id":"openstack%2Fmanila-tempest-plugin~977884","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I6ef431f501b99b0e6dbc2e058fb7d09e9ff78eff","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-03-06 21:38:32.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2026-03-06 21:38:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6ef431f501b99b0e6dbc2e058fb7d09e9ff78eff","subject":"Add required snapshot extra-specs in scenario tests","status":"MERGED","created":"2026-02-24 21:43:26.000000000","updated":"2026-03-06 21:38:32.000000000","submitted":"2026-03-06 21:38:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977884","meta_rev_id":"2c3c763c307736e89aabb2711a1053568269c1b4","_number":977884,"virtual_id_number":977884,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-06 21:38:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2026-02-24 21:51:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2026-03-06 20:47:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-24 21:51:50.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2026-02-25 00:09:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-27 00:13:17.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-03-05 20:44:13.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"ec7d97f44ea2c37e237d9be1fb7edbe180a36a51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-24 21:43:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"081267f1ad9fdcef66a60bcacf32728860352135","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-24 21:51:50.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e515194880b1f212d4a61ec426797bdd1e736ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 00:09:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/22a28da5c3ea44da87a3a21e106d73b3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7ccb1eaaceb44e6a1269f1e6b80b5e3 : FAILURE in 3m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2e86781cd9d0422b82b4acdfbe63a845 : SUCCESS in 56m 03s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/155f01460ba94944b6cb18d4762fbddf : FAILURE in 2h 17m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/446e8f19796040cf801140183b95a299 : FAILURE in 1h 11m 49s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e7ea9783b77f41f68725c0f9799c48a8 : FAILURE in 26m 53s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e180bbb8afdc439496146d31f3fa902d : SUCCESS in 1h 30m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/073cdd89aecf47a8b3b98b7b59bbb0b8 : FAILURE in 58m 37s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2d6486d0c67a4a29841cabb5d3e614ab : SUCCESS in 46m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/aa4e94869e6c4239a00ed7485bc443cb : SUCCESS in 39m 18s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/20a315bc631143a1b1cd89450c6fd7b0 : SUCCESS in 1h 09m 09s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/4ffd0f285de84041babb86e6865e4e33 : FAILURE in 19m 25s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/ff7bbfb1842e4d69906fb18f94feed75 : SUCCESS in 35m 00s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/465c08113afe427bb3d32b8cba09271a : SUCCESS in 30m 55s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/398e827802be48018991474ca9c96639 : FAILURE in 1h 01m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c64a1da3aa3e61e02248ca6256d7e5f777801c3c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-25 07:19:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae588494837cbe0fe88a224f64673542ad33c0ea","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-26 21:54:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ec8469cb8700ef8ab81da43cda68e1213e10e76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 23:42:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb2922c8829046c5810d9f7fdcfa6b6b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb6a5ca9ad354999ac5339deb6b3529a : SUCCESS in 2m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/05ef662862084465a60cf6e5b4dc5f2a : SUCCESS in 46m 32s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7f711fd311da4b7985a47f246db84d83 : SUCCESS in 1h 46m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/370941a9ae994fc096cc718ad06fe132 : FAILURE in 1h 07m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/7d65583be4e74aeaab1eaf03b1a501ea : FAILURE in 41m 04s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/0953ee5dbd954ac59c6c382ae41a5f98 : SUCCESS in 53m 39s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/65d186d074454740bd8179b6c5518c32 : FAILURE in 46m 37s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cb46d359ebc34cc09105fc7f2edfd8dd : SUCCESS in 33m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ab558e965d204f96ad08e7ce04bab0d0 : SUCCESS in 34m 29s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/0d89a5562ebb4c5cbc5d579c9d085081 : SUCCESS in 43m 13s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/e6296ed167e2434fa3e3c2a83f926404 : SUCCESS in 1h 11m 59s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/81f220e43f2144f996aac7ecf80a4ccd : SUCCESS in 1h 30m 05s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a7c2b0b07aac455daf01567da54ecbd9 : SUCCESS in 31m 53s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a303ce80139b47f1814522ae39d4ed18 : FAILURE in 1h 06m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7fac4e6773e69d55e3bb0b36c0da5749b62c0b61","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-02-27 00:13:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss finger://managesf.services.openstack.netapp.com:7979/5a68d7ef36b94a3c98156dbd7c25d90b : DISK_FULL in 29m 20s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"6076bbac17acc805f27fea69dc41ce524e7b8a66","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-03-06 20:47:21.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1fc4ab23333c5892a1d4de62baf78e32bb6f601c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 20:47:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5a277b59b5b98705018aabd76c01b0a080450e8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 21:38:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e53840c86cd54f26873a94539a336df6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7abe35c4e0249ac8798c47a44b285fd : SUCCESS in 2m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2c10649fe87f499e9056f3bd76be818f : SUCCESS in 34m 28s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/97709a46b2b74c4ba0fc5a0c3438932b : SUCCESS in 33m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/754b7e2e640b44beb3e2c8b62c58e630 : SUCCESS in 45m 59s","accounts_in_message":[],"_revision_number":1},{"id":"2c3c763c307736e89aabb2711a1053568269c1b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 21:38:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"24447b889c4bb9a8d8a5b75fffd3474ed2f927d0","revisions":{"24447b889c4bb9a8d8a5b75fffd3474ed2f927d0":{"kind":"REWORK","_number":1,"created":"2026-02-24 21:43:26.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/84/977884/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/84/977884/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/84/977884/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/84/977884/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/84/977884/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/84/977884/1"}}},"commit":{"parents":[{"commit":"f99c487d8d1e78184d4c85e226c04346c7c75cf2","subject":"Use official service type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f99c487d8d1e78184d4c85e226c04346c7c75cf2"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-02-24 21:42:43.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-02-24 21:42:43.000000000","tz":-480},"subject":"Add required snapshot extra-specs in scenario tests","message":"Add required snapshot extra-specs in scenario tests\n\nTests must request the specs they need to test; but\nthey aren\u0027t, currently.\n\nChange-Id: I6ef431f501b99b0e6dbc2e058fb7d09e9ff78eff\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/24447b889c4bb9a8d8a5b75fffd3474ed2f927d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/24447b889c4bb9a8d8a5b75fffd3474ed2f927d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
