)]}'
{"id":"openstack%2Fmanila-tempest-plugin~977914","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I207204c0ab7ca1cefa252fa3409cab2049425b39","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-03-06 22:09:01.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2026-03-06 19:08:35.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}},"hashtags":[],"change_id":"I207204c0ab7ca1cefa252fa3409cab2049425b39","subject":"[CI] Set snapshot cloning spec in CephFS jobs","status":"MERGED","created":"2026-02-25 07:19:13.000000000","updated":"2026-03-06 22:09:01.000000000","submitted":"2026-03-06 22:09:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977914","meta_rev_id":"6007f5dc6819bc8b9a6ce475918489dc06587e2e","_number":977914,"virtual_id_number":977914,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-06 22:09:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2026-02-25 17:21:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2026-03-06 19:08:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2026-03-06 20:47:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-25 09:29:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-25 17:21:31.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2026-02-27 00:23:08.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-03-05 20:42:54.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"73425725ca787de2dc8cb8cb7c6e1c4673dbf815","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-25 07:19:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d56b276aeea304537a99b6f3a2862d810ebe5411","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 09:29:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbb1f6fcd6294b74b4894d3e099d923e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/905ff04fd26f498f8522485c4ff46b14 : FAILURE in 2m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/82f2ea0dd59b4cada929db480d71d4b2 : SUCCESS in 1h 20m 12s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5985ed1d5a6f4c6d8629bf76316bad15 : SUCCESS in 2h 05m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f6695e95cb104821aa942f000a1fe54a : FAILURE in 53m 55s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c5d819552c864023953d3590f2f8d0a2 : SUCCESS in 29m 41s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/5f7de59c8eda4188933a6f771a0724cd : SUCCESS in 1h 32m 25s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9f5d0837d1734b2d8c04f9db48fdb007 : FAILURE in 1h 13m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/30ea7ba4e06e4a3fabf63829bbb6783c : SUCCESS in 35m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/eb7a97a22a0c4b3eb05d0141f9325aa9 : SUCCESS in 29m 33s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/26c58e983af04cad94f59a91aac09efc : SUCCESS in 1h 10m 05s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/3d16f6056a01419fbad016b928ad6729 : SUCCESS in 55m 39s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/d2f9c0e26b994f0ba4512e57564b2669 : SUCCESS in 32m 51s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/61781e83479140409caad10f482eac62 : SUCCESS in 32m 15s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3dbcbcc631c04f1696154ba12a1558c4 : FAILURE in 1h 03m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e49e7727783cb925433b4d1f8a1e70c0ad32afe3","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-25 17:21:31.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e628d97110664d6efacae4baedc840fc64e61a99","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-26 21:54:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"150d4df12e6018e683e6d3b21521ca0ed389fdad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 00:07:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d21ac89ac4fc4a9aaf18d253bc0fcaea\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7269f9bbde744231a585e9d032af99e2 : SUCCESS in 2m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4e1b55918b6d4a7bafaaa772812025e4 : SUCCESS in 1h 13m 08s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/fe42a864b5fb492e8b5aa1be286c1cc3 : SUCCESS in 2h 06m 42s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3b6fc69ac8564fdaad291e08406edb1c : FAILURE in 37m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ee78af46c8214b32805b2d52b0169a05 : FAILURE in 11m 12s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/0e6402cf401d4c8c8d82a340a6cdade4 : SUCCESS in 46m 33s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2ad6f7c56c534277b05e1a88dd9b1f13 : FAILURE in 48m 08s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ae0e193ab51a4ec683124d7be546ce8a : SUCCESS in 26m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8066634543354da69634d7f5aa7cb33e : SUCCESS in 32m 31s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/83da021c161e4deda26be543d85902b9 : SUCCESS in 1h 13m 08s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/d6be7795e03b4ce997250c5fb8215dd8 : SUCCESS in 1h 05m 52s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/785fc9ca52e3457fb2384902cd309875 : SUCCESS in 1h 15m 21s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/1435b4f3f23e46db87745c9a6a94d2f2 : SUCCESS in 33m 05s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1ab16a55159e41968e4744e2037595db : FAILURE in 29m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"50cb6db45b99392409f8f3bf2cefd6807c06d460","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-02-27 00:23:08.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss finger://managesf.services.openstack.netapp.com:7979/403db434fb8442149ec579857bb4b818 : DISK_FULL in 14m 00s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"cd056523b40410ad174822689cd10603f006895e","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-03-05 20:40:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c6817c1ccc52019d1fdfe30d21a69b8d757bdbc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 22:57:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f617e09c08d483f913fd626a1b842b5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ee69c6b2bb24e9382cc18dc32ec0eda : SUCCESS in 3m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4fb8cbdc5ac04ddc9e5afeddd1f3e9cd : SUCCESS in 1h 16m 25s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/34e6322076074278bf607efe3b98b43a : SUCCESS in 2h 09m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/20beb413637446468a39dd595148b83f : FAILURE in 1h 11m 33s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/52f95a613e524449a96362a8dadc0341 : SUCCESS in 28m 00s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/5d8ca9569cf84524a3d38ecadd735a7a : SUCCESS in 45m 01s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4a6d74f8bc0c4fcdbf05d76c3f559a2f : FAILURE in 1h 17m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/10611d43eaed44018888b9f7abeee220 : SUCCESS in 21m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ed0ad56a89744401b5ff730771c0cb07 : SUCCESS in 27m 14s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/079f42f9840e42b6a60794a45ecab923 : SUCCESS in 38m 14s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/fc722e92bfdd44ab84b12bf2ee479186 : SUCCESS in 1h 16m 16s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/df11f54d1ea04157a2eab43ea3f44aa5 : SUCCESS in 1h 09m 46s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/2aff167d3b5e418aa41a64e0efd1db1a : SUCCESS in 31m 42s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/450c28468f7b41008f4ab4acd7da6bf3 : FAILURE in 1h 05m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"397ee9779aae736937d69ea1ed684234f432fabd","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-05 23:12:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/14/977914/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/f2d8f90/ : RETRY_LIMIT in 7m 32s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"2542be75032ea334b85917964a3cbcf7b6ac9951","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2026-03-06 19:08:35.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"269284193f9def3cdd76f9833f71783dd9832cf0","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-03-06 20:47:25.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"86ec93fa29cdb9af112430d23b2d5a7d6a64b466","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 20:48:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b4ee993f0eebaaab66afaedcd31dd40ffd2ad87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 22:09:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f250b263ad5345a786686084f2abebad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a3a5b9cf5a147dbac4db89cbff02b09 : SUCCESS in 3m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ecfa34c5522a4f3a9ab10ed1918e87b4 : SUCCESS in 1h 15m 49s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/41f853c5527f40e5a8ae21813ccfa0a5 : SUCCESS in 38m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/38cc018d88c34e69b76f41086f64f2d2 : SUCCESS in 44m 07s","accounts_in_message":[],"_revision_number":1},{"id":"6007f5dc6819bc8b9a6ce475918489dc06587e2e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 22:09:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c018347377014115a3bd8e367e18cb3d990585df","revisions":{"c018347377014115a3bd8e367e18cb3d990585df":{"kind":"REWORK","_number":1,"created":"2026-02-25 07:19:13.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/14/977914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/14/977914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/977914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/977914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/977914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/977914/1"}}},"commit":{"parents":[{"commit":"24447b889c4bb9a8d8a5b75fffd3474ed2f927d0","subject":"Add required snapshot extra-specs in scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/24447b889c4bb9a8d8a5b75fffd3474ed2f927d0"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-02-25 07:16:28.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-02-25 07:16:28.000000000","tz":-480},"subject":"[CI] Set snapshot cloning spec in CephFS jobs","message":"[CI] Set snapshot cloning spec in CephFS jobs\n\nSome of our test jobs have the default share type\nenabled. When this is the case, the tests don\u0027t\ncreate share types, they just use the default type.\n\nCephFS has supported create_share_from_snapshot_support,\ni.e., snapshot cloning since the Wallaby release,\nso we can enable this in the default share type.\n\nChange-Id: I207204c0ab7ca1cefa252fa3409cab2049425b39\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c018347377014115a3bd8e367e18cb3d990585df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c018347377014115a3bd8e367e18cb3d990585df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
