)]}'
{"id":"openstack%2Fmanila-tempest-plugin~981671","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I744a0a0ffed17b3a1962498e0416e560c3f2c96d","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1959329","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-03-31 21:02:55.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-04-07 16:46:21.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"38471":{"account":{"_account_id":38471,"name":"Denver Baraka","display_name":"Denver Baraka","email":"denverbaraka@gmail.com","username":"denver-baraka"},"last_update":"2026-04-07 17:56:29.000000000","reason":"Change was submitted"}},"hashtags":["outreachy-2026-prospect"],"change_id":"I744a0a0ffed17b3a1962498e0416e560c3f2c96d","subject":"Add share server subnet filter tempest tests","status":"MERGED","created":"2026-03-22 08:49:30.000000000","updated":"2026-04-07 17:56:29.000000000","submitted":"2026-04-07 17:56:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"981671-bug/1959329","meta_rev_id":"fe2ea072df617057507ef1f82898267c34f379ba","_number":981671,"virtual_id_number":981671,"owner":{"_account_id":38471,"name":"Denver Baraka","display_name":"Denver Baraka","email":"denverbaraka@gmail.com","username":"denver-baraka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-07 17:56:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2026-03-31 21:02:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2026-04-07 16:46:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2026-04-07 16:46:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-22 08:50:26.000000000","updated_by":{"_account_id":38471,"name":"Denver Baraka","display_name":"Denver Baraka","email":"denverbaraka@gmail.com","username":"denver-baraka"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2026-03-22 08:50:26.000000000","updated_by":{"_account_id":38471,"name":"Denver Baraka","display_name":"Denver Baraka","email":"denverbaraka@gmail.com","username":"denver-baraka"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-03-22 10:50:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-22 12:13:21.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"045cfe87efae538d89787ec26cb0a471061ecc1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38471,"name":"Denver Baraka","display_name":"Denver Baraka","email":"denverbaraka@gmail.com","username":"denver-baraka"},"date":"2026-03-22 08:49:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"189507bc01f9dd86dcb74d43d36e53e9965de0e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-22 10:50:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd81d0419cb644a8a1746e654f4b1dbe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3075f63baf7344d19d121502c5ad9d7f : SUCCESS in 4m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c61f0a6528104aebb151b2b031f3ac4d : SUCCESS in 47m 40s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/90e07dfdf9fd458cb9e9f0bd0f18157b : SUCCESS in 1h 54m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6eaf4251f2654e77b4a016d6a1c9b833 : FAILURE in 1h 23m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/69675aed4ac8492081ef6a80bf722e47 : SUCCESS in 1h 01m 27s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e86dc30b03f4409482cf4966fcaa4275 : SUCCESS in 1h 24m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9b694f96cbaf4ba7909e3dac6705fb69 : FAILURE in 1h 19m 55s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cdef501cf0af46918e8ccb178008e49a : SUCCESS in 24m 47s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/70b3f722004f4b379297e571116a61bd : SUCCESS in 27m 14s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/8401488c9b284349a21078dca9feff9d : SUCCESS in 48m 13s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/c32a23e813b14dc18abcc5207881ff3b : SUCCESS in 38m 11s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/85ac13a250bb4bd685fe893f34771f77 : SUCCESS in 49m 26s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/c11fa2f1efb846c99c9988a7bda316b6 : SUCCESS in 47m 15s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/1463500ffcef4c06b008833501a50d68 : SUCCESS in 32m 56s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8fc5552973264884b7eb2822dc81fe58 : FAILURE in 30m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"439aadea43b16579484f60c14cb95d561abb7a33","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-22 12:13:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/71/981671/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/bbe6821/ : SUCCESS in 1h 21m 55s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"bcf409477d5302783905eaea2addd2c42410ed54","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-31 20:05:28.000000000","message":"Hashtag added: outreachy-2026-prospect","accounts_in_message":[],"_revision_number":1},{"id":"cb9310b340348808e0f3efaee9ef25a7fa1668a8","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-03-31 21:02:55.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe9e01c1457f7ea687dc39d4a948a70c9402f43d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-07 16:46:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"afab5d3409fad4013dc7ce3caa160d39aef0a038","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-07 16:46:24.000000000","message":"Topic bug-1959329 removed","accounts_in_message":[],"_revision_number":1},{"id":"cd32ab2ea6489cf0ec9bf87ae9ec8807e52f769c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-07 16:46:27.000000000","message":"Topic set to bug/1959329","accounts_in_message":[],"_revision_number":1},{"id":"6ca62c9ad94f22cf43e46dfae5aa4358111a43ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 16:46:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b2fde9d362b8d6b197b2e95ad5252f89af74dcee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 17:56:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e48901b842b8482690003027f68eacbc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16fa5d388ff94f648ba11931f2e43aaa : SUCCESS in 2m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d7dff3b5f8284603b1dd1ca51e4df756 : SUCCESS in 1h 05m 19s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d50924f89a8543e5b11829bce7238341 : SUCCESS in 27m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/75bd0747e4044f48a74d8804d3c84f5f : SUCCESS in 31m 48s","accounts_in_message":[],"_revision_number":1},{"id":"fe2ea072df617057507ef1f82898267c34f379ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 17:56:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"11e69e46bd5b8e28b61c0fd57687c29854c782ea","revisions":{"11e69e46bd5b8e28b61c0fd57687c29854c782ea":{"kind":"REWORK","_number":1,"created":"2026-03-22 08:49:30.000000000","uploader":{"_account_id":38471,"name":"Denver Baraka","display_name":"Denver Baraka","email":"denverbaraka@gmail.com","username":"denver-baraka"},"ref":"refs/changes/71/981671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/71/981671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/71/981671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/71/981671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/71/981671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/71/981671/1"}}},"commit":{"parents":[{"commit":"12145b00c527273c2dd30f453e9cb70aab57f6b8","subject":"Merge \"Add stable/2026.1 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/12145b00c527273c2dd30f453e9cb70aab57f6b8"}]}],"author":{"name":"Denver Baraka","email":"denverbaraka@gmail.com","date":"2026-03-22 08:41:52.000000000","tz":180},"committer":{"name":"Denver Baraka","email":"denverbaraka@gmail.com","date":"2026-03-22 08:41:52.000000000","tz":180},"subject":"Add share server subnet filter tempest tests","message":"Add share server subnet filter tempest tests\n\nThe list share servers API can filter by [share_network_subnet_id]\nbut \u0027manila-tempest-plugin\u0027 was missing tests for this feature.\n\nThis patch adds a positive and a negative test to verify that\nthe [share_network_subnet_id] filter works correctly in API\nmicroversions 2.51 and above.\n\nTest Plan:\nPASS: tox -e pep8\n\nCloses-Bug: #1959329\nChange-Id: I744a0a0ffed17b3a1962498e0416e560c3f2c96d\nSigned-off-by: Denver Baraka \u003cdenverbaraka@gmail.com\u003e\nAssisted-by: Copilot (Writing commit message)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/11e69e46bd5b8e28b61c0fd57687c29854c782ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/11e69e46bd5b8e28b61c0fd57687c29854c782ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
