)]}'
{"id":"openstack%2Fmanila-tempest-plugin~983154","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Iaa0e33dde629d4321adc3317bb9fa4ffa1d4b853","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"24993":{"account":{"_account_id":24993,"name":"Carlos","email":"carlos.carrero@veritas.com"},"last_update":"2026-06-04 19:49:40.000000000","reason":"Change was submitted"},"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2026-06-04 19:49:40.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-06-04 19:08:24.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"38371":{"account":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"last_update":"2026-05-26 04:30:57.000000000","reason":"\u003cGERRIT_ACCOUNT_38371\u003e replied on the change","reason_account":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2026-05-26 09:00:22.000000000","reason":"\u003cGERRIT_ACCOUNT_36180\u003e replied on the change","reason_account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"}},"38373":{"account":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"last_update":"2026-06-04 19:49:40.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-06-04 19:49:40.000000000","reason":"Change was submitted"},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-05-26 14:09:39.000000000","reason":"\u003cGERRIT_ACCOUNT_38059\u003e replied on the change","reason_account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}},"38479":{"account":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"last_update":"2026-06-04 19:49:40.000000000","reason":"Change was submitted"},"38623":{"account":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"last_update":"2026-06-04 19:49:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaa0e33dde629d4321adc3317bb9fa4ffa1d4b853","subject":"Add replication_policy metadata tests for SnapMirror policies","status":"MERGED","created":"2026-04-02 04:31:58.000000000","updated":"2026-06-04 19:49:40.000000000","submitted":"2026-06-04 19:49:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983154","meta_rev_id":"b51f746d9e0dd097139b3dda6aabb017096c0ae3","_number":983154,"virtual_id_number":983154,"owner":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-04 19:49:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"recommended":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"all":[{"value":1,"date":"2026-05-26 04:30:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"value":2,"date":"2026-06-04 19:08:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2026-05-26 09:00:22.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":2,"date":"2026-05-26 14:09:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"value":1,"date":"2026-06-04 19:08:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24993,"name":"Carlos","email":"carlos.carrero@veritas.com"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24993,"name":"Carlos","email":"carlos.carrero@veritas.com"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-02 04:35:54.000000000","updated_by":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"reviewer":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-04-02 04:35:54.000000000","updated_by":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-04-02 04:35:54.000000000","updated_by":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"reviewer":{"_account_id":24993,"name":"Carlos","email":"carlos.carrero@veritas.com"},"state":"REVIEWER"},{"updated":"2026-04-02 04:35:54.000000000","updated_by":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"reviewer":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-04-02 04:35:54.000000000","updated_by":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-04-02 04:35:54.000000000","updated_by":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-04-02 04:35:54.000000000","updated_by":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-04-02 04:35:54.000000000","updated_by":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"REVIEWER"},{"updated":"2026-04-02 04:35:54.000000000","updated_by":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"reviewer":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-04-02 06:41:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-11 05:21:19.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"374a7a460f40e9e42e203226f278469f4404243f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-04-02 04:31:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c14413a4b2ef8865b49a36d360bfa5f1dc6db783","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-04-02 04:35:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bee229897f97b0fb534ab1159870f98f52f5bfe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 06:41:56.000000000","message":"Patch Set 1: Verified-1\n\n(30 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/755e24b0991e40178c99b8afa239a7c5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a0e238309814ec3b2d859876e00f013 : FAILURE in 2m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/440e497b8ecb44858f63d1782a3ebb1e : SUCCESS in 1h 16m 55s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/712ea976d0dc4d7aa8580877d186369b : FAILURE in 2h 09m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e4b16901a44c492aafb3802ba609a13a : FAILURE in 40m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/dc95683c98154a6aa0f03ccbeacf1ce2 : SUCCESS in 1h 03m 35s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/04eaabbb4fa94782bc4c24f869b79069 : SUCCESS in 1h 14m 36s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f0152a2ad4a74680bddb070284333432 : FAILURE in 1h 00m 01s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a99fe431829a4129be16cc30728040e6 : FAILURE in 31m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/94110fb3fec141d1b0382b9692ee5334 : FAILURE in 33m 57s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/4f606ed42c624db695b249f85cb98110 : SUCCESS in 38m 42s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/e02dc30d28b64e2cac055058cb7b9282 : SUCCESS in 52m 21s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/c5eb1471021b4698bfbdc8ace0163633 : SUCCESS in 33m 20s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/e8ba74098f4a4d7ebe2081f2cfe97c14 : SUCCESS in 49m 10s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a8d6e65b0b1f43dda3fb6503cfd5bf2e : SUCCESS in 22m 54s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9f4d9022c1784fad9d286f1ae3bcf817 : FAILURE in 44m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5c28996f56de51f4f1776c1069d62c367a3fbe71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-04-10 07:34:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8ee7e40156f59b07bade7fe9a53665c590f186ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 09:37:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/87053f44a0bf46048aa59f547181a09d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a82669280064754ad222bb80a620523 : SUCCESS in 3m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/63132c3d83524044bc2ee365de8c84f9 : SUCCESS in 1h 10m 00s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7158274e928444559bb030cb7e8102c0 : SUCCESS in 1h 57m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f76b6de26e514abd99e9d15c862a13c9 : SUCCESS in 1h 34m 53s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8fdb420bfe394768aad5f8d6964f68ae : SUCCESS in 31m 22s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e324bdc044d6490981f24b7dd14fb45f : SUCCESS in 1h 41m 08s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1e39012c5e874edbbc649c84a96f868d : FAILURE in 1h 01m 46s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/98e008c169294a008f79297c36556c7d : FAILURE in 22m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5ecd50b4fd8a4ec5a871bb5efda2066e : FAILURE in 34m 49s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/bdff66c90469477995b2bf9658d6f91b : SUCCESS in 36m 19s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/c715edea5bbb4d8a85ca1a526ae61602 : SUCCESS in 37m 43s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/107057d558e04507a86872fa3e7ede87 : SUCCESS in 39m 56s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/c59b7a7aa05c4fe98a46b38066614026 : SUCCESS in 30m 05s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/99e92e2a2458476fbc0e9e6c09dc447c : SUCCESS in 31m 39s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/db89109a4edc4cd7b1a5d84130ffeba5 : FAILURE in 12m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"def7e8ee58e5da020451b87d6b1558b93df46cd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-04-10 11:31:31.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"137cbf86a9befcdcd482706e8afd3e848d98d3dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 13:24:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a170b7abd3d484d8cd83e2620d38bbe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6759db1ab0d40d89ef262ac35f5e7f0 : SUCCESS in 2m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a6b8b39fd32a49409af0e33a251dfcf9 : SUCCESS in 43m 15s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/992ffb3ad35c495792a012c28bb90c6d : SUCCESS in 1h 51m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2e5b665a68614bdea6210aaf6f824543 : SUCCESS in 41m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f57e2066a6e6427aba6b5af827ad7b8e : SUCCESS in 33m 02s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/d7d0bdb39bb44239855a9e1eca1f55e6 : SUCCESS in 1h 32m 50s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/119a19b1735a4dd4ad12a3cac6e52b30 : SUCCESS in 1h 23m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d2742766a0454873aedd905e9d1d8b77 : FAILURE in 25m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4effc297036d4f77b9e63c1008ac39ae : FAILURE in 38m 05s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/97f9bcfdc61c477cbbcd342b2550e7da : SUCCESS in 1h 24m 35s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/a083df7d315049ab99e257eaef5f4d59 : SUCCESS in 38m 33s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/a2f354c135464121b19c994abd4909ad : FAILURE in 1h 10m 58s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/74e2d4d6bd7442e0b3df54a14f4ab07d : FAILURE in 42m 41s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/88029b05ac3d4aa9a5c2d90712a7ac48 : SUCCESS in 22m 39s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c6210deb95d84ee89161b5cdc60d26d7 : FAILURE in 18m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a4797096fecf1a96769188e8b78b8ee969826554","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-04-11 03:02:35.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"d2718c3cbacdec3ac3b31f5bb6e8b7ae608823ba","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-04-11 03:07:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bb999716686c6c1eec73237915b9a11dda4cc672","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 05:16:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d48ef642fe3c4d38b2ac00b063f7ba4f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0ad9b9e9dda4a89b371fc66f561b294 : SUCCESS in 2m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bbd75dbf9c8a4a4e91a0a6afd37ded3c : SUCCESS in 34m 56s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d29389741f3049e188c17fbf5bba4c67 : SUCCESS in 2h 08m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bbc42479b644441a9599cc4e540a9f80 : SUCCESS in 1h 27m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/1d5463f6673f4106854b52801e44b8b1 : SUCCESS in 1h 02m 14s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/dd963090a99942a1a023b977593e93c7 : SUCCESS in 1h 10m 58s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/680a4b77350f46b4a6bb8eff79f67897 : FAILURE in 1h 14m 36s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/368e34c49b58408d83a7317d5bf0df87 : SUCCESS in 36m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/00b4f3b539cc402882f4629049cdcb53 : SUCCESS in 34m 11s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/47d9629f5ec94402bba846da0fa8f18c : SUCCESS in 52m 35s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/626af0136d0b452489ff492177f43806 : SUCCESS in 50m 04s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/4dbbb1085c454fc6a9f903224079ecdd : SUCCESS in 54m 11s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/da247095e32c434398dc463a9e02f3f5 : SUCCESS in 1h 07m 09s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/360e6422943742ceb0be71a88bce0294 : SUCCESS in 34m 29s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0a8bcecb192e46fb86ace2709820d623 : FAILURE in 21m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dc0c4d49c060f09ccbb9a51833c6e727580c4213","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-04-11 05:21:19.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/983154/4/upstream-check-manila/manila-tempest-plugin-ontap-dhss/3b09bfd/ : POST_FAILURE in 1h 44m 48s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/983154/4/upstream-check-manila/manila-tempest-plugin-ontap-no-dhss/1495cee/ : POST_FAILURE in 1h 57m 11s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"7b5214aedfe0436dcca8d984d9b17cc9edd1a75f","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-04-13 04:56:14.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9a2f81f186f78f6a388cefe4784352f18c62947c","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-04-13 06:01:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c39a3b5d0d649b384c37e740949b6adce9b27ce8","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-04-13 06:49:53.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"718008cebb9911f52c83c8e4e0a93e83120ff1cb","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-04-13 08:32:44.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss finger://managesf.services.openstack.netapp.com:7979/5658fe6196274ee386a75fed429237f7 : POST_FAILURE in 1h 43m 04s\n- manila-tempest-plugin-ontap-no-dhss finger://managesf.services.openstack.netapp.com:7979/ea70b9340a914f048169f46f06c15c95 : POST_FAILURE in 2h 13m 18s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"b24c916357fc4d080d9fbecfc054932a7192f1bc","author":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"date":"2026-04-14 05:32:11.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ca3702496591bd9556a1d3fac39b6cd4fa33acd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-04-14 09:33:36.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"ed2813e54d6c30401d44aa977fd30c8c372d27ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 11:12:16.000000000","message":"Patch Set 5: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/643e9cdda8424f89a9f3c90eebcc4d0f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1099b06cc8a64e23a3213563161a6eff : FAILURE in 2m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/dc15b5218abd4f93b6d2f67733f4a70a : SUCCESS in 1h 14m 46s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/dcc1f4994cee48b0a4a9918995961cf8 : SUCCESS in 1h 33m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f27846c9d5044bf2a03d0da3a65e7abf : SUCCESS in 1h 12m 07s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a547a7a82b0a4ea389b9dbeb2f54c94e : SUCCESS in 1h 02m 47s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/726263af50a84ce1b5c5000369503758 : FAILURE in 16m 37s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fb6bbd621f31405fbdc9169aa7f5a504 : FAILURE in 1h 21m 58s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a21f7120b9344fa28697850e552bfef9 : SUCCESS in 27m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3b3e4cc5831a4f1dab3eb473ba42060d : SUCCESS in 46m 15s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/8e9f274f43644b2c9e0358e7c21acff8 : SUCCESS in 52m 13s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/dc624f180f88416e80705d90da88f3ea : SUCCESS in 1h 08m 46s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/51e5be2c803943359ce73af47dbdb7c5 : SUCCESS in 58m 12s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/5a855f71a0a94c3fa762e13a90dc1b9f : SUCCESS in 1h 09m 05s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/2103334b9ccc472997dcaae23a261362 : SUCCESS in 35m 24s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/481f6a9a43154f26af7f2cad0b04bf54 : FAILURE in 20m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e8fc21c370c672a80fe889129777284c20f1169f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-04-15 01:33:57.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"7aed7fb542221e75c6dd4141c6c2a6957f1b3648","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-04-15 01:51:58.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"bfe2f6c82f0b96436dbe6d9c260277c0c8964bff","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-04-15 02:45:35.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss finger://managesf.services.openstack.netapp.com:7979/f510e35caa584c7b8c07130cfd670058 : POST_FAILURE in 33m 56s\n- manila-tempest-plugin-ontap-no-dhss finger://managesf.services.openstack.netapp.com:7979/0fe3847458974ff5bc0f4a4bc2d2cce8 : POST_FAILURE in 34m 26s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"cc11c63c1d4981a7e44c55abfb5045de452969ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 03:55:54.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6bc56a7a61a4e2cbe2d9de79035adf4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35402aadb0bf4feaa75b32790ba21db9 : SUCCESS in 2m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fa6eb5ea22e04f63a94933ce17137128 : SUCCESS in 37m 30s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c8156e353efe45fe85d0fe6917e292e0 : FAILURE in 2h 16m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/58bc70bc920e418fb670958590f161e7 : SUCCESS in 1h 37m 43s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f62543b43801422a89cf256366eba92d : SUCCESS in 29m 06s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/caf6466d34984f8d9fecea662def5f39 : SUCCESS in 42m 53s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b7d5d45b1e874ad2ba317f458305bf2e : FAILURE in 1h 22m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/53e944131c7f46bd8f7d5a4a68e26478 : SUCCESS in 20m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8f4fbf33e817461e8278b4d4323d29c8 : SUCCESS in 40m 29s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/46cf36664cfe493d8740de162b1c8a5d : FAILURE in 17m 40s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/98ae830c9a644fa1b3f9c1a64e5032c7 : SUCCESS in 1h 13m 14s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/1f13208f7b834bf89fc2735bf3d62b76 : SUCCESS in 1h 17m 41s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/8792004629874820aa59a541892fe5ae : SUCCESS in 1h 02m 56s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/bd3022016473462a9e2ec0f3cabead55 : SUCCESS in 32m 52s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f033e24c311a4eb39cde0c0ef710f519 : FAILURE in 16m 36s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"748da821fbf2c0ad5e83adbb60e22f9c67a087a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-04-15 06:12:14.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"3964a93a91a6cf3ff0acd738a6aa59c3072f838e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 07:43:57.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31acf3c738e14a16a86914d62edbbb6e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfb36a4710c2435195d2f5d9fc5801cc : SUCCESS in 2m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/29e7896197074b8cac4c9686911f3284 : SUCCESS in 1h 17m 09s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b48fbfcf3f5447e0838cf2ae3955781a : SUCCESS in 54m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8e2c3e5d5ee346728d7cd9f11d7ccdf3 : SUCCESS in 33m 22s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ee27fe8d594e4befa46604a8c465ba08 : SUCCESS in 43m 48s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/7dafb1195bba47ff95c74713a14cccaa : SUCCESS in 44m 55s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/17c688b150464d27a1a44deb86c8ba49 : FAILURE in 1h 09m 14s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4f4a175392d24c6abf456189dc2a45bf : SUCCESS in 21m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/782886cf929f408bacdfb15231dc403f : SUCCESS in 28m 33s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/83801644605c46b0ba43712aa41f7be8 : SUCCESS in 1h 10m 26s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/6cc7fb5773cf4f8bb36599bd6385bbdc : SUCCESS in 35m 45s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/41dd80ddd945465ea1e98c94b99a9f5e : SUCCESS in 1h 26m 16s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/29a519017dd44b6fb0b7310e1fc4148c : SUCCESS in 1h 16m 26s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/05ec9e3b308542aa93a0cb24ffc18d2f : SUCCESS in 31m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/42c768f20d724b4abe3ff2241cc79c79 : FAILURE in 20m 23s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fd41e51d6cfa96f2005845e8f74963b8061b9da8","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-04-15 09:02:30.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/983154/7/upstream-check-manila/manila-tempest-plugin-ontap-dhss/472db2e/ : SUCCESS in 1h 12m 27s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/983154/7/upstream-check-manila/manila-tempest-plugin-ontap-no-dhss/2234876/ : FAILURE in 1h 15m 56s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"47de4f4d6a5677658ef27aa705fe862908759381","author":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"date":"2026-04-15 11:48:53.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"81232ac417355e7c64f7a471e51f83cb251e34ff","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-04-21 11:32:17.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3b45241458c3351ef3312b09d855f56c7c347ce6","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-19 21:11:07.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"51461bde4d8ffb1ef4bd847529de04981ecd414b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-05-22 04:28:55.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"6de0187f57a9181a2c38820f29208314553e585f","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-05-22 04:34:42.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"31c0b25c4c6da7ec321103806618e0dff28b0b65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 06:17:54.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c40fc7bdc0a4c689215181bceac9d04\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/530598dcd85044b98a2c7512d5cc7e5e : SUCCESS in 2m 08s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2b52d4088ebf4fd782b3b7890c6356b4 : SUCCESS in 48m 44s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b249fe74d2fc4ad1b6c1ae2af925ffa5 : FAILURE in 1h 47m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/077f7559c2c14df3a82af5a77003f30f : SUCCESS in 51m 43s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e9f532b1e26a470d998f593156bf5dd3 : SUCCESS in 43m 12s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/86fa2fa658544654a93dd092082a8d8d : SUCCESS in 1h 32m 35s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/33e96d007f3e49c4bd7c385e9c8cebc2 : FAILURE in 54m 33s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2886a333b41344109c868e5bb664e383 : SUCCESS in 32m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5e6363ef301b49f79c3f1e36deaee4d9 : SUCCESS in 30m 10s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/50cb0fe5601248f3a75c28914a0a0fc4 : SUCCESS in 31m 42s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/e69dbce3891241068664788528235927 : FAILURE in 1h 36m 07s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/de677a3cddf748f3b432c1a73191bac7 : SUCCESS in 32m 59s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/dfdb72f65b384cd183b68c676b1e6161 : SUCCESS in 31m 23s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e9e5b56f1bf84468885cd983f4a7fc15 : FAILURE in 9m 57s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2ce5b706acff9749797e8b24a0026b519da64ada","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-05-22 11:25:11.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8d26982b7fee7b865ad2d230b6b8f8c17b905392","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 13:51:36.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fefa70970f44acc9fa49b90137a1ecb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e21d838807674ef9a07e2993f5e81248 : SUCCESS in 3m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d065f02d18cc4d50b3c069646611b50f : SUCCESS in 1h 15m 50s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/554066555c8c4759b4ac09a31c2641fc : FAILURE in 2h 20m 57s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a4472f1e5d3a445da6fefc96f6573862 : SUCCESS in 1h 26m 05s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b6e27bf545ca40e983fd2c7c34ff8203 : SUCCESS in 34m 50s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a692ddfdfc26450a9e6647fc16ef8d42 : SUCCESS in 1h 57m 00s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ef3387a8eda6452b95691b1e6aa22436 : FAILURE in 1h 20m 42s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/208aa9574d57472d8fb6ef6478910858 : SUCCESS in 21m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/170f32088a254016ba7d59c2e0b0604c : SUCCESS in 48m 43s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/2beacf6da6dc4c3699a735028566403e : SUCCESS in 1h 04m 33s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/b64e109ec1f144f8874b41d5f6304420 : SUCCESS in 1h 16m 07s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/f155a9af4ed345bebe1fd9ba6525dc77 : SUCCESS in 32m 06s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/7e3d009415a1437e8dc2d05ae6b3435e : SUCCESS in 19m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/130813b188d64dafadeecb18c51f2f57 : FAILURE in 11m 35s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"db2e53374ba3445b90ed095c56a3e79a0c7c8b9a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-05-22 15:07:20.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/983154/8/upstream-check-manila/manila-tempest-plugin-ontap-dhss/8ea25f7/ : SUCCESS in 1h 13m 21s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"5d01300617aecf1d30931f502452504530add76c","author":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"date":"2026-05-26 04:30:57.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"45c3e534a839f24d0ece316c8e79e63e7c3fd41f","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-05-26 09:00:22.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8d141bd335f8a31d925dfdb6efe2697aeb0acd90","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-05-26 14:09:39.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6989bb48678767a92f2ab944f368c908a32ca51f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-06-04 19:08:24.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"f7e7a14d9da7eb4625d82b421aa01e4fc3efa38a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 19:08:53.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"49263742a6e559fdb1b9bd204d608c07b4430947","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 19:49:40.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5493a4d388fc49229b8eb422b9a028a1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e98871367874a3ea9631219ac94df48 : SUCCESS in 2m 16s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e642720010a2468fb786246385527c9b : SUCCESS in 40m 02s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/12a048311a234e51997702267482f2b6 : SUCCESS in 23m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/38c567d389a54d7f8d680f55fce66b75 : SUCCESS in 25m 38s","accounts_in_message":[],"_revision_number":8},{"id":"b51f746d9e0dd097139b3dda6aabb017096c0ae3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 19:49:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"66d0a19ce9f564071cd2ee504221764b041213c7","revisions":{"4511cc1c173487e329a70464f69bbe7d5391368e":{"kind":"REWORK","_number":1,"created":"2026-04-02 04:31:58.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/54/983154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/54/983154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/1"}}},"commit":{"parents":[{"commit":"fd065035885b60e15e899cde36f8442150bfe2ea","subject":"Add tempest tests for share replica metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fd065035885b60e15e899cde36f8442150bfe2ea"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2026-04-02 04:29:12.000000000","tz":-240},"committer":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2026-04-02 04:31:24.000000000","tz":-240},"subject":"Add replication_policy metadata tests for SnapMirror policies","message":"Add replication_policy metadata tests for SnapMirror policies\n\nAdd tempest tests for the replication_policy replica metadata key\nintroduced in microversion 2.95, covering the three SnapMirror\npolicies: MirrorAllSnapshots, Sync, and StrictSync.\n\nChange-Id: Iaa0e33dde629d4321adc3317bb9fa4ffa1d4b853\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4511cc1c173487e329a70464f69bbe7d5391368e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4511cc1c173487e329a70464f69bbe7d5391368e"}]},"branch":"refs/heads/master"},"b99c39293e4bf1fc5c2126ad773dae145d549307":{"kind":"REWORK","_number":2,"created":"2026-04-10 07:34:06.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/54/983154/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/54/983154/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/2"}}},"commit":{"parents":[{"commit":"fd065035885b60e15e899cde36f8442150bfe2ea","subject":"Add tempest tests for share replica metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fd065035885b60e15e899cde36f8442150bfe2ea"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2026-04-02 04:29:12.000000000","tz":-240},"committer":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2026-04-07 06:07:21.000000000","tz":-240},"subject":"Add replication_policy metadata tests for SnapMirror policies","message":"Add replication_policy metadata tests for SnapMirror policies\n\nAdd tempest tests for the replication_policy replica metadata key\nintroduced in microversion 2.95.\n\nChange-Id: Iaa0e33dde629d4321adc3317bb9fa4ffa1d4b853\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b99c39293e4bf1fc5c2126ad773dae145d549307"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b99c39293e4bf1fc5c2126ad773dae145d549307"}]},"branch":"refs/heads/master"},"9c6189bfa8928ce449d29ee413c2d8c811d1d566":{"kind":"REWORK","_number":3,"created":"2026-04-10 11:31:31.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/54/983154/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/54/983154/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/3"}}},"commit":{"parents":[{"commit":"d03e1521c2307f7340252f7797c9df071bb67236","subject":"Merge \"Add share server subnet filter tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d03e1521c2307f7340252f7797c9df071bb67236"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2026-04-02 04:29:12.000000000","tz":-240},"committer":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2026-04-10 10:34:13.000000000","tz":-240},"subject":"Add replication_policy metadata tests for SnapMirror policies","message":"Add replication_policy metadata tests for SnapMirror policies\n\nAdd tempest tests for the replication_policy replica metadata key\nintroduced in microversion 2.95.\n\nChange-Id: Iaa0e33dde629d4321adc3317bb9fa4ffa1d4b853\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c6189bfa8928ce449d29ee413c2d8c811d1d566"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c6189bfa8928ce449d29ee413c2d8c811d1d566"}]},"branch":"refs/heads/master"},"8b4d8f6eff9c3efa1bd7265b147fc70efd6ec137":{"kind":"REWORK","_number":4,"created":"2026-04-11 03:02:35.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/54/983154/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/54/983154/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/4"}}},"commit":{"parents":[{"commit":"d03e1521c2307f7340252f7797c9df071bb67236","subject":"Merge \"Add share server subnet filter tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d03e1521c2307f7340252f7797c9df071bb67236"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2026-04-02 04:29:12.000000000","tz":-240},"committer":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2026-04-11 03:02:25.000000000","tz":-240},"subject":"Add replication_policy metadata tests for SnapMirror policies","message":"Add replication_policy metadata tests for SnapMirror policies\n\nAdd tempest tests for the replication_policy replica metadata key\nintroduced in microversion 2.95.\n\nChange-Id: Iaa0e33dde629d4321adc3317bb9fa4ffa1d4b853\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8b4d8f6eff9c3efa1bd7265b147fc70efd6ec137"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8b4d8f6eff9c3efa1bd7265b147fc70efd6ec137"}]},"branch":"refs/heads/master"},"e30321caf1c6cec67702573462efc0dbe9132ed7":{"kind":"REWORK","_number":5,"created":"2026-04-14 09:33:36.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/54/983154/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/54/983154/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/5"}}},"commit":{"parents":[{"commit":"d03e1521c2307f7340252f7797c9df071bb67236","subject":"Merge \"Add share server subnet filter tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d03e1521c2307f7340252f7797c9df071bb67236"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2026-04-02 04:29:12.000000000","tz":-240},"committer":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2026-04-14 09:33:15.000000000","tz":-240},"subject":"Add replication_policy metadata tests for SnapMirror policies","message":"Add replication_policy metadata tests for SnapMirror policies\n\nAdd tempest tests for the replication_policy replica metadata key\nintroduced in microversion 2.95.\n\nChange-Id: Iaa0e33dde629d4321adc3317bb9fa4ffa1d4b853\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e30321caf1c6cec67702573462efc0dbe9132ed7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e30321caf1c6cec67702573462efc0dbe9132ed7"}]},"branch":"refs/heads/master"},"b88d3e0b01d690f865f4d754daca24a49a0eca6c":{"kind":"REWORK","_number":6,"created":"2026-04-15 01:33:57.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/54/983154/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/54/983154/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/6"}}},"commit":{"parents":[{"commit":"d03e1521c2307f7340252f7797c9df071bb67236","subject":"Merge \"Add share server subnet filter tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d03e1521c2307f7340252f7797c9df071bb67236"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2026-04-02 04:29:12.000000000","tz":-240},"committer":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2026-04-15 01:33:42.000000000","tz":-240},"subject":"Add replication_policy metadata tests for SnapMirror policies","message":"Add replication_policy metadata tests for SnapMirror policies\n\nAdd tempest tests for the replication_policy replica metadata key\nintroduced in microversion 2.95.\n\nChange-Id: Iaa0e33dde629d4321adc3317bb9fa4ffa1d4b853\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b88d3e0b01d690f865f4d754daca24a49a0eca6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b88d3e0b01d690f865f4d754daca24a49a0eca6c"}]},"branch":"refs/heads/master"},"47dacb2c575474a0888c42f60f2c7490b3cebfeb":{"kind":"REWORK","_number":7,"created":"2026-04-15 06:12:14.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/54/983154/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/54/983154/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/7"}}},"commit":{"parents":[{"commit":"d03e1521c2307f7340252f7797c9df071bb67236","subject":"Merge \"Add share server subnet filter tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d03e1521c2307f7340252f7797c9df071bb67236"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2026-04-02 04:29:12.000000000","tz":-240},"committer":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2026-04-15 06:12:05.000000000","tz":-240},"subject":"Add replication_policy metadata tests for SnapMirror policies","message":"Add replication_policy metadata tests for SnapMirror policies\n\nAdd tempest tests for the replication_policy replica metadata key\nintroduced in microversion 2.95.\n\nChange-Id: Iaa0e33dde629d4321adc3317bb9fa4ffa1d4b853\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/47dacb2c575474a0888c42f60f2c7490b3cebfeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/47dacb2c575474a0888c42f60f2c7490b3cebfeb"}]},"branch":"refs/heads/master"},"66d0a19ce9f564071cd2ee504221764b041213c7":{"kind":"REWORK","_number":8,"created":"2026-05-22 04:28:55.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/54/983154/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/54/983154/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/54/983154/8"}}},"commit":{"parents":[{"commit":"d03e1521c2307f7340252f7797c9df071bb67236","subject":"Merge \"Add share server subnet filter tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d03e1521c2307f7340252f7797c9df071bb67236"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2026-04-02 04:29:12.000000000","tz":-240},"committer":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2026-05-21 11:04:21.000000000","tz":-240},"subject":"Add replication_policy metadata tests for SnapMirror policies","message":"Add replication_policy metadata tests for SnapMirror policies\n\nAdd tempest tests for the replication_policy replica metadata key\nintroduced in microversion 2.95.\n\nChange-Id: Iaa0e33dde629d4321adc3317bb9fa4ffa1d4b853\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/66d0a19ce9f564071cd2ee504221764b041213c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/66d0a19ce9f564071cd2ee504221764b041213c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
