)]}'
{"id":"openstack%2Fmanila-tempest-plugin~986039","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I0ff8921245e22b977d1bd88f8705c35274a8d6ff","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-27 08:41:51.000000000","reason":"Someone else replied on the change"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2026-08-27 01:19:17.000000000","reason":"Vote got outdated and was removed: Code-Review+2"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I0ff8921245e22b977d1bd88f8705c35274a8d6ff","subject":"Drop direct reference to [identity] auth_version option","status":"NEW","created":"2026-04-24 07:20:57.000000000","updated":"2026-08-27 08:41:51.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6b0fa7b9555d5d2839a4b655023ed7324030c810","_number":986039,"virtual_id_number":986039,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-27 08:32:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-24 09:43:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-27 13:51:58.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2026-08-27 08:41:51.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"ced9ae4207704fbe1e1da267216adad07b749a5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-24 07:20:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aeb3756d69de41f69162d7fcfbf35101e83dfa09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-24 07:23:55.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6212c3ee43aa1501f1602adea861ee133705cca0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 09:43:12.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c0ce1ab1f8d4953a2a6ffba8d2bea0b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e8c1aa0339e430cb5c7b2b370969c5a : FAILURE in 1m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4a95a32b988c45e2b86884bd7d899ba9 : FAILURE in 1h 13m 18s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/59269b59d0e54abbb6206d987c16b50a : FAILURE in 2h 15m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f9aeaa7bb2f24c5ca32d68a3d11c205c : FAILURE in 29m 43s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/529e8b71217945e79062476fdc20dc33 : FAILURE in 1h 16m 06s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f58a80fab20d466a9b3d909b6c18d6f7 : FAILURE in 1h 30m 15s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e50e672dd2d34726816872b88269ab2e : FAILURE in 1h 07m 43s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/de2b91e7b2794a65b9478ce36cb36e5c : FAILURE in 20m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/889578b0ab064b61bb60d28a67912e4b : FAILURE in 26m 00s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/b0abb975ba37432387655a12183de15b : FAILURE in 1h 12m 48s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/700fee6e5dce4372b7ae2b2748879bc6 : FAILURE in 30m 08s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/a14335b81b874145b46ff9e0f047cba6 : FAILURE in 1h 05m 02s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/4d2db49f03a14d02b1c57ade5cdf6da1 : FAILURE in 1h 09m 35s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3d47ea8777574d26be10783479db6411 : FAILURE in 20m 12s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ed7e55e6cc964bc7a8b1d30bc6595854 : FAILURE in 27m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cce0449273cb6d15da35196b0c24199c5a81c2d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-24 15:32:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"3db00b83b0a783027977e08ceea0a5601bb54a42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 17:43:48.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f938089bfae44233ad377f8a774b3875\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4ef59121ccb4d1ba9144f8e0908093f : FAILURE in 2m 03s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9293310e593c4d33afb23c1c9a0162ae : SUCCESS in 34m 23s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d6c143d89d394bc989e48bbbcdebb874 : FAILURE in 2h 04m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/17b9ed4a24094eb4b1f0a6483695ec92 : SUCCESS in 1h 27m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/0ec6a28bc57f4a55a14d18f947574208 : FAILURE in 1h 34m 24s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/44362787fd5a45e7bc7290a3e8d0772f : FAILURE in 1h 58m 38s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7e042cae34d849eca4427b3d4b1a8cf6 : FAILURE in 1h 11m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d0a5ca1ca4104410a990aeb762c79598 : SUCCESS in 36m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/90040e288cc7470484715117d1d4b17a : FAILURE in 41m 20s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/0da2674663564bf6947b155784e84d88 : SUCCESS in 1h 10m 49s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/2ac368a5b55b4ae0adaa4f8262392e64 : SUCCESS in 1h 15m 14s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/5f1258980a9f4acabcc02556d93d809b : SUCCESS in 1h 10m 14s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/097cf6530a7e4641afdeac93c59242b6 : SUCCESS in 1h 28m 18s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/59208fdce3654b5f8fe836019a56d454 : SUCCESS in 22m 58s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/57c0256e043f4bc3b26c51ea9e4d829f : FAILURE in 22m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"415be280e9f3afb71e274eb83cb2be2f852da77f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-25 09:57:10.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"894a78748b8ad51c613685069bed7f6e7c514909","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-25 11:55:43.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fc296464ddb44f3ac2a043e175d794c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b4d8bcb901f4f739106c47ed4dc155a : FAILURE in 2m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2e32f92c9b484caa96752d027dce7012 : SUCCESS in 1h 10m 55s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d18231e59efe4417b9a7adfee3f56877 : SUCCESS in 1h 45m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2ecaf27c59424044b7f9ce3cb91bb742 : SUCCESS in 1h 42m 17s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/5284e7017629462ab446eb2763c8adbf : FAILURE in 1h 17m 03s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/3301d3a2ac774dbab0e2ce1621e85856 : FAILURE in 1h 50m 19s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9c777be298ad495ba9e4710fa304035f : FAILURE in 1h 21m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/61e4d418dbd945dabeb3effb95c755aa : SUCCESS in 35m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/282af711fd184f999cf593e4550f4548 : FAILURE in 40m 17s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/67d0dbd62e8141d09db0e52b2ce1b30f : SUCCESS in 58m 43s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/fd09eb822c974cf3b158eb4e0821be82 : SUCCESS in 1h 19m 49s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/0c27f4596cb7475d87565748f2d74be8 : SUCCESS in 1h 00m 07s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/fd433d1e29014e8ab2647abb39134c5a : SUCCESS in 1h 05m 43s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/0f070288df4e4c5c80276ca676dfa27a : SUCCESS in 38m 18s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/194faaac11864c559ccdffd45b6531aa : FAILURE in 18m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"736aeca3c7ec8f0929ec3d1ae80a949363fe3140","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-25 12:47:06.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"3cd53799d5c7a73f4c910f88a0139af2db4888e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-25 15:11:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6c1ef8c16d6426a8b82f1dc81faa982\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ec69715664f4d9a8aa1abf82ebcfc31 : SUCCESS in 2m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/40bffa58018442609bc5e37495ef9bb3 : SUCCESS in 48m 03s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8c9d2d1aa4604db6ac21ed0954911057 : SUCCESS in 2h 19m 11s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/47c743439f5c4a97a581d52ff524714f : SUCCESS in 1h 44m 17s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b141a92c02a3451e99e8d842cf666f88 : FAILURE in 1h 08m 17s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f0d3af9a97f5471b9160d8ceaa99182c : FAILURE in 1h 23m 19s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8f1fa8b5a6cd4a2ca301e212a005aa04 : FAILURE in 1h 13m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/56a02a0cd2f740e798a953a9908f4ecd : SUCCESS in 31m 53s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/82d78e2a1d31493d8dfef0d4eb903612 : SUCCESS in 23m 01s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/2f20c7575ebe49328fd8884d04c60b50 : SUCCESS in 1h 19m 45s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/1153cc061415478ca1a6919f38ce0d83 : SUCCESS in 28m 49s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/ffcb332195204be6a537cde5cc469a76 : SUCCESS in 1h 19m 16s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/9ad14745d36d4629835f6e4d1ca79864 : SUCCESS in 1h 20m 46s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/aec3e107e26f4f298ee96f136753a421 : SUCCESS in 24m 43s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/15ac3d7752d34e94958005fbe820be7e : FAILURE in 10m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"097f66f7caf715a674d5defc419bb0a5c813c4cd","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2026-04-27 13:51:58.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"14c58b49be4418ae483869517e8815e362de17a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-27 01:19:17.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"919d522ebf235f2bd41629a3292383c9c9951f80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 03:20:31.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f47df4b03fd4471d896dc5ef915566a1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3a200dffd2843baa984923b79b99923 : FAILURE in 2m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a560c2ca8e71423f8943955dea2fd3f3 : SUCCESS in 1h 10m 26s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b36344bc155b4fe3bf9049e1a8057717 : SUCCESS in 1h 19m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f04c03f6d6184eaeac2175516f45d2eb : SUCCESS in 1h 08m 05s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a162afcf15ae44f9ac8e23029e898ab1 : FAILURE in 48m 36s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/9942a7a78bf946bd81899c8348911820 : FAILURE in 1h 54m 07s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a3685c21077d4285841ef17968a47f9f : SUCCESS in 1h 12m 47s (non-voting)\n- manila-tempest-plugin-lustre https://zuul.opendev.org/t/openstack/build/bf4fbecfdef94a9f888514944c4a2621 : SUCCESS in 30m 14s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7011301c74204f9aa97cbd50597d6f54 : SUCCESS in 29m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cc71fbe369c54af88361f2adb3ffa0c7 : SUCCESS in 23m 48s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/b02242783d2c48b291381902ef374c06 : SUCCESS in 42m 35s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/b12c7680a5cb4ba2bd08337a7bc7fd63 : SUCCESS in 1h 04m 50s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/1bb0f499279a41c5ab6226cb761aa17b : SUCCESS in 43m 04s\n- manila-tempest-plugin-dummy-no-dhss-old-rbac https://zuul.opendev.org/t/openstack/build/40ad875a1cd84f5d8008d6fcdfdb1092 : SUCCESS in 22m 06s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d2380ab0fa9a46dba7d37b8e473bf75d : SUCCESS in 1h 00m 38s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ade29d39b38cb972eadf780546f7bfdf97b016e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-27 06:27:43.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"ba79654b1abba7ee49b4539c590fab4b89d43058","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 08:32:21.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7251529a264404cbc073ea95bc977e8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19418fd3a1f04cc285e861f807822b7e : SUCCESS in 2m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2c5077a6cae042249f78d3d5244c9aea : SUCCESS in 1h 01m 55s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d085c424dd664488bf7a59e4b06de814 : SUCCESS in 1h 59m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/223163c50f214b06bc285cd9e021f8c3 : SUCCESS in 1h 27m 51s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8dedb15463be4fa8966415174ca56ea2 : FAILURE in 29m 52s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8abeb1b09fe04f8283388772ff389e52 : FAILURE in 1h 18m 46s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1b4018a6660a448d8aff5db42e751058 : SUCCESS in 1h 00m 57s (non-voting)\n- manila-tempest-plugin-lustre https://zuul.opendev.org/t/openstack/build/feccbb3a5d06428e81dbfd53b6b2a1a2 : SUCCESS in 54m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0b39516aee7f4e2094b9ad69e5c80d1f : SUCCESS in 37m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9968d7a849674106ae18979bdd357d72 : SUCCESS in 35m 18s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/38cd40bdd7854a89a97fa6ff2f373353 : SUCCESS in 1h 06m 17s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/4971c77f6a2f442db9eccf411d57701b : SUCCESS in 1h 00m 46s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/de26174f6b794c0db50b15c87adb4a6f : SUCCESS in 1h 05m 00s\n- manila-tempest-plugin-dummy-no-dhss-old-rbac https://zuul.opendev.org/t/openstack/build/19880874e3674ac1bc6c0f5b4410df2a : SUCCESS in 22m 23s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/72607a825a234c10a113906582c5d1bb : SUCCESS in 28m 06s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6b0fa7b9555d5d2839a4b655023ed7324030c810","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-08-27 08:41:51.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/39/986039/7/upstream-check-manila/manila-tempest-plugin-ontap-dhss/c33db13/ : RETRY_LIMIT in 6m 56s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"b7947f1cbc837024ffbda3e7fd49c2268e2dbabd","revisions":{"1e99b9f9759a9f59f6e92b22f8f3daf5df8e8017":{"kind":"REWORK","_number":1,"created":"2026-04-24 07:20:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/986039/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/39/986039/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/1"}}},"commit":{"parents":[{"commit":"c42be469136bef44587426464c2b24acd4b69309","subject":"Merge \"Wait for floating IP association before SSH in scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c42be469136bef44587426464c2b24acd4b69309"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-24 07:20:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-24 07:20:47.000000000","tz":540},"subject":"Drop direct reference to [identity] uri option","message":"Drop direct reference to [identity] uri option\n\n... to avoid compatibility problem caused by this option, which is\nbeing removed from tempest.\n\nChange-Id: I0ff8921245e22b977d1bd88f8705c35274a8d6ff\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1e99b9f9759a9f59f6e92b22f8f3daf5df8e8017"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1e99b9f9759a9f59f6e92b22f8f3daf5df8e8017"}]},"branch":"refs/heads/master"},"96e297f452601e2a5c5aa0454db65da862d5a792":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-04-24 07:23:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/986039/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/39/986039/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/2"}}},"commit":{"parents":[{"commit":"c42be469136bef44587426464c2b24acd4b69309","subject":"Merge \"Wait for floating IP association before SSH in scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c42be469136bef44587426464c2b24acd4b69309"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-24 07:20:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-24 07:23:44.000000000","tz":540},"subject":"Drop direct reference to [identity] auth_version option","message":"Drop direct reference to [identity] auth_version option\n\n... to avoid compatibility problem caused by this option, which is\nbeing removed from tempest.\n\nChange-Id: I0ff8921245e22b977d1bd88f8705c35274a8d6ff\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/96e297f452601e2a5c5aa0454db65da862d5a792"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/96e297f452601e2a5c5aa0454db65da862d5a792"}]},"branch":"refs/heads/master"},"ecb7201c77041f676d976ab201134d2bcc121f15":{"kind":"REWORK","_number":3,"created":"2026-04-24 15:32:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/986039/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/39/986039/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/3"}}},"commit":{"parents":[{"commit":"c42be469136bef44587426464c2b24acd4b69309","subject":"Merge \"Wait for floating IP association before SSH in scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c42be469136bef44587426464c2b24acd4b69309"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-24 07:20:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-24 15:32:08.000000000","tz":540},"subject":"Drop direct reference to [identity] auth_version option","message":"Drop direct reference to [identity] auth_version option\n\n... to avoid compatibility problem caused by this option, which is\nbeing removed from tempest.\n\nChange-Id: I0ff8921245e22b977d1bd88f8705c35274a8d6ff\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ecb7201c77041f676d976ab201134d2bcc121f15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ecb7201c77041f676d976ab201134d2bcc121f15"}]},"branch":"refs/heads/master"},"14c4866b18adf7d0ce9614d0a32739f4e4e78f48":{"kind":"REWORK","_number":4,"created":"2026-04-25 09:57:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/986039/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/39/986039/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/4"}}},"commit":{"parents":[{"commit":"c42be469136bef44587426464c2b24acd4b69309","subject":"Merge \"Wait for floating IP association before SSH in scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c42be469136bef44587426464c2b24acd4b69309"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-24 07:20:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-25 09:56:59.000000000","tz":540},"subject":"Drop direct reference to [identity] auth_version option","message":"Drop direct reference to [identity] auth_version option\n\n... to avoid compatibility problem caused by this option, which is\nbeing removed from tempest.\n\nChange-Id: I0ff8921245e22b977d1bd88f8705c35274a8d6ff\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/14c4866b18adf7d0ce9614d0a32739f4e4e78f48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/14c4866b18adf7d0ce9614d0a32739f4e4e78f48"}]},"branch":"refs/heads/master"},"8d2ab90f5c337e29ecbcae91eb6c6f65ea55a6b9":{"kind":"REWORK","_number":5,"created":"2026-04-25 12:47:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/986039/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/39/986039/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/5"}}},"commit":{"parents":[{"commit":"c42be469136bef44587426464c2b24acd4b69309","subject":"Merge \"Wait for floating IP association before SSH in scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c42be469136bef44587426464c2b24acd4b69309"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-24 07:20:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-25 12:46:48.000000000","tz":540},"subject":"Drop direct reference to [identity] auth_version option","message":"Drop direct reference to [identity] auth_version option\n\n... to avoid compatibility problem caused by this option, which is\nbeing removed from tempest.\n\nChange-Id: I0ff8921245e22b977d1bd88f8705c35274a8d6ff\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8d2ab90f5c337e29ecbcae91eb6c6f65ea55a6b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8d2ab90f5c337e29ecbcae91eb6c6f65ea55a6b9"}]},"branch":"refs/heads/master"},"976a2b6ca3fc08ffae3b155fbd2285646b16665e":{"kind":"REWORK","_number":6,"created":"2026-08-27 01:19:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/986039/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/39/986039/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/6"}}},"commit":{"parents":[{"commit":"d753c5552b822754c1d0256fd6713ef3bb58bc27","subject":"Merge \"Use manila-test-image for non-generic driver jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d753c5552b822754c1d0256fd6713ef3bb58bc27"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-24 07:20:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-27 01:18:52.000000000","tz":540},"subject":"Drop direct reference to [identity] auth_version option","message":"Drop direct reference to [identity] auth_version option\n\n... to avoid compatibility problem caused by this option, which is\nbeing removed from tempest.\n\nChange-Id: I0ff8921245e22b977d1bd88f8705c35274a8d6ff\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/976a2b6ca3fc08ffae3b155fbd2285646b16665e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/976a2b6ca3fc08ffae3b155fbd2285646b16665e"}]},"branch":"refs/heads/master"},"b7947f1cbc837024ffbda3e7fd49c2268e2dbabd":{"kind":"REWORK","_number":7,"created":"2026-08-27 06:27:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/986039/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/39/986039/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/39/986039/7"}}},"commit":{"parents":[{"commit":"d753c5552b822754c1d0256fd6713ef3bb58bc27","subject":"Merge \"Use manila-test-image for non-generic driver jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d753c5552b822754c1d0256fd6713ef3bb58bc27"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-24 07:20:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-27 06:27:43.000000000","tz":0},"subject":"Drop direct reference to [identity] auth_version option","message":"Drop direct reference to [identity] auth_version option\n\n... to avoid compatibility problem caused by this option, which is\nbeing removed from tempest.\n\nChange-Id: I0ff8921245e22b977d1bd88f8705c35274a8d6ff\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b7947f1cbc837024ffbda3e7fd49c2268e2dbabd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b7947f1cbc837024ffbda3e7fd49c2268e2dbabd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
