)]}'
{"id":"openstack%2Fmanila-tempest-plugin~986761","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Icbff80bb1f34c5f3ea959fadd390a29b0b5fff25","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bp/graduate-share-migration-feature","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-05-21 21:15:09.000000000","reason":"Change was submitted"},"38371":{"account":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"last_update":"2026-05-21 21:15:09.000000000","reason":"Change was submitted"},"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2026-05-21 21:15:09.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-05-21 19:44:03.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2026-05-21 08:36:41.000000000","reason":"\u003cGERRIT_ACCOUNT_36180\u003e replied on the change","reason_account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"}},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-05-21 21:15:09.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-05-21 21:15:09.000000000","reason":"Change was submitted"},"38479":{"account":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"last_update":"2026-05-21 21:15:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icbff80bb1f34c5f3ea959fadd390a29b0b5fff25","subject":"Graduate share migration feature","status":"MERGED","created":"2026-04-30 09:42:10.000000000","updated":"2026-05-21 21:15:09.000000000","submitted":"2026-05-21 21:15:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986761-bp/graduate-share-migration-feature","meta_rev_id":"2fc96876debd1b2a6515c445d254390c9e4c059c","_number":986761,"virtual_id_number":986761,"owner":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-21 21:15:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2026-05-21 19:44:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2026-05-21 08:36:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2026-05-21 19:44:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-30 13:25:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-05 05:08:44.000000000","updated_by":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-05-05 05:08:44.000000000","updated_by":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"REVIEWER"},{"updated":"2026-05-05 05:08:44.000000000","updated_by":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-05-05 05:08:44.000000000","updated_by":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"reviewer":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-05-05 05:08:44.000000000","updated_by":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-05-05 05:08:44.000000000","updated_by":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-05-05 05:08:44.000000000","updated_by":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2026-05-20 10:50:08.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"cd9af4a5bc257df7d020ef77596fa5f509ca205c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-04-30 09:42:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f83eb131e3838b0cb53c4998c3ab8429977a14a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-04-30 10:56:12.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9b1ab614f10a35433bb7627a362d4566bf058c0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-04-30 10:59:34.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"9d6a3db23b757400b26ca24673a5fdcf0a841540","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-04-30 11:57:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4e0b2f08145f8597882485ed4a94ff95d927f8fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 13:25:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2568ae6c47334dad948c75f0a315138e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/780ae4b89ee14ed28cefc17792b6ec92 : SUCCESS in 3m 16s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c64681f3c4a143459d1c31085a69b015 : FAILURE in 30m 02s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/23376bd6b74f46b081ed98eafea8537e : FAILURE in 35m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/08ca864ee200436ea6a141ad61084a97 : FAILURE in 33m 06s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/04b267d2bb6346eda2632755d56cd84c : FAILURE in 39m 47s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e2b5913f841a49349059d91461653c98 : FAILURE in 50m 02s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d547ed1b12824d3e94de4e5b2a1c0101 : FAILURE in 27m 26s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/aaed8ea7e3764a8fad9621171ba2d16b : FAILURE in 16m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/643380b133bd430aa0924e6134015e28 : FAILURE in 16m 56s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/94bfd6c6326541fab3506ff880f5128c : SUCCESS in 1h 21m 47s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/1d799de5946049f19579a038ffd1ca30 : SUCCESS in 1h 07m 24s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/5273e5a148bc40aeb7d3217e3b4c3d07 : SUCCESS in 33m 34s\n- manila-tempest-plugin-lvm-2024.2 https://zuul.opendev.org/t/openstack/build/9722853a07e14e8c8f0c4ad2e2857f38 : SUCCESS in 1h 00m 01s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ab3926fc51294cdabf1643b61971abf6 : FAILURE in 15m 51s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/75549b035ebd4db9bd956ec6b9aa6b2e : FAILURE in 21m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2af1d5d5f8880937e6d129d0ca84424d2784baea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-05-05 02:36:14.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"6b6d25c4b311e90d44fbb1818a23f792c70d1c5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 03:12:50.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/manila/+/986760 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"5658849fc24950072143c772d0c8f89fa8b400bb","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-05-05 05:07:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a140dc62157ad1b8b87051304eb729f5f7d719cd","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-05-05 11:03:36.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d865bb1da232eb1ef3084f01918f9bc85850b357","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-19 06:38:18.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ca5e33d2df61a43e2291243f44353a050cfdfd13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-05-20 07:37:00.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"1626d9c485ed54697ec0bd540e1d883dd67d0486","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-05-20 07:37:09.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"77ca50178f7913eef539dfd824d9037ebf07a90e","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-05-20 07:37:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9a405824dd535f14cdaf55d71f46d6a4b66ef273","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 09:27:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dc861d8bed944f8bb495976565d79ef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bad959f0a854e28a111e802441ba70d : SUCCESS in 3m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4aa968d1c95f428fb5fef6e30e71fde8 : SUCCESS in 35m 57s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ca7ae3a0e5bc46048df775cf010492c6 : SUCCESS in 1h 41m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/12dd6df9fe4a4c1596247ac6ab867493 : SUCCESS in 1h 14m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/1e2ddd3c015f47d0a1d562abe2b4d61a : SUCCESS in 27m 19s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/b12621a352d94854bdc9b3eaeb061c4e : FAILURE in 1h 40m 57s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/217901860ac34024826f8073a36cec09 : FAILURE in 1h 26m 45s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bb7fb696661d421f90fde16b64e3284f : SUCCESS in 31m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2aec8eeba332437aa2d25929805a965c : SUCCESS in 40m 02s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/6c32feeecbf04607af205d3eb4818e66 : SUCCESS in 1h 14m 30s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/1e1431dc4dc64a969f8bec261ba042bd : SUCCESS in 1h 08m 19s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/a5f9c072d637427c991be7bfa5bd31ed : SUCCESS in 53m 35s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/415537f161b64869a4dc342443355695 : SUCCESS in 20m 08s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b3e97ecb2dee4d008545ebd7a226f435 : FAILURE in 17m 38s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d98df688e810ec07e0e275cebd0b6fb7b1ee45d3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-05-20 10:50:08.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/61/986761/6/upstream-check-manila/manila-tempest-plugin-ontap-dhss/15127cc/ : SUCCESS in 1h 21m 35s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"81df6ad9972520530c298f37873982819fa3836c","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-05-21 08:36:41.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6b5626a6cd6656ff2e781526225d16668d50cb40","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-21 19:44:03.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"956cf3fbf570533036580fdf0fca6effbd217100","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 19:44:32.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"5dfc70313490c5badaf2885377a8218f47d82378","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 21:15:09.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da83986fd4bb46e9ad71bd264b2d3bea\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f1a4d1b104e4e819d0011d7797d377e : SUCCESS in 3m 15s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/24013d2cc02f41408b9136f1e9f891bc : SUCCESS in 1h 24m 50s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5f12ac79fa8a4a0aa13ec8d37a1a5f80 : SUCCESS in 25m 42s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/655af3e8effa49dba4e54202bd378c80 : SUCCESS in 35m 28s","accounts_in_message":[],"_revision_number":6},{"id":"2fc96876debd1b2a6515c445d254390c9e4c059c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 21:15:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"dd1b6422b9de8feaac147a9f6828f6af06c7091b","revisions":{"82d524fbcc234464f87c37554ba54655510c0972":{"kind":"REWORK","_number":1,"created":"2026-04-30 09:42:10.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/61/986761/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/986761/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/986761/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/986761/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/986761/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/986761/1"}}},"commit":{"parents":[{"commit":"d03e1521c2307f7340252f7797c9df071bb67236","subject":"Merge \"Add share server subnet filter tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d03e1521c2307f7340252f7797c9df071bb67236"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2026-04-30 09:40:18.000000000","tz":-240},"committer":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2026-04-30 09:40:18.000000000","tz":-240},"subject":"Graduate share migration feature","message":"Graduate share migration feature\n\nRemove experimental flag from share migration tests.\n\nDocImpact\nPartially-implements: bp graduate-share-migration-feature\n\nChange-Id: Icbff80bb1f34c5f3ea959fadd390a29b0b5fff25\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/82d524fbcc234464f87c37554ba54655510c0972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/82d524fbcc234464f87c37554ba54655510c0972"}]},"branch":"refs/heads/master"},"1e5a8606e560f5a2facc18f399858c3f2e8977eb":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-04-30 10:56:12.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/61/986761/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/986761/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/986761/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/986761/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/986761/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/986761/2"}}},"commit":{"parents":[{"commit":"d03e1521c2307f7340252f7797c9df071bb67236","subject":"Merge \"Add share server subnet filter tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d03e1521c2307f7340252f7797c9df071bb67236"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2026-04-30 09:40:18.000000000","tz":-240},"committer":{"name":"Medha Choudhary","email":"medhaj06@gmail.com","date":"2026-04-30 10:56:12.000000000","tz":0},"subject":"Graduate share migration feature","message":"Graduate share migration feature\n\nRemove experimental flag from share migration tests.\n\nDocImpact Partially-implements: bp graduate-share-migration-feature\n\nChange-Id: Icbff80bb1f34c5f3ea959fadd390a29b0b5fff25\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1e5a8606e560f5a2facc18f399858c3f2e8977eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1e5a8606e560f5a2facc18f399858c3f2e8977eb"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5ea3ba0d3647b9d4b822d34143f9a4f4050329ef":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-04-30 10:59:34.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/61/986761/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/986761/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/986761/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/986761/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/986761/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/986761/3"}}},"commit":{"parents":[{"commit":"d03e1521c2307f7340252f7797c9df071bb67236","subject":"Merge \"Add share server subnet filter tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d03e1521c2307f7340252f7797c9df071bb67236"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2026-04-30 09:40:18.000000000","tz":-240},"committer":{"name":"Medha Choudhary","email":"medhaj06@gmail.com","date":"2026-04-30 10:59:34.000000000","tz":0},"subject":"Graduate share migration feature","message":"Graduate share migration feature\n\nThis patch removes the experimental flag from the tests related\nto the share migration feature.\n\nPartially-implements: bp graduate-share-migration-feature\n\nChange-Id: Icbff80bb1f34c5f3ea959fadd390a29b0b5fff25\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\nDepends-On: Iaca76f4cd73c0373f26a56f9e9447f52e59a9c68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5ea3ba0d3647b9d4b822d34143f9a4f4050329ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5ea3ba0d3647b9d4b822d34143f9a4f4050329ef"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7f79830b826b243ac5c1b4786ec0e46e9fc3422f":{"kind":"REWORK","_number":4,"created":"2026-04-30 11:57:44.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/61/986761/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/986761/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/986761/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/986761/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/986761/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/986761/4"}}},"commit":{"parents":[{"commit":"d03e1521c2307f7340252f7797c9df071bb67236","subject":"Merge \"Add share server subnet filter tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d03e1521c2307f7340252f7797c9df071bb67236"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2026-04-30 09:40:18.000000000","tz":-240},"committer":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2026-04-30 11:57:05.000000000","tz":-240},"subject":"Graduate share migration feature","message":"Graduate share migration feature\n\nRemove experimental flag from share migration tests.\n\nDocImpact\nPartially-implements: bp graduate-share-migration-feature\n\nChange-Id: Icbff80bb1f34c5f3ea959fadd390a29b0b5fff25\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7f79830b826b243ac5c1b4786ec0e46e9fc3422f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7f79830b826b243ac5c1b4786ec0e46e9fc3422f"}]},"branch":"refs/heads/master"},"f5e706ef320fa4495a2da7bec3f2094ced7cabb9":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-05-05 02:36:14.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/61/986761/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/986761/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/986761/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/986761/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/986761/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/986761/5"}}},"commit":{"parents":[{"commit":"d03e1521c2307f7340252f7797c9df071bb67236","subject":"Merge \"Add share server subnet filter tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d03e1521c2307f7340252f7797c9df071bb67236"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2026-04-30 09:40:18.000000000","tz":-240},"committer":{"name":"Medha Choudhary","email":"medhaj06@gmail.com","date":"2026-05-05 02:36:14.000000000","tz":0},"subject":"Graduate share migration feature","message":"Graduate share migration feature\n\nRemove experimental flag from share migration tests.\n\nDocImpact\nPartially-implements: bp graduate-share-migration-feature\nDepends-On: Iaca76f4cd73c0373f26a56f9e9447f52e59a9c68\nChange-Id: Icbff80bb1f34c5f3ea959fadd390a29b0b5fff25\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f5e706ef320fa4495a2da7bec3f2094ced7cabb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f5e706ef320fa4495a2da7bec3f2094ced7cabb9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"dd1b6422b9de8feaac147a9f6828f6af06c7091b":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-05-20 07:37:00.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/61/986761/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/986761/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/986761/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/986761/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/986761/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/986761/6"}}},"commit":{"parents":[{"commit":"d03e1521c2307f7340252f7797c9df071bb67236","subject":"Merge \"Add share server subnet filter tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d03e1521c2307f7340252f7797c9df071bb67236"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2026-04-30 09:40:18.000000000","tz":-240},"committer":{"name":"Medha Choudhary","email":"medhaj06@gmail.com","date":"2026-05-20 07:37:00.000000000","tz":0},"subject":"Graduate share migration feature","message":"Graduate share migration feature\n\nRemove experimental flag from share migration tests.\n\nPartially-implements: bp graduate-share-migration-feature\nDepends-On: Iaca76f4cd73c0373f26a56f9e9447f52e59a9c68\nChange-Id: Icbff80bb1f34c5f3ea959fadd390a29b0b5fff25\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dd1b6422b9de8feaac147a9f6828f6af06c7091b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dd1b6422b9de8feaac147a9f6828f6af06c7091b"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
