)]}'
{"id":"openstack%2Fmanila-tempest-plugin~987445","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I76cdfa4379ee1cc5305c13ece35500b4873e785b","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-05-06 14:46:50.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I76cdfa4379ee1cc5305c13ece35500b4873e785b","subject":"Refactor option handling","status":"NEW","created":"2026-05-06 03:45:08.000000000","updated":"2026-08-27 04:42:31.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1b1a63dfa61455bbe7ece1b91b32e2d454e9361a","_number":987445,"virtual_id_number":987445,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-27 03:17:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-06 05:03:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-06 14:46:50.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"42b89d2497aae4c22171ae1238e59751b1fdd099","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-06 03:45:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b05233102fa0c6d2f94a5b25898ccf69b65c9ffc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 05:03:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ffb5e9e730b43c3b5ad5401da399bf9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0946249c965647ffbb528089b94bd130 : SUCCESS in 1m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0e685718f5a04a9eb801edf5bc16904d : SUCCESS in 41m 12s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/0983fac80ebe46cb8b290b820ea84247 : SUCCESS in 49m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/04a10b9a16fd46ff8c6da6a7559ac544 : SUCCESS in 40m 41s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f6137aa783924453be80b3753b6ffcaf : FAILURE in 46m 27s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/37f35c47ca9e4f1a97d9c4d77b49eb78 : FAILURE in 1h 16m 38s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f9703a82f1784a9f8ea1135f608b429d : FAILURE in 35m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fed18bbb850b4d0fba8903f4c7ea88d1 : SUCCESS in 36m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0fdbe06ed84b4caa80f6b3fb5efe92a0 : SUCCESS in 24m 11s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/dc5d64daef154495b77c216190b9e54e : SUCCESS in 30m 30s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/1a612abefbc04917aa1f54c0f3293db0 : SUCCESS in 1h 03m 14s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/d1ca6535f1974a54a710e749876139fa : FAILURE in 43m 21s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a47e86b3116a4cc887638780b7c54a29 : SUCCESS in 23m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/52cf8e385a814f3bad2a3ad3b84f3b1e : FAILURE in 8m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"de9cc306c51183ab174be79cf58cd4c0f1596341","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-06 12:21:15.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"50f3eeb49776606d40e6ba256b8e53ce631affc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 14:29:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70903e9a99e44962b92924fae119b024\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adb0406986294811bebf02a3fdedbd80 : SUCCESS in 3m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4f145f5998124ebebe67027e6afcc0a2 : SUCCESS in 1h 27m 26s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/08de2ee960964402b52f997a5f1f4cb3 : SUCCESS in 2h 04m 24s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/443dc91ff3c04b99aa9148a9359a82e3 : SUCCESS in 1h 24m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b23a67a0f7a94b029c552c4ec441a7f9 : SUCCESS in 1h 01m 27s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/9908161d92c14c04bb47b66d397400a7 : SUCCESS in 1h 23m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6837c017aa084d8281f3947226010e8e : FAILURE in 1h 07m 58s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f9a6173255e1448db39cafa0938f40c2 : SUCCESS in 35m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8a8f11ff2ec746578135dd1a3d5ae839 : SUCCESS in 48m 24s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/f55d89d80e804ae098f9b260dca09f2f : SUCCESS in 1h 11m 13s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/8413e28fb1894fd7ad484e78f9a60bb2 : SUCCESS in 1h 07m 15s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/9c3d583423944f3fbe0ebd40d6c2221d : SUCCESS in 1h 05m 51s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/e3b76b7b44744b5c8e591b29a1af59a7 : SUCCESS in 18m 03s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/504cb6f559bb47af943270d475a2d645 : FAILURE in 19m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"55dce4372a5a124fcfe173a7a5cd265b252d344a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-05-06 14:46:50.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/45/987445/2/upstream-check-manila/manila-tempest-plugin-ontap-dhss/7f5b61b/ : FAILURE in 16m 21s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"c610fab9a60b3b636d69d7392160e0a405a1b9d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-27 01:11:32.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"8a5cda2bc8d9197c587d1296142b264c9f43fdff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 03:17:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2223c99859dc4956849504a1e1ab8a98\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57e97fe23e71494292b366b1b3ae5461 : SUCCESS in 3m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6448833a56df4fc990db61104c8cce60 : SUCCESS in 1h 07m 19s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/88c3d928ef864ec680dbd50f418c00ed : FAILURE in 1h 45m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b745123df07e4e52876fd137d8dd4410 : SUCCESS in 1h 13m 38s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/76bef5a21f0c4d11aec36830a69a56ba : FAILURE in 20m 46s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/876d4c18f9bb4915ac8b34690790a3fe : FAILURE in 2h 00m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e4e28ffa05544fe2b258096763c5d93b : SUCCESS in 49m 11s (non-voting)\n- manila-tempest-plugin-lustre https://zuul.opendev.org/t/openstack/build/f14eae98b93e48e784558690a6dca9a8 : SUCCESS in 28m 06s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f401060389714753b226c4387daeb410 : SUCCESS in 34m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/25de40de6c90487cbb52508fb581f3e1 : SUCCESS in 33m 30s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/1a06cc0d1e144818acef3eb059e594c1 : SUCCESS in 1h 02m 25s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/6d4a17ede73f4e1a97a7fb0f7cde4c3a : SUCCESS in 42m 49s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/f6ad21620b5345609690d4318b44d86a : SUCCESS in 1h 06m 15s\n- manila-tempest-plugin-dummy-no-dhss-old-rbac https://zuul.opendev.org/t/openstack/build/39bb57470434495984085e05becbe288 : SUCCESS in 34m 15s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/33c12323762b476d94ba4dbd38d53980 : SUCCESS in 40m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1b1a63dfa61455bbe7ece1b91b32e2d454e9361a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-08-27 04:42:31.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/45/987445/3/upstream-check-manila/manila-tempest-plugin-ontap-dhss/7cb10c8/ : SUCCESS in 1h 22m 39s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8fcd591cbbce670b72fecccb743df981106b75f0","revisions":{"ff7b2d946f7e3621e2e9564cbf1a72e827f424ce":{"kind":"REWORK","_number":1,"created":"2026-05-06 03:45:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/987445/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/45/987445/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/987445/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/987445/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/987445/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/987445/1"}}},"commit":{"parents":[{"commit":"c42be469136bef44587426464c2b24acd4b69309","subject":"Merge \"Wait for floating IP association before SSH in scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c42be469136bef44587426464c2b24acd4b69309"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-06 03:44:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-06 03:44:36.000000000","tz":540},"subject":"Refactor option handling","message":"Refactor option handling\n\n... using the utilities provided by tempest, to reduce duplicated\ndefinitions and hard-coded strings.\n\nChange-Id: I76cdfa4379ee1cc5305c13ece35500b4873e785b\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ff7b2d946f7e3621e2e9564cbf1a72e827f424ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ff7b2d946f7e3621e2e9564cbf1a72e827f424ce"}]},"branch":"refs/heads/master"},"97c4bcdc9e60f82af389bb8ea45e8b338aaa2c86":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-06 12:21:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/987445/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/45/987445/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/987445/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/987445/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/987445/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/987445/2"}}},"commit":{"parents":[{"commit":"74c619c35751024cd377253e4656b9bde37577c8","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/74c619c35751024cd377253e4656b9bde37577c8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-06 03:44:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-06 12:21:15.000000000","tz":0},"subject":"Refactor option handling","message":"Refactor option handling\n\n... using the utilities provided by tempest, to reduce duplicated\ndefinitions and hard-coded strings.\n\nChange-Id: I76cdfa4379ee1cc5305c13ece35500b4873e785b\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/97c4bcdc9e60f82af389bb8ea45e8b338aaa2c86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/97c4bcdc9e60f82af389bb8ea45e8b338aaa2c86"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"ff7b2d946f7e3621e2e9564cbf1a72e827f424ce","theirs":"74c619c35751024cd377253e4656b9bde37577c8","contains_conflicts":false}},"8fcd591cbbce670b72fecccb743df981106b75f0":{"kind":"REWORK","_number":3,"created":"2026-08-27 01:11:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/987445/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/45/987445/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/987445/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/987445/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/987445/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/45/987445/3"}}},"commit":{"parents":[{"commit":"d753c5552b822754c1d0256fd6713ef3bb58bc27","subject":"Merge \"Use manila-test-image for non-generic driver jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d753c5552b822754c1d0256fd6713ef3bb58bc27"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-06 03:44:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-27 01:11:19.000000000","tz":540},"subject":"Refactor option handling","message":"Refactor option handling\n\n... using the utilities provided by tempest, to reduce duplicated\ndefinitions and hard-coded strings.\n\nChange-Id: I76cdfa4379ee1cc5305c13ece35500b4873e785b\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8fcd591cbbce670b72fecccb743df981106b75f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8fcd591cbbce670b72fecccb743df981106b75f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
