)]}'
{"id":"openstack%2Fmanila-tempest-plugin~990327","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I66c4dabfb81fb6376de991653ab9ee04832c8e16","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-06-01 17:31:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I66c4dabfb81fb6376de991653ab9ee04832c8e16","subject":"Use native devstack options to tune cinder quotas","status":"MERGED","created":"2026-05-27 16:19:33.000000000","updated":"2026-06-01 17:31:49.000000000","submitted":"2026-06-01 17:31:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990327","meta_rev_id":"aa7dde3f1bd19266ab55fec7d974718d6ca050fe","_number":990327,"virtual_id_number":990327,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-01 17:31:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2026-06-01 16:32:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2026-06-01 16:32:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-27 18:21:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-28 05:36:44.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-06-01 16:32:59.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"6b93b02b6ba820ffe870dd8247f33e51cabe8e71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-27 16:19:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ae5ba4e86bf55d6b4b9f3b4473eea93546fc88e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 18:21:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/085464051c8a49dd977cd890884d85a3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/744ce16a3dd9498084ffa2c213dac336 : SUCCESS in 4m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d0c3798e13884115a75a19b439bb3a91 : FAILURE in 1h 09m 32s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1edc6a292a47456a915a8cc60c7c2f69 : SUCCESS in 1h 57m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/69f0b634683c49e08540bd0a8ef188ca : SUCCESS in 51m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/dfcae2cb29f6471f8aa99be5f88684de : SUCCESS in 1h 06m 43s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/79afb82889ce4751935b50c19b6a2c59 : SUCCESS in 1h 37m 14s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/21a02bd49a924507b4739e8830032d7f : FAILURE in 1h 27m 41s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2ddd11f14290437e91eb15f37369cf42 : SUCCESS in 24m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b3a34a1ee51b467491866dbca5dc0331 : SUCCESS in 35m 09s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/8ae8ae32f5fb4705815a9b5ae3dfe054 : SUCCESS in 1h 17m 08s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/c24577824549410884e8b57f3bdaf2f6 : SUCCESS in 29m 16s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/5466a9ef136a44c2a04b0ca2ad50d5d4 : SUCCESS in 1h 07m 51s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ed5aa38cf2724c81b34a1a04cce4a6d8 : SUCCESS in 32m 29s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f406c61e931b4f1a99977b2e2f9521ca : FAILURE in 18m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"45defdd4e9261be19df0533a541a7efd6bd1a8ca","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-28 02:04:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"36eb7874dbf13a52e5cf4bd0ab1fb70a9400e722","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 04:25:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35ccbd41d65b43b5be4a8d59cd52cb14\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/705d3253d47741e0b6c6a344b96b5a01 : SUCCESS in 3m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/173c01abfb234241be47ba555ef088ae : SUCCESS in 1h 02m 27s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e3739f58335344f4acb1454eb4fa3962 : SUCCESS in 1h 44m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/840cf3960bd944579bfe41b3880f9c29 : SUCCESS in 58m 22s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/6c6cb23cd9e34ea78e0aa0688a7094d3 : SUCCESS in 42m 23s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/60f36af93942499ca1ba2dd067b786f0 : SUCCESS in 1h 07m 51s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d5abc7a5991348e3af729bb0530742fd : FAILURE in 1h 17m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/698f85f35f5345e2b5c29b0efeb18341 : SUCCESS in 37m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3ebec03e44dd49f687afece79bcd21a0 : SUCCESS in 31m 10s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/2e00015816da425bbea292139d89be47 : SUCCESS in 1h 13m 42s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/8fe42b39787a460389163e9a4650251b : SUCCESS in 1h 04m 12s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/e63ac6ecbc724d50ac56c6a7e8da31b9 : SUCCESS in 52m 46s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/5162a36f6b194824a56e1071e1426320 : SUCCESS in 30m 20s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7557ec0caf1b4a9da97a326a0da7662f : FAILURE in 20m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"834f59013ea130ca4dc354a6aaa9ae60c24f692a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-05-28 05:36:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/27/990327/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/c69f165/ : SUCCESS in 1h 09m 14s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"4ba674775b1e99bcaaa7cb24d6e6f21f575173d6","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-06-01 16:32:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"20dc9f437655a1e1edbc697a7f2009d90248d66d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 16:33:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e99f3d512cf953c63cecd0d40ec12670922a02b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 17:31:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a1b612a918642f9925b941da026c41e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e0789664c9446c489a6fce287749f44 : SUCCESS in 3m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7b7d7b40939c465da1f8c2adff2faa66 : SUCCESS in 57m 01s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/211fb96d60e6464480314bf7b418eff1 : SUCCESS in 41m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7806ef10b94e447f9f1d5aa7008872b8 : SUCCESS in 42m 25s","accounts_in_message":[],"_revision_number":1},{"id":"aa7dde3f1bd19266ab55fec7d974718d6ca050fe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 17:31:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b21fa887c55776b6d08a8a38c7e8fc04f67f0025","revisions":{"b21fa887c55776b6d08a8a38c7e8fc04f67f0025":{"kind":"REWORK","_number":1,"created":"2026-05-27 16:19:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/990327/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/27/990327/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/990327/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/990327/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/990327/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/990327/1"}}},"commit":{"parents":[{"commit":"c42be469136bef44587426464c2b24acd4b69309","subject":"Merge \"Wait for floating IP association before SSH in scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c42be469136bef44587426464c2b24acd4b69309"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-27 16:18:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-27 16:18:32.000000000","tz":540},"subject":"Use native devstack options to tune cinder quotas","message":"Use native devstack options to tune cinder quotas\n\nThis is the prep work to remove the explicit override from manila\u0027s\ndevstack plugin, which conflict with the core devstack functionality\nto manage quota options.\n\nChange-Id: I66c4dabfb81fb6376de991653ab9ee04832c8e16\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b21fa887c55776b6d08a8a38c7e8fc04f67f0025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b21fa887c55776b6d08a8a38c7e8fc04f67f0025"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
