)]}'
{"id":"openstack%2Fmanila-tempest-plugin~992312","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ib17fbd012dda802882aec982c94462b6ae874685","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-06-08 20:48:26.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ib17fbd012dda802882aec982c94462b6ae874685","subject":"Remove MANILA_INSTALL_TEMPEST_PLUGIN_SYSTEMWIDE","status":"NEW","created":"2026-06-08 16:45:27.000000000","updated":"2026-06-08 20:48:26.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dc0ef72a5c286d264b17085533341f92cd188dd9","_number":992312,"virtual_id_number":992312,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-08 19:10:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-08 19:10:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-08 20:48:26.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"e3b6da6ee81296b14f1af5785e6818edaa3a7ae0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-08 16:45:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8775f0294ff77fc912c85b8e700b02002775d41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 19:10:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/564b2698d69445d5b421c4321fbaf19c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c3599b6b9b24608b84faa820f858a41 : SUCCESS in 4m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/66e0e0d554ce4ec3856c9c2d96aa8a6b : SUCCESS in 40m 15s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/871f6582dce24615b7500c6f18b99175 : FAILURE in 2h 18m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0e29c4a8788c4ba6a3f3915256bfa77f : SUCCESS in 1h 31m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ab5b3f81cc6744aca7629db068681421 : SUCCESS in 1h 06m 51s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/895b7313d06a451e813cd11482014036 : SUCCESS in 1h 28m 11s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f7e89c764cce4b629c2a91119aab2342 : FAILURE in 1h 15m 37s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e1885fca6b0c41d69dce42803ebed7ef : SUCCESS in 48m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e6796057ec5e4da0bc2a90d562ced140 : SUCCESS in 45m 02s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/d3b5ef01c9d04c7ebcade4126c63135a : SUCCESS in 1h 02m 56s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/45106a3fef5444c3a45bc8d008fe531c : SUCCESS in 41m 55s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/319f46aeaa794e57abea97f20d44bacd : SUCCESS in 40m 10s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/8c35e127a93c412fb7ea37c0673b1f59 : SUCCESS in 31m 42s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9e83446355bd4946813bf5967db646cd : FAILURE in 9m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dc0ef72a5c286d264b17085533341f92cd188dd9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-06-08 20:48:26.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/12/992312/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/d1c8e00/ : SUCCESS in 1h 36m 04s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"70cf6236e0236957bba543450a414f3f16e29f32","revisions":{"70cf6236e0236957bba543450a414f3f16e29f32":{"kind":"REWORK","_number":1,"created":"2026-06-08 16:45:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/992312/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/12/992312/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/992312/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/992312/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/992312/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/12/992312/1"}}},"commit":{"parents":[{"commit":"afb832b53c04aee8c121fd9048f1480fe6459918","subject":"Merge \"Add replication_policy metadata tests for SnapMirror policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/afb832b53c04aee8c121fd9048f1480fe6459918"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-08 16:42:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-08 16:42:32.000000000","tz":540},"subject":"Remove MANILA_INSTALL_TEMPEST_PLUGIN_SYSTEMWIDE","message":"Remove MANILA_INSTALL_TEMPEST_PLUGIN_SYSTEMWIDE\n\nIt was removed from manila\u0027s devstack plugin during xena cycle.\n\n[1] 3b6ea369044f4bade319bdf3334d5b855e0b74ba\n\nChange-Id: Ib17fbd012dda802882aec982c94462b6ae874685\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/70cf6236e0236957bba543450a414f3f16e29f32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/70cf6236e0236957bba543450a414f3f16e29f32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
